)]}'
{"id":"openstack%2Fkolla~939196","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I3f45178c07a9cd054e572ff80d0e19fb014d69d3","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-14 12:14:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3f45178c07a9cd054e572ff80d0e19fb014d69d3","subject":"ironic: Fix dnsmasq.log permissions","status":"MERGED","created":"2025-01-14 05:20:25.000000000","updated":"2025-01-14 12:15:12.000000000","submitted":"2025-01-14 12:14:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939196","cherry_pick_of_change":939195,"cherry_pick_of_patch_set":1,"meta_rev_id":"a3147e3a730b97a17c6b25ecfeac2018f3940335","_number":939196,"virtual_id_number":939196,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 12:14:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-14 08:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-01-14 06:52:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-01-14 09:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-01-14 09:56:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 06:52:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-14 07:27:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 09:56:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"fbc01dd0540afc273509cc9663665d88ebf65f8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 05:20:25.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"5c53fa147fe67c7a3c9bb026d57d1e95946bcaca","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-14 06:52:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f662e3da66244e7866057fc13b1c7f9fee03bede","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 07:27:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26b84b05e75246749c7e5754b7c7a08a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6f855b9a2d7040a5b922f8b7d8324719 : SUCCESS in 19m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/261a26fa302b4cf5a7023d626bd91eb8 : SUCCESS in 1h 04m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ee337ae8c2214f7688b6267b269d0403 : SUCCESS in 18m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ba61980c261046dcb3da956d9ae7e7b3 : SUCCESS in 29m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e6edc80d21d547b8bd8684350ce39129 : SUCCESS in 1h 07m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7d56dbe6d74b4255a07848639dc81f77 : SUCCESS in 1h 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d0d4a218f9e4fc88ea514f3c196dd90 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eac0323bb4d45eab64962b12a342d94 : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d61a63f8770c4b2aa7d8d91ab8a72f46 : SUCCESS in 2m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ead9f870ae145dc85c87e602e70f076 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/211bcc2725ba4b1e9a17f048d0a98f0f : SUCCESS in 2m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f099f078fbb4edf9d337809e286bd81 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140ef5023c634e24987c41f350eb27cb : SUCCESS in 4m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36882e3d2027484fa26fe2363535482e : SUCCESS in 24m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a17fe09af803437496bb146e112992b2 : SUCCESS in 24m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0f6806e75bf24b7a8a235af6ca0a414f : SUCCESS in 1h 04m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/834ff72d2a98431c9ee7640e6c568ae7 : SUCCESS in 1h 36m 52s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/84ca6a364c68457ab9bc81f59dbfd4be : SUCCESS in 1h 16m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ad8972b24a9c49488d5fa138dfc91f7c : SUCCESS in 11m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7317a66fe6ee49c0ab60ccb21d2ddf58 : SUCCESS in 1h 03m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c0a487d14ec544788262d18985511bf4 : SUCCESS in 52m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e555ea47535c4ff98db1c79be853af08 : SUCCESS in 1h 21m 23s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/e1c2fd3ded74451fb72ef4850a56209b : SUCCESS in 1h 01m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"22da1ac5e1323502b9aa7697090f09f98dc1554d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 08:11:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f6cf7a9e9fd3cb821912f67b4598baff33f9b63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 08:25:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e6fe1c1565245ea9956d2ae91855dbe\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/226e61be52914d1ab5f5227ada3e0187 : TIMED_OUT in 3h 04m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c07466b4f0084402850439e68a583fc1 : SUCCESS in 2h 35m 28s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3072ad87582948f4aa29398780e58c8c : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8539c19b6bf34834a68f3d0abd230c04 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/52594994584d453fb5ed6acd7c571b52 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/08f46dc11ffb4686a0f630f80e135114 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"91a45e82f846865ff667ff867a51e191ed956a95","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-14 09:56:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"88532cbe4091bd268d9dec9e9043bd11393d8037","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 09:56:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7247b9f571ee90b084a5fc2aaef0eaf9949f3e91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:14:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17e723e956d24579826f925aa94db7a8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a9ad2f34f5540dab564ec846ae9b08d : SUCCESS in 21m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/631fea3f032b4e20b32b12af8f782c39 : SUCCESS in 29m 21s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/be4fe9c02eaa4a84b103f621c04f2dac : SUCCESS in 1h 08m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4c3aa68a91be423592074d91b37d4eee : SUCCESS in 1h 39m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a26cce0b2da4cf7980e73064ce981da : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d09c51ab73743a08081bc57cb2cec3c : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6073a51594254849bb22c75f6d52083b : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a87f7f594fbb4e5f80971cc36b63f6b7 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2aee70e89574434932c35182384f6e8 : SUCCESS in 5m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3fea328b088646d19fed59c5299cb1dd : SUCCESS in 27m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9b9649f8e1ce403bb72bbdf4aa0a7e2f : SUCCESS in 30m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9b1724c718724dd49efe9e07576c4fbb : SUCCESS in 1h 06m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bbee606f24384b31bcaec891786dbed8 : SUCCESS in 1h 34m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/48c789571fc34c539714ce7c2313fcc9 : SUCCESS in 19m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e3359ec1dbc94a3cb816d79d21dc46e5 : SUCCESS in 1h 18m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06f2fd8445fc4f6c8bfa2b5c3bd31d5d : SUCCESS in 54m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f4f25a9f594041e0b94ab5fa440ca8dc : SUCCESS in 1h 32m 58s","accounts_in_message":[],"_revision_number":1},{"id":"6d27e61fc708db19863a7aaddd1328ebeac00109","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:14:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a3147e3a730b97a17c6b25ecfeac2018f3940335","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 12:15:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80209a9655db427eaff4833f0c0cd8a7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9272fce35c6148e48e86d3846b3bd0d5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d45819b051226fcb1d098c19f7bc8c421fd59817","revisions":{"d45819b051226fcb1d098c19f7bc8c421fd59817":{"kind":"REWORK","_number":1,"created":"2025-01-14 05:20:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/96/939196/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/939196/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/939196/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/939196/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/939196/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/939196/1"}}},"commit":{"parents":[{"commit":"1b880e8f9962692544c3774deb69292946165c58","subject":"neutron-dhcp-agent: change dnsmasq.log rights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b880e8f9962692544c3774deb69292946165c58"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 13:21:24.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-14 05:20:25.000000000","tz":0},"subject":"ironic: Fix dnsmasq.log permissions","message":"ironic: Fix dnsmasq.log permissions\n\nCloses-Bug: #2093881\nChange-Id: I3f45178c07a9cd054e572ff80d0e19fb014d69d3\n(cherry picked from commit ac2bbcc9f19f039cc2414fe090029f37cf781176)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d45819b051226fcb1d098c19f7bc8c421fd59817"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d45819b051226fcb1d098c19f7bc8c421fd59817"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
