)]}'
{"id":"openstack%2Fkolla~939200","triplet_id":"openstack%2Fkolla~stable%2F2023.2~Iedfb2df027c0cefc45668a76d45ce62193aba9ef","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-01-14 11:44:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iedfb2df027c0cefc45668a76d45ce62193aba9ef","subject":"openvswitch: Fix logs dir permission","status":"MERGED","created":"2025-01-14 05:21:25.000000000","updated":"2025-01-14 11:45:09.000000000","submitted":"2025-01-14 11:44:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939200","cherry_pick_of_change":939055,"cherry_pick_of_patch_set":6,"meta_rev_id":"d6996755b041c38793b86c120453692db6b8971b","_number":939200,"virtual_id_number":939200,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-14 11:44:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-01-14 08:51:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-01-14 09:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-01-14 09:54:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-01-14 09:54:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 06:51:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-01-14 07:10:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-14 09:16:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-01-14 09:54:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"1e361fcbbab55ba5e5bbdaf4104e57da58534a0c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 05:21:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* docker/openvswitch/openvswitch-db-server/extend_start.sh","accounts_in_message":[],"_revision_number":1},{"id":"f24a0a25df5357d28e3fb16d1f5c4c25b112b4fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 05:22:25.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"71ec598da782a9043555fa500c165fa020cecfe5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 07:10:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc0bbd6a6ef469983ce7aa7ac3a7b0f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/83fcb4e2c69b4a2e92d1d4eaf61282de : SUCCESS in 21m 53s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b5ede6c0a16e4d6889e204e3c202d8fe : SUCCESS in 1h 06m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/93b3a2d78d3c43b5a0be58718112a8f5 : SUCCESS in 14m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6b6a50fedfda4537a44d9fba9d692dfd : SUCCESS in 29m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/760fd7e39a954510b28958b31d9b6deb : SUCCESS in 1h 03m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb5980310cc24422a6fcef8ff0a16f98 : SUCCESS in 1h 42m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1f6976193a74c029d4d188577e449ce : SUCCESS in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b354cba13c24ff1891e239501ddf974 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5dc2dcca899342da886b00deb61cd155 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/291ff532adee45b78d49d4990e21914e : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5d37ae996ba948ecb077fa0ed6ec250c : SUCCESS in 2m 54s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4414d66aee75486f93e66b996f4d9fc9 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41175a4dd204c9889cec293d2a06b34 : SUCCESS in 4m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6025f2c5d126465db5fc63f1403b1f74 : SUCCESS in 26m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ccfc80181fe4bdda48b07784597fba1 : SUCCESS in 31m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b36b5577b3d7415591ec6ec5270d6076 : SUCCESS in 1h 08m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a287ab44364847d18673af2fdf5eb32a : SUCCESS in 52m 11s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3f5dd3d600744315827cfe4d30bf676a : SUCCESS in 1h 12m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8d418f9360344e39a7412c0fa65efa23 : SUCCESS in 1h 13m 43s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0b78f161bfc042c3a80197c6c11e99ab : SUCCESS in 16m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b8508a58f4f543e7972a001102baa4d0 : SUCCESS in 1h 16m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a47a317144274d9db0f9f1ed0314f8dc : SUCCESS in 54m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/906ec584a4114b02860148cd333bdcca : SUCCESS in 1h 22m 02s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2740e85118354a939896da6dfcfcd49b : SUCCESS in 44m 46s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b10e02d7960f4a589bdcf1544d69a199 : SUCCESS in 55m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0c94078536720caa450af96c98eb81731619578","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 08:51:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"7fa8d357916db61fb5924e7de00715b1899bf8b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-14 08:51:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bdedc4a52bedbdcfc0efa928cc2ade4711a6709c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 08:51:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53788f187fd9462980bb247b635498b5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ee078c76b69543109e2315b81026716c : TIMED_OUT in 3h 02m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a72534b10d79494ea952e0753c1f12f0 : SUCCESS in 2h 29m 53s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b16d31ebd3a342f9bec10bd0ac8fdcc9 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/beaab38b2df34282814ce35659af467b : SUCCESS in 6m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bc7a7b3d6aca42e88a6f842a6a53345f : SUCCESS in 5m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/78bd3c39067841f196bf16a17c65705f : SUCCESS in 5m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f14ec9a54275e460b62036b3ceca53d796c252a1","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-14 09:16:06.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"7cdc274f17a5271283ab750b30f41fd02f5510e9","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-14 09:54:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6251d01db32414ba04d27b77d4882ea875cf615e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 09:55:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dc067c3d39db3767ffab517911a2bcf9125e68a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 11:44:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b171da9b474040549ec414d611e75e07\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec5d7319da634afbbccecb8fe821ff24 : SUCCESS in 19m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/34fc427b6a1c4f6fbef973004200e19a : SUCCESS in 24m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a9fe657b9f894438a1beced75d40d44d : SUCCESS in 1h 19m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c8929630d98f44b3a3b8771f0e1467f1 : SUCCESS in 59m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51b09c1080843a996956891cb986044 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c76e65264ad4161844da047ae395f11 : SUCCESS in 2m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7141af0b68d94e09b29a088837cac884 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/387f968b97a34f959cb784864068e889 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96ee789dc33e4dcca94fbcc4008c2bc1 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e4b4cd2f3bf341c6adb6482cbed2b312 : SUCCESS in 27m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d806d0326aad46dd84efe58bd1a6d555 : SUCCESS in 39m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5adaee382ace4c14a5acc661c20932cb : SUCCESS in 1h 16m 49s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/54ab004181164fc28effc3ffc753307e : SUCCESS in 1h 42m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9069b8ed51684bb09b2f7ee8eaafc867 : SUCCESS in 18m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b8534e8304884ca38c9d3a1095621752 : SUCCESS in 57m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9683470e5ac94afd8d68e77b564ca0f5 : SUCCESS in 51m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1cbb4e84af6c4d0e925fed4a25dd243f : SUCCESS in 48m 25s","accounts_in_message":[],"_revision_number":2},{"id":"4bd05a119f2a2219238a76bd6271d64b5eab3b38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 11:44:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d6996755b041c38793b86c120453692db6b8971b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 11:45:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca7ad070b137494c823929800c7d1401\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7d1033561f4b20b071538f56bab23b : SUCCESS in 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"daca47545e4dd54d30ec8c2f3f93074dd9ac776e","revisions":{"3ed6b7319362756f4a5951b75f8cd8eb9d79ff22":{"kind":"REWORK","_number":1,"created":"2025-01-14 05:21:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/939200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/939200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/939200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/939200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/939200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/939200/1"}}},"commit":{"parents":[{"commit":"1b880e8f9962692544c3774deb69292946165c58","subject":"neutron-dhcp-agent: change dnsmasq.log rights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b880e8f9962692544c3774deb69292946165c58"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-14 05:21:25.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nCloses-Bug: #1795609\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n(cherry picked from commit a9c94abf530f3ed36b9b3f10c89549f04af26df4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ed6b7319362756f4a5951b75f8cd8eb9d79ff22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ed6b7319362756f4a5951b75f8cd8eb9d79ff22"}]},"branch":"refs/heads/stable/2023.2"},"daca47545e4dd54d30ec8c2f3f93074dd9ac776e":{"kind":"REWORK","_number":2,"created":"2025-01-14 05:22:25.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/939200/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/939200/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/939200/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/939200/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/939200/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/939200/2"}}},"commit":{"parents":[{"commit":"1b880e8f9962692544c3774deb69292946165c58","subject":"neutron-dhcp-agent: change dnsmasq.log rights","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1b880e8f9962692544c3774deb69292946165c58"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-13 08:54:25.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-14 05:22:25.000000000","tz":0},"subject":"openvswitch: Fix logs dir permission","message":"openvswitch: Fix logs dir permission\n\nCloses-Bug: #1795609\n\nChange-Id: Iedfb2df027c0cefc45668a76d45ce62193aba9ef\n(cherry picked from commit a9c94abf530f3ed36b9b3f10c89549f04af26df4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daca47545e4dd54d30ec8c2f3f93074dd9ac776e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/daca47545e4dd54d30ec8c2f3f93074dd9ac776e"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
