)]}'
{"id":"openstack%2Fkolla~939256","triplet_id":"openstack%2Fkolla~master~I17811f2225edd64fb20cc44f0d40e14e9edb1413","project":"openstack/kolla","branch":"master","topic":"kolla_ironic_pxe_filter","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-10-03 12:50:59.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-10-03 12:50:59.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-10-03 12:50:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I17811f2225edd64fb20cc44f0d40e14e9edb1413","subject":"Adds ironic-pxe-filter container","status":"MERGED","created":"2025-01-14 16:24:23.000000000","updated":"2025-10-03 12:52:21.000000000","submitted":"2025-10-03 12:50:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":24,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939256-kolla_ironic_pxe_filter","meta_rev_id":"c1cd612a24110cc0e7cecdc9931453f66887ab4a","_number":939256,"virtual_id_number":939256,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-03 12:50:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-03 08:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-03 09:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-03 09:11:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-10-03 08:07:28.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-10-03 08:07:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-14 17:02:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-01-14 18:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 14:45:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-08-29 12:53:52.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-09-11 17:18:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-09-19 12:03:40.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-10-03 09:11:14.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"61b7aa310208c260a94cd4279dd1285e31099338","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-14 16:24:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1f72a47bb68fed2df63f39c4f0d9364186da545c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 18:28:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a05509da8b431f8ec86ee836d34d90\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3d01303fe1e241c5be5a3c18ae82d9c7 : SUCCESS in 25m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f6ec18e4eafd470a9cce986ebcaf0b41 : SUCCESS in 58m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4dd1b67b2f66401eab6158a62a238d57 : SUCCESS in 12m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8c6516c0cce54f87b5e6edc59267404a : SUCCESS in 23m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cd1dc5c88cd24c3987d54138dbcf465e : SUCCESS in 1h 13m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf6a8b96ffc04b4e935172be0de9d653 : SUCCESS in 1h 40m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb9bc53780c4d948471b6316725b837 : SUCCESS in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f466974684d9497f99d6b18c0a011475 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d9a349f2b3e145129d9dfa748d5288c3 : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f35860671a9d4b03b877cad9768e2305 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469e25fc4e4e40f2a9ddfb1a7e4b01fe : SUCCESS in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ba1da5e80a94cc4afe1c174a327cb77 : SUCCESS in 33m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d8a6568a8b2e451e94fa97f227e41f20 : SUCCESS in 37m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7d47a6dd6b654e5592217f764f47f09d : SUCCESS in 1h 18m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/526978dabc0745438e952a58cb5971b2 : SUCCESS in 1h 55m 02s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/fc39b0490f4645e79499de63ad0426a9 : SUCCESS in 1h 20m 28s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fdbe1a86c4ee408cb79003c86ad25a22 : SUCCESS in 28m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/527d17275b654a99a65ef6bb6f269fb1 : SUCCESS in 33m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/87b27068057e46cca38cb95bfeffb0af : SUCCESS in 1h 02m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/77bedbb5a26e4c6c9d7ebc115fedcc74 : SUCCESS in 1h 31m 58s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/28b9e011ed234ccabbb900a8141dd93f : SUCCESS in 1h 08m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b17bbe330af2a5723b80ca8ce9bb307e834dfd9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-14 19:37:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcc11796a323441ca240cac158b8d5d1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fb9d05b4bebd4013ae487a9970116427 : TIMED_OUT in 3h 03m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e9ee51267a49412db0a634f3877fa165 : FAILURE in 1h 31m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f21a71a9fc5249cd8b598c46ca339c19 : SUCCESS in 2h 22m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12b62458a71f407e9eeecd7a90589e14 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/925b0e5ecbbb4889805bc45c7e321f11 : SUCCESS in 5m 08s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/175b9dace7054eb1af343a232b92afa0 : FAILURE in 2h 46m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2034d686c5f344dda8e1a06fedad32f0 : TIMED_OUT in 3h 01m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"89fea617c0b8d6100257a74f29ce77cd2e1dbbe5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-01-15 09:53:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"edbf8c2f43f0bbc383259e37358dc47d7703b2fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 11:33:04.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70cf2e7335294adcb6b32043160f80d0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6aa46046bb134e7c923f6a9b1400e019 : SUCCESS in 28m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f5e6da754b994bd6bf9f7191507246f9 : SUCCESS in 1h 03m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec0be5ad423a4f7eba8f08f919d25a82 : SUCCESS in 15m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2d3a188542194cddb9935d58075def14 : SUCCESS in 26m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/83091d0a920f4017a07fc9953611cd38 : SUCCESS in 1h 12m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2ba768583dff4c79b22b23368658b4ce : SUCCESS in 1h 29m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b1501a8db3149deaa6dfd3c8f10b05f : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cb8a560022242f783da5aa274da27b2 : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0180b63c2ad7414c9d2f40ec192f8645 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/110f2dad7cdb4a1ba2efd2387b6c6704 : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28aa783868a4476fa271159b1aae5ed1 : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/786842f77d4a4660906d2cccb3299643 : SUCCESS in 31m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2b2369cb2d504f928350b8b1b5499e69 : SUCCESS in 37m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/877263d2251b4962be4eceb7089fe15c : SUCCESS in 1h 04m 41s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d87ddf68aeb44e95b87e0417ca323286 : SUCCESS in 1h 25m 27s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/38ecb60c88e749939aec635f13c1f5f9 : FAILURE in 39m 03s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f76a9a26119a43e98d57239dd228d6f1 : SUCCESS in 24m 45s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1b0006372abd4bcdb5a38580e117656f : SUCCESS in 25m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ea390ef9a172407188ae2b1b5ab26b4e : SUCCESS in 1h 10m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/858c897159ed43df98103bce0388d2b9 : SUCCESS in 1h 26m 12s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d94a8c039392438bbd1a0c1d6c44f8b4 : FAILURE in 29m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7ea8f48cad983468ba26dcdf2f1b0ae2931250ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-15 13:00:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df50ba1cd8cd4f86a4a7eaef843ccd18\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2b47f77e93534f2ca069da293545a05a : TIMED_OUT in 3h 01m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/454ba0aef9bc4bfab342b72880d25dd7 : FAILURE in 1h 48m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e94546cff40c4373950deb23e7e2ed87 : SUCCESS in 2h 22m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d361f55045c54dd283752a233fd2770a : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/05b4e9c809254701bc4237e2ed6c29eb : SUCCESS in 6m 28s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8e50be25827844d38d61feb8bbf41cf7 : FAILURE in 2h 13m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8a00f0936c7f4528bd6509a95b30409f : FAILURE in 23m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"998a21881be80e59243c2ea4b1d9069cff22dec8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-12 17:53:19.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"b5959d8cb4eb879b09fbe7de06f6d6ec1fd12055","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-05-12 17:53:32.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c0dfb39f2400e86e24a73890c17b0e1630f74d8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:01:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a92a56894d346e78a6ba62bf35572bd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/08b8c84a5bba48bc803f4262e13313d4 : SUCCESS in 44m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d31e6ad33f324e70a049e20938316934 : SUCCESS in 40m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d263c96f9c14eb09ebf9ddf4ae8b017 : SUCCESS in 1h 04m 55s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c3122db929aa4c1bbcd85fc08b36c33a : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/806c6b5772e649bbb1eb78cc359f6184 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8b8d0dec0b234e679e03b924f9200553 : FAILURE in 6m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/461ad46ba4b24d4498d631c3e71984d1 : SUCCESS in 43m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1129ebacc0c6467598f4089070646fa1 : SUCCESS in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"6453c2356f0e71b3114b0da8054790fea0cb16cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 19:33:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452c5aa616e84bda9c22f23ce0064a24\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7ead0f7b8e414144a7d73d0451e0dcaf : SUCCESS in 26m 21s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/151fcccb30b44963b327fa2125434970 : SUCCESS in 1h 08m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d84834b39a14a6197990b898c09997c : SUCCESS in 19m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f1d4a016d3f64d8d945727cc8fbd759e : SUCCESS in 26m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a334fa68367d4b0e9ed5ba2c01a79b36 : SUCCESS in 50m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/71da23d1a5ed4fb0873dc22196c1d9ca : SUCCESS in 1h 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d76e8f50f77437f974f0d3c23ffa3e7 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e32c50f21d014e7099c7dc046a7f90b7 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e8135743dbd74e86a578160698f23d8b : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e2afeb5ffea14d86b466101647f1f842 : FAILURE in 4m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fbacd1887134d94aaf5439ecd19e7a3 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36b822082fe4ccfa246721ce91283b4 : SUCCESS in 6m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eb8fae5b89cf45d8bc1038b8c660a557 : SUCCESS in 2m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5370a00d5ff44582a39cd4945c29f799 : SUCCESS in 19m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b37919eb849144f7ace369a0b28cedce : SUCCESS in 1h 01m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ca3e5a7ba21541d7acef42fc2f970374 : SUCCESS in 1h 01m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0b9d8cf411fd4737a1a31ff1075cde81 : SUCCESS in 1h 18m 48s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/453fc2f5e6904c63b152ce863c4e0db4 : FAILURE in 36m 46s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d8ff967258974304b114e7f15b4d50b0 : SUCCESS in 21m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6469cf9a323b46f9ba0b3c6dad392ef4 : SUCCESS in 27m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/86bd0e178dc246edb5afb819fa9eb761 : SUCCESS in 1h 00m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/73cda80efbea468098e8a990b45e50dc : SUCCESS in 1h 25m 46s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1c462d898f754f66a30409a23a8483a5 : FAILURE in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ecd2c3964678d50b69d8ca299ac945262ab86fa1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-10 14:45:00.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a99ee21608053291fa30395a6bbcb48deec7129e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-08-20 11:33:25.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"955320a67abf9f93b5926760cab5875c4e9eaadc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-05 13:35:03.000000000","message":"Patch Set 5: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"73bd75ed17e04852969ab96a06a8f273786eeb89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:52:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea4c46ea6f334b36abc473e4627c8cac\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/148d28e19a6e43dc95cccfa345ed28d1 : SUCCESS in 20m 58s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/802494ceb62c467abf7afa092d9f4573 : SUCCESS in 37m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b27d25587314043be66053c46246cbf : SUCCESS in 12m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cc6c915f23fc4feaa4cb8e0009576304 : SUCCESS in 16m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9a0360c788cf4c2fa45eaf7c27726228 : SUCCESS in 49m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0e366e7371064c3595bfce20e98417a5 : SUCCESS in 1h 05m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a427da0ee7f442ba82867fb8c9485b63 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af74d5d87d574e10b166ef802fd10903 : SUCCESS in 1m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cbeb0ab4bee47c88fd17f57e6ff7685 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1c3f8d8430146b99f1017da1784cd2b : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82990eb6b9d9451687672e94ef146fa8 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8315c4d7454541dab5d0d1f72ba06e3e : SUCCESS in 6m 11s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7acef75d94bb49e68f65ec65a29f00b7 : SUCCESS in 1m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0f94b84b03214cbbbf571f6d793ce696 : SUCCESS in 12m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a85ba5e0655c468097ac92ac89c33615 : SUCCESS in 28m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae76e9820aef4362aef31d20b464297f : SUCCESS in 53m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8a91881cc14d4f52878054d83164b085 : SUCCESS in 1h 16m 09s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ac3944f2e3f84397898ead5b04fda63b : FAILURE in 39m 09s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fbdc85f0ada81b90adccf3048d7cb4eadc5433f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 16:27:57.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f427bcf04334c788d4d920748be8cae\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/e65be98713b148659a9c5974e812d290 : SUCCESS in 48m 35s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a308d5541c1d418798abf576c49e657e : FAILURE in 2h 28m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/441694bc944941c1a921b8293b085553 : SUCCESS in 45m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/29b3fd1aec7d4dc1884841d29b4559ef : SUCCESS in 1h 31m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3edddceed1904ce7858e2bca42fb7d4e : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7544d8c9b7de4b1c874c9c66da4a21e0 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bc9075606aa74bda91e8f80bfba1a69a : SUCCESS in 10m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f92a5e08009846a0a88af846a16975fa : SUCCESS in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"ab01bd43d3124b56da8173f0fce529476a8452cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:43:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a59670131398ecf9548b2586792b5c8de6d6d435","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 16:43:56.000000000","message":"Topic set to kolla_ironic_pxe_filter","accounts_in_message":[],"_revision_number":5},{"id":"71dca0c35945d4128577f531332fbc553cc70a40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 17:55:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01574dc89f6a40c2a80ea8677da33cbd\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/2103e31ea6e842969e27e5fd368642b9 : SUCCESS in 13m 31s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/7fdf33f5be5745019f91d56dba7d4512 : SUCCESS in 1h 03m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c4792d3a56c44b9095ae7c48f1547661 : SUCCESS in 19m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/284d7e0f4b424b42b33e5c6297281645 : SUCCESS in 29m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/990d7a77f728456aa7fba3485502a7b0 : SUCCESS in 44m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3aa93756be0b4dab88480af49826fab3 : SUCCESS in 1h 03m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9df3a3ccfca74af3b8dd9584282a93b3 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1a5d4f129cf42a5b1fd454dc4aa41cf : SUCCESS in 1m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1427fc28775b4a418beca59492f119fa : SUCCESS in 3m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/126cb49d2af3422191c92277143da020 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/901123b442be40c59d4eb9f99245c1c3 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9c9c3830bb48ce97bfe06e5e99cd13 : SUCCESS in 6m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/15025845a99b43b3b1076990cba96c9f : SUCCESS in 3m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4d25cde870e842c6ada7a7f626f70c8e : SUCCESS in 11m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d5fb256fb3e74ceb8de96dc488ae44ca : POST_FAILURE in 21m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9002d4aa9638474aa1fc373bdeb762d7 : SUCCESS in 50m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fb94249c96bb4ed2a0b798295e0fda57 : SUCCESS in 1h 11m 24s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/42657790daff40688369532b6e1091d7 : FAILURE in 46m 45s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5f3ce743f9413df0b38e20a66bdbb2b537697955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-09 17:58:07.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"85ac757ed4093088d3969ea317cddf1f6758bfd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 19:24:44.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/612f5d040d624463a847c80112879f0f\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/b9d66b472ee94523907e929745a177d4 : SUCCESS in 17m 58s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/783633a829a64d488c48d2bbb41ea34e : SUCCESS in 34m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/31b0a32970e54d6886c64a278917efdd : SUCCESS in 11m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/037b6794d1ba467c869ccd7723922255 : SUCCESS in 14m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7fb98f104dd74116834fd72798a16187 : SUCCESS in 44m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/acbb1aa47f3646babd99c3fe63e16d43 : SUCCESS in 1h 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d2ca84686ab46668704aaca8c1ad27b : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4caa1d37c9342f6a3fa0250657f6cdd : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/971555bb18af446281c0815fa08e3fc3 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7a9bed506594afaa5041021efabefff : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeffa9c57c664f8d999f958b71330892 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/255799102ed341f5aa68a6c7f48eea6f : SUCCESS in 6m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a820cfef54ce45e6aa560e7268670bc9 : SUCCESS in 3m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/35eafd1f898c4ab5a357a5c44d1b24a4 : SUCCESS in 13m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3c7d1b0a47ec452db5b3ccf6c528a932 : SUCCESS in 12m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e21603d7f54b45ad856590896b106f06 : SUCCESS in 53m 04s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3648afe9aa774b19963bde855043b0d6 : SUCCESS in 1h 16m 06s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1ac2ed3bb3934071ba29ad22c9c208d4 : SUCCESS in 1h 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"5dd47a7416259e596b5ce307c347cb02669904d7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 23:42:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76e74d3ca823428887b8c72d5c7e4872\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/6e785b841cac4a47a345595dc7e7b306 : SUCCESS in 21m 49s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/0cb46601827c458dbd4b1bea70bdb228 : SUCCESS in 2h 08m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41de8a26a8bf4c2684edd78d5bd14d55 : SUCCESS in 25m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/00e77280352147aa86f0fbb156d4acbd : FAILURE in 2h 25m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5098d795b42c4efca4db9fe20a854f14 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2b2626b8cde546159642c457fef0ec4d : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2977b10c0043411b830695d5bbcfae19 : SUCCESS in 11m 18s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/da22676859df4bceb420838b160b3e0b : SUCCESS in 51m 38s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c303bc9b750d5d78c3bd04c4b57333a4d9e39857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-11 14:16:19.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d8ea99f6eab176ec2ea8ccda799ca4133c2aa9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 15:38:12.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67ec5b207325404ca71db1e04d82e299\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/54b16fb262ff48e2a736a203883968ef : SUCCESS in 18m 59s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/c0a5a9a1aef44c3791ca689877f5e335 : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7793b2068e5f43d19cb7586d2cef9318 : SUCCESS in 22m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7be6a23332724bc8b6a01cf84f58f347 : SUCCESS in 31m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/aa0ca782f5f84412b6186525b7a2e9e7 : SUCCESS in 51m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb5984bcf9c04f1d9a6f6ecdab389406 : SUCCESS in 1h 13m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2113c15ef3434abdbbbee764c7d440d7 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/447cef7656004ff5aed9516fead2fee9 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b7d400b61fee43999d504e5fb4ddd89f : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/896557a2818448b18dea872df982b2b1 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c72d1233ea6047da8df4f23af9bc643f : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b463e67392f741969a9c07c576a176e4 : SUCCESS in 6m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e25e89c2f10e47cb9f925b1ed2deebe6 : SUCCESS in 2m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bdbebf21f878484489ecf832bdab481a : SUCCESS in 21m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/94276106517649759c1c8083f569d926 : SUCCESS in 27m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7aef9588676849be9d229ba7c7364262 : SUCCESS in 55m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a31edc9602224797b72204aa91b3d2e0 : SUCCESS in 1h 20m 05s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0c0d725a0b354b5680e1f78e3902b79e : FAILURE in 1h 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd4b5b4a373f8a4a9b483784b88e0340b868a9ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-11 17:18:29.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"07d0ba4d1d825c9d55a23100da68548f34f14fff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 18:59:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fb7d243e06a46cb8d25014bd7e476ad\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/38b9820483264b459e6360cc8113e0d2 : SUCCESS in 10m 41s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/ecde47a405c84e4c9b262e2d6f7101d2 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3fadd10cb21346f1ac39ab703911719f : SUCCESS in 20m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4097233633f34292b24f363ba21d66cb : SUCCESS in 31m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/79f720c7a72e4f1aa212dea0f6f8ec52 : SUCCESS in 44m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/889b609709244d80a0ac416992ca1720 : SUCCESS in 1h 06m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e385280f5cf14150b67535edd78e63e5 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/69438107b8734019ab21a5112b6db6e5 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/87de8f249e0640a4948c08e09c8fc1b4 : SUCCESS in 2m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73efc3ff67464b959cb5b7022a73ee07 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edf830c871554831a0c070914f7f6293 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bb22ca837c041f2b795eef96e834709 : SUCCESS in 7m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/86dcd2895bb244e08e43166870172d48 : SUCCESS in 2m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/123894c09e054055a5756b90f9e839c6 : SUCCESS in 18m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/abd53c8c3f0d4eba8da314f2f87f9ac2 : SUCCESS in 26m 04s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/94fbcfa0325c415b9e43c999699ac6f6 : SUCCESS in 49m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6fa72514bc0246dd88256bbd061bafc4 : SUCCESS in 1h 10m 32s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/eb678b4f682540f79244e0318b372dc7 : SUCCESS in 1h 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"233e94a15a60ec5cbba2ee38e4c2caca0a6f2ce2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 19:01:13.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65e18187c8c942cba89e72ef49a62183\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/72a681de6ba540cba48a90a11560ed02 : SUCCESS in 31m 40s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/24cef1ab39eb42fea646a2975651d93c : SUCCESS in 1h 06m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9a70588510d240db812ef30373fbc286 : SUCCESS in 37m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/57a8bce58d914fc8a64a7f5ce5778b27 : SUCCESS in 1h 35m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ba9f64cf209f40cc90e944ea87408828 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67eb90fed11c46bfae32ecf26adba04f : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/521648c63bdb463b95cc41906b8919d0 : SUCCESS in 9m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5fc18699c4ea43079c5f8180ef09b12c : SUCCESS in 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d72fdfa9ceb100678cf8f8ede4ff1f34c475f43e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 08:56:32.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1225c7affe016f31c2b76d16bfac4eac3ab9fcab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-12 08:57:57.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"1a527f286c9d3ac3b71a296b6941e3deb715ab54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 10:20:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bc254e8c75e49fb9b576d466799e1da\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7ebbccd4854442daa9b68580d54eacf7 : SUCCESS in 22m 04s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/fe98aa21773046ac81a6b45c97ca0ae9 : SUCCESS in 35m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ae35bee31a694ebbb1ba7831826f8bff : SUCCESS in 11m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f1b08e924a314bc0b75639ec58140b71 : SUCCESS in 16m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d0af0b9c10cb47d1ad715ff03349edbc : SUCCESS in 43m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6e962b6b621e4f129189b85c6b820574 : SUCCESS in 1h 05m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0d8c7ab1f843ddacdd9c74711fc928 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7baecd7ac674214a1b81f922a46b7f0 : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3d89d55b3d74720aab4f5d87b7eb5b1 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23efbf6601ae4313b61898c78c0ac566 : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4f5305e12934c8d95e4e91e76c5d6d8 : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9ec4f53d8334f08a8bbafd94361fad6 : SUCCESS in 8m 26s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1231f55a9db445b08e0c7df1928b805c : SUCCESS in 3m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/990e037e92ac4862ac0739563435dc06 : SUCCESS in 21m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7563a3c5b6f54cff9065e8aa535d7a52 : SUCCESS in 27m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ed6376fc9e1e469fabc352fd6c7b2195 : SUCCESS in 49m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/60fb1e99d62e4dcfa84222c9a77b0260 : SUCCESS in 1h 12m 03s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d4a6d7d05b4e46d78aea2f917e8c18ee : FAILURE in 1h 20m 20s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"a35e3604bd81e1c0200b65edac96d9207d0156fe","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 12:27:25.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"fb9b1d785d55ad8d24098fea829b8b6feaad0b49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 13:23:36.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/905b572c1db742ab986f50f5fcf03c6b\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/2a0b00f6249347b8ba020d307bd5ab5d : SUCCESS in 46m 20s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/8b61a69db0074d0ba3d415a43375dbd9 : TIMED_OUT in 2h 32m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/745553f170b14955a9b4035ffef3d081 : SUCCESS in 48m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/36135f311486434586fc2a54a5b73b62 : SUCCESS in 1h 38m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e50519d923a54ded994ae512661d1edd : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c85d537ad7b4cde8e6b4cfc2f1734f0 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e0a50647fef84dc1b0f47166889d8405 : SUCCESS in 9m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eff90566f9a9466280e10ae0a2fef360 : SUCCESS in 1h 08m 00s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"d39d1b3cfea906d6c409cf7c9d15150a7e39702c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 17:30:41.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"50382c892b13437ac27b31732a8e3e27b929c02c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 17:35:47.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"cf62d3cb917d96fa4fd28ebdfb07332e0fa499f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 17:58:44.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"b076f069ae440677ace33ba7932e63d24a62e9a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 17:59:03.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"0e447cfd7e00a7ee6bb81da16dc4c1d3ec85c66e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-12 18:05:37.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"48f7fe9d15a7acb3405b9fe893838cb3142585c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 19:11:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f356e12b5c8148fa854ea9c6c45ffd87\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/88f7bbed79e74fa78913ecd21e01516c : SUCCESS in 18m 12s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/2ff2e783bfc14c87ad2ce9cf4464dc8e : SUCCESS in 59m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f19d317ea8b74962af0b578433f40553 : SUCCESS in 10m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a5d3e482ee3c4e37b17df5670f12c281 : SUCCESS in 27m 31s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/96b82d6cb7b1414b9870db7b8b08c74d : SUCCESS in 44m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5bc6166f620f45459a68cdbc855a77de : SUCCESS in 1h 07m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44819904f6c74b3cbf66f159a66ac5c5 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cfd2d4b04f7e4314b14ed231409a9062 : SUCCESS in 1m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41f083d120ac46a7855c1ea733e3ae30 : SUCCESS in 1m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f3d9059a72b040a08e383f9e762ab1f3 : SUCCESS in 5m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e45f69ec1ba44ffb956f565f6e5b7bd : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1802de576949099b64f692a14d2243 : SUCCESS in 7m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a6833abb9abb4d8096c63479f9c60c5e : SUCCESS in 3m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0c7d5aa782341cd8c75499066f252bb : SUCCESS in 20m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/651aef94c8e94da3946e5c61a4faf7af : SUCCESS in 18m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/75fc4b179ab0449cab27e0f3d2fbff8c : SUCCESS in 47m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f31ebc6250aa49c3ad9dbdcc3be64f3c : SUCCESS in 1h 12m 04s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0055b1f513f34c579be1f8f76fa5d635 : SUCCESS in 1h 04m 00s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d2ef33f70e9bbaf2a8c5d9b05f3752950745fbe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 19:57:39.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb3ae9de66914977a9cce26e635d65d6\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/a1d3923fb9594315985141538054b6f3 : SUCCESS in 45m 20s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/305e13a18ba04ca2a6aa1bafb57b6f68 : SUCCESS in 1h 56m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba0b8bc3ab15418a85abe00bf60772cc : SUCCESS in 40m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1895d6cfb3b64fc4a40ad2d9f4c9b7fd : SUCCESS in 1h 44m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/937469886a1f496395e23fb7e079288d : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/186a0722bd06480aab48fd2af1e0e895 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b79afb8f8461461192c37938faeaaa9c : SUCCESS in 8m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/58e5ef3e0bfc4610b7b27aa168980d52 : SUCCESS in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a802ea5d021dc9b758843226265a9f229803cae9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-16 05:10:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"9859b4e64e4a3077de575792b841ebb248133a13","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-17 14:29:06.000000000","message":"Patch Set 10: Review-Priority+2","accounts_in_message":[],"_revision_number":10},{"id":"e0bba0530bace3bc7c48d00b82659ee708dc237b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 10:17:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"a0fb5d83ab88b634011c855539e066b009806e11","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 10:18:24.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e9a625008a314cabeba02b8bffc68b125ab1170e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 10:18:46.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"3bf114b3e829edb63a606408fa82c254d58987f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 10:19:25.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"72723df73487bc5e3a9b368b0b5cc92c86fcc328","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 10:22:18.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"88973bf820e8776ea8eb2a61e8ce6843e45d1b22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 10:34:14.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"88d95b86fdf88b90090e31c757d7eefc44cf796e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 10:34:27.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"d2d1e1d8500b516d2a777178795d64714b61848f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 11:08:55.000000000","message":"Patch Set 12: Published edit on patch set 11.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":12},{"id":"620e5ce5ff026c25a444a512b866e0f3d8b6da75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 11:09:55.000000000","message":"Patch Set 13: Published edit on patch set 12.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":13},{"id":"7a9cf6ae5e8fbc3b752fc5c9f14587d5219c5cd1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 11:36:18.000000000","message":"Patch Set 13: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":13},{"id":"9ae26ce199e9d8b9768885d7fda074f6f87b3d01","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-19 11:36:26.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"98f1dbb8002c6f593dc39da712a9a672bbe40252","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 12:03:40.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0f3cc64dd0711e6ca5fbb2a3d1a82e5d42c4eb2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 12:23:53.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d138f1f0d4fa44d7a056f651214563da\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7fead5a3ebef4f62b2ad07856b2dfef7 : FAILURE in 8m 38s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/0231f22fe583434eb03f0520fd97b639 : FAILURE in 19m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5068400b818c43b7a02a017d295e9b62 : SUCCESS in 18m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7d1ffb230fc247088d2b4f7ec4a62bf4 : SUCCESS in 27m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9ed06a6187dd40e3a696939ff70640ae : SUCCESS in 44m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3dfd2eb8a33d4457aba35d9455469c13 : SUCCESS in 1h 06m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09a3a0e2208549c1b75462be08c32d05 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2ea707b6ac0f429e85a24abce0950ed0 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3346bfae331c4504b161ea51f6007982 : SUCCESS in 3m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97779a16de6f4557812e66abaaf61209 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7f3e58ab1884308895095dbd8f3493b : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab298ba6fd84408788a0c80ae0cb701c : SUCCESS in 6m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c890bca14a1c4481a396710b68bc3e0b : SUCCESS in 4m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0cae1ec642b745d79ecdc15329d7b2c7 : SUCCESS in 19m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a713556ef3664664981591038fc82b97 : SUCCESS in 28m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/791a521c6f724d1e87362aeda6c09e17 : SUCCESS in 50m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2a1dc662e134481aac3d2daed0f41c24 : SUCCESS in 1h 13m 14s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/ac7d5003777243d5b06b8acaf3e90550 : FAILURE in 38m 44s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"30c728dfb3a92c2af08947abc4f7148300010d6d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 14:30:25.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0baf9ea7ac6d44878dde0188c3b235f5\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/80da0598d6494ee0b748db7f07d670fc : FAILURE in 22m 45s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/dfed4f67e51845a7980d9490435b0fea : FAILURE in 24m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2adf1979d00c42ab831c2185be4455b3 : SUCCESS in 43m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62086a1002a8431883d3f79cbe0cf292 : SUCCESS in 53m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/84b14615876a460484f473e50eca33e6 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e9c53f1a425543db992b54ed9d8d9f3f : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5b7dc888f2f84a20b2af0b9c64ee1cdd : SUCCESS in 10m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eda713b61a24409dbb229b0f0b6fdc52 : SUCCESS in 46m 08s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"105ee0ea0fef232e273b49012c4f90f9360e4fbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 15:49:27.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"1509f4aee18497784b27de0f7078d4ec6c99af50","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 15:50:12.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d882b0916ac985e02e38b583546bea51136ccd2b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-09-19 15:51:15.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0958db188c2a49cdeeb7c6cbba246b5f6af05e78","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-09-19 16:25:12.000000000","message":"Patch Set 14: Code-Review+1","accounts_in_message":[],"_revision_number":14},{"id":"2842524e616bbce94bf4fa15ef8bd450a4c71679","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 17:02:58.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64af949558644785aa35d0adc34448c9\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/7840bfb0472b4dd6a6b084fc2ba67cca : FAILURE in 15m 02s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/de3eb720244640fdbdd0d2fc24f31cb0 : FAILURE in 11m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f6927c2eaed24c009d00235009114786 : SUCCESS in 21m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7ca98f08f8eb43e58cfadf4476fe3b38 : SUCCESS in 28m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8744bff4cf9e4254b3bc96fe929f5504 : SUCCESS in 57m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5f7fd03670744b77b6e8ee2c7f5c77b8 : SUCCESS in 1h 06m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfdabb87db4a4d3eb4a421aeb8b48d9a : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27e3e360b0e940959a298ef7457a2ab9 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/21ad763246ba41a5befab24fdd596af1 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9531e56a01e041cf8929d48ffae3c1a4 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eba1cf0ad2a042e982ddb5ea02edd231 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d2eb6cf6434e5389565af25624948c : SUCCESS in 4m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/09be6910f2ec47709c942faae266de6d : SUCCESS in 4m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/06879173d78d4e35acdf84511e8dcd06 : SUCCESS in 21m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4ecbcadc887d42bbb818e31daa0c27aa : SUCCESS in 13m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ed6585c2b2a2435298077fc0e3e7fada : SUCCESS in 50m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6e3de816e339414dbc35f3461a329d0c : SUCCESS in 1h 12m 37s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c490628a1bfa45269718c2b83c874a74 : SUCCESS in 53m 41s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"19b4726bde8f3512918567fde44a0b016f749a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-19 18:24:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b0f98c4768b4a02848161411d9e2d77\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/00aaca30ad4a4207be6988091048c762 : FAILURE in 15m 00s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/01e095d9a76843e4b2d575a4eec6edd5 : FAILURE in 23m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/40b700b5e05f4e55b519286d03a81059 : SUCCESS in 47m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3e44e3f26e6c4c618b482a735d907bc0 : SUCCESS in 57m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f26188cd62854d2fa1ddafbd8abdc8f5 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a4423576d4e14e20a6e6a6f2c7883cb5 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/200dc448101a46b0bf1271e2ec89c892 : SUCCESS in 10m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ee9cfdedb26485db38586d07a00809c : SUCCESS in 35m 09s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"15ae78b31f47ddb9022c8fd6c70c25f424b459e4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-09-30 11:03:55.000000000","message":"Patch Set 14: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":14},{"id":"81478d899785ffae7d39d0ba71603b5b813dfb5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 08:00:08.000000000","message":"Uploaded patch set 15.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"d2a9cc2e564cda2d716fbc126e6fc3e429792f3f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-10-03 08:07:28.000000000","message":"Patch Set 15: Backport-Candidate-1 Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":15},{"id":"4006589f7aa4110c6f50fb7d76fb1f443c08667e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 08:46:04.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2433cf5de8fb43de802adaa4f51f3d8a\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/22ab31e4f4a04250b97ece0af4c96199 : SUCCESS in 43m 58s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/56734b0ecd66432ab8948b842fd91b9f : SUCCESS in 43m 34s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/c78a114a38994114bacbbafddb287ebd : SUCCESS in 41m 11s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/fd1708ef26f7418792f1d179ca020a79 : FAILURE in 43m 14s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9cb121a7010e423dbd707f14a8b0cbc7 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b2973f1984d1481ebf6ce5a3518e7265 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ea17a554ee924575a191cce233e3e6ed : SUCCESS in 7m 31s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"69fe1c621d126ff72ebf2aba193a0ffeeb7bc3ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 09:02:00.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94eabd497cc64c74b9512c8d6e8fe806\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/a19d593585f34d70be06a6b4276a82fc : SUCCESS in 16m 29s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/c139c3e851d846f680ab8f03aecdce33 : SUCCESS in 16m 02s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/dbc1185f264b4c41b53ccb4301fd833e : SUCCESS in 30m 09s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/390f6f0617ab4eae9d26fa413091cfda : SUCCESS in 12m 00s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/d40739c10e3c489d8de3415f7a76a88e : SUCCESS in 56m 47s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/12d9fa3087044e1a9acdce55eeda9555 : SUCCESS in 14m 57s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/5cb485f36ecd40b281742f2c10ea689c : SUCCESS in 24m 55s\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/cd5ed21787494b44ad39e337e24fe346 : FAILURE in 4m 02s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/c66776ca4c15401793317b2e894fbb2f : FAILURE in 3m 39s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7e7c7eb555044f7cabd386499afae1cc : SUCCESS in 2m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d496cd776534cb98642c2ada5bb7c95 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cef675d5dec4db08ae4063ab2cfb65a : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56ddea683826437d9e500c6bc5e1ffcd : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6ef0aeea27544bdc954b0de4a29935d9 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b591e0144554c259f3c4bee94d09aed : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc16011c79654e4986fe83b4a20b7399 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16342fcf48df4b55af5432a913ab8e97 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":15},{"id":"390242a19f33f0a828a0bb97f6d7a815fde671bc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-03 09:11:14.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"c252f6d553d2650a1694736625b919af43a403b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 09:11:25.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"9be7aacb76eca5021a9a38588727cf5d9d5c6789","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 09:47:08.000000000","message":"Patch Set 15: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7873833220214983ac7106cbed61d6db\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/b6fd0d39c3574c80aad92518ee5ccab3 : FAILURE in 11m 26s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/c87b14863f6147b38ed11ef3cea2a26e : SUCCESS in 27m 34s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/f3f843a291a94f67b69232c478b658e5 : SUCCESS in 14m 51s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/70705ba238bf4fe4a4eb3518c7796e59 : SUCCESS in 35m 07s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7eb815b1d7874105800c0109bbe29763 : SUCCESS in 17m 59s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/f5d6ed625f4a492f9fdc5c5fbfcf19a2 : SUCCESS in 27m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eae354a564dd41f58d7e463c0a2f954a : SUCCESS in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8400d263f89d4bdeba4304b866e5054a : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ffc7f433ff384c889453e15d2caf3ca1 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0a0ed96611d340d59f35d6cec0560369 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d351db3a734569bbc52e8851d4c1dc : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ecf19e902fc4b1da849c300959cd199 : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":15},{"id":"23789bfd982e300ead3de1416c9174b940b9dc3a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-10-03 10:39:35.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"438310ec25da90fd273a11e604cc4e50fa47d572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 11:54:48.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af27bede12b5479bb210ec33fe00dc22\n\n- kolla-build-centos-10s https://zuul.opendev.org/t/openstack/build/b453afd455a94879ac0e396a3e658deb : SUCCESS in 21m 40s (non-voting)\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/4804a1ca1ccf43e1804274d973a46344 : SUCCESS in 20m 17s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/073fd6ddc652473da7bee5d832c88d2e : SUCCESS in 31m 12s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/b151b68ba5fa42c098f203a989636e6b : SUCCESS in 19m 08s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a555c9f8c0e74ef2aed3587590faab53 : SUCCESS in 48m 37s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d97970b8e5bb429982d9663f130853d7 : SUCCESS in 20m 09s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/1b32ad97a11443fab147e291f0144100 : SUCCESS in 12m 14s\n- kolla-ansible-debian-bookworm-ironic https://zuul.opendev.org/t/openstack/build/c275c8b5dc7c4ad3a350dc4fb146312f : SUCCESS in 57m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/4267a230f8cd43a1b421a11d35fecbd2 : SUCCESS in 1h 13m 11s (non-voting)\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1c313cc7f6b74ad1a6a6695404f0238f : SUCCESS in 2m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c540de715cc8487d9a58c7c5b7b2657f : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dece7aaba23b4e648277dc119d5f67fb : SUCCESS in 1m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1813adce4e9b4ab0a1383d6f1b9defb8 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/484a9bf493094eb4aebda6400bfd1ef0 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94797ad96bf24a67b9d2986e97fbb912 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ce8e92138f4319b6646f79c34ddaf8 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/267c8607af3142dcae5070b2120478ce : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":15},{"id":"30a3c441d65c5e82760d1d947187fae06fa29783","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 11:55:01.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ed45d8acb01a16e627d93d138e35e54b9451cd8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 12:50:58.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb8c859890a482ead3eb0dc43e75dca\n\n- kolla-build-debian-bookworm https://zuul.opendev.org/t/openstack/build/3a2007ab78ca4082a3590bbfee2d5d8f : SUCCESS in 18m 29s\n- kolla-build-debian-bookworm-podman https://zuul.opendev.org/t/openstack/build/eeae2598ee024e4fbc752928f4893d73 : SUCCESS in 24m 51s\n- kolla-build-rocky-10 https://zuul.opendev.org/t/openstack/build/921d83cacfb3452a982574721c46ab02 : SUCCESS in 20m 11s\n- kolla-build-rocky-10-podman https://zuul.opendev.org/t/openstack/build/a5dce50860db4fc9bfb5cb42e0b228e7 : SUCCESS in 52m 40s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e13d555abbb24d72b3e7ee9d4c7f4037 : SUCCESS in 18m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ca10d96efa04abe81f28f11f6410ef5 : SUCCESS in 24m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c4dcc652600b4269bd28d002cacf2424 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daf96a1cf4ec472eb8bb7fa973483990 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03ff8c88557541feaeb32a0fea289e62 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd93e42fb59f49ec825db819e54be5e0 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9b4c550fea4534b411f018df923e99 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a932b7e822f84a34a2a58d241a148e49 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":15},{"id":"23df140dd4e82ed7c217f11411c65816b18e5c7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 12:50:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"f9a47702134ae40fa116067c9c2c5b17e43b682a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 12:51:49.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b940d0280a84cc8be8bc06f1bd6191f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa5a35f2295144daa1abd0317ca2f46c : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6297010d0c544be8f00e0ad9ab173d2 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":15},{"id":"c1cd612a24110cc0e7cecdc9931453f66887ab4a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-03 12:52:21.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/701c33cdf1bf40b68b2be8640ab42b3f\n\n- kolla-build-centos-10s-arm64 https://zuul.opendev.org/t/openstack/build/1bf05035877e4d48b5fdcc61babfced4 : SUCCESS in 1h 28m 54s (non-voting)\n- kolla-build-debian-bookworm-arm64 https://zuul.opendev.org/t/openstack/build/c31a958fc9584783ae2f88b46b0db4a9 : SUCCESS in 42m 45s (non-voting)\n- kolla-build-rocky-10-arm64 https://zuul.opendev.org/t/openstack/build/0b94126f35fa47e88d3eba3b3bf4fcb6 : SUCCESS in 25m 02s (non-voting)\n- kolla-build-ubuntu-noble-arm64 https://zuul.opendev.org/t/openstack/build/4a5289bd7d004c2caa0d1c7b02001a92 : FAILURE in 1h 30m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b01d766576c4c5ab368995a588dc31e : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/79b7ff1544be4f54bafda37704b19d6d : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e55f5a2e690a47438c59c8d15f4075ca : SUCCESS in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"bacaf8f49412271ebcd3ed5294e2b1435d74ed29","revisions":{"bf2faf8c80634974383080cdbd83ee34bb9ad460":{"kind":"REWORK","_number":1,"created":"2025-01-14 16:24:23.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/1"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:24:21.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf2faf8c80634974383080cdbd83ee34bb9ad460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf2faf8c80634974383080cdbd83ee34bb9ad460"}]},"branch":"refs/heads/master"},"9d335e93232b0ca342f8a5a24c7a282c500ade81":{"kind":"REWORK","_number":2,"created":"2025-01-15 09:53:08.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/2"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-15 09:53:02.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d335e93232b0ca342f8a5a24c7a282c500ade81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9d335e93232b0ca342f8a5a24c7a282c500ade81"}]},"branch":"refs/heads/master"},"b6521609c2be1d1121a097dafb34495915a3e635":{"kind":"REWORK","_number":3,"created":"2025-05-12 17:53:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/3"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-05-12 17:53:19.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6521609c2be1d1121a097dafb34495915a3e635"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6521609c2be1d1121a097dafb34495915a3e635"}]},"branch":"refs/heads/master"},"55045c594df5ea53323a1428d63eb4d7166e2c29":{"kind":"REWORK","_number":4,"created":"2025-05-12 17:53:32.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/4"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-05-12 17:53:32.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55045c594df5ea53323a1428d63eb4d7166e2c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55045c594df5ea53323a1428d63eb4d7166e2c29"}]},"branch":"refs/heads/master"},"a64871f33d4fdf011c82b76d5b16d80e51c50c1c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-09-05 13:35:03.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/939256/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/5"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-05 13:35:03.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a64871f33d4fdf011c82b76d5b16d80e51c50c1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a64871f33d4fdf011c82b76d5b16d80e51c50c1c"}]},"branch":"refs/heads/master","description":"Edit commit message"},"aaadc2a74850423891acedd61ea9d87d10a60f32":{"kind":"REWORK","_number":6,"created":"2025-09-09 17:58:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/939256/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/6"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-09-09 17:58:04.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaadc2a74850423891acedd61ea9d87d10a60f32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aaadc2a74850423891acedd61ea9d87d10a60f32"}]},"branch":"refs/heads/master"},"94e53c2bdf232fae54e74054a1b8143b2790dec8":{"kind":"REWORK","_number":7,"created":"2025-09-11 14:16:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/7"}}},"commit":{"parents":[{"commit":"f86849dcbe641dd78b73c6d6b626cffd8f67710c","subject":"Merge \"openvswitch: Fix logs dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f86849dcbe641dd78b73c6d6b626cffd8f67710c"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-11 14:16:19.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94e53c2bdf232fae54e74054a1b8143b2790dec8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/94e53c2bdf232fae54e74054a1b8143b2790dec8"}]},"branch":"refs/heads/master"},"1e219f43d7ebb12ef1ac5a413bf5bf2f942f5198":{"kind":"REWORK","_number":8,"created":"2025-09-11 17:18:29.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/8"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-11 17:18:19.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e219f43d7ebb12ef1ac5a413bf5bf2f942f5198"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e219f43d7ebb12ef1ac5a413bf5bf2f942f5198"}]},"branch":"refs/heads/master"},"4174986ff66097b966a3e01a9dc6139b9dd493f9":{"kind":"REWORK","_number":9,"created":"2025-09-12 17:58:44.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/9"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-12 17:58:44.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4174986ff66097b966a3e01a9dc6139b9dd493f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4174986ff66097b966a3e01a9dc6139b9dd493f9"}]},"branch":"refs/heads/master"},"15c941a7edc954802a4e2b1bea24e22531d2b908":{"kind":"REWORK","_number":10,"created":"2025-09-12 17:59:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/10"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-12 17:59:02.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15c941a7edc954802a4e2b1bea24e22531d2b908"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/15c941a7edc954802a4e2b1bea24e22531d2b908"}]},"branch":"refs/heads/master"},"f2f5ac6a7709ef9b2898866570e2eba62c6d1a49":{"kind":"REWORK","_number":11,"created":"2025-09-19 10:34:14.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/11"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 10:34:14.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2f5ac6a7709ef9b2898866570e2eba62c6d1a49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2f5ac6a7709ef9b2898866570e2eba62c6d1a49"}]},"branch":"refs/heads/master"},"96810ba18f1f3341c576bcc8ec055e095e188d90":{"kind":"REWORK","_number":12,"created":"2025-09-19 11:08:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/12"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 11:08:55.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96810ba18f1f3341c576bcc8ec055e095e188d90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/96810ba18f1f3341c576bcc8ec055e095e188d90"}]},"branch":"refs/heads/master"},"dbbce823eee02c072f1bedb0e18d61b71eb25727":{"kind":"REWORK","_number":13,"created":"2025-09-19 11:09:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/13"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 11:09:55.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbbce823eee02c072f1bedb0e18d61b71eb25727"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbbce823eee02c072f1bedb0e18d61b71eb25727"}]},"branch":"refs/heads/master"},"af098800192010d50d22f5ae59d6b91510e647ed":{"kind":"REWORK","_number":14,"created":"2025-09-19 15:49:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/56/939256/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/14"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-09-19 15:49:27.000000000","tz":0},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af098800192010d50d22f5ae59d6b91510e647ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/af098800192010d50d22f5ae59d6b91510e647ed"}]},"branch":"refs/heads/master"},"bacaf8f49412271ebcd3ed5294e2b1435d74ed29":{"kind":"REWORK","_number":15,"created":"2025-10-03 08:00:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/56/939256/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/56/939256/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/56/939256/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/56/939256/15"}}},"commit":{"parents":[{"commit":"0d51f0ab0349d51ec3faa3061ddd89cde2b2364e","subject":"Merge \"Prometheus and core components update\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d51f0ab0349d51ec3faa3061ddd89cde2b2364e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2025-01-14 16:21:04.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-10-03 08:00:05.000000000","tz":120},"subject":"Adds ironic-pxe-filter container","message":"Adds ironic-pxe-filter container\n\nUsed in conjunction with the inspector implementation built into Ironic\nand based on the Ironic documentation[1].\n\n[1] https://docs.openstack.org/ironic/latest/admin/inspection/pxe_filter.html\n\nPartial-Bug: #2094790\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/959850\nChange-Id: I17811f2225edd64fb20cc44f0d40e14e9edb1413\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\nSigned-off-by: Will Szumski \u003cwill@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bacaf8f49412271ebcd3ed5294e2b1435d74ed29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bacaf8f49412271ebcd3ed5294e2b1435d74ed29"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
