)]}'
{"id":"openstack%2Fkolla~940143","triplet_id":"openstack%2Fkolla~stable%2F2024.2~Idfc26fe380a546d0fd712a531e5913090b01d808","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-02-17 11:12:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idfc26fe380a546d0fd712a531e5913090b01d808","subject":"Fix genconfig","status":"MERGED","created":"2025-01-24 15:43:14.000000000","updated":"2025-02-17 11:15:01.000000000","submitted":"2025-02-17 11:12:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940143","cherry_pick_of_change":938546,"cherry_pick_of_patch_set":11,"meta_rev_id":"5968060b42ec619d1252f714204cb4cf55b145e9","_number":940143,"virtual_id_number":940143,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-17 11:12:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-02-17 09:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-01-25 19:11:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-02-17 09:24:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-02-17 09:24:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 17:46:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-25 19:11:19.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-17 09:24:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"75ae2dc0e5c03a82f4398ee0c92685a88d9a0479","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-01-24 15:43:14.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0be1f629410c0c1ace045df7af7a6352d18890b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 17:46:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9378136b65c148e6989b66882bfa26ea\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b884445f90554ce79fef56949f4afb9d : SUCCESS in 23m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ceaee1e34657480c8c5659ef9616b871 : SUCCESS in 1h 04m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b9f3babc36543439272a156c2e27d3b : SUCCESS in 20m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6fa7033e6efb4247841e42e7d0ae5829 : SUCCESS in 31m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a2db633f5c94423a87175bb20777c51f : SUCCESS in 1h 10m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/53e2b4524f684eba92a65f944990fddf : SUCCESS in 1h 29m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c703608437a4bf8a802617d6bbf615c : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb1109fbda154d84a6d824cbab777897 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a775d7ba13104b7390581db6c7f755cf : SUCCESS in 3m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60f57c79b3d84e2898cab0cb9108bd01 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/21f4dc4828e445b190f7bd9b98559dae : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29526475407e4b37b819480eae6915e1 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/048c231ba6b043528fb1bb20524028c9 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2b3b919d29aa437e96f8de34e46df871 : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/373b4eb53ec945dab1a815ec48cb47fe : SUCCESS in 22m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a8d9246080847f38ac8ef59950f90e4 : SUCCESS in 27m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8b937d7bc7c64ab8a340f34f94227bd7 : SUCCESS in 1h 03m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/82b7b84672f34479b43dad3759055944 : SUCCESS in 1h 30m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8990500699f845eba00c9dc59e656309 : SUCCESS in 22m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a3cfaa5050da41008a25a67447e5136b : SUCCESS in 31m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7763a625e52e4d98b891fc2a3bdd609d : SUCCESS in 1h 11m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b1dc1e1f83834259813d40781a5ace15 : SUCCESS in 1h 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"61f4e1cb259429ed0bccbaf980269f7004248b56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 22:23:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e4652bc4f894c0595e3f32b542a6909\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/815006103968487b96b6383cf0e77ed9 : FAILURE in 1h 30m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f02a75b484064745835fec6e1ffd40d4 : FAILURE in 1h 32m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0875e07f7a584e138d8f5b2ad964fc23 : SUCCESS in 2h 18m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e875c253d9fa4b27aae4f5973bc9eb05 : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/531d143ba4fc497597866e6278f8ca97 : SUCCESS in 6m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1388acf32edf42f7be5c5b43580a197d : FAILURE in 2h 25m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bfff8e0048514f1184f605ffdd40ae88 : TIMED_OUT in 3h 01m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"edebda156f6a7e0fe538552dc8e7bb545095acd3","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-01-25 19:11:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"08bb86f8a90c598e34ab69a648c5abe2bd6b57d4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-17 09:23:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f44016409c1e6127ddd530447611d80f182d550","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-17 09:24:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"bd07a71e6ab21b01e826268f49bc42c97e1c421b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 09:24:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5c93678da992289965cd7b6015749ed6e407fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 11:12:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93fc246f45b34999a2b6020cc8c6818f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e6c392ad3eb047c6a9410b92c5cd15cb : SUCCESS in 19m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4f213b7c8ce04009837829d72a12c0bb : SUCCESS in 31m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1cec9715c7f54cad8006516290210b96 : SUCCESS in 1h 09m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6ced917061654d5ba3a228440fb27f94 : SUCCESS in 1h 37m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f210ac7d50aa4dc8b5f2be2b361d2e4a : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfa8526dea434643b7783458826a0404 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0159ee7b831c490c8805a315f6d406f9 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66b3125ba7524f7396a1612bb96f7bae : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1215402b73834484ab6368474c431a0d : SUCCESS in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4af5cc720c4242bd814091f72ffa2c2b : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/76b356e839ba48d499091423b35a55d8 : SUCCESS in 23m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/de2bb5d93db34275a05d024a806cd4c3 : SUCCESS in 22m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0ba2446e8e16434c9529ccb33301df00 : SUCCESS in 1h 03m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b3c1a26034d04f2b92ca601514f0eae8 : SUCCESS in 1h 15m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/75543ecf860645308650d139af5f2e0e : SUCCESS in 23m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5291dc5dff5f4e508135335118561dcc : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1c21149c8a114d2494f604f5f7fdcefc : SUCCESS in 57m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/926c26cfcf4d487b9e2da29d4f000940 : SUCCESS in 1h 16m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ed92425d89b6abf64fba4dca2dfb89547b4fe1e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 11:12:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5968060b42ec619d1252f714204cb4cf55b145e9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-17 11:15:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca413662c41f45a69df6b6e4b79a5cce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4f67c47fc6e4b33adf781b99b322231 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1a714d419947d894387bad9e933398 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4110115b2317cfe2ca05e950ad7598140d2b370e","revisions":{"4110115b2317cfe2ca05e950ad7598140d2b370e":{"kind":"REWORK","_number":1,"created":"2025-01-24 15:43:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/43/940143/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/43/940143/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/940143/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/940143/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/43/940143/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/43/940143/1"}}},"commit":{"parents":[{"commit":"033a23c53b30cd58290960929d17af790c336243","subject":"Merge \"openvswitch: Fix logs dir permission\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/033a23c53b30cd58290960929d17af790c336243"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-07 10:57:12.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-01-24 15:43:14.000000000","tz":0},"subject":"Fix genconfig","message":"Fix genconfig\n\nCurrently tox -e genconfig fails with stevedore issues,\nbecause tox is not doing sdist.\n\nCloses-Bug: #2051575\n\nChange-Id: Idfc26fe380a546d0fd712a531e5913090b01d808\n(cherry picked from commit d3bb06d5ce3fb9df53fd6e043a3fea9bb1faeae0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4110115b2317cfe2ca05e950ad7598140d2b370e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4110115b2317cfe2ca05e950ad7598140d2b370e"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
