)]}'
{"id":"openstack%2Fkolla~940442","triplet_id":"openstack%2Fkolla~master~I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-24 09:19:12.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-03-24 11:09:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05","subject":"tests: Add cross-arch test for etcd","status":"MERGED","created":"2025-01-30 12:32:56.000000000","updated":"2025-03-24 11:10:22.000000000","submitted":"2025-03-24 11:09:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940442","meta_rev_id":"f5a039728eb3c90a1e0e1569e032e4a39912e853","_number":940442,"virtual_id_number":940442,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 11:09:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-24 09:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-03-24 09:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-03-24 09:19:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-24 09:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-30 14:49:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-20 09:31:05.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"2476e01e19f55682dabe413963c04da1eb4ef2a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-30 12:32:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fa6e64de4b324a168061e473d6f6577a64e14d10","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-30 12:52:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d9c955c55e657fb0b87a27b9603fe31ebdf98408","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-01-30 13:09:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8d715d01c7213591e9543ce35dfbd97832c2c44b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 14:49:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/41b5260c6c614a959ff594843755a242\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/27753a954259454f969b4b5fe46519b8 : SUCCESS in 29m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d637110fbe9a4b158431e9ca0abb7b91 : SUCCESS in 1h 33m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5f0df765908d4cf09fc31fd0dc8e6f5a : SUCCESS in 20m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/140c76e30925415d9f1f32b103a3b506 : SUCCESS in 26m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/feb2ebab44b44f5bb795d4825ebbae2c : SUCCESS in 1h 04m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2705d9d00fcc4e21ae6152eaf7d63000 : SUCCESS in 1h 25m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43c5750e171b4aed9a4a8ccb90427fe8 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c3d426015354cbca1a4c19307437a1c : FAILURE in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85836d699a3345188b3fe888d40239a9 : FAILURE in 3m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9407b0bdc13b4d738634193657e3e5be : FAILURE in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b5e3256f3854e1a95ea363831b0105b : SUCCESS in 4m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6bd9c70ee73f46b8bb987684e54cca36 : SUCCESS in 3m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/69bfb846aeb24394a33c22a0fd4e3c76 : SUCCESS in 26m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4b8e5f563ad64f8da604ded4063784c9 : SUCCESS in 31m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1f3cb0487d5047868fdfd44be66aee29 : SUCCESS in 1h 09m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2731af79d8954e9eac672d00dcc3c85d : SUCCESS in 53m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/486e9e2e34744e26acc93103ea7b7285 : SUCCESS in 24m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/44dfffafba044b4d80b6fa8a685b5c37 : SUCCESS in 30m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a73344a8f7054c52b98eb86ad145eb12 : SUCCESS in 1h 03m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8b8d66c0578342b9af54a397a471f57f : SUCCESS in 1h 33m 41s","accounts_in_message":[],"_revision_number":2},{"id":"48bc01e0100040cc4d9e95ad38956fecbe884649","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-30 16:56:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df1b654889b148e885b12da717572d0a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e2201906e94f4bf48a442242414036b1 : FAILURE in 2h 18m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/695da000e4ca41d79231f8f361fce406 : FAILURE in 1h 22m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7a894c303e6d447e8fbd5550de62215c : SUCCESS in 2h 25m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0d84f5119d914d25b7bac15694918ea0 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/297f3c9a78e0404f8b4e7c92613c99e4 : SUCCESS in 5m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/bb57f79f4fd644f5847dd3a382469cb9 : TIMED_OUT in 3h 01m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/206e7ec9703b4d8493b23228add17d1f : FAILURE in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5105901d95e5eefe8e075935ae96a557ff62fbdc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 09:31:05.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"8b747c09a56fc229de8f6c9e1f3fd014ecf92103","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-21 15:09:19.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4348a40e77267ad9ed515cd008cebd18181adc21","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-21 15:09:34.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"c60e0b03b12ceac278eca21f6312904a63efb88d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:48:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fa073041ad3e4014815288e32944e058\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5a84451c09834348adadacf2d8fd58b5 : SUCCESS in 24m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d7b76e1054744e27aaded105d01ce6cf : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3a48fcd1cfbd4390ab29a6930b06ef26 : SUCCESS in 22m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/01c446e7805f4453a2fce13cac4905dc : SUCCESS in 23m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3b5dee5e2372442385e38492bb5879da : SUCCESS in 1h 05m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b1e0a2b109ac4c0bb52b50147750bd1c : SUCCESS in 53m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/118a434b3fcb41f49ffcf9e9e2599ac6 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b203c06cabdf461b91baabbdddba797a : FAILURE in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/20bdd0e061d94aea9b3cd4496955a29a : FAILURE in 2m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/850bcf4dd92340378ed871d380221449 : FAILURE in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/374b9f760bed498ca601b8d3b1da283d : SUCCESS in 5m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e96cc94dd16243ccbc97d5fe69d80dfb : SUCCESS in 5m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/84b69ad4a3864f809ed4e35675e1dd95 : SUCCESS in 26m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4d23ff4a864c43a0bef3504b1cee662e : SUCCESS in 1h 04m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/31ec460f54f4446396dc909ff20b12bd : SUCCESS in 1h 18m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/20831f2add3d4f6488bb7769177773f4 : SUCCESS in 52m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e225cade95b2410c9b7c77f29e600d55 : SUCCESS in 29m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4dd6b7142f66497bbbc9c0df8aa4663c : SUCCESS in 25m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f63fd859b5fb483694ba2620170391df : SUCCESS in 1h 14m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d6506e2d7e5745d28e620163d341fae3 : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":3},{"id":"b32c21e53082eb772eb3e1cd4f813500e0863217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-21 17:06:48.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c05968ae3b5c738731879f99ce48a81a17b39019","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 18:48:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4a6e98d5456492eb787d12f3b02612d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f03d92ab297c4ab3a543e183ca6c6a03 : SUCCESS in 31m 43s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5e45ca906e394f1381047c8fadf3fab9 : SUCCESS in 59m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/122060915f1e4cc999fcf03a2331024d : SUCCESS in 26m 31s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/555275996d964a658ef7bb7a8159cc16 : SUCCESS in 36m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d0633d48a9034db7b294a9f97ad5f711 : SUCCESS in 1h 02m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/04c45e2c17024d239680c34dd71032de : SUCCESS in 1h 35m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a72c123c7bc46dc8a179620383cb43a : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd5087642a0e45d4a77a95813d8cb69a : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe9f92ff714a461890fb299bb2948a50 : SUCCESS in 3m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cd6134ca9ca42d4923f41cf745ffdf6 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bac68d3e4d914671b85a5ebb385204de : SUCCESS in 5m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/666cee65c5c641da9e735404be725df3 : SUCCESS in 3m 03s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cbda8843d20a4a5681623249635c15ce : SUCCESS in 15m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5b8c19dd7b4744a8a5c6c694293b66f7 : SUCCESS in 1h 36m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7f16370edc8c445ba57e241748461052 : SUCCESS in 1h 12m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e4598341e9a54c50ac241f198199fb45 : SUCCESS in 1h 13m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2d06ebfadb2e49a9b3a1c76bbae1bd13 : SUCCESS in 21m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8da1eead0eb84e5f8e0e2fb1a7d3834a : SUCCESS in 17m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/854c49488ba74a9dbb6c5386688688d4 : SUCCESS in 1h 10m 41s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b967468b0a0a48788f6bbfc4439df8eb : SUCCESS in 1h 30m 45s","accounts_in_message":[],"_revision_number":4},{"id":"27c0f6bf37072208372d2c7b560a78df18e82527","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 19:51:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff5ba79e5efd4facb68244395ad00c16\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7e45378bd1c8486aaf922f24af9db0e8 : SUCCESS in 2h 11m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5af7d7e643d94bfcb6c12fba0a99602f : SUCCESS in 1h 20m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f8c32f69fb234353b54417c65ea501f0 : SUCCESS in 58m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ea18f38821245609242e3bd668955f2 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/47ccbaec4a0e424db5feedde08c9c9ad : SUCCESS in 4m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f3931ab06ed043b7b580b9ddbe8727da : SUCCESS in 2h 39m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f9a3eb6681934b5fa3934104ee56b572 : FAILURE in 24m 01s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"70ff7d53aae9238d339d645048a1db5587d6ad2c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-24 09:19:12.000000000","message":"Patch Set 4: Backport-Candidate+1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"427adca3b4f5fb3a211068f6ff6f7ad26356f0ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 09:19:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"593725e003707856165df21f4f8e64b6fb00fa61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:09:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10b9cca110a34febae0e9d7f36157f70\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8d26d36dbe946538a43f68e090caf03 : SUCCESS in 13m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7df90e9dacc1481fb7a14b7b76d878f9 : SUCCESS in 32m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fada9603823c42419ac2b7151eaa438c : SUCCESS in 37m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8fca1bafacdd432fb3b9c7608f23b8b9 : SUCCESS in 1h 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b5a725d72e84d64a1b538d19f536853 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6139ccffc6a440a3af737f7d30e7dfca : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66cc2e97ea294dcc99dd2693f7850f97 : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b26a8a36b0754c60a99f5f1ee7f6a718 : SUCCESS in 5m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cc2bb93fe2a649e98084137fd81b4713 : SUCCESS in 2m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/054a1601584943f3acad0609736f1e86 : SUCCESS in 16m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6a57cbd320024ebea5bbe04ccf3aa2e0 : SUCCESS in 1h 38m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7238f002fa4741b39bfb92bad6a9f507 : SUCCESS in 38m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/16a046271d324d4e97b76b5732a7cb28 : SUCCESS in 51m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f9702fa7235a4ba6837b9b51925dba81 : SUCCESS in 15m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/50a53e5530d14a7f8148c69cee8a4de5 : SUCCESS in 17m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8a6d6384e2444beeb96f70923d28fa42 : SUCCESS in 57m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/beef79022f214cbbb34b8e25f65589aa : SUCCESS in 1h 29m 01s","accounts_in_message":[],"_revision_number":4},{"id":"cc45af03aba7fb93b75ff6546b963f18f764a603","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:09:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f5a039728eb3c90a1e0e1569e032e4a39912e853","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:10:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a39a0617cf844729a1a03e097cd50a0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b52b517a52214846a89d6fd03b17af11 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9c6f6ae2d9b1d412bde2821c48ff1f414066f566","revisions":{"f020e1c78276f916f3b5a75342a1cea71d7cc206":{"kind":"REWORK","_number":1,"created":"2025-01-30 12:32:56.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/42/940442/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/940442/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/940442/1"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f020e1c78276f916f3b5a75342a1cea71d7cc206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f020e1c78276f916f3b5a75342a1cea71d7cc206"}]},"branch":"refs/heads/master"},"79f248ea6c40f008c51e107063ad7224fde10915":{"kind":"REWORK","_number":2,"created":"2025-01-30 13:09:14.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/42/940442/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/940442/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/940442/2"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 13:06:35.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79f248ea6c40f008c51e107063ad7224fde10915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/79f248ea6c40f008c51e107063ad7224fde10915"}]},"branch":"refs/heads/master"},"a0089c9e42c85dbfe47b8b29d490c5100c53500f":{"kind":"REWORK","_number":3,"created":"2025-03-21 15:09:19.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/42/940442/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/940442/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/940442/3"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-21 14:59:45.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nRelated-Bug: #2103810\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0089c9e42c85dbfe47b8b29d490c5100c53500f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0089c9e42c85dbfe47b8b29d490c5100c53500f"}]},"branch":"refs/heads/master"},"9c6f6ae2d9b1d412bde2821c48ff1f414066f566":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-03-21 17:06:48.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/42/940442/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/42/940442/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/42/940442/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/42/940442/4"}}},"commit":{"parents":[{"commit":"915189d10be83136d54f4216b81310b3dd16b82d","subject":"fix sources.py based cross-arch builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/915189d10be83136d54f4216b81310b3dd16b82d"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-21 17:06:48.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nRelated-Bug: #2103810\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c6f6ae2d9b1d412bde2821c48ff1f414066f566"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c6f6ae2d9b1d412bde2821c48ff1f414066f566"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
