)]}'
{"id":"openstack%2Fkolla~940514","triplet_id":"openstack%2Fkolla~master~Ic46b895775edf5e5fb2b637be49e2de1eb4adf36","project":"openstack/kolla","branch":"master","topic":"ironic-fix","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-02-07 13:56:16.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-02-07 13:56:16.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-02-07 11:46:44.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-02-07 08:59:47.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ic46b895775edf5e5fb2b637be49e2de1eb4adf36","subject":"Fix permissions for ironic metrics","status":"MERGED","created":"2025-01-31 11:52:19.000000000","updated":"2025-02-07 13:58:00.000000000","submitted":"2025-02-07 13:56:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940514-ironic-fix","meta_rev_id":"4dc4089099232dd53ccfde0f1a7166afec29c8da","_number":940514,"virtual_id_number":940514,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 13:56:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-07 08:49:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-07 08:59:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-07 08:59:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-07 08:49:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-07 08:49:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 12:27:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-03 11:03:09.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-02-03 11:03:09.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-03 11:03:09.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"9c6e8d1da53e60160c205dcc7fd04d0d06004ac1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 11:52:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7500d80583f22aeb0ff620754cc60d5ab7b03e7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 11:54:01.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"78760b6633291e0439eef0f76c491628bc44ec07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 11:55:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0fb47ebe7c04db7e65284702f84df17ceab0a99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 12:27:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/483e1b59ed4940328894718b1f46a69d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6745bae7af1743548fedaedc5cbe1f35 : FAILURE in 14m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/232677aa76d641619b7ff96cc8671971 : FAILURE in 16m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/445f6a8105664f6389569327e26771bb : FAILURE in 9m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4e4bd30fe2d2426fbd54071c819dd8c2 : FAILURE in 16m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4b6e8fa080d34ea98b7d824aa9587352 : FAILURE in 9m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8b0f884ed7d448bd9411c588906e6ad9 : FAILURE in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/091f9f31f2024a0b9f72217dbbf4f4b5 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37e86c32b92c4c329926f0ccee1e7036 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1514dba051fa440aa503b63148986cc2 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cc2ae676545498ea394e581aab2229f : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e15f2343804f42b1593a6ecf0bbd0d : SUCCESS in 4m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/72e2c5c76fea4a9d9d2126c0b00ece9a : SUCCESS in 2m 57s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2a46c5064f4745e69cf6c418442f60e7 : FAILURE in 19m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3ced999fec82475a99aaec84bb6cf1db : FAILURE in 19m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f61f7adf82a048e3885573f265280e8b : FAILURE in 23m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/49d8edf197a94631a03add91e386d534 : FAILURE in 15m 14s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/04a67a04befa45828b11fa1c472febbb : FAILURE in 15m 35s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fe94a6077efc46bf89316ab1f7b5e40f : FAILURE in 19m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b50f0d9737624f7eadadf8791a80bb86 : FAILURE in 15m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/880df25785ed4cf8a15e62261cde0c9a : FAILURE in 13m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5f00bcad79e94059ae4867664fdfbba0 : FAILURE in 14m 56s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/bee6143e310240e2aac97048f887e6c1 : FAILURE in 13m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd18809a140849990ae1c689e88d0de9d206f432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 12:34:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"d774de8d1e50b4dd4703c50a3426f4ef5aae595a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 13:29:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ccf30fd31dee767ebcac035e1a2a9ee42a16cc9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 14:28:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"651d3b594d4d31872a0b925f641337c9d2a293ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 14:49:11.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"2926b4007a67701713b604ab7e075084c55541b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:41:14.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/794c3554633549649258f2b04f13e368\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c18f54bcbd7c49fc95d0ea505bfee93f : SUCCESS in 20m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ba4c0cb82b5c4de89f5b5a9e6263d086 : SUCCESS in 1h 05m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9b6b2fa9d06e4cf8be88566a0e2b4d45 : SUCCESS in 21m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e4024fd6eadc46a690292f6c45e31957 : SUCCESS in 24m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/feaa8d2640924f7f86178bcf6ed25df0 : SUCCESS in 1h 06m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f974297b8e8b4793a82bc5f45bebb17c : SUCCESS in 1h 23m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f50181b5b96447629bdd16e25121f406 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e8dace7706042ca931c59961823be01 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e298db2e00f246a3a9cb5490aabcd495 : SUCCESS in 3m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c726f4b28651426d855a9938fbb213a7 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643ec16791234b44852f4fd68acefa24 : SUCCESS in 5m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/27d12a1f3a99400c8de60dda10b1dbda : SUCCESS in 4m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7deedbd45fee4c139ce4a579f2298aac : SUCCESS in 29m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ead7b37930f140d8bd19adc92f0ee204 : SUCCESS in 29m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/55d4ec6b66f84edd9ee46d4bb0bff698 : SUCCESS in 1h 14m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2753012e38f84e348cb8f04015348a60 : SUCCESS in 1h 37m 10s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5e7e232116e2443d952d74da95ee5d7e : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b4a08105104f445481b9fa76dc34f250 : SUCCESS in 22m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8423b434521d470689d640b9cdc8423a : SUCCESS in 23m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9320fab5ca1f4ed7ac30880a35accd91 : SUCCESS in 1h 07m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/38048972d52f45e3ab68e3cff16a736d : SUCCESS in 1h 41m 10s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f9a9469484e44111a80293bbbe951a4c : SUCCESS in 1h 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"8b91284da3ff00fa9f41745efed053cf285126c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 17:56:31.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/734fc3e384554dd39b3bf19312c34067\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9413c009acba49259cd2559cd21123fe : FAILURE in 2h 05m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/97e5e523b9324c298daf17256b16068e : FAILURE in 1h 11m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4f155fb024bd4c1ba55d4a3d0818fe39 : SUCCESS in 2h 13m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74ccfa205eae4a4c9902d3cc51e023e5 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5b7f854befa4922af53576dba66f65e : SUCCESS in 6m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5e8d1b1f0773406885188d668353ebb8 : FAILURE in 2h 54m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/823eac2a8c7c4d2cac51d1eb4efa0ee8 : TIMED_OUT in 3h 01m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"248cd0d46a8487500e490e99b13b84a7e710c960","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-03 10:43:48.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"b4ab8d4f3c645a00af245b9d38c94e3993c93185","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 12:16:24.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23de1a6d3d3d41ff93e19ab8fc7ef3f8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/75a14d6b27c84da185e8439ec4747ad9 : SUCCESS in 26m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dffb3a85e33b4aada17d3e3d63058e89 : SUCCESS in 38m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/28f4bad9d08b4d088d4af54b7163329d : SUCCESS in 16m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ba0d80ed0e644d659eddc54dc01ccdd0 : SUCCESS in 25m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/684b1e34562f4114b99fe763677f88d1 : SUCCESS in 1h 03m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4d75ed4a764e4449a8bab542abf21249 : SUCCESS in 1h 24m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8928ea220036426cbe1391e496029b12 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2266f15fa2194103aafc6cdf6bbb4335 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3bef848a34534f1e9219f845ab82b9f9 : SUCCESS in 2m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3e6e449e8ef45e69040f254f41f27e4 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0fa0ffede7446208b5866e7f951944b : SUCCESS in 4m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d69c04e51dd24a6baef14c795b8684d7 : SUCCESS in 2m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e0136e1476be49f8b637a2532d392af5 : SUCCESS in 23m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/df341b810c964e9dae583535e550f9de : SUCCESS in 20m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ac41dfd3ae1e40a79e0379936cfc714a : SUCCESS in 1h 10m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7a15a3ce6fc649e28471ff6229a59eb3 : SUCCESS in 1h 24m 07s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f5f97cb7fa0f468aabe028254807fad3 : SUCCESS in 1h 19m 40s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/758f706ae3544bc7bc0b45899e3cb15a : SUCCESS in 28m 14s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/57491db2d9184e1fb4651f38a33cbd24 : SUCCESS in 29m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5df30ea7081b4fa7b51c9de1d43e6a08 : SUCCESS in 1h 05m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e615364e7f6c4f8494d529f909178dd8 : SUCCESS in 1h 27m 55s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d032f252c7a949dfacb583594c3db411 : SUCCESS in 1h 01m 36s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"28c2037228a24e8bcdaeb814872ac05e1ffca09f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 15:04:06.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/977cec62ccbe4d78ba83c320b9791817\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/44448153524f4fc580f80bd49dd2c08e : FAILURE in 3h 02m 03s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9717964b674748b4b4467e6bbaf37926 : FAILURE in 2h 06m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ea05856d7be14d47a4a79b45a02fabdf : SUCCESS in 2h 05m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0efcc1e1ddfe47c690e4e732fcbaf245 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/30321ffa993b451698ba220eec621bce : SUCCESS in 5m 21s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/16f4b9d0bcc54e3cae07a462539bd8ee : TIMED_OUT in 3h 02m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/be1c4cdc104a481e854dcbda518c1ca4 : TIMED_OUT in 3h 02m 24s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"376aa36ccd5a4f25b3f77f2ce75a40fb02c45468","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-06 05:54:30.000000000","message":"Patch Set 8: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"8297a82e533355fc28bcf8e09458b9e90198b2e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:16:49.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"36a8402a96146c6d66255c291639223fe2a1e411","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:17:03.000000000","message":"Patch Set 8:\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6c13d9ac98459d7f02d71a1ab39bd28e5b378fc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:24:49.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3516ad19037823a0a12e1d329ccea74c0978d550","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:25:42.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"0c16e360cf575c2f1c1b3581e74a7b712a3ab2a9","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:30:07.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"7890d43f7666fac55ec157e9c2086a0d099cad7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:31:48.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8d2c6352f0ad871a3113a7a258b8d5da16973147","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 08:20:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23aa6933422144daa9ce5dc6b0e9dfa3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7cefa19178734126b873fc051d812eca : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bb41f5ce29cf4b85bd0039844ae4338d : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7a236f86560d4280a0911d1f3d5933c5 : SUCCESS in 21m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/72e02854ff4042c4861db0e68fc6dedc : SUCCESS in 26m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6f3bbc0ddea846d2820e5be1aadbaa18 : SUCCESS in 1h 02m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f8fd1e4688f64835af9a54ddb9bb52b2 : SUCCESS in 52m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e1fc76b7060480f871ef275108bc3bd : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27720ef7342f4b8ca4446f78b7593c26 : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/18ee3a46947849a79e371aead231c75d : SUCCESS in 4m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef898709741644bfa339a00b3187c620 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd17068ddc644a38a6867b3167cc99e : SUCCESS in 4m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/83f7aef3708447909b69fa2b38e69dcf : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6f7ad02f8a63442d841e1a65af84e187 : SUCCESS in 39m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fb2ba2e4985b454f95e25dc60210436b : SUCCESS in 44m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd3436edd5444d3b96cf488f71c9f7aa : SUCCESS in 1h 12m 52s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/33f1622a6ad044838d04554ce561bda7 : SUCCESS in 1h 39m 59s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/d9c30be22c7141078cc2ec4b38a45a84 : FAILURE in 59m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/82ce32979074414b8b757e3365152e3b : SUCCESS in 33m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb8982b6d5fa48faa2388459bdb67674 : SUCCESS in 36m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/add0de96243447afb3c0835fb2548fef : SUCCESS in 1h 20m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5d5aa906a2544311b7249f410643ee0c : SUCCESS in 1h 31m 06s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/1d0544944caf408891e017365e498834 : FAILURE in 52m 31s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"5819a2bf04e6a1cda75ba570cd65567ac1f4160d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 09:40:27.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f8590da7954462188b70f76ebb69515\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/66088aac74a949489b23218b190a2b70 : TIMED_OUT in 3h 01m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/378e493c38244ba78a8938b529a1c84d : TIMED_OUT in 3h 01m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c65e3be9e7834e1485e5be2810508d57 : SUCCESS in 2h 01m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/943868946c184910bcec589a12881a80 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98634ed01748477790f44864e95463c9 : SUCCESS in 4m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e7d586b988c2482a9ef05fa6a3e7da49 : TIMED_OUT in 3h 02m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/af4471d0880e4dd7a7b0302bd387b998 : TIMED_OUT in 3h 01m 43s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"a399b804e75080eb05f68f5d6e797a529ce091b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 12:42:00.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"25a6409bd2ac87c042ade09fea04d9263acabc20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 12:43:34.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"6554c9f98c923ef7f8e0ab61c55bd9aef79dfaee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 14:34:09.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d7daa0137be4998a7e6a8504af96310\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8ff3f1f48d9b40cdb05d51ff3b4e9f8a : SUCCESS in 44m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/331b7a008d974b8982ade2e224eeb895 : SUCCESS in 1h 32m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c6f758350d8448eca6a176d4a492d1f8 : SUCCESS in 19m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2f6bb0fccc6e44a99ffc39d5bb608e10 : SUCCESS in 18m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/279ec5a4d24341be9f6bbbbd3cdf156d : SUCCESS in 1h 12m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3b77c8d275c34c1da6b261e2899edb22 : SUCCESS in 1h 26m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/278a7a090cad4d6bb0b6726f4d49fefd : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06bf744307d046c1b95e88101c2554a4 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5bd073e7cf6c4f448b50f74f41890090 : SUCCESS in 6m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9829556734084e368253244f173c7f85 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9b1a9cde0947589a152e75050730be : SUCCESS in 4m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/53b7d2b62a0a4e41a1ac6d825fa4bcc8 : SUCCESS in 2m 39s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a142cdfb8670422dad50fb581af878e2 : SUCCESS in 17m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/eeff49a4988e408dafa4e17f9a38076d : SUCCESS in 37m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c51c2ba85a20489cbc4b3c28cf2bd680 : SUCCESS in 1h 07m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cf9c8f7869e64aeea08c3afa0ae15ec2 : SUCCESS in 1h 18m 16s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9b7994fca3154b578721f99bda32bd06 : SUCCESS in 1h 16m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/57e34abd1c1e4721aefaff8586b6f7e5 : SUCCESS in 15m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5236063d92134c5b9829414ac0699c0a : SUCCESS in 25m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7ac48b4f781e4eb48b5ba4ef75bd0603 : SUCCESS in 1h 26m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/392a946420ee4f0c8f6016dede411ec8 : SUCCESS in 1h 44m 47s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/12b5ebe88c684cda9c7e746aac178b1c : SUCCESS in 1h 39m 11s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"118cc548a668158126a9146e62d8ba8e74406bd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 16:25:55.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb5460abf3147ea8a9fb26da74b1883\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/0f7e151bd95b46a6b6f2c65de9df8be1 : TIMED_OUT in 3h 01m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ecda6555048e4638b8af606b4b491282 : TIMED_OUT in 3h 02m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1b03cc221cb45b385e473337ffb81af : FAILURE in 2h 28m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7729ecf59304c47a3cd2f8d0b636a96 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/768845d035604b76a7e3b63fbd62b08a : SUCCESS in 5m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c8df970ca4874e86870ec74b13b77a8a : FAILURE in 2h 21m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bda6136a1c2f40fa96632a197a6d77e3 : TIMED_OUT in 3h 01m 41s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"b96e3c4af9e731a6b8e7b7cf7f9c3e21fca5535b","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 22:39:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"d1bd497b28db0c33426da0f18a289475a0933ea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-07 08:49:31.000000000","message":"Patch Set 14: Published edit on patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"466167cd6bacf1fcb74d8c8e8e319dbd7792e87f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-07 08:49:45.000000000","message":"Patch Set 14: Backport-Candidate+1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":14},{"id":"616a82138dd8e2dfa94609125385315e39319b99","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-07 08:59:47.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":14},{"id":"f4b2f9fbe560a914edd5130b1c51c08d8bfa883b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 10:40:01.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5178688d275a4b719f79b99e7887b946\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e03b89b5cbbc4137bf3b48a2feae32e0 : SUCCESS in 28m 00s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/22f2856e387f42118efae304e1fdb555 : SUCCESS in 1h 01m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/088c6386981e457bbab1866d5b7fd447 : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ba41e68a6bb84e8b9ef18ea86ea3ef8a : SUCCESS in 19m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9a5d083bc2284323a9fdb75e2916fb8e : SUCCESS in 57m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8566ce6a09194d6490c12471df928df7 : SUCCESS in 1h 36m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c577f8cbd66465c9849cdc175cf7f36 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b869351f7f845b385e82780ca74750b : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d939b6fde49d4e5f9ed61580f83f8ab0 : SUCCESS in 3m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/840d4f8107c344ec9b03d07bf37877ef : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14bdd5f1bc9a4d96a81c6ff776788868 : SUCCESS in 5m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/06145e84dfd24b698e805dc9bcbb94a9 : SUCCESS in 3m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4887d39bbcdd4dbfb2b6c6f07e2e4bda : SUCCESS in 29m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f5af8f9cf1254e9c92a15cba5418da55 : SUCCESS in 32m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f72c963eecca46a0be85cee449d2f556 : SUCCESS in 55m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e1e84e9c4aa4486dad497118d14316f1 : SUCCESS in 1h 42m 39s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/ae4546c7579e47e38604389935f19151 : SUCCESS in 1h 27m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/39afc405929f4a54897b064b7adcc021 : SUCCESS in 24m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b532ad80e3264de5ac2eb539fd678207 : SUCCESS in 29m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d1cd534db374478fb5746767fe88d634 : SUCCESS in 1h 01m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/da7b1c5b2805491c97c69a16ae2428e3 : SUCCESS in 1h 27m 26s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/c2aaeb5283f54b9d9bbd7e9a19c39552 : SUCCESS in 1h 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"cb823624c34187cec0cc4e7a81feecb634258fdb","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-07 11:46:44.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6f792e1e378da075c519222ba8ba7196760d5640","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 12:04:24.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":14},{"id":"699666e2c24222a46d9dc8f3425ec255d26a8bcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 12:27:34.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f8716608ef47389f408dabfce64c10\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9c117894ad1043a4b59011e5e1e8b06e : FAILURE in 1h 47m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/72f638456bdb42b2963a2c2b185946b1 : FAILURE in 1h 39m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ed1241285e88450c8635c0ed322a7c15 : SUCCESS in 1h 32m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e67304902e3645cf9eff357e8e5c444f : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2304d24f138044db9840c73337cde3d8 : SUCCESS in 5m 33s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b275a3fa9cdd49b5a197d2ab98d7041a : TIMED_OUT in 3h 02m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4541f35fb93149dab3b4859be2ec9841 : TIMED_OUT in 3h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b169deb6d2ef1890e8373f60b86d8c3beaad1a0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 13:56:15.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5a09a4d6fdf4930bcdd35a46e4a107d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f600ab3228e141ffa210ed9e02f7c19f : SUCCESS in 12m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f493bdc363924ac9b31279139a68bdca : SUCCESS in 27m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8c7954b4c9534c76831e41c896137f76 : SUCCESS in 57m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/63a2e48c32e2457ab25a0be893a733b1 : SUCCESS in 1h 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c90f7a95012841f8978e5522643918bc : SUCCESS in 2m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f51f69ad478b43ceb069e5929684e395 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25b86bc0d7144728bc8affcadc5b0eee : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79f168ea1d3d4fc385db9912eac0fcb5 : SUCCESS in 4m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0b1842c1ca554860bb5d0e9ba02043c8 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dcd8d0c706c740c4b72b70081c51475d : SUCCESS in 22m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/05b19fd4faaa41cd924291dbd853ec07 : SUCCESS in 30m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/08c3d9b985b4468884a3476d18f6df55 : SUCCESS in 38m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bbea0942cf3044b4a164c5fce8185322 : SUCCESS in 1h 33m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/19532b4b4b6445fd866f8e99c50d3e56 : SUCCESS in 19m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0b42436b0ba24a768aa96edf90c7e347 : SUCCESS in 23m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18575ab8cd294115aae5764f8791aaa2 : SUCCESS in 1h 14m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f70b4d78a86d4064aa1d836375bdc5e2 : SUCCESS in 56m 07s","accounts_in_message":[],"_revision_number":14},{"id":"1c0a0ade8d7e4e368b9f7972834d403881c43cbd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 13:56:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":14},{"id":"4dc4089099232dd53ccfde0f1a7166afec29c8da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 13:58:00.000000000","message":"Patch Set 14:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6057be035b2e4eda8bed764a3923ea07\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/373a3f22f09844b59e6c339408bac760 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"ed7c8399c4a2a43e9a4165db416b98f5be8d538c","revisions":{"a43fa1274e8348c7745baafc8cd33b69e1950e95":{"kind":"REWORK","_number":1,"created":"2025-01-31 11:52:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/1"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097097\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43fa1274e8348c7745baafc8cd33b69e1950e95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a43fa1274e8348c7745baafc8cd33b69e1950e95"}]},"branch":"refs/heads/master"},"3db1fe3c2f0aa92c6ca27500d916e6ae8acf0c07":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-01-31 11:54:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/2"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:53:48.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db1fe3c2f0aa92c6ca27500d916e6ae8acf0c07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db1fe3c2f0aa92c6ca27500d916e6ae8acf0c07"}]},"branch":"refs/heads/master"},"934d0cfad5d159a1fcd7ae51e35d6338ed811cdd":{"kind":"REWORK","_number":3,"created":"2025-01-31 11:55:02.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/3"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:54:54.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/934d0cfad5d159a1fcd7ae51e35d6338ed811cdd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/934d0cfad5d159a1fcd7ae51e35d6338ed811cdd"}]},"branch":"refs/heads/master"},"9278e618be811e01d590e8b4e5eeb6495944b14c":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-01-31 12:34:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/4"}}},"commit":{"parents":[{"commit":"9c04b5db8551c5dbcefec69f83e59dee967191fc","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c04b5db8551c5dbcefec69f83e59dee967191fc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 12:34:40.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9278e618be811e01d590e8b4e5eeb6495944b14c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9278e618be811e01d590e8b4e5eeb6495944b14c"}]},"branch":"refs/heads/master"},"7a19f18235d0a73bea229d43ed309ef7ba505e98":{"kind":"REWORK","_number":5,"created":"2025-01-31 13:29:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/5"}}},"commit":{"parents":[{"commit":"9c04b5db8551c5dbcefec69f83e59dee967191fc","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c04b5db8551c5dbcefec69f83e59dee967191fc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 13:29:13.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a19f18235d0a73bea229d43ed309ef7ba505e98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7a19f18235d0a73bea229d43ed309ef7ba505e98"}]},"branch":"refs/heads/master"},"b818e0a375340c0a9dccbf15100115aa2e1d1abb":{"kind":"REWORK","_number":6,"created":"2025-01-31 14:28:19.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/6"}}},"commit":{"parents":[{"commit":"9c04b5db8551c5dbcefec69f83e59dee967191fc","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c04b5db8551c5dbcefec69f83e59dee967191fc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 14:28:08.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b818e0a375340c0a9dccbf15100115aa2e1d1abb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b818e0a375340c0a9dccbf15100115aa2e1d1abb"}]},"branch":"refs/heads/master"},"473828ef9f90ea82edc1fc14d30585b01ab7e435":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-01-31 14:49:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/7"}}},"commit":{"parents":[{"commit":"ff55c29259ed985bc8e5a01a5d3f62011c138cd2","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff55c29259ed985bc8e5a01a5d3f62011c138cd2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 14:48:54.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/473828ef9f90ea82edc1fc14d30585b01ab7e435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/473828ef9f90ea82edc1fc14d30585b01ab7e435"}]},"branch":"refs/heads/master"},"40377aec9e28d018e4bd01cb52ee8bb7afea5ead":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-02-03 10:43:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/8"}}},"commit":{"parents":[{"commit":"59df8bc257b2154887af0878392aa44032e01fb2","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59df8bc257b2154887af0878392aa44032e01fb2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:43:38.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40377aec9e28d018e4bd01cb52ee8bb7afea5ead"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/40377aec9e28d018e4bd01cb52ee8bb7afea5ead"}]},"branch":"refs/heads/master"},"107f9cc63cbbde804e03da93eb762e2ecf1ce1d2":{"kind":"REWORK","_number":9,"created":"2025-02-06 06:16:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/9"}}},"commit":{"parents":[{"commit":"59df8bc257b2154887af0878392aa44032e01fb2","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59df8bc257b2154887af0878392aa44032e01fb2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 06:16:38.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/107f9cc63cbbde804e03da93eb762e2ecf1ce1d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/107f9cc63cbbde804e03da93eb762e2ecf1ce1d2"}]},"branch":"refs/heads/master"},"028b7eadcb5ea2ca64abbd65b3f09a803ed608d9":{"kind":"REWORK","_number":10,"created":"2025-02-06 06:24:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/10"}}},"commit":{"parents":[{"commit":"59df8bc257b2154887af0878392aa44032e01fb2","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59df8bc257b2154887af0878392aa44032e01fb2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 06:24:41.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/028b7eadcb5ea2ca64abbd65b3f09a803ed608d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/028b7eadcb5ea2ca64abbd65b3f09a803ed608d9"}]},"branch":"refs/heads/master"},"4acc6f635a6c44d0a46420f58a6f61f3edfffe8b":{"kind":"REWORK","_number":11,"created":"2025-02-06 06:31:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/11"}}},"commit":{"parents":[{"commit":"91d613174330c213d210498edd7db149a98ec376","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d613174330c213d210498edd7db149a98ec376"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 06:31:39.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4acc6f635a6c44d0a46420f58a6f61f3edfffe8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4acc6f635a6c44d0a46420f58a6f61f3edfffe8b"}]},"branch":"refs/heads/master"},"db1ffc821866555247af66fb90f33669e191139a":{"kind":"REWORK","_number":12,"created":"2025-02-06 12:42:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/12"}}},"commit":{"parents":[{"commit":"91d613174330c213d210498edd7db149a98ec376","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d613174330c213d210498edd7db149a98ec376"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 12:41:46.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db1ffc821866555247af66fb90f33669e191139a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/db1ffc821866555247af66fb90f33669e191139a"}]},"branch":"refs/heads/master"},"679e8b0f75c3d747cc80039ad1889be3f4e1aad1":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-02-06 12:43:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/14/940514/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/13"}}},"commit":{"parents":[{"commit":"13d6a664f063b5c65ce957bac21652111a13d6b7","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13d6a664f063b5c65ce957bac21652111a13d6b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 12:43:24.000000000","tz":60},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/679e8b0f75c3d747cc80039ad1889be3f4e1aad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/679e8b0f75c3d747cc80039ad1889be3f4e1aad1"}]},"branch":"refs/heads/master"},"ed7c8399c4a2a43e9a4165db416b98f5be8d538c":{"kind":"REWORK","_number":14,"created":"2025-02-07 08:49:31.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/14/940514/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/940514/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/940514/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/940514/14"}}},"commit":{"parents":[{"commit":"13d6a664f063b5c65ce957bac21652111a13d6b7","subject":"Install pycadf from pypi package","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13d6a664f063b5c65ce957bac21652111a13d6b7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 11:48:45.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2025-02-07 08:49:31.000000000","tz":0},"subject":"Fix permissions for ironic metrics","message":"Fix permissions for ironic metrics\n\nThis patch adds creation of metrics folder, set\npermissions  and  SetGID bit.\n\nCloses-Bug: #2097098\nChange-Id: Ic46b895775edf5e5fb2b637be49e2de1eb4adf36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed7c8399c4a2a43e9a4165db416b98f5be8d538c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed7c8399c4a2a43e9a4165db416b98f5be8d538c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
