)]}'
{"id":"openstack%2Fkolla~940518","triplet_id":"openstack%2Fkolla~master~I7c086ed5da0e3713ac15aaeaceb38f4a05b84bf0","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-02-01 21:53:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c086ed5da0e3713ac15aaeaceb38f4a05b84bf0","subject":"Bump pycadf version to 4.0.1","status":"MERGED","created":"2025-01-31 12:34:54.000000000","updated":"2025-02-01 21:54:49.000000000","submitted":"2025-02-01 21:53:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940518","meta_rev_id":"b715ad7c8ef8631c8a5e96a93dffd07c432c0529","_number":940518,"virtual_id_number":940518,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-01 21:53:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-01 20:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-01 20:38:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-02-01 20:38:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-01 20:38:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 16:19:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-01 20:38:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"4070dcb5e12bd22c214e7c44020935c00b57a7f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 12:34:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e68f858dc9ac546aba44978127acd203707ff166","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-01-31 14:49:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"af1ce349bc99196ebed3c834a3918657138407a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 16:19:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9029b19276ad41f581204915515ffa3b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1b57eee457c044d195dbba6bcd267cda : SUCCESS in 22m 49s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e4e386c1d35946f5be8375ba995cab96 : SUCCESS in 58m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d1629e446eb74722af11b03c024500cf : SUCCESS in 19m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f7f05c47314246ad98f339fceffa05be : SUCCESS in 29m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/393663fc50a84754b64e2c32c93332af : SUCCESS in 59m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/20a841a394a54bcd836e7c3d88a6335c : SUCCESS in 1h 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6d45de3d954712be9013121b15349c : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4207e984573e42758f1ed59d67130181 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e09b99c90562458495445c9494f0d90f : SUCCESS in 3m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f648ed97870f487c8afab15a7d79d9e0 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e114d45e3e85404184dbe89f6fb138ae : SUCCESS in 4m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/726769a14aff4b429296b1613c166593 : SUCCESS in 2m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c916f5ad1877403c85d894f11f871cb9 : SUCCESS in 22m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d00ba6a96c61454c99e166dbaafd5d70 : SUCCESS in 31m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/af3160a7d7bf450aaf3ef165dd5b13e9 : SUCCESS in 1h 02m 09s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/77a38e5c7ac14fe7a963cdda1d7b105f : SUCCESS in 1h 28m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/44c19e4b19fb4880809256806d0a86f0 : SUCCESS in 33m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cbd7ec401eb64be298a2eb3766a4742a : SUCCESS in 23m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6325d6b7141444f19f6d7e6e058e8f71 : SUCCESS in 1h 05m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a65b365c9a5b4b0b9f0be602ae89de79 : SUCCESS in 1h 15m 35s","accounts_in_message":[],"_revision_number":2},{"id":"1d8fc1e02a1a065ec23d7d8c037b2879a9e32dd1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 17:52:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fcbdb6ec89d4e3eb6a802f1dc2eead8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/13e1a98f46fd417285c6f52570d1f3d4 : FAILURE in 2h 13m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9009366158a14254b8452d840ba64140 : FAILURE in 2h 01m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe3fe14c10fd4fd986087a72a4c47b49 : SUCCESS in 2h 08m 34s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8b2fdfd1bc5f42f69eeb13a388d6a280 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eec3f95237fd40a0962fc8351191707f : SUCCESS in 4m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/96a6cbb5552445d39915a7a4a8df78d8 : TIMED_OUT in 3h 01m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8570286597fe45468d26547cfe81bdcc : TIMED_OUT in 3h 02m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d139b2adf7b0b50fca205126e9d45660aea29b68","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-01 20:38:49.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bbd495b1e4f9d140632fd0cd17403bc1dedcaa0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 20:39:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9143f3932ee4de2aefa7a02d8ccd533087f25088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 21:53:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae8c1825037d44918fc344a36df153e8\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e767c890f63544cab87082b8c3d0781b : SUCCESS in 25m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5dce5de176894beb93deed8a85b05aa5 : SUCCESS in 20m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/44f119f7bcba44a99973b05c04bdde37 : SUCCESS in 49m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9e04cf5663f049d1bee1ebfa6122b95d : SUCCESS in 52m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0187b1001f6d421c96b5aec59f1dfbce : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bba18cb55a6a474ab6def36b1f330825 : SUCCESS in 2m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8b06d51a2e74e11b0dede4295e285c4 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3636aedb8c413d8ba9fc766654b15e : SUCCESS in 5m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9ac80631dbbb4e448d00d919a4e9405d : SUCCESS in 2m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/968b0531b0ce493c9832cb29fd91b925 : SUCCESS in 24m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4911615724d64ec7bc2f8eb654008016 : SUCCESS in 18m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6d68f374a5f9497cb7de8228753403ab : SUCCESS in 52m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3dac6a7d09004fb7900e36e4cded5630 : SUCCESS in 1h 09m 19s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c06832afe70044ddb60f2046744eec1e : SUCCESS in 18m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b7696de8354f405999f0a9afac8a2a57 : SUCCESS in 32m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e056ff5faff8446c8e3be002a1df7427 : SUCCESS in 35m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/41bb0084acfd4735a1a4fef48e37f250 : SUCCESS in 1h 13m 19s","accounts_in_message":[],"_revision_number":2},{"id":"545b6f0708afd4e54db9a524b341a63159154ec9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 21:53:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b715ad7c8ef8631c8a5e96a93dffd07c432c0529","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-01 21:54:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2542a9df13b34bae8471aabb70e37dfa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0bb05d73714d2a932fd3c6b2f3ace2 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ff55c29259ed985bc8e5a01a5d3f62011c138cd2","revisions":{"9c04b5db8551c5dbcefec69f83e59dee967191fc":{"kind":"REWORK","_number":1,"created":"2025-01-31 12:34:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/18/940518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/940518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/940518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/940518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/940518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/940518/1"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 12:34:18.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 12:34:40.000000000","tz":60},"subject":"Bump pycadf version to 4.0.1","message":"Bump pycadf version to 4.0.1\n\nChange-Id: I7c086ed5da0e3713ac15aaeaceb38f4a05b84bf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c04b5db8551c5dbcefec69f83e59dee967191fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9c04b5db8551c5dbcefec69f83e59dee967191fc"}]},"branch":"refs/heads/master"},"ff55c29259ed985bc8e5a01a5d3f62011c138cd2":{"kind":"REWORK","_number":2,"created":"2025-01-31 14:49:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/18/940518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/18/940518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/940518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/940518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/18/940518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/18/940518/2"}}},"commit":{"parents":[{"commit":"5e24daa33269f11760962f51abfeed0f0ecad486","subject":"Merge \"cinder: Fix log dir permission\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5e24daa33269f11760962f51abfeed0f0ecad486"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 12:34:18.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-01-31 14:47:38.000000000","tz":60},"subject":"Bump pycadf version to 4.0.1","message":"Bump pycadf version to 4.0.1\n\nThis patch also removing copy of\nceilometer_api_audit_map.conf file as it is\nnow removed [1].\n\n[1] https://review.opendev.org/c/openstack/pycadf/+/930971\n\nChange-Id: I7c086ed5da0e3713ac15aaeaceb38f4a05b84bf0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff55c29259ed985bc8e5a01a5d3f62011c138cd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff55c29259ed985bc8e5a01a5d3f62011c138cd2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
