)]}'
{"id":"openstack%2Fkolla~940605","triplet_id":"openstack%2Fkolla~master~I022aa4c403f08385beb83b5d0c81f8abb42eb4a7","project":"openstack/kolla","branch":"master","topic":"ironic-fix","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-02-06 05:42:25.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-02-07 13:56:13.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-02-07 13:56:13.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-02-07 12:03:57.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I022aa4c403f08385beb83b5d0c81f8abb42eb4a7","subject":"Install pycadf from pypi package","status":"MERGED","created":"2025-02-03 10:42:24.000000000","updated":"2025-02-07 13:57:19.000000000","submitted":"2025-02-07 13:56:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940605-ironic-fix","meta_rev_id":"02a777f56dee17f80074551a0b822e390d30e2a5","_number":940605,"virtual_id_number":940605,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 13:56:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-06 20:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-02-07 12:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-02-07 12:03:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-02-06 20:20:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-02-07 12:03:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-06 20:20:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-03 11:02:52.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-02-03 11:02:52.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-02-03 11:02:52.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-03 13:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1d6bc0116afdb1f52883ac3c6170babc2d88827d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-03 10:42:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae5814e840679235784c2815a85b81d465c61a37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-03 10:43:00.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4a7b7ef17e2ac975f71ca9ea95fa5d3a606363ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:31:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e2f6880d8954f6ba34503c1b7c452aa\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/318def9bf4c049d6ad40efb51df01f9f : SUCCESS in 28m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/42a944e12fa64d648c07e2eae9ebb9b2 : SUCCESS in 1h 04m 29s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ec3ed6bb6d1249359ad83fc9c74cc37d : SUCCESS in 21m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d309726365884490af4f87a3c3f0b5c1 : SUCCESS in 23m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9c1d48c33ffa4333afbf8481d4f343c3 : SUCCESS in 1h 01m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d7e0e53a7eb24cd9a61e26ec80b54746 : SUCCESS in 1h 26m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70831fe088df429b84538b28c6a77e22 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08782a89ed924061a7a07e0b82a82858 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/11b99cbdcb3d479ca1b0fd276fc791f7 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbcd5013000c4875aa8f455375588146 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9745a6dbd4104eab8862dbc07737af52 : SUCCESS in 5m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/91cdda982e73470ca9df3e739c8c48e8 : SUCCESS in 2m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5d2f63c946cc459fac8060cb525a06bc : SUCCESS in 32m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/48c0c1824b564284b2e684bfdea26be9 : SUCCESS in 30m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d45d646d301a43c3ae6fbeac71439588 : SUCCESS in 1h 06m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dd34568c167d4400b330a34e559633ea : SUCCESS in 1h 34m 31s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d5b7c674eb3b4d6eaea8e0c4d6325b6b : SUCCESS in 1h 45m 15s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/72b4d4c8386d45bab08b3627c14e8370 : FAILURE in 1h 37m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/149dbd76507644eca9f0ebc5ac53aef4 : SUCCESS in 1h 15m 14s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/f7333f53a46a439eb180c0e6be232990 : SUCCESS in 1h 02m 50s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c39f1f9a08cc4e43955a4d579b5e195a : SUCCESS in 1h 14m 18s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/65946ed4a15c44febfdfad3532cfeeb2 : SUCCESS in 49m 02s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/c9f9de66df7a4607b6580f77f5b88aab : SUCCESS in 58m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7b856e19dcf444bbb9c6dfe54f1edd20 : SUCCESS in 28m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/05a6f03eafff438e9c24e88aee77b32b : SUCCESS in 30m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae417097ce33461ebdf3339bc07c83d6 : SUCCESS in 1h 07m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a94777930b4d4573ba76b5463b1083a9 : FAILURE in 53m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5b1b7d7a417944d684f9af30ff335778 : SUCCESS in 1h 04m 43s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/03282d013e154a42b958551a96f02bf3 : SUCCESS in 2h 41m 46s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/96772ce408404bb9ba9893eea23811e5 : SUCCESS in 1h 12m 03s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/33a5e64621b94830a5872b209bddea42 : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/4cc579ef784e41e49ba145ebd5b17789 : FAILURE in 1h 36m 33s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/474dded9c1de4a26bc0228b0c092d602 : SUCCESS in 1h 15m 47s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/34d57ac71c994506bcb321efba08caaa : SUCCESS in 41m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"830f2c826bac32d5b50725378eeb34ea7950c00a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-03 13:44:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"36ec2f5cfeb65d2f8c8df50e552fdfe03ca3061b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 13:50:05.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eedc99b117304e4aa006cf357e6748a7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2bc54bd0ca374d2085631cdfc5013136 : FAILURE in 1h 49m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8183e2d447740e5b0b6e13712d3c722 : TIMED_OUT in 3h 02m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d72e2989524d454383088615f8c2b002 : SUCCESS in 2h 01m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b4911554ed954ef4819d2cab6e08efbb : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b08993ac247c4d16825b99a6003b5008 : SUCCESS in 5m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d2580740e573418a8168d6af6ba9d56e : FAILURE in 2h 50m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e2e85b281e4647ce9c9dc1ca1bff0348 : TIMED_OUT in 3h 01m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c4c3ddeb04cf92d83015c93a99acf674f8898a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-03 16:11:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/651b22108b1a462891048c7e6281280a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dc040bc9200f4a978f3bf9a1312e4103 : SUCCESS in 37m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d6a5ce7fe55041479fe2efd863c1a10f : SUCCESS in 1h 07m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d07cd50e152a40c7a217c8a2fc720bac : SUCCESS in 27m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/171efe829876497cb57e5fe8f8e2ef30 : SUCCESS in 29m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9c2b6c82c30647a4976413059c53dc71 : SUCCESS in 1h 10m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/66d01d724e3846a6905f38bc6a004dd5 : SUCCESS in 1h 39m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d916cdef025747b2aa1c7866e437be6d : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0357debb708a4c96929774d52c5feaa6 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8dbc5def591640c4b874da31f0ab736f : SUCCESS in 4m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5c3b69b9d124ca28c677919854e4211 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb33244f04d4e76bfe5bbfef542cb4f : SUCCESS in 5m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24692003d31f4ae8b30c403cf55c170c : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f0c3cc0f09454943930b498aac746a8b : SUCCESS in 22m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/075fc5cc56be4bd8bc8c3c827c7e179a : SUCCESS in 30m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3cf4f805e5bc4a61939faa123adea717 : SUCCESS in 1h 09m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c915d402de894ef59122844ce5235961 : SUCCESS in 1h 27m 40s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/cd3ffabf0d1e4921bc0c646440230367 : SUCCESS in 1h 42m 55s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b02b113c5b2b4aa18113f9681347c924 : FAILURE in 1h 35m 18s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/899b5d626d1c44feb725c37c23accc7a : SUCCESS in 1h 12m 21s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1efd36732c8842cf9399b5bfdf221d94 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/239013c35aa042e8bc1adf614d91b839 : SUCCESS in 1h 09m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f87722d85f7344fa99c755dbf0ecc346 : SUCCESS in 1h 20m 25s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/b9d2f028025d423fa447431529cf10e0 : SUCCESS in 52m 39s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/479acd4cc4ce4c2cb987ba63920e11d2 : SUCCESS in 21m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/01747fe649844ee1acbdc55c5ed55650 : SUCCESS in 27m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/009d254ea5994f918cb51a6c755b6539 : SUCCESS in 1h 05m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1ac31ed93184da89cd0a3a2d86ae2eb : SUCCESS in 1h 23m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/445f5de55bbe424b8a6eab13a41fd2d8 : SUCCESS in 1h 44m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/928bb200a1d24a85867ed449a9da084d : SUCCESS in 2h 24m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/45a4e126fe1c40b684b01b8e64b199de : SUCCESS in 43m 29s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/79a822aa90b34fea949e864b410216fd : SUCCESS in 1h 03m 52s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1047ed03a66147508f4f4cfad92f0e81 : FAILURE in 1h 00m 09s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ed635cd9c029467abfb699b09aa5b35a : SUCCESS in 1h 31m 09s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/9b7bc12be7114f8dbf6af0ac5313cd6a : SUCCESS in 35m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d95fbbde76f82052e9dfa2c628e29afdca57a1c7","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-06 05:42:25.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"bca7c6b6b337c26e4fbfbdab13e4c14ccea07980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 06:31:48.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent tree, and commit message as Patch Set 2.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"07017e25a369533f014bb67934163ca59a3110b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 09:20:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f35f58ebbba499692ae8dff2aa7ba6b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/374af32a310f429f84e4d8fc6875b2e6 : SUCCESS in 47m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a47759bb196b4bf9bdc4e6748759351c : SUCCESS in 1h 22m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ebb141e5cc774b47ace33b766039df91 : SUCCESS in 26m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/aa2edf59d8dc4e1e8352a4417054050a : SUCCESS in 28m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dbd4154d17984d4db9d5db28b056283b : SUCCESS in 1h 13m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/66ae0bfb1e7944d3952b9d285715fbd4 : SUCCESS in 1h 34m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d481185c2395413e8c8be6acceebe26a : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5463d16b47a740b4bf01764021f40022 : SUCCESS in 2m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f56bb0924eef447e85770b3be1b1ddeb : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d741bbe90304c2c8f4e850270b18f57 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed215c32f7a446f89a716536cc9a34a9 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b2d9b15cd4d54d248793f915cf6743b8 : SUCCESS in 2m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f6ebca0a1d4c497b84135c93683fc26c : SUCCESS in 23m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e4602e0914cc488dac89ab865b88bd32 : SUCCESS in 37m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d6cbb026d24f49479815fcf49d349e54 : SUCCESS in 41m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a4e8e877a3e54480bd79b175929891c7 : SUCCESS in 1h 28m 05s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5ced8e37033a453aaa9fb1dc0c40336a : SUCCESS in 1h 50m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fb670383038c496698da6f18dc3c821b : FAILURE in 1h 47m 21s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/42f575aa7fd34612b257e457e9050f24 : SUCCESS in 1h 32m 08s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/81546995ce074445b108490ce9d20126 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a5999a29c4b04058832cf1d5e79abc95 : SUCCESS in 1h 15m 58s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c183db52141d4e89b5c0ea1d544623a9 : SUCCESS in 1h 36m 32s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/a1e206e4f2ab4667a197809112ddf16e : SUCCESS in 37m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/067d7ead7f62475088c96c224ce1beba : SUCCESS in 32m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e8ddbf2b7e37463ab7db3cc3e83b8d43 : SUCCESS in 36m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f31eb33b69a34e228b2b9dba8354a8f0 : SUCCESS in 1h 07m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b18ee8810a5f40c08c418e1b801674bf : SUCCESS in 1h 29m 26s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/296fedab0d8c4caf8fb1c49d12d59ded : SUCCESS in 1h 58m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eac85328aab347e18e3254da391fa43f : SUCCESS in 2h 27m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/90717b38ee7a42fb86c19620f7f3cc80 : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/5e51dfe9922c433aaf5a6a3817dd0f3a : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0abf02fd6b4c497f8606c7d145c23144 : FAILURE in 59m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1842a236b0bf4809b5ad29b33b89a30b : SUCCESS in 1h 20m 10s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/a2b2ad9eb5da464682aa864e3fd6fc41 : SUCCESS in 42m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9e91ec98fcd3ff269632151a2510565d1eaaa648","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 09:39:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1acbf9b608b453cb51dffc0add2feb9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/55c65e2750704a2395422d10489c8cb4 : TIMED_OUT in 3h 01m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eadbfd90d4a3492e820cdc787cf76cb3 : FAILURE in 1h 51m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64d38aaa805e4ce0b8dee407338a2c11 : SUCCESS in 2h 02m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/610c8f79d4ff4233ad1f107588672ff3 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2bef0ba2387543bcadefc54693da87ba : SUCCESS in 5m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d811e18bb5e84638aec1442733778663 : TIMED_OUT in 3h 01m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/71211c7aa94e4d79b9a3baa9fa999d4f : FAILURE in 24m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"77386a0518019d00fdab61129e48d603dec23df6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 12:43:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"cfe2c739c07cb8d1c29db4192105ad132e63c577","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 15:58:16.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e5a1af6501942ed93ed238e8f51dbe7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a99ee815d53541d0bbef40e9149bcbcb : FAILURE in 6m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a9f5d172a5c3480bbff16b7849938cbb : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b86b5c96365c460bab395124a2ed86a7 : FAILURE in 6m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4db821f5327c4b2daab1f3708dc6dbd9 : FAILURE in 5m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/08fb91401bdc4644bebe5740318b223c : SUCCESS in 1h 16m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7320934463494edf979d765b71055b1d : SUCCESS in 1h 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0490c8eb011e4be39ea7d1bd501e8278 : SUCCESS in 2m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63a068c2d9aa44de91f669950386b7bc : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/582d3adcde404e879a58b14de4f7608a : SUCCESS in 3m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef7e880ad66146e3bd0919108aa3b27c : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/189680683f4046b58102c5ea3b577f11 : SUCCESS in 4m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7cb64099b5eb42e2b4101284e9e68ec0 : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0dbcb854a22c446db437caaa71ed9adb : SUCCESS in 25m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e8f5883387604bd0a7682d4f28afbce7 : SUCCESS in 33m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/73f6d1118c4d48ebb9d98e651f1b9aea : SUCCESS in 1h 07m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8f109e6460d2445bad7ea3c2e17de441 : SUCCESS in 1h 50m 40s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/c10efa27ddbc48b8bf0133e3be0e7a20 : SUCCESS in 1h 40m 29s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8f05023466a648788d7db854fa002c65 : FAILURE in 1h 41m 22s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/329aec45dab14e438554962f0a860e1b : SUCCESS in 1h 17m 23s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a02b626e915b423d9e7aa8f4e13eed38 : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0e335abfd8c141ebacf8f9b4c4b46da9 : SUCCESS in 1h 21m 59s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2f44342d4920409c9b90cd344872a41c : SUCCESS in 1h 21m 16s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/e7ba2cd72999467b8ea70a31469ec1cb : SUCCESS in 1h 08m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c2102f5d195345008ce35a337e63ec11 : SUCCESS in 26m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6c766ce4a01e437d87082bfea196d2c6 : SUCCESS in 26m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0906ee93a21b4856ab77737206ccbfaf : SUCCESS in 1h 08m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/dabe1bd5ef9f4d6cbda3ed52dce55624 : SUCCESS in 1h 22m 28s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/55d757a3aee24b9596cffdcc372770f1 : SUCCESS in 1h 46m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2c042549cf8e4fc8b1a658b930160568 : SUCCESS in 2h 37m 53s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/95fb098437f74710b206bc858df50ffe : SUCCESS in 1h 31m 14s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/f7b2634963264355bb496d4ff32c8b3e : SUCCESS in 58m 58s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/1b5357cb0726416f95ef6d1dddb10b00 : FAILURE in 57m 12s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/aba5b2386b3f4b7285a6c1c392d435a7 : SUCCESS in 1h 13m 43s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/0ce23fd4d5894d7bbeca14f55f27bc5a : SUCCESS in 42m 38s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93cc0440368f3e8451961dd0e87f76d5e16ecc79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 15:59:21.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b04d98ec96444cd599165bebc5f1d411\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f79e1aea9d584bdeaf5fbf1329343459 : TIMED_OUT in 3h 03m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6a5c1a99fce44f26984a0d50c207912b : FAILURE in 1h 32m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/81d4f3a9df6a40cba036d3fa1c9e79dc : SUCCESS in 2h 55m 53s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5956fa5673254f94b41f5e858664c84e : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9186526ffbbc41efb044e14e58a143b2 : SUCCESS in 5m 27s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/61ec8afa45294e82b4c298349dd12d17 : TIMED_OUT in 3h 01m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b5bffd4023ce4fa3b1a558a90820a832 : TIMED_OUT in 3h 01m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35060bf2b94b37b6f22bd69fccbdeda7ffa3b81f","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-06 20:20:11.000000000","message":"Patch Set 4: Backport-Candidate-1 Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9e91d2bdae9f8bde9f8ee2f7f0a058615fff5fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 22:29:30.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6e474a2cb34789b4a47be583962fc8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/652a00f7e5704e3a8805823abbe27303 : SUCCESS in 27m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b419fe45269641c883b95d41eb1b0f7c : SUCCESS in 40m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ae635bfd038642e1acb60cfd24fcf386 : SUCCESS in 13m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49d2d29bf0254d5488a0cc9edf04bc11 : SUCCESS in 29m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2b6681d622db467d80d5647810186302 : SUCCESS in 1h 04m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4813858dd9ed4685ac5f1a66de0f00ab : SUCCESS in 1h 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12e5876365a54c1b9a6e35753c8fa5ae : SUCCESS in 2m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98290b61d8ec454089b5d50fa8f5ec5f : SUCCESS in 3m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/344587e26b544a859b2631dcf5148454 : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3060654e08248488aca2e8d115bf9f4 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea2c64c16308467881ba6bda0c6bd23b : SUCCESS in 5m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a9e9ac92a6034e008cf2f216214f0fa8 : SUCCESS in 2m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ad28642dabfa461381204c63b7c53fc2 : SUCCESS in 26m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/42176841693f437fa0dba4040b314ead : SUCCESS in 34m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/55d0f589dce745e18b50f2bd4906ce7f : SUCCESS in 1h 12m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e2685f3959774702a04dbedeeb44511c : SUCCESS in 1h 40m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/fd7153a7912e453c89e9cefc0f4a399c : SUCCESS in 1h 58m 53s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/888a20547bb54456b13898a00855bdab : FAILURE in 1h 25m 50s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6cdc0550bb734a6bae3a6b4af8743b42 : SUCCESS in 1h 22m 37s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c508128a906842088c30d605faa51b1f : SUCCESS in 50m 02s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a8c2f43021da431c8b6991d3fdb48c0a : SUCCESS in 1h 09m 52s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a7f8e2b23ea347929709765032d0acbc : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-rocky9-kvm https://zuul.opendev.org/t/openstack/build/80ee953114424b11a03529e5347440dc : SUCCESS in 36m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/928ed4017188458fbd6358036c495470 : SUCCESS in 25m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/89b62bb1a26d4ebbb1aaca6cd3b75d07 : SUCCESS in 34m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8167a0fc64474408aa95964187492bcf : SUCCESS in 38m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/aebfcf0c36e04ccf89fdf1cec1231c2e : SUCCESS in 1h 24m 19s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6ce495eda209483bafb075eaf9a447fa : SUCCESS in 1h 49m 02s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/22df38fe75a6442ba3ce83d38dfa9fc0 : SUCCESS in 1h 35m 19s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8758ac9837eb4cb1a79823cf4f1e72cb : SUCCESS in 1h 04m 45s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/be62505b9e534a27bca3783b46891c19 : SUCCESS in 57m 57s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/a7a102dadd3e4a3e8863c0c8fcd00c69 : SUCCESS in 1h 08m 43s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9f7885e697864656a80415a7dbb48d8c : SUCCESS in 48m 36s (non-voting)\n- kolla-ansible-ubuntu-kvm https://zuul.opendev.org/t/openstack/build/203cefd9ef1e448cb31a48e729d1c39d : SUCCESS in 33m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3441271db3dffadd4c2b0a3d034e6357991214e8","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-06 22:39:22.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2ee364f767b4b4e29613168f77987a015f8c7089","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-06 23:22:32.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f9b58986de041c68b3107c0ffcb3049\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fa35d8a54c3c4c34aa6e51586b4848a5 : FAILURE in 2h 28m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/774e6b8333094ec4b158fbf38733de40 : FAILURE in 2h 42m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dc95092feddf430498525929decf6e2e : SUCCESS in 1h 53m 47s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3577f9008b964d3cae3df1d573e85076 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a04943b95ce47d2b56e95716d6a08eb : SUCCESS in 5m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/14a10cb004d44b9e8d96d25d6c6d12bc : FAILURE in 2h 01m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/35ddfe6b0762418f93b81f8030bd2208 : TIMED_OUT in 3h 01m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fdc1eadc0e42923980c19e504dd2f8b223c57cb0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-07 11:46:59.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"375c7c9786ef8378f6f24120d8aeef35cbdbdfbc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-07 12:03:57.000000000","message":"Patch Set 4: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d7a9c1e395711826745dc8aaf7cfee820210abe6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 12:04:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"6743156d33325be5ee996112145c10ef5fcfef5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 13:56:12.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec6926904b6f44299633a77e01ff456b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1fef721c27d498bb3b5a08976fbf4d0 : SUCCESS in 14m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0ac11ff4bc4c474a8208c6b061970ef9 : SUCCESS in 24m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b5424e0b51264017bcc451b6fb8eda0c : SUCCESS in 1h 03m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8916e09316c14d37bf5ebfc4cce8e215 : SUCCESS in 1h 28m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6ae4b81f0441e69dab138022273774 : SUCCESS in 2m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6cd9c7e53ce34002b682abda67d4ac8e : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f3866ad82fc43c8b43516a2aecada39 : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc98dda85b3f44b092b45e5a405f18a0 : SUCCESS in 3m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bfde8475225b415698680e9987c01a72 : SUCCESS in 2m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f0e482ab710b4786af801e74312de30e : SUCCESS in 26m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/71606e7219174887b5dd1b74dc5cc2cf : SUCCESS in 36m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/61c0b674eb65451c9ab22c122b77697a : SUCCESS in 37m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/68365ffb27d4441a85347544ff026874 : SUCCESS in 1h 39m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/59fe2fabb03946399300608b9a40e6ac : SUCCESS in 14m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/093b480a21c4441089c3d912dbb34da7 : SUCCESS in 26m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f80e1fcf023d4822bf1a013a4063fd92 : SUCCESS in 37m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2bf9acb7aaa8407aa01a07b7af73ca28 : SUCCESS in 1h 39m 51s","accounts_in_message":[],"_revision_number":4},{"id":"3e6fa60b2b5f26d2b8db841f7a4118d6b96e775c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 13:56:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"02a777f56dee17f80074551a0b822e390d30e2a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 13:57:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf0042eb5cf24924906bbbaca7e94194\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24ce9a6bcc794787ab0cefc77ec73854 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"13d6a664f063b5c65ce957bac21652111a13d6b7","revisions":{"b3c1b273225d23f3e6b456ec15363e81d1d90c9b":{"kind":"REWORK","_number":1,"created":"2025-02-03 10:42:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/940605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/940605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/940605/1"}}},"commit":{"parents":[{"commit":"ff55c29259ed985bc8e5a01a5d3f62011c138cd2","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff55c29259ed985bc8e5a01a5d3f62011c138cd2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:40:04.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:40:04.000000000","tz":60},"subject":"Instal pycadf from pypi package","message":"Instal pycadf from pypi package\n\nPreviously, the `pycadf` package did not include\nconfiguration files, so they were manually installed\nfrom source. However, this was fixed in pycadf\u0027s upstream\nin [1], allowing us to rely on the package itself.\n\nAdditionally, the `ceilometer_api_audit_map.conf` file\nhas been removed, as it was dropped in pycadf\u0027s upstream\ncommit here [2]\n\n[1] https://review.opendev.org/c/openstack/pycadf/+/936741\n[2] https://review.opendev.org/c/openstack/pycadf/+/930971\n\nChange-Id: I022aa4c403f08385beb83b5d0c81f8abb42eb4a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3c1b273225d23f3e6b456ec15363e81d1d90c9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b3c1b273225d23f3e6b456ec15363e81d1d90c9b"}]},"branch":"refs/heads/master"},"59df8bc257b2154887af0878392aa44032e01fb2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-03 10:43:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/940605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/940605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/940605/2"}}},"commit":{"parents":[{"commit":"ff55c29259ed985bc8e5a01a5d3f62011c138cd2","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff55c29259ed985bc8e5a01a5d3f62011c138cd2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:40:04.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:42:49.000000000","tz":60},"subject":"Install pycadf from pypi package","message":"Install pycadf from pypi package\n\nPreviously, the `pycadf` package did not include\nconfiguration files, so they were manually installed\nfrom source. However, this was fixed in pycadf\u0027s upstream\nin [1], allowing us to rely on the package itself.\n\nAdditionally, the `ceilometer_api_audit_map.conf` file\nhas been removed, as it was dropped in pycadf\u0027s upstream\ncommit here [2]\n\n[1] https://review.opendev.org/c/openstack/pycadf/+/936741\n[2] https://review.opendev.org/c/openstack/pycadf/+/930971\n\nChange-Id: I022aa4c403f08385beb83b5d0c81f8abb42eb4a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59df8bc257b2154887af0878392aa44032e01fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/59df8bc257b2154887af0878392aa44032e01fb2"}]},"branch":"refs/heads/master"},"91d613174330c213d210498edd7db149a98ec376":{"kind":"NO_CHANGE","_number":3,"created":"2025-02-06 06:31:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/940605/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/940605/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/940605/3"}}},"commit":{"parents":[{"commit":"ff55c29259ed985bc8e5a01a5d3f62011c138cd2","subject":"Bump pycadf version to 4.0.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff55c29259ed985bc8e5a01a5d3f62011c138cd2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:40:04.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 06:30:57.000000000","tz":60},"subject":"Install pycadf from pypi package","message":"Install pycadf from pypi package\n\nPreviously, the `pycadf` package did not include\nconfiguration files, so they were manually installed\nfrom source. However, this was fixed in pycadf\u0027s upstream\nin [1], allowing us to rely on the package itself.\n\nAdditionally, the `ceilometer_api_audit_map.conf` file\nhas been removed, as it was dropped in pycadf\u0027s upstream\ncommit here [2]\n\n[1] https://review.opendev.org/c/openstack/pycadf/+/936741\n[2] https://review.opendev.org/c/openstack/pycadf/+/930971\n\nChange-Id: I022aa4c403f08385beb83b5d0c81f8abb42eb4a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d613174330c213d210498edd7db149a98ec376"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/91d613174330c213d210498edd7db149a98ec376"}]},"branch":"refs/heads/master"},"13d6a664f063b5c65ce957bac21652111a13d6b7":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-06 12:43:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/05/940605/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/05/940605/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/05/940605/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/05/940605/4"}}},"commit":{"parents":[{"commit":"dbdc53fa8f534948d0221b1359a8eeda60be6509","subject":"Merge \"Prometheus update all exporters to the latest versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dbdc53fa8f534948d0221b1359a8eeda60be6509"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-03 10:40:04.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-06 12:43:24.000000000","tz":60},"subject":"Install pycadf from pypi package","message":"Install pycadf from pypi package\n\nPreviously, the `pycadf` package did not include\nconfiguration files, so they were manually installed\nfrom source. However, this was fixed in pycadf\u0027s upstream\nin [1], allowing us to rely on the package itself.\n\nAdditionally, the `ceilometer_api_audit_map.conf` file\nhas been removed, as it was dropped in pycadf\u0027s upstream\ncommit here [2]\n\n[1] https://review.opendev.org/c/openstack/pycadf/+/936741\n[2] https://review.opendev.org/c/openstack/pycadf/+/930971\n\nChange-Id: I022aa4c403f08385beb83b5d0c81f8abb42eb4a7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13d6a664f063b5c65ce957bac21652111a13d6b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/13d6a664f063b5c65ce957bac21652111a13d6b7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
