)]}'
{"id":"openstack%2Fkolla~941240","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I06e61ab388d9ecfa78e5d1fe1ae9cb8602d1cd36","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-19 17:33:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I06e61ab388d9ecfa78e5d1fe1ae9cb8602d1cd36","subject":"Adds missing nvme package in cinder-backup","status":"MERGED","created":"2025-02-11 13:13:33.000000000","updated":"2025-03-19 17:34:51.000000000","submitted":"2025-03-19 17:33:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941240","cherry_pick_of_change":936191,"cherry_pick_of_patch_set":3,"meta_rev_id":"5041030661b78371a81fbf270a34ff38964ab9da","_number":941240,"virtual_id_number":941240,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 17:33:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-19 14:30:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-12 06:54:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-03-19 15:31:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-03-19 15:31:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 13:13:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"state":"CC"},{"updated":"2025-02-11 15:37:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 06:54:26.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-19 14:30:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-19 15:31:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"94a9dbbecf0c30413038617f857269a6e2a381ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-11 13:13:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0e2e8be510033ba05ab22497aef48bcf0243eb49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 15:37:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a5403d14f0541b9874f47d1acc8ea56\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0f67e90ad10c4cfab7566d0d4c1982bc : SUCCESS in 25m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e5a2880d826b4958975f7d5ad5893fe6 : SUCCESS in 1h 18m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/abe55d21b2b94b9f88c60da05fff2555 : SUCCESS in 21m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6cc91ab1e5214109a644deef3ed3100f : SUCCESS in 30m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22ebb134095f491c939267290f9d2963 : SUCCESS in 34m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d6cc84b825d14658a457b213f2eddacc : SUCCESS in 1h 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0986e5a5c1e84e699c56edda613b30cf : SUCCESS in 2m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d2733a3e074b48568e4e6eddc4516706 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/75aef693c50649ae962050d769c1f064 : SUCCESS in 2m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6dd06575cc5841d2af4e88df773e0b8a : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a70ff9f5e0f4f6e9d15a445602d64df : SUCCESS in 4m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/48b91f75be54403b864886e77e80a674 : SUCCESS in 28m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bd765b85426c42fbb922328114cc036c : SUCCESS in 16m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bb57f3c77ef74fdbb61ba6602f0740a6 : SUCCESS in 51m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/123aa759499d4fbc9dba2b5d3f26143e : SUCCESS in 1h 36m 24s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/01a805a651214c3b9da5df853e9e9f48 : SUCCESS in 1h 45m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ab26ea02c0ad4fb5be035603ef229501 : FAILURE in 18m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fd0cdfbbb90443db84ec96924e4313a2 : SUCCESS in 18m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dfa98a278a55427380e1725bbe3c0814 : SUCCESS in 51m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18861d51a2ec4d239c94304e9f0a4389 : SUCCESS in 49m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bbdcaa5e44a54a9090077927a078675f : SUCCESS in 1h 29m 00s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b07f1cdf5ef6475ea330547ca7d78d6d : SUCCESS in 1h 29m 41s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a871b41e4b1c4554bdeb09f8dfa42a62 : SUCCESS in 1h 16m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e9284b7bad3f9be73e891e547dfe3f1b00df423","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:30:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/709e6d5415df4ebabf97452756c58c7e\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1e9be98bf6c41d1a5b6093c6e258988 : TIMED_OUT in 3h 03m 16s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7a19fcc89a164d19abb09b65687e12a1 : SUCCESS in 2h 31m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e7124823e1ac47bfa8cbd462563a4033 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73e9334849604225b4137e65372ecfd9 : SUCCESS in 3m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5946df75dd1f7ed8ff4cc880390f8ae3c1078198","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-12 06:54:26.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ffb5cbcb00f49c2ff7eb063554bb2a2215ccdd44","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-19 14:30:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a89e985cd06f7a81e9be9a3836c310a414c0e6eb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-19 15:31:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b3e31fe1cd6890477633d8003a342731dc804bbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 15:31:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"08b1d3e74a5742300374d59bcfbb8136e71b9fd2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:33:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deab91447d40493a93db370f61d12492\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5d5a6674bb3f43cfbabe99ae9fdda262 : SUCCESS in 16m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/545037b02121407db5c22cd54a59c4e4 : SUCCESS in 33m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47fd0a97b9c84851875e0192544ea98b : SUCCESS in 57m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0f659c61a0f8453fa870a0c0bed76a38 : SUCCESS in 1h 38m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6edef2bc8c1e4457b34b00539fe281ab : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6349423e633b4a1d80e4efc66584f5a3 : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7708cd4fe0c46e6a3287c7cfd777a0e : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e7dfe61f4a464d863e1dd90817db12 : SUCCESS in 4m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/16ab13b2ba4446718ac1af116a35c4f8 : SUCCESS in 2m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a5a644a85959487db4c778b40d23e9e4 : SUCCESS in 24m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9d266d9525ca492fa202dcc5c7e8e4aa : SUCCESS in 24m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/eb37a9914f0e4f33937ec5da8c0c401d : SUCCESS in 1h 06m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f239403c8eb646c0b672d21e26e1af3e : SUCCESS in 1h 22m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f62a2b525fb34e12aa8b8732d7fbf508 : SUCCESS in 17m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/942300e55b28422eb2d819937c035624 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06536956381f4dc085a7dd3ca07e00da : SUCCESS in 55m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2772113edb84425e80877976c95fdbd8 : SUCCESS in 1h 16m 14s","accounts_in_message":[],"_revision_number":1},{"id":"06efabe2acf0bcf50bb8dc03e46a9a0b37f8ec65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:33:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5041030661b78371a81fbf270a34ff38964ab9da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:34:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dd21ee89ee4422b974a321588927f77\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b14aa20f0742eda2ce3299f88c2d32 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e99c27ac4171467070be4d3ade543daa4de84725","revisions":{"e99c27ac4171467070be4d3ade543daa4de84725":{"kind":"REWORK","_number":1,"created":"2025-02-11 13:13:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/40/941240/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/941240/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/941240/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/941240/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/941240/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/941240/1"}}},"commit":{"parents":[{"commit":"efc4b72d181a1dbdc334b9e1c31efb8cde4dae24","subject":"Merge \"ironic: Fix dnsmasq.log permissions\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efc4b72d181a1dbdc334b9e1c31efb8cde4dae24"}]}],"author":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2024-11-25 18:26:25.000000000","tz":-180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-11 13:13:33.000000000","tz":0},"subject":"Adds missing nvme package in cinder-backup","message":"Adds missing nvme package in cinder-backup\n\nThe package moved to the cinder-base image since it already added to\nthe cinder-volume in I84c7633765749a5e9ced983505ef290c32154ccc.\n\nCloses-Bug: #2089587\nChange-Id: I06e61ab388d9ecfa78e5d1fe1ae9cb8602d1cd36\nSigned-off-by: FelipeAFV \u003cffigueroa@whitestack.com\u003e\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4a7af468b6c24322fd89b97933b1fee84e7b8d85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e99c27ac4171467070be4d3ade543daa4de84725"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e99c27ac4171467070be4d3ade543daa4de84725"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
