)]}'
{"id":"openstack%2Fkolla~941241","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I06e61ab388d9ecfa78e5d1fe1ae9cb8602d1cd36","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-19 17:33:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I06e61ab388d9ecfa78e5d1fe1ae9cb8602d1cd36","subject":"Adds missing nvme package in cinder-backup","status":"MERGED","created":"2025-02-11 13:13:42.000000000","updated":"2025-03-19 17:34:39.000000000","submitted":"2025-03-19 17:33:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941241","cherry_pick_of_change":936191,"cherry_pick_of_patch_set":3,"meta_rev_id":"6b4ba49fcab2b27235edb6a45d01a60b978933f5","_number":941241,"virtual_id_number":941241,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 17:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-19 14:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-12 06:54:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-03-19 15:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-03-19 15:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"CC":[{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-11 13:13:42.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":35765,"name":"Felipe Figueroa Vergara","display_name":"Felipe Figueroa","email":"ffigueroa@whitestack.com","username":"felipeafv","status":"Whitestack"},"state":"CC"},{"updated":"2025-02-11 14:53:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-12 06:54:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-19 14:30:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-19 15:31:18.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"b1a0a50e90f5e14721a851e2b06b12fdd2335c02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-02-11 13:13:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"52cb444556f9a89bf11d2487e17c0b145ed979b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 14:53:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/400ecb89572b4d2c888fe58c408376e4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2b979dd5c4094771a490dfd3b8c29f5f : SUCCESS in 30m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3956108a90284eb1abb6c33803cf0900 : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/37a793cac00b407ca93671258c39fa40 : SUCCESS in 24m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3bfd681dd3294991b7298ee757588089 : SUCCESS in 27m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cbb07be727234b06b7b26ee9d73b3054 : SUCCESS in 59m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1cf816a65274464bb27807292a184b24 : SUCCESS in 1h 32m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e38d24de4bf349f2b3ff98b8d89def36 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7e0021b794d5492e9d2e4ed7554566eb : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a2dfb401d3849c98be8991f24c675b0 : SUCCESS in 2m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3049dd0180184b46923a64071e50d73c : SUCCESS in 2m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c2495a3897a647aa86d93fb7ea682aa2 : SUCCESS in 2m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46b9611eb2ce4e1eb9b8f37caa6af3bd : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a0ad96bbe94acb9c84d32dbe0dfbd9 : SUCCESS in 5m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/12a4bc1309144efda792f2c61b3646a1 : SUCCESS in 20m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cf435c496f744b4aabfa8a6d20e764ba : SUCCESS in 18m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a5f5f630751644e7844205d04efeb53c : SUCCESS in 1h 16m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/84ba8ea4bfe740e3b178719fe34fd6c5 : SUCCESS in 1h 30m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7f7e9eb1e43c4ad38544d176e50f019c : SUCCESS in 1h 38m 58s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9177f9739a3d4a699dbe6d914f6b4911 : FAILURE in 19m 30s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/de9cd2989a3d40aa8e83ee5d4db20fae : SUCCESS in 16m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/af77f342a97c491e98b2a8ca79d37d8a : SUCCESS in 1h 13m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c4e9641960c340a1aef93190d014c841 : SUCCESS in 1h 00m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cf7c57df742c473fa98764f2d08c1cbe : SUCCESS in 1h 30m 08s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e106d6a7d8b14a759e32063f48b8e77f : SUCCESS in 55m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6548675c3aed481f8d86a8df2e7899ea : SUCCESS in 1h 26m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d80732bd239506910df705b1423c24cf24a13bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-11 16:34:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25fd64edcf814156a72d8a4756996209\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/24c77056419849e5ae52711602b8a221 : TIMED_OUT in 3h 03m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9b03acbf94ae4ef1a45c5f5ea0771e7d : SUCCESS in 2h 34m 11s\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/30a02b420fc34a1dbe8dc4430c2a3090 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a1aa6f4e7f304c28807054d076469d30 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e330c977df11410992b264c9f225f7a6 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/96bffae005ea410d86db82676ba48f1c : SUCCESS in 4m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39306db1f61663cb4f548dafad0ab94a51ac0830","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-12 06:54:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e57695eb7323abf77ebe043817210964bdae5982","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-19 14:30:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a72b022ed204af82ad268f606fcaebd2fcbc044","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-19 15:31:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35c34c0f82e38b0b0fc9691394a0a3fc3bc58f76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 15:32:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ccf932a64e2fd644dac0f9a25928a373230cf43e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:33:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c939f791ce354169a6861733e3f297a7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/74ced97888f74136aba4580085611ffc : SUCCESS in 24m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4f270f0665d14eb3b80154e568808339 : SUCCESS in 27m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f736335e207d4e9faf08d6a8eab171cd : SUCCESS in 35m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f055e5ef8a3e4affa0ecc21443ee8ff2 : SUCCESS in 1h 30m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745c1750d8594494819dbb59f948b292 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e4bca31531347b89c691d81233e282f : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4021f121a949454c8df8f7617f723c96 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4b6cfefbf8d4183905071e96b43a44c : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db39f6e89ee94f4b9e7fddc8d3efedfc : SUCCESS in 3m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3298936214874ac1b7ea5e10933f6011 : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/bfab047050b540e7873ca79f811486b6 : SUCCESS in 31m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f3395ed19c1b4356b5d37f7e8c024450 : SUCCESS in 36m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4575bbfc7c624aaab4f4a8ca214d7408 : SUCCESS in 56m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/934529b8410e4aa58fc2d546ebe7faf9 : SUCCESS in 1h 31m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/64ad89e745024020963d0c62fc3365da : SUCCESS in 16m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9733ee26d0314857908605ead3aea699 : SUCCESS in 56m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c126d88f3a154034ac743ace27d150fb : SUCCESS in 53m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c1a76dbd882a4817aa88e7b02c86b8a4 : SUCCESS in 1h 20m 37s","accounts_in_message":[],"_revision_number":1},{"id":"dbb8b58cc7693ecc3dcda15a16f237da54f87d05","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:33:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6b4ba49fcab2b27235edb6a45d01a60b978933f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 17:34:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2451590b507a4abb8e03e62ce216e6e1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a9948b9e1e74560997b611519856b65 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e1875c8a9badd8e9dd2227849bcfa6ddd4fb4f7","revisions":{"6e1875c8a9badd8e9dd2227849bcfa6ddd4fb4f7":{"kind":"REWORK","_number":1,"created":"2025-02-11 13:13:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/41/941241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/941241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/941241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/941241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/941241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/941241/1"}}},"commit":{"parents":[{"commit":"e841eb3fe4745888e87cffef95d9649ce5026f15","subject":"Merge \"ironic: Fix dnsmasq.log permissions\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e841eb3fe4745888e87cffef95d9649ce5026f15"}]}],"author":{"name":"FelipeAFV","email":"ffigueroa@whitestack.com","date":"2024-11-25 18:26:25.000000000","tz":-180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-02-11 13:13:42.000000000","tz":0},"subject":"Adds missing nvme package in cinder-backup","message":"Adds missing nvme package in cinder-backup\n\nThe package moved to the cinder-base image since it already added to\nthe cinder-volume in I84c7633765749a5e9ced983505ef290c32154ccc.\n\nCloses-Bug: #2089587\nChange-Id: I06e61ab388d9ecfa78e5d1fe1ae9cb8602d1cd36\nSigned-off-by: FelipeAFV \u003cffigueroa@whitestack.com\u003e\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 4a7af468b6c24322fd89b97933b1fee84e7b8d85)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e1875c8a9badd8e9dd2227849bcfa6ddd4fb4f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e1875c8a9badd8e9dd2227849bcfa6ddd4fb4f7"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
