)]}'
{"id":"openstack%2Fkolla~942299","triplet_id":"openstack%2Fkolla~master~Iec21fc2363e03133ec77326da7392bc25b40132a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-02-20 13:24:08.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}},"hashtags":[],"change_id":"Iec21fc2363e03133ec77326da7392bc25b40132a","subject":"Remove Git remote origin after cloning sources","status":"MERGED","created":"2025-02-19 22:25:46.000000000","updated":"2025-02-20 17:37:09.000000000","submitted":"2025-02-20 17:35:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942299","meta_rev_id":"ee49ca4f0058ca7a2ee2f55a3a94ccd54712b1f3","_number":942299,"virtual_id_number":942299,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 17:35:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-02-20 10:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-20 07:26:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-02-20 10:58:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-02-20 10:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-02-20 10:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-02-20 08:12:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-20 00:29:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-20 07:26:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-20 08:12:00.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-02-20 08:12:06.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-20 10:54:05.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-02-20 10:58:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"2c99f143921a9f9b0a8704d7e6311ddc9125fe88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-19 22:25:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a473bbee6b1c3bcb60bad3105c1495068b9d699","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-19 22:30:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cc0e3c3a900ddd61552cde365eb4a7d5a67ca596","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-19 22:43:26.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"65710ffc2ae131b9bfb7d41aeb12e436236819cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 00:29:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d471932caf4477e89f198d8079abb8f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/71c9c085032b4af9a020ab0e10a83399 : RETRY_LIMIT in 5m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1318ecfdf99f440babf6fed43ee12418 : SUCCESS in 1h 12m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b99938df9334087a50eb6fcaece502e : RETRY_LIMIT in 2m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9b163fea6943413f97d7f8e01f2442a5 : SUCCESS in 35m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5d3d985857a14d27ab3d29d7f0e54a46 : SUCCESS in 1h 11m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e95cd6739a1e4e1fb16e2a49205a86de : SUCCESS in 1h 38m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa0617078734d81a5b2fa65c425dcdf : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3845b85041954d469652b8d92ffc572e : SUCCESS in 3m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea43e4bc7148496c97bd2a5a0ab09544 : SUCCESS in 4m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/868ffadac6654e0f9ff27cd33c315da9 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a1534e56d934a6f93f7d029dc70275c : SUCCESS in 4m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4419388d8e964eacbcbdc9628f9326a4 : SUCCESS in 2m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/97327935836645d8ab835ed72da927e1 : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/345b2aa213f04c57abd2bc0c94198840 : RETRY_LIMIT in 2m 55s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ef0e34cc4e444ea59f1e1f1e23ca40f4 : SUCCESS in 53m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f43b85a7ea40409badf115fbc1ba97a6 : SUCCESS in 1h 14m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/77c9ea5814f04fb3bb003994d4849142 : SUCCESS in 1h 18m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f6f1878a0834bf482ffd4910fd0d1c1 : RETRY_LIMIT in 3m 07s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b5d396363c9e44af84a81e951663697a : SUCCESS in 32m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1a68591b135b48bb81068dd985adacf7 : SUCCESS in 37m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/54ad18cdef224c77a0d66ee4ed1c736f : SUCCESS in 1h 34m 52s","accounts_in_message":[],"_revision_number":3},{"id":"0715b1049a9a6f5df20274ef986ff7a51079b3c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 01:48:44.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49980c0e10bf43979846034492a2f670\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/17ba0b99084e4bfcbb70ba3fa6e31653 : RETRY_LIMIT in 5m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/12df9e1aab4641d89afa5f9eb83d8de4 : RETRY_LIMIT in 2m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/634f184ff21a4a288fcafea2f35037b0 : TIMED_OUT in 3h 01m 24s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f3e6619d55d42debc488dccc5dff36a : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1245253a1dd14a62be6cb6aaef96135b : SUCCESS in 12m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/198f925e49494eaf9cfff411a4fe7f28 : RETRY_LIMIT in 3m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3d4e4e0cf33c4ff7bfec471f176a922d : RETRY_LIMIT in 2m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3e52d6133f7103ce6e01ed1cf3d10e7e04d9c778","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-20 07:26:04.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"cd93d86a8a3829dd55824dff6c586116cfb2fa64","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-20 08:12:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3a95671bb1da32442ad46716d0755368da06deab","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-20 08:12:06.000000000","message":"Patch Set 3: Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"0d726125c9fc8e3cfc67cb4b0cd64204db6039ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-20 10:18:43.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1fc4a9bf03a7f55e249176be2e85cb587d24ed5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:56:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a4c4bc6ea642f4beac37c00615bd03\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e499b88e487740bd86b10e75b7040479 : SUCCESS in 32m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d988d0c27d6244a282c02f5fc31cbca1 : SUCCESS in 1h 09m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/66a407e1f3ed4be88579d4a704ec6aba : SUCCESS in 16m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d5f2cb36a48b4bb0be8340ac7d3daa46 : SUCCESS in 43m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4107bef99f7e4795b9619b0953cfce98 : SUCCESS in 1h 25m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9a109a6deb4240a498ad2127b6b2efad : SUCCESS in 2h 25m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d293c65e6144ed286a5e6eb1e0e7d16 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a87d153d0b924a44b4e8d58f04d24659 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2077c1548bd14b50a70911dbe01122fa : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/415786c8950945b4a5d228297557e73c : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc40276c25ba40fdbc9c4c6c95bf7993 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc5668b8054e464b8b9aa6c2e38f6809 : SUCCESS in 3m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d83e21ff88aa438195c3e04a1215f2d6 : SUCCESS in 2m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d402a87b02004eac918b253cdca21f38 : SUCCESS in 23m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/103e719a90094d069c52a4f878b77005 : SUCCESS in 1h 21m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/37a987154b314db096b60015eb0a9282 : SUCCESS in 1h 06m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f425091f50e94e32bee280781f6fb157 : SUCCESS in 1h 30m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/026fbd87e81541a79b356945abcc9dd5 : SUCCESS in 25m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/10081537784f427bae9db7e5eaf99f2a : SUCCESS in 33m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/18a6e5a2a2444658879c2fdccfba8456 : SUCCESS in 1h 35m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9db246cb39844f5cb78c9e1736add5b4 : SUCCESS in 2h 12m 31s","accounts_in_message":[],"_revision_number":3},{"id":"c2876122e3c97772421c57c1d0a2d74ae107c640","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-20 10:58:08.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b41ea0357f338f892be7ca7bc0d1b1d53b9a32c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 10:58:32.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0238ab15ec6075b68b6c1bdb8ec2e3295b404c97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 11:47:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/350dc4254ea143f6b1d98cbce0b96bba\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e606ff466e5e4b5fb7b9024ae11f74f8 : FAILURE in 7m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4d770f46133c40bcb5c515987b85d6e2 : FAILURE in 4m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec55d0be582a405bb634ebca9e9a5467 : TIMED_OUT in 3h 01m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5573a215bb644c8f986afd7f24bd1604 : SUCCESS in 11m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/380d3c88424842a7a945344b7918676f : SUCCESS in 11m 40s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/26ae4201d7b14f01bce9b6bbbcee6cc0 : FAILURE in 4m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b1ea8b28dab4430d9ef437af5a638746 : FAILURE in 4m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"17524923494ef621e2b6c9163423417af32959e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 13:12:25.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbf6c335773b4c9a9d7f12b1c0697263\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/50e5eaac8886424f865e22e2f0cfe081 : SUCCESS in 22m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d97d5b3b867e45a48afac4d4ae076ee9 : SUCCESS in 24m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3c48a93069494908a09a8d57e497559d : FAILURE in 59m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/dcf77be443854d36bfc2f1bb1200f8d3 : SUCCESS in 1h 20m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db186ca28ce945c8a43366c89f90f6d9 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa523555edc6433cbbacea5d790dd11f : SUCCESS in 4m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cac22b375eff4f9f83d41dd3ca48df65 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5cbc95f20b04be7aa00f2ab648fc26c : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24a4a461d9f3416bb9af6e4889c46b4e : SUCCESS in 3m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d2828cf082a94524b33f24f3a139c322 : SUCCESS in 2m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/25f4576b7c63424ab442c60764cb8423 : SUCCESS in 39m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c180e80181da4086985661999df397d6 : SUCCESS in 1h 47m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/82c2773e5c274af7b0bec7620368a467 : SUCCESS in 1h 00m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ccd9dd861171461fa7355a6e86fdd2f4 : SUCCESS in 1h 33m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a432ec1c1df648cba9e62d205779acb1 : SUCCESS in 28m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/639167b25cf944dea3e4c4c302d67ee1 : SUCCESS in 1h 01m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2edccc5b943445afa7d679da3e991fe0 : SUCCESS in 1h 35m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/024b240fb69a429480a0c54b1522bc7d : SUCCESS in 1h 43m 54s","accounts_in_message":[],"_revision_number":3},{"id":"fca013567cadd26fbd91d5cb48d4c4f60b2acca6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-02-20 13:24:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"50812046349abc76e2cf9e39d51b52351659c450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 15:10:23.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae891e0468f24e3c8ffbff89591e2d89\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/527d4f731aae484a8ee36fe2d82031ff : SUCCESS in 31m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/91f86ebc48ce46c2b6db35901898f2aa : SUCCESS in 38m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3c6e94f7970944b18019f628e72f00e7 : SUCCESS in 12m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fc8b1db25e644595b9108032df895f41 : SUCCESS in 39m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e7c281098f3a41209a29746f7f88ef5b : SUCCESS in 1h 10m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6213036faeab43f2af40106383412fcb : SUCCESS in 1h 31m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f6c72d274c4434b9c807ec4c154d80e : SUCCESS in 2m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9215d5f963c94ce39a9154835d5f0fea : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f6d81846a0c046619bb6f702e3731f16 : SUCCESS in 3m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fedecde92f74cf1bbef043507955551 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e6ddbade7e4309b048e747a6f36078 : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f031d334ccf0480384182882b727898c : SUCCESS in 4m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/70c6866d196543c2bc764699402a87a7 : SUCCESS in 2m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/64e8fd8ef6c94f2dbff23df552c0ddf5 : SUCCESS in 24m 12s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/57d5005deb204bd3b8ce661fa638bb12 : SUCCESS in 51m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/85a39b3e5a49489fb3857ad387cac637 : SUCCESS in 1h 15m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7de0d4a7035049bcaab3a7ba0fc1def8 : SUCCESS in 52m 31s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7f6fe230f42a4ff2877c2b591b3723a7 : SUCCESS in 23m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/df3b196e37b945d1b6aca71a6d9e8530 : SUCCESS in 16m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/25f60562c9584a2fa4f372a588164122 : SUCCESS in 1h 00m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b3eedd975fe04928913c3f405a2158b4 : SUCCESS in 1h 43m 43s","accounts_in_message":[],"_revision_number":3},{"id":"e8bfbaf480092555d58be260678a2afca410f204","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 15:10:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a4ea1cd3e757d42666b0645998243677eef3f656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:32:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d738996fcf924bcdbf699fed9df20684\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3b438292354b41998febf7bc5cc557f7 : FAILURE in 6m 42s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fba4ded87d7b4c33baf56989e33ebee9 : FAILURE in 4m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62c2a385c5a84dab88e24ca2a866bb73 : TIMED_OUT in 3h 03m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ebf600b3b3494aaebdb8a3944f0d23ff : SUCCESS in 11m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fcb6a4f85d034fef88edb3147c292fc7 : SUCCESS in 7m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/379e6d80153f431faa7e9b1bc662e2f3 : FAILURE in 4m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8dd79e34bcd744d7906137df25f0d875 : FAILURE in 5m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c606d2a481af27cd18db6d52e802e5cfa92285f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:35:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c6f18c9102f49e9b9b9b9ac644f93b3\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fe9e38662a91456998fd6a2e28488cb0 : SUCCESS in 19m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d05bb16b01c642939b6b88055e542916 : SUCCESS in 19m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/91e79a4828f94a49a204c42ac3c16484 : SUCCESS in 1h 35m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d3806082cacf4d9da0759f7c0d4ae970 : SUCCESS in 2h 11m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff03a5e81c842d8ab67f3ffeb76454c : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/284b934151d24fd4b24acd6aa1335c88 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/734d882b10f5497489e3bdd5394f1640 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16ce87eddd946b2a5a942f8e17154c6 : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84c73866144a47cb9bcbd4a758712b24 : SUCCESS in 4m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0a83a0f2f8e74e50b1e25e11ae24575e : SUCCESS in 4m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e5950a650cc04950b3b46971a11f4715 : SUCCESS in 29m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3daa6bc6f8904422842a9fa1f404e884 : SUCCESS in 1h 27m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e9fc96736a6443af9e2b214ecbe47eb9 : SUCCESS in 1h 28m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/612ad1575606434f9024fd6f4f2dca8e : SUCCESS in 1h 38m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/400e49017f394337808f2180a4503cb0 : SUCCESS in 25m 57s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6eca53ea55604c69a28ca94c677397ec : SUCCESS in 58m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c8b6a7962917407a85eadd46df5767c8 : SUCCESS in 1h 12m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/721e08f0cab14683b1827ac694507c77 : SUCCESS in 2h 18m 44s","accounts_in_message":[],"_revision_number":3},{"id":"95fc13bd652a563f80c43e8e2f2973b19baa97c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:35:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"ee49ca4f0058ca7a2ee2f55a3a94ccd54712b1f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 17:37:09.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86d77724bd324e80822955da1a354573\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8d0076acfa4a3d901b9e78f0fb1f16 : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81fe6733c7fb4828936add555c0d6275 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"405f95019354b9be45a750e246127bd9c68d2508","revisions":{"08d5d04a9ac9143fd735c31b9bbf17143760e886":{"kind":"REWORK","_number":1,"created":"2025-02-19 22:25:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/99/942299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/942299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/942299/1"}}},"commit":{"parents":[{"commit":"6d44736756e857e7893e572d5f62a5f7f91c4e86","subject":"Merge \"CI: Move pre tasks into roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d44736756e857e7893e572d5f62a5f7f91c4e86"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-19 22:22:07.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-19 22:22:07.000000000","tz":60},"subject":"Remove Git remote origin after cloning sources","message":"Remove Git remote origin after cloning sources\n\nWhen kolla is cloning a Git repository during the build\nprocess, the `.git/config` file retains the remote\norigin URL, which may expose sensitive credentials if\nauthentication tokens are embedded in the URL.\n\nThis fix ensures that the remote is removed after cloning\nto prevent unintended exposure of credentials.\n\nCloses-Bug: #2098904\nChange-Id: Iec21fc2363e03133ec77326da7392bc25b40132a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08d5d04a9ac9143fd735c31b9bbf17143760e886"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/08d5d04a9ac9143fd735c31b9bbf17143760e886"}]},"branch":"refs/heads/master"},"365695ab100f54a393d9a299be2787ed25b05f01":{"kind":"REWORK","_number":2,"created":"2025-02-19 22:30:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/99/942299/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/942299/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/942299/2"}}},"commit":{"parents":[{"commit":"6d44736756e857e7893e572d5f62a5f7f91c4e86","subject":"Merge \"CI: Move pre tasks into roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d44736756e857e7893e572d5f62a5f7f91c4e86"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-19 22:22:07.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-19 22:30:12.000000000","tz":60},"subject":"Remove Git remote origin after cloning sources","message":"Remove Git remote origin after cloning sources\n\nWhen kolla is cloning a Git repository during the build\nprocess, the `.git/config` file retains the remote\norigin URL, which may expose sensitive credentials if\nauthentication tokens are embedded in the URL.\n\nThis fix ensures that the remote is removed after cloning\nto prevent unintended exposure of credentials.\n\nCloses-Bug: #2098904\nChange-Id: Iec21fc2363e03133ec77326da7392bc25b40132a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/365695ab100f54a393d9a299be2787ed25b05f01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/365695ab100f54a393d9a299be2787ed25b05f01"}]},"branch":"refs/heads/master"},"405f95019354b9be45a750e246127bd9c68d2508":{"kind":"REWORK","_number":3,"created":"2025-02-19 22:43:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/99/942299/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/99/942299/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/99/942299/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/99/942299/3"}}},"commit":{"parents":[{"commit":"6d44736756e857e7893e572d5f62a5f7f91c4e86","subject":"Merge \"CI: Move pre tasks into roles\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6d44736756e857e7893e572d5f62a5f7f91c4e86"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-19 22:22:07.000000000","tz":60},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-02-19 22:43:19.000000000","tz":60},"subject":"Remove Git remote origin after cloning sources","message":"Remove Git remote origin after cloning sources\n\nWhen kolla is cloning a Git repository during the build\nprocess, the `.git/config` file retains the remote\norigin URL, which may expose sensitive credentials if\nauthentication tokens are embedded in the URL.\n\nThis fix ensures that the remote is removed after cloning\nto prevent unintended exposure of credentials.\n\nCloses-Bug: #2098904\nChange-Id: Iec21fc2363e03133ec77326da7392bc25b40132a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/405f95019354b9be45a750e246127bd9c68d2508"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/405f95019354b9be45a750e246127bd9c68d2508"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
