)]}'
{"id":"openstack%2Fkolla~942422","triplet_id":"openstack%2Fkolla~master~I57e05a235359547a4477dc869e92047818dc5c15","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-02-21 16:49:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I57e05a235359547a4477dc869e92047818dc5c15","subject":"CI: Add rocky support to process_build_logs.py","status":"MERGED","created":"2025-02-21 10:59:17.000000000","updated":"2025-02-21 16:50:41.000000000","submitted":"2025-02-21 16:49:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942422","meta_rev_id":"74d89f93d670d8f8721a345aeab3318e9bfbc7fb","_number":942422,"virtual_id_number":942422,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-21 16:49:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"recommended":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":1,"date":"2025-02-21 13:17:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-02-21 15:05:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-02-21 15:05:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-02-21 15:05:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 12:36:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 13:17:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-21 15:05:32.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"651878c82766f4ffc7b2f0b631babe679f9bfc18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 10:59:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e24279d4aaff50af20ab1fb95e74c6f6195c2b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 12:36:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415134f468024152aea99c1c55f2999c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c360b9cbae904ee298b682d71f23501b : SUCCESS in 25m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ba6707adad8349f59774753ed1ef0fb7 : SUCCESS in 1h 15m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/81aa28d60e7c48b2a89fbe50dac2ca7c : SUCCESS in 18m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/40744b42e9ab4d76acad5cfa8f1ca115 : SUCCESS in 19m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cf92f14a7d734ab4a84b86f77355ee8d : SUCCESS in 55m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0aed71bf9b44990be69311073017394 : SUCCESS in 53m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/283370af35b94d038c6d3e13336a022f : SUCCESS in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbc2e624862947f8be41d487d88bbd38 : SUCCESS in 3m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b280e8fa31e44ea1a33accfda8c88330 : SUCCESS in 2m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5deeb863329e4e9fa031b7da090e050f : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0f1c8a0c6b24241a6def85125e98834 : SUCCESS in 5m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/87e14ea5512945e886720f7ed7556e3c : SUCCESS in 4m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dc901d1db1284be29eafbd38e5b16217 : SUCCESS in 24m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bd69855bd80c4d0ba94d4fea47bcf9bc : SUCCESS in 52m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/945e7ad46c8947718e46f1096b9d9506 : SUCCESS in 1h 15m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2269f2421f39426c9674a8ea538ca66b : SUCCESS in 1h 19m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0ccd61ad56c346c79e29664aa149e618 : SUCCESS in 23m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d338ce0f2b324b978969593abde00eb1 : SUCCESS in 34m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9b62ea0b0784424aa695e93cbedabfa8 : SUCCESS in 1h 10m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c7699e8bd65249bb884421bbf04feee5 : SUCCESS in 1h 30m 45s","accounts_in_message":[],"_revision_number":1},{"id":"de47fcd037c262c9c8de19f39f43ef0eb86d7c81","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-21 13:17:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"568cb148b7bc81c137fff580588dd59ed40ec2fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 14:05:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c2d9c398643465780400973aaa01034\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a2a84069988a4e668b042fd84801e4c3 : FAILURE in 6m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e5338129cb3246549fe4a493b1041d16 : FAILURE in 3m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3bacaae8cff247d6ac7d6f3ba3af6ed6 : TIMED_OUT in 3h 01m 44s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/69d60ac691c042929dcb7b20219eeb58 : SUCCESS in 10m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/495512e8db6b426bae1c4376bcf8df95 : SUCCESS in 10m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eca6709056d14da2aa25ed65cac8ead3 : FAILURE in 4m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ed4aacdc45c14777b2237c78057df0e5 : FAILURE in 3m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3f396a7f7cc73f8dda45b532ec6a9855cb63a3fe","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-21 15:05:32.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"588bda08696acebdeb396c3743c8549a00e3e801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 15:05:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"33859b15940490b6baddaaf157a42120bd8c0385","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:49:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72daf2ba968b46d3acca765ac8fe0184\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/244d3e0b042e4eacbf11af372054a832 : SUCCESS in 21m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/54ff8174a2274e22b07519d868abdda6 : SUCCESS in 38m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a37438955d584f8a98ebd52f9ecf4e3f : SUCCESS in 1h 15m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fdb57bb01dbc409794056c162dc2d1ec : SUCCESS in 1h 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e80f2c74b1d04558a8c5903804ce6397 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/441a585e20c645a291e2c905cb0560f8 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9e5aa18d42b423ba404503752a0561f : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f767efab0fce43b2b79491792aa142b9 : SUCCESS in 6m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c8022cdd221b4f72a814609f399f8a10 : SUCCESS in 7m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/42d403382491435694c76b872bcdd5eb : SUCCESS in 23m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/927e7678f4a9421081c8955ca687e97f : SUCCESS in 52m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5059811bc3ef4a87885651c663737475 : SUCCESS in 1h 18m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dc4bbc409aa4460eabea429ae2eb23a1 : SUCCESS in 1h 30m 14s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a90cbb7693d64919ad24e228e27e9c42 : SUCCESS in 16m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/36bed419c2f04e98818e8f21990555c2 : SUCCESS in 32m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e1832b9cf044614a787cd83a0239ee1 : SUCCESS in 36m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8c5a1f017f0c4d0b94a11254c4577ae3 : SUCCESS in 53m 22s","accounts_in_message":[],"_revision_number":1},{"id":"76cf4d7a56d73c33c953254b424b9d4a76a99beb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:49:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"74d89f93d670d8f8721a345aeab3318e9bfbc7fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 16:50:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4a0c57dd344ce4852e43044008bef5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1436679e58334dc889229f71e483fedc : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"498b793c869384f6564fc0e2d422200321974745","revisions":{"498b793c869384f6564fc0e2d422200321974745":{"kind":"REWORK","_number":1,"created":"2025-02-21 10:59:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/942422/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/942422/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/942422/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/942422/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/942422/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/942422/1"}}},"commit":{"parents":[{"commit":"405f95019354b9be45a750e246127bd9c68d2508","subject":"Remove Git remote origin after cloning sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/405f95019354b9be45a750e246127bd9c68d2508"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-02-21 10:59:07.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-02-21 10:59:16.000000000","tz":60},"subject":"CI: Add rocky support to process_build_logs.py","message":"CI: Add rocky support to process_build_logs.py\n\nChange-Id: I57e05a235359547a4477dc869e92047818dc5c15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498b793c869384f6564fc0e2d422200321974745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/498b793c869384f6564fc0e2d422200321974745"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
