)]}'
{"id":"openstack%2Fkolla~942479","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I9359df23a41a600dd46dbc1812a54be57fd33f71","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-03-08 13:54:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9359df23a41a600dd46dbc1812a54be57fd33f71","subject":"Prometheus update all exporters to the latest versions","status":"MERGED","created":"2025-02-21 16:54:04.000000000","updated":"2025-03-08 13:55:26.000000000","submitted":"2025-03-08 13:54:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942479","cherry_pick_of_change":929457,"cherry_pick_of_patch_set":10,"meta_rev_id":"6d45d1f23d81fc1b076dd429f062371c8cf27cd9","_number":942479,"virtual_id_number":942479,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-08 13:54:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2025-02-27 15:58:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2025-02-21 17:14:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-21 19:38:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-03-08 12:39:34.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-03-08 12:39:34.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-03-08 12:39:34.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 16:54:04.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2025-02-21 17:14:34.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-21 18:30:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 19:38:46.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-27 15:58:52.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2025-03-08 12:39:34.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"a3236e2f49d85ed0743b6f6deb43d415d0a0562e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-02-21 16:54:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8b1a49821933b20cef28294545c52c17910ad96a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 17:14:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3534da22c4878dac4357a857c15755c903548849","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 18:30:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5682a65aecb340d0b36cebdf451db94b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3b93e934dd8649b0ac69e5eecf86b14d : SUCCESS in 28m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5e0ab059a6fa45d2a34c22080bc7a701 : SUCCESS in 1h 04m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ae4c5a0b9a024e63bfa4e9b3acdee5c6 : SUCCESS in 15m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/56cc0cfc4f8e4df8945dac3cce24a61d : SUCCESS in 34m 34s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/417e86e63c4044828470424ea78dcfc6 : SUCCESS in 1h 00m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/054f6ad5d7e94cfd85ea2780ebb00f3c : SUCCESS in 1h 29m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9252baeb4b31402bb60733459f3c1ef3 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58333a9eb8eb49acb65fd014c539e42e : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0fa1b768d1ec411f87fac001ae5c3cc3 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/197ea80e54194dca9cf9321d5b874446 : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc5105b3cd84436db13ccc28fdb8e8f4 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c548d20e8b6540b0997c840ec8b22e70 : SUCCESS in 4m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4d9d3a5423c43fcb1ba73cad745389d : SUCCESS in 3m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/620d029058124957a9274c205e0377ad : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/19ce05dbb469489ebb04849d411ea728 : SUCCESS in 22m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/94ba325e4e10402594579ae88587b5c8 : SUCCESS in 16m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c6e4fff3699a4896aed01cdbabce1027 : SUCCESS in 1h 17m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b8e91992e61444e5be62be326671b74a : SUCCESS in 1h 19m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0671f74e392d4871a0e03d043c5bd134 : SUCCESS in 20m 20s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2b0acc0dace9424bb4c325b879feb066 : SUCCESS in 27m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8264f92c92334f30a2b8702056945c73 : SUCCESS in 34m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a7329b94d17c4ff79015d5847ce5c4f7 : SUCCESS in 1h 20m 20s","accounts_in_message":[],"_revision_number":1},{"id":"54fff45703c8f29f7c7785a431bd6d31546d10f7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-21 19:38:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"462d3f711921da5b4639cf21a7de679c3c55ede8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:01:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afa1c00bbf9d48a59e3924f793278890\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6c5133baf148484b844bbc5c310013b3 : FAILURE in 2h 07m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/232385e1e45e42cd8c0e7265b5da9c8b : FAILURE in 1h 13m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6a33bc3022524673b686ad0d6346a9c6 : TIMED_OUT in 3h 02m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bbcce56eb60248228a1e219c3c79ec63 : SUCCESS in 26m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1fb8ad2605364dcdb150dafbcfffc8df : SUCCESS in 5m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c87fb83178d947969271e5fba5223a26 : TIMED_OUT in 3h 01m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/dcdd704055d1496eacab2c15ecb68f9f : TIMED_OUT in 3h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b402561f536221ec44a1301897c1a39ac0225842","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-02-27 15:58:52.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"295201c53daec86edb1e1d21aa3793089312e1d0","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2025-03-08 12:39:34.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1536a71a94005376f3d887cf25a419102ea08d7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 12:39:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1483fa50f89884d300f440f51a1852039ec77ae0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 13:54:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47b8836c55b94f0e86f7064ca0097442\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8d187eab47c24efa8e7fba723a990e47 : SUCCESS in 17m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/149992999b694f50bea634884cb855bb : SUCCESS in 22m 08s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3bec3886e76b463f802ec9c7da1dccf4 : SUCCESS in 53m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/24847b4fbf4f4f2caf65c61c7ec5695d : SUCCESS in 1h 11m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/061ee3c646274dd599912885bd68ca1f : SUCCESS in 2m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91f65f4d1c45477caec5b862a4b50c04 : SUCCESS in 2m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b6af9c383bc4ea09d4c84210bdd7f8a : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b210f47b9df48e18ff671a3e1d80218 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5680b64f79c4fca9d89bd6bda9a66bb : SUCCESS in 2m 27s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ad389ed405a24bf999109c4ec0b9169e : SUCCESS in 1m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3a844198e41e4025b79c4c2c0b85562b : SUCCESS in 13m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aed916dcba7e49e0bdee3e5733496e91 : SUCCESS in 16m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/021c70eccdb74c3ca0b17a70c86d88e6 : SUCCESS in 31m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c2650e926b0a494f801750aedc514142 : SUCCESS in 47m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/61afe0daa37048f690f8504b3793d753 : SUCCESS in 13m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/671fca8419d44bf2acfc70b46eb8640a : SUCCESS in 17m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ff9451173e13437790e6f72cba20c6e7 : SUCCESS in 1h 09m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0e7c1fe5eac646c7b60c23175c7c9fd7 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":1},{"id":"f00174585eb1562789f61fe5aa8ae4fa26321445","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 13:54:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6d45d1f23d81fc1b076dd429f062371c8cf27cd9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 13:55:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8aa5bf32db35453aad7d9175cbbaf1a1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d947626d2a340eaa1e198066943ca18 : SUCCESS in 35s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fcd02d16d83143aa90e9df1bc8587e78 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0a897ab28e200a3016bf1f06ee2a83a87f4ab541","revisions":{"0a897ab28e200a3016bf1f06ee2a83a87f4ab541":{"kind":"REWORK","_number":1,"created":"2025-02-21 16:54:04.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/79/942479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/79/942479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/942479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/942479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/79/942479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/79/942479/1"}}},"commit":{"parents":[{"commit":"4110115b2317cfe2ca05e950ad7598140d2b370e","subject":"Fix genconfig","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4110115b2317cfe2ca05e950ad7598140d2b370e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-02-21 16:54:04.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-alertmanager: 0.27.0 -\u003e 0.28.0\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-cadvisor: 0.49.1 -\u003e 0.49.2\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus 2.50.1 -\u003e 2.55.1\n\nCo-authored-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n(cherry picked from commit 0f5c679dc93d25d0ab8818924adfd8037a18b289)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a897ab28e200a3016bf1f06ee2a83a87f4ab541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a897ab28e200a3016bf1f06ee2a83a87f4ab541"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
