)]}'
{"id":"openstack%2Fkolla~942480","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I9359df23a41a600dd46dbc1812a54be57fd33f71","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"17669":{"account":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"last_update":"2025-03-08 14:21:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9359df23a41a600dd46dbc1812a54be57fd33f71","subject":"Prometheus update all exporters to the latest versions","status":"MERGED","created":"2025-02-21 16:54:23.000000000","updated":"2025-03-08 14:22:19.000000000","submitted":"2025-03-08 14:21:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942480","cherry_pick_of_change":929457,"cherry_pick_of_patch_set":10,"meta_rev_id":"415e449c62bc7584182ef077a457ed350cb0b851","_number":942480,"virtual_id_number":942480,"owner":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-08 14:21:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"all":[{"value":1,"date":"2025-02-27 15:58:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2025-02-21 17:14:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-21 19:39:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-03-08 12:40:11.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-03-08 12:40:11.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-03-08 12:40:11.000000000","_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-21 16:54:23.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2025-02-21 17:14:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-02-21 19:33:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-21 19:39:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-27 15:58:57.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2025-03-08 12:40:11.000000000","updated_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"reviewer":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"state":"REVIEWER"}],"messages":[{"id":"5eb56c48fca86819c6a5a14be28bf3d02507211a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2025-02-21 16:54:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a2947f970f8c230e0bb84065aa1cbecde4c41739","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-02-21 17:14:39.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"79145815d9c2080bf7be04eb5a81aabd56d0cf75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 19:33:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/138502f8fdb040909f29539ab0da3334\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1715e9e345164ada83a63972c0e13244 : SUCCESS in 24m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/51262558635c4bd1b60942d0e2485618 : SUCCESS in 1h 14m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/efd8914c56174b7d99e09558ad281b5f : SUCCESS in 22m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/61641a6e8afc470cb71484e56cdf45b3 : SUCCESS in 30m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6b6516e9f1544b53a28c2d67557cf8c2 : SUCCESS in 1h 10m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fbba2e46435b4bb2a955de42a4954a4c : SUCCESS in 1h 40m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f2291549fee49e7b7e983b3c7ca0e3e : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/364102e48b4b431bb8f90304f30df82b : SUCCESS in 2m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/20bd97aed0c0468d9f1b25bec652c75d : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f477e707c5f4474917f72089e21fcce : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbad130accb4edab05926d7514e0c3f : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8d8a49381dc415c8017a2b9e2308af9 : SUCCESS in 2m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c6bd47f7c998468896c7ebaddd1b2e77 : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f82862fa9bee47109faf45721abce8ff : SUCCESS in 19m 23s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d8b662fb82c498e8a7f78b3810543ff : SUCCESS in 30m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/022c084b1b164a88a2248ddb04385128 : SUCCESS in 1h 05m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/76505a25ddc14593a4eb25d32c776b71 : SUCCESS in 2h 31m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d7348dd606b34f90a4b32a034d9269b2 : SUCCESS in 16m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a57cc72f1cb842339fc3d62771ba839b : SUCCESS in 51m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/964c0670e9c74a9483d996d52097a95a : SUCCESS in 56m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/801f98dbc63e4c8c8faa7699564d4d05 : SUCCESS in 1h 14m 53s","accounts_in_message":[],"_revision_number":1},{"id":"7a3fbf6f19a41c5787b62cbb987ab4c64b2ba942","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-21 19:39:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4f1812859056a5540320234adb12d50fe2119283","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-21 20:01:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f1c8fb7d4134245b2f3cda7d3cb0c57\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/640cc690e76d4672a17397d2239d58ba : TIMED_OUT in 3h 02m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/60f3c07b304b4f0793318fa10b6cab04 : TIMED_OUT in 3h 03m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/be3975ff72c445838f5937ea824be7c1 : SUCCESS in 24m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3347fba8a36e4a58ae57e686ebadd31c : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f593c0eb52d1f1b1c02e7ca383200e1f85515c56","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-02-27 15:58:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"80b648765bd394a442e864992511e6df501b377b","author":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"},"date":"2025-03-08 12:40:11.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"507359951f4b8ab735df6d3bf9fabc85422e7b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 12:40:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85e237f3b4eea02eab5226ee862e731262fd813b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 14:21:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ced40cb47344cf5a86ad98ae91e2d01\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4910f67e60bd4022906bf3cafea40395 : SUCCESS in 20m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/099b14b964884180920bc7bfda44c199 : SUCCESS in 22m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/77729291edbb4b56b96a18ee5c2ca55b : SUCCESS in 1h 04m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b6773915cc134b84abcaae8809008975 : SUCCESS in 1h 22m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf1451ea4754f21843729d5d2517644 : SUCCESS in 2m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60ee9c44c79347e38803b9f4f5de1b6b : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02585c710cd948acb8b4a977b01a775f : SUCCESS in 2m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adf0283387144e968aa3c77d7b780c81 : SUCCESS in 3m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a8d535d37a748fc97dcdb1ad2c99520 : SUCCESS in 2m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1f1bcac040814a29a09ce8430650e8d5 : SUCCESS in 2m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1d2861e98e4f4b69b65a72f9d49cc1e5 : SUCCESS in 16m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/707a0300d60a49c3acb9a451e6508d34 : SUCCESS in 20m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1c55758f407b428dbaeb6c529e7f58df : SUCCESS in 49m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cd7992187f18423ca9a37c6e4820ba3a : SUCCESS in 1h 35m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c8b1a9047d61452a8211b21be3e9a07c : SUCCESS in 12m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ac77d32a76c14d3096a4e91299404db7 : SUCCESS in 48m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c185f605539943499609ad02c3c6d52d : SUCCESS in 53m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/228c2bb012c944a4ab9c4970fd3aeb5a : SUCCESS in 38m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c10a48de6eb3adb550ea7b9a43876a86310223b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 14:21:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"415e449c62bc7584182ef077a457ed350cb0b851","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-08 14:22:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65ab61abdfa74fd0b63004e214731869\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f19fbbc61ba54ccc82efb1e045bfcb5c : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fae68ceaa9a04438b90b6d80d9b53cd0 : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ac6af71cfba2a2f15c2eec4cb8dca712162e999","revisions":{"9ac6af71cfba2a2f15c2eec4cb8dca712162e999":{"kind":"REWORK","_number":1,"created":"2025-02-21 16:54:23.000000000","uploader":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"ref":"refs/changes/80/942480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/80/942480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/942480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/942480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/80/942480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/80/942480/1"}}},"commit":{"parents":[{"commit":"fddc922c3a2faeed207053acdc2e07b182b73b70","subject":"Fix genconfig","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fddc922c3a2faeed207053acdc2e07b182b73b70"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2024-11-19 19:40:23.000000000","tz":60},"committer":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2025-02-21 16:54:23.000000000","tz":0},"subject":"Prometheus update all exporters to the latest versions","message":"Prometheus update all exporters to the latest versions\n\nprometheus-alertmanager: 0.27.0 -\u003e 0.28.0\nprometheus-blackbox-exporter: 0.24.0 -\u003e 0.25.0\nprometheus-cadvisor: 0.49.1 -\u003e 0.49.2\nprometheus-elasticsearch-exporter: 1.7.0 -\u003e 1.8.0\nprometheus-memcached-exporter: 0.14.2 -\u003e 0.15.0\nprometheus-mtail: 3.0.0-rc54 -\u003e 3.0.8\nprometheus-mysqld-exporter: 0.15.1 -\u003e 0.16.0\nprometheus-node-exporter: 1.7.0 -\u003e 1.8.2\nprometheus 2.50.1 -\u003e 2.55.1\n\nCo-authored-by: Piotr Parczewski \u003cpiotr@stackhpc.com\u003e\n\nChange-Id: I9359df23a41a600dd46dbc1812a54be57fd33f71\n(cherry picked from commit 0f5c679dc93d25d0ab8818924adfd8037a18b289)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ac6af71cfba2a2f15c2eec4cb8dca712162e999"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ac6af71cfba2a2f15c2eec4cb8dca712162e999"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":32553,"name":"Sven Kieske","email":"sven_oss@posteo.de","username":"skieske"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
