)]}'
{"id":"openstack%2Fkolla~942684","triplet_id":"openstack%2Fkolla~master~I6c15cf0fd7b8f1cc3339936c58ad8a8581cfc570","project":"openstack/kolla","branch":"master","topic":"uca_epoxy","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-02-25 15:11:32.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-02-25 13:21:18.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I6c15cf0fd7b8f1cc3339936c58ad8a8581cfc570","subject":"[release] Use UCA Epoxy","status":"MERGED","created":"2025-02-25 09:55:40.000000000","updated":"2025-02-25 15:12:24.000000000","submitted":"2025-02-25 15:11:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942684-uca_epoxy","meta_rev_id":"aa3e801790a6544cd1a8147a5f8105ebed26dc9e","_number":942684,"virtual_id_number":942684,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-25 15:11:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-25 09:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-02-25 10:56:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-02-25 13:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-02-25 13:21:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-02-25 09:56:22.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-02-25 09:56:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-25 09:56:22.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-02-25 10:56:34.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-02-25 11:49:13.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-02-25 13:10:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-25 13:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1c748b1238a8184d33eea02cf789e52f7fc306b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-02-25 09:55:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33d23ac35efa41e1fe87ff8ae584e6be90aa41b9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-25 09:56:22.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"686aeabd56536fca1ea68fa5f9921bbdf3631bfd","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-25 10:56:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4c766495b643ceead2cf0f20630d901e31e1eb83","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 13:10:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48df5eaa92af419c8d78bdc5dca11033\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/320fe2760d854111be109ee17b7147a9 : FAILURE in 6m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d47cfb87b81b40f38f4457e632aec575 : FAILURE in 3m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9bca4b22b5eb41a49df6fa1ff54a0f7c : TIMED_OUT in 3h 03m 30s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70f38289cd1347c4a2657ccc96a708fe : SUCCESS in 12m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c694bc4fab014eea9ed543cff9ac7668 : SUCCESS in 8m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/910f61d24bf143239b2911720c9fd62e : FAILURE in 4m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/176291955c284fe09c42af5b0b4b2d1a : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd35a5b7d08e20032ba3a0698c253ea535f1494e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 13:20:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7269e21a3503409da1c9dc20b6ffc111\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cfbb3524b5f54d419043a7c50e866a9b : SUCCESS in 29m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1085cfedb81a48e1a89baa6f8194757e : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cab3dc3e27aa4a1897e5680946576db6 : SUCCESS in 22m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d5a2cc6b8c604b8289b000e7543cb366 : SUCCESS in 32m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/78e5f1a57a3147769ac816c8f2a1bbd3 : SUCCESS in 1h 07m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2492788b87e24d51899b9cfc5f3743d2 : SUCCESS in 1h 41m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e8530bd5bc94942aef2d4e1e04a5063 : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d402291cd78c412a8227cbc8e674bc50 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/017a7bca59ec4b5aa0a514ffadbb303b : SUCCESS in 3m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed60f0a455e648a088b5c4971d9c62a5 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c8d9bf47a14d6d965a11dfd1f1a7a1 : SUCCESS in 4m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dc52f5900dd3495a9ab94189a07ef000 : SUCCESS in 4m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/14e70d610ebe42caa40ab276940a4e05 : SUCCESS in 20m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1152fcbcdbbc40e0acb2919e52092af6 : SUCCESS in 1h 20m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b54a3feb14f642eb82bbe54f0f5a7a0f : SUCCESS in 1h 11m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ddbbaeffbed34516b7263fe06ae8f628 : SUCCESS in 1h 31m 03s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/40bc467a74554800993512bfd09a453e : SUCCESS in 1h 50m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e692582d0a0a409a98886d1a74acae80 : TIMED_OUT in 3h 03m 09s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/bbca0e16e73b4148ad79151753bd6bee : SUCCESS in 1h 16m 43s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/13b6090c954c4d4ba765ea3a262b9918 : SUCCESS in 29m 40s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/337b70bad6274b3b826aa6104f236516 : SUCCESS in 1h 06m 52s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a5e45ac4ea694fdca95ca66e2b9a115b : SUCCESS in 1h 16m 21s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/58ace929334842879ea235b4d365bff6 : SUCCESS in 1h 15m 50s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/498c1eacc9ab4d5b89bd79077e05cded : SUCCESS in 33m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2babdd8aeed44ffba55ad33257264f77 : SUCCESS in 21m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/62438431e428444b96db79be7256911a : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5d6da76a68f148fabe4fb73607c5818e : SUCCESS in 59m 08s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a337afec99934aa783091d8cb161e914 : SUCCESS in 1h 33m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e1742afed5ba4d4286f59b9350e2803d : SUCCESS in 1h 52m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/af3605499f5444f1973e2e71cd163a32 : FAILURE in 2h 16m 02s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/f5e5940029c842f597b932a770e1d67e : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/94d1180fed444af9967cb544fd58565b : SUCCESS in 26m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/804e77ba83df4f34837fbd1f1c457182 : SUCCESS in 1h 07m 23s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/23fbc4235fae49a9a29c7c64db81173a : FAILURE in 57m 17s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e438276c1a3e4c72ba47250c0de80819 : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ac9306de11044ee6951b391b77c06bdb : SUCCESS in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fdde6a95d3fe331c38f645e59b50a88a97cf7f67","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-02-25 13:21:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ca35b882b94e51efd6182e9ca5ab750dca749bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 13:21:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9a0a9cfa98682368ae5891f42690f4ee438e3020","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 15:11:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82bacb9d67664fdeb675df24fac7cae4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d42afded862340ab9440d66737065b3a : SUCCESS in 19m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/70efcf4446fa49d99c23a44cda4443a1 : SUCCESS in 28m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7fbac939c8154b958c7e7bbd92b6714e : SUCCESS in 1h 02m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0011643839e34382883f92e1b60d99c4 : SUCCESS in 1h 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b86f34b8c3456faa947428776da602 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a5cb1bfd11f4c74ac5b871a492e55b8 : SUCCESS in 3m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e429fdef6f4d48b4a0d967ab0473097e : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66f4bc94efbe414d93f7cac761c00b6b : SUCCESS in 6m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fd36cfb0579049b79f89f0f9a7a797b4 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/240b6d70b68f4dd09bb31d7aa9b20a6b : SUCCESS in 25m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/89769dcaa8084fe3a44568f667124e8b : SUCCESS in 1h 43m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5ee21c725ec94befa2f9cab777c35204 : SUCCESS in 1h 12m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e697f9d55fea4ad398fa60fe9ad61d9d : SUCCESS in 51m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99dd3a32a71149c6acffcb41b8343783 : SUCCESS in 21m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dc7aeb82f385438dac52dee484f69528 : SUCCESS in 29m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/15bc0189f5c14eddaa56e714759e42fa : SUCCESS in 1h 12m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6302a63783414d18b6aafacb42030585 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a5d678de942ff95ffdb710d295aba201f1c94560","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 15:11:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aa3e801790a6544cd1a8147a5f8105ebed26dc9e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-25 15:12:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35931ff05da14e5aaf06dd4ffe073deb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6f27070394b475d83908c5ba439f084 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9efdcc01ca5a66304568a6582a182e4171e3e625","revisions":{"9efdcc01ca5a66304568a6582a182e4171e3e625":{"kind":"REWORK","_number":1,"created":"2025-02-25 09:55:40.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/84/942684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/942684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/942684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/942684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/942684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/942684/1"}}},"commit":{"parents":[{"commit":"48423b908fcd64457776bfc5ae8f39d3ad69e172","subject":"Merge \"CI: Use ansible-lint for CI roles and playbooks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/48423b908fcd64457776bfc5ae8f39d3ad69e172"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-02-25 09:55:15.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-02-25 09:55:38.000000000","tz":60},"subject":"[release] Use UCA Epoxy","message":"[release] Use UCA Epoxy\n\nChange-Id: I6c15cf0fd7b8f1cc3339936c58ad8a8581cfc570\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
