)]}'
{"id":"openstack%2Fkolla~942785","triplet_id":"openstack%2Fkolla~master~I00ccc47f7a993f13e2b2feb3f6a2565dad32b80c","project":"openstack/kolla","branch":"master","topic":"ironic-mtools","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-03-21 13:44:54.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I00ccc47f7a993f13e2b2feb3f6a2565dad32b80c","subject":"Add mtools package to ironic-conductor image","status":"MERGED","created":"2025-02-26 11:57:30.000000000","updated":"2025-03-21 15:35:35.000000000","submitted":"2025-03-21 15:34:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942785-ironic-mtools","meta_rev_id":"135520db99e82774c53ea147da0221170f4721ca","_number":942785,"virtual_id_number":942785,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 15:34:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-21 10:18:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-26 14:03:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-03-19 13:16:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-21 10:18:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-02-26 14:03:39.000000000","_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-02-27 10:10:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 13:46:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-26 14:03:39.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-19 13:16:07.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-03-21 10:18:43.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"626913a6d2371f3ed431011531ec4c5d9c48699d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-26 11:57:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"443e476d78cec282cac9e5511bae7c89046457a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 13:46:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7724eabcbb3946ad8eace78489092e7d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b2f545337e674d708a1c3c6c8336789d : SUCCESS in 31m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2fcc37c21944474b9070b6e28ebe27c1 : SUCCESS in 1h 16m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e071b4d6be3c409e8f9f9ecdea296f28 : SUCCESS in 24m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ddce63e0353447998d7191cb900de69c : SUCCESS in 31m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ba6b3afd99a946dd8a1dfb63f47997be : SUCCESS in 57m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0fdf0930b77b49d5b048f5d3b4dd5c92 : SUCCESS in 1h 38m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30224e1c97c44269feb94045be94121 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f2b22eaeea4540cd90a0819d28a313ba : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e86a75902d6430989e1d4d131c3b9b1 : SUCCESS in 2m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dddcb5dc215a46d08d52b137b6f8f8e7 : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0a5064ca9ce4abfa8e60e3877364f7e : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd0e35ec2365474c952fbe81b8729319 : SUCCESS in 2m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b89ce6f2b7eb4264a5b131b82ce64887 : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ebfaa35b13c649b69c2df686432c473e : SUCCESS in 33m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7cd4c9f02a5e443ea88e521b239c7a7a : SUCCESS in 1h 42m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd3e5eda59b740febf242e73f4d919da : SUCCESS in 1h 06m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/069f9f85328b424d9dec16421b18916e : SUCCESS in 1h 38m 00s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/07ab0e29bd574f3e869907a3b2d67b8b : SUCCESS in 1h 15m 46s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/27d6ebd09f754e1fbda6eb874c1a5b7d : SUCCESS in 25m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e6bfb92a0c2f4caeb120d0ee5f81676a : SUCCESS in 29m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dae92b29d1174792ab1cefd2c34af9c8 : SUCCESS in 1h 00m 18s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b9d07e15ade14d2695d2f9f61ff1860e : SUCCESS in 1h 39m 46s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/291d5c6710a14658bcb313e66bc299be : SUCCESS in 1h 07m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"376d5efdda6fab18d4486e9f150eeb933a869562","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-02-26 14:03:39.000000000","message":"Patch Set 1: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"40b009951362f4699fb214e61ccafaa5983dc127","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 14:38:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e03f2f3cfba54c9a9a2dc27973245a88\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6a731de41d0640399022f7822e1b03c1 : FAILURE in 6m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/85289b60bdf84cf7bca8ec3b72a4b1b1 : FAILURE in 4m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2679b8c5898444f4b7771be34972b1c0 : SUCCESS in 2h 37m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/064944413a0c4ab79abf42c3d85a9069 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee64a6fc32e44128b25117224c9e2b5a : SUCCESS in 5m 50s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/daae0091acb940feaa6e11c827301cb7 : FAILURE in 4m 03s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8baabbdc0ac649ae85deaece3f4529a0 : FAILURE in 4m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d5f41be2b6e8c3a8d437fd4c0acc84fb3115c0c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-27 10:10:58.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"8513df758deb24a1e72eabc3ae08e251f0546234","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-19 13:16:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4096cfd61c9405159ac94980468692d4a42b859d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-21 10:18:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77c157bc7f424bd1f69e66852e7490d6a6e5f830","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 10:19:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"206431b7c665cedc7713c04b1caa220ade273302","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:02:37.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e7f356fae7f412d950e0f8c9c35bc4d\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2bd69c2dbaaf46809ae5836fc506f59c : SUCCESS in 40m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b970d32e8ddb4ceca55f6e8ecb4582b6 : SUCCESS in 46m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a8371f1e53054973ac10b9f1f8f8a4c2 : SUCCESS in 1h 25m 59s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/05b1bf1e98ae4b2197438945a018fcca : FAILURE in 43m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fee34390affe4ae8b81bf2ab40103a1c : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69f0ada580bd4cdca36a0bf4bb3f664b : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5abfd1b8012548a7b267b6df96a6fafe : SUCCESS in 1m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9438a595a904a2e85d1e52c582e2385 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74d3feacabbc4371ba43ebbfb727e83b : SUCCESS in 2m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f675b46784a94f8d90deb90bae47ce84 : SUCCESS in 2m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0b98ebf2a2144423be42d5cfc9ace1d4 : SUCCESS in 15m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3b301dfed2b64d8a92f7c0dfb0d27f93 : SUCCESS in 1h 13m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/86178e3ea5d642109a46b051d07c6664 : SUCCESS in 1h 06m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/841a06ed11ae4132b07c702a8da0d30d : SUCCESS in 1h 38m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7830b6f7619b461091b934389cea1642 : SUCCESS in 25m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2d2830bfaa1d45bf8b3c1f4fa3884682 : SUCCESS in 51m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3442f749e6704250b04d19748fa65e3a : SUCCESS in 1h 16m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3bc88be8554a49d0b6700bbfbd78813b : SUCCESS in 1h 26m 52s","accounts_in_message":[],"_revision_number":1},{"id":"01e70e9b519f184b36d1f55be5e6cc2d64370073","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-21 12:04:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7139ba8f39095216712cd31555314c134c3790d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 13:42:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/416e7f4f7d2a476ebcb0424f48b779ba\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5b050f8e3ef74238a09ace9258fcd88f : SUCCESS in 36m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2afb6388568148e0860b02c991187294 : SUCCESS in 32m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a55ad48f05184015aeda08eee580f640 : SUCCESS in 17m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b67b5449062e4d3c89a6c7716e8a56bc : SUCCESS in 31m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8306dce98e724d91919017f4ace9e996 : SUCCESS in 1h 07m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6f345df9696048368f8f2040aa6f63cb : SUCCESS in 58m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28bf0b2db49c4b85af848ecde299eedd : SUCCESS in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/30b78f29669e41c5a59a8ef5e226c4d8 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f44b8823677941718ebb8600e7549e9b : SUCCESS in 4m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efe571909ae449179831a68176370ef9 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b85bfe002be49f1b42769eb3efe174e : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20f937ba998548aaaac55fe5e90c3776 : SUCCESS in 2m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5db0f0b151e3497eab7119d623b32f29 : SUCCESS in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/026cf9a9420c48ea9c2304be88ea0d1a : SUCCESS in 21m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b005f7ba33344623aea26668951ccf7e : SUCCESS in 1h 19m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2d6282546d3c4dcbb6eac6a50fa61b5c : SUCCESS in 1h 06m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d75b60b3b208455ca87e820620e14fb7 : SUCCESS in 1h 18m 29s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f04a7edb03f74e969603001b697a6c86 : SUCCESS in 1h 20m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4c124136b025409395dba0c9dddf5047 : SUCCESS in 24m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d8faf3ed6e264b44b0e48c5ae9117e53 : SUCCESS in 24m 35s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b7dbf8c2c1384bcab85a293308d0da3c : SUCCESS in 1h 03m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7d1c6bead23a48dcb0f329a5565cc239 : SUCCESS in 1h 24m 56s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8bbd8facbb0041868136d3666430a54c : SUCCESS in 1h 18m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c60bd96ec38bd4972e4963381cd8731e921f68eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 13:42:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b093b68d37a53c947f42a45c8a9ac3b84377ad8d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 15:09:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28f2931c958249ae9ea85a58c2ce6cb3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b06211b0cd1a479b800178e5fc65753e : FAILURE in 2h 23m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/54c3dc5212304ee8ae307c507a176e17 : FAILURE in 35m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07796ea977054785be0ac90e1d07dfcb : FAILURE in 6m 52s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74d35d01e0064658bd002278d02508ca : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a72fccacf1f848308954bb0a0de969d5 : SUCCESS in 4m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e110addf083c413fbd8e6d873943f910 : FAILURE in 2h 02m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9163fb1cd555467690a4521243c10ec5 : TIMED_OUT in 3h 01m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31bee5a208a4fa31f1669766c9313aca3e8af039","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 15:34:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d1e76fd11364a2fbf62d2572e34a2cb\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1e6b23e900664257ae8809da107001f7 : SUCCESS in 10m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/72cdc05f83ef49b59ab985fdf0b271e1 : SUCCESS in 29m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/69eba0f37fae4d37b6618b40b214c1b0 : SUCCESS in 58m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5f26f2f2938b42bb858ea006965f4f2b : SUCCESS in 52m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced2e96ae3eb4702a23f4cfdd39716f0 : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b215ffef548e4025978f73fd0e7a8abf : SUCCESS in 1m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e8fe079a63849248f76318504e432ff : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f861d94eccc4c3c870af70b2c627c1f : SUCCESS in 6m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ee7072e0d48431dab5708f1bd80cdf8 : SUCCESS in 3m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b845067ba93542a38ea35635156752e2 : SUCCESS in 2m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/54408112436c4918aa87dc94b3499813 : SUCCESS in 23m 29s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/4adabe7213594925bf19c955241ed567 : SUCCESS in 1h 39m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/34761927eaa249ecb06a4a41ffb9d2ba : SUCCESS in 1h 06m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b5c52d55b5114595b9183a9274d3e3da : SUCCESS in 1h 29m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/738d22cbc4484b97a99e887bbd1016fd : SUCCESS in 25m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5abf0ba32a174a17af7c2b5f86569e3d : SUCCESS in 31m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0b65de5dac3946cb8dda9d5a439b5ded : SUCCESS in 1h 01m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1658ba68bda34c63875339b15dbaf30a : SUCCESS in 1h 38m 38s","accounts_in_message":[],"_revision_number":1},{"id":"4b375c1d83e8f009ebc31cfd53b4e4619ce32a65","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 15:34:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"135520db99e82774c53ea147da0221170f4721ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 15:35:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16b7fd46dbce4b568822d8ff73e9aa22\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3bf4b3ef1a14328b83018db5e27e087 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1714029fbeef4b1dba3eaeb968bc8a29 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2837d8538a74cbcca391cff1279a2e47264c88d","revisions":{"d2837d8538a74cbcca391cff1279a2e47264c88d":{"kind":"REWORK","_number":1,"created":"2025-02-26 11:57:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/942785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/85/942785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/942785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/942785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/85/942785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/85/942785/1"}}},"commit":{"parents":[{"commit":"9efdcc01ca5a66304568a6582a182e4171e3e625","subject":"[release] Use UCA Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-26 11:55:26.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2025-02-26 11:57:11.000000000","tz":60},"subject":"Add mtools package to ironic-conductor image","message":"Add mtools package to ironic-conductor image\n\nIn certain deployments ironic-conductor needs to create floppy images to\nmount in virtual media. This needs the mtools package installed.\n\nChange-Id: I00ccc47f7a993f13e2b2feb3f6a2565dad32b80c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2837d8538a74cbcca391cff1279a2e47264c88d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d2837d8538a74cbcca391cff1279a2e47264c88d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
