)]}'
{"id":"openstack%2Fkolla~944100","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I3ba85069c265844bd2f8b5e6d2b14f32e4898cfd","project":"openstack/kolla","branch":"stable/2024.1","topic":"kolla-ubuntu-noble-stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-31 21:24:22.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"34076":{"account":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"last_update":"2025-03-31 22:55:11.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-03-31 22:55:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3ba85069c265844bd2f8b5e6d2b14f32e4898cfd","subject":"Added systemd-standalone-sysusers package for Debuntu","status":"MERGED","created":"2025-03-12 10:17:41.000000000","updated":"2025-03-31 22:56:09.000000000","submitted":"2025-03-31 22:55:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944100-kolla-ubuntu-noble-stable/2024.1","cherry_pick_of_change":930931,"cherry_pick_of_patch_set":2,"meta_rev_id":"1a5a8dbfb837e3def289d38d91d68109348120d6","_number":944100,"virtual_id_number":944100,"owner":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-31 22:55:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-31 10:22:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-13 11:52:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-13 11:42:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-31 21:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-13 11:52:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-12 10:17:41.000000000","updated_by":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"CC"},{"updated":"2025-03-12 12:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-12 13:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 08:00:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-03-13 10:26:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-03-13 10:26:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-13 11:42:04.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-31 10:22:04.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"c7a5a6bddf77f8606fbc3e29715b58e7f0d94cb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"date":"2025-03-12 10:17:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0daed048301d5973d4684e03073932df59173e3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 12:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03a1a2faf5444636a9ba3b10619c69d5\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f57affc83744090ac1d46ee708c4a8f : SUCCESS in 2h 11m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/26c4679f58664c3f9f8d687d8b341542 : SUCCESS in 1h 25m 10s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/27d703bb63b84d47b66f6767997d8d4a : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/689428733431429aacf23c2020fb4052 : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16bd23010a918fe9c86688275f7aadde046f79c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-12 13:03:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce3b14157c54cc7aa2b7db938593275\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/cd544762e3984968be12e686d3661254 : SUCCESS in 24m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2e68737dbeb74f9092f2fd9e89e83e09 : SUCCESS in 34m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7bb34dea92424f2ea74dbd2bf792deb3 : SUCCESS in 24m 58s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3a2adc1fabc4412c8874548669f98dbe : SUCCESS in 33m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7549b22c2ccb4761b339912b4cefca51 : SUCCESS in 37m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2eeafbd297b04e52adfa471b21961afd : SUCCESS in 1h 25m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f98d0627699465585d08b377a0e4a01 : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9548f2dc029f4f2d918e124737910410 : SUCCESS in 1m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac3ab503c7f646e9af90b5ee10f4883a : SUCCESS in 4m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0367961b0614c94934dd989fce1f6ed : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9deb16a19f5647828ce4150770f6d9dc : SUCCESS in 3m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f7cfea086e0454884414dfcc1f4fffa : SUCCESS in 2m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3687195e22ef4348a0b36dc497218806 : SUCCESS in 20m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a9547e7c111146d4a07478df4678c800 : SUCCESS in 27m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5d1390bf685f48d2972a0aec52d0c9b3 : SUCCESS in 1h 03m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f5340424b4a046c18beb2bd3588e7331 : SUCCESS in 53m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/76f94571f5364849b9cbd1cdaca15aa4 : SUCCESS in 50m 59s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7261cdc46c954543ae9091761344e8db : FAILURE in 21m 38s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/0bac3c6730b7471c81ad2ab4a754e34d : FAILURE in 27m 51s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/195f4b81c9fb43f3a580222303f355e7 : SUCCESS in 1h 25m 52s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c1bae6cec1eb4818a4445b222e6b79f0 : SUCCESS in 28m 02s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/b8612d6d49c547a7bdb1d1df268f6082 : SUCCESS in 35m 12s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/141b70390ca04ee684bd0694707afaa4 : SUCCESS in 1h 03m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/7b857d6cc620455eb0220f5aa3b419d8 : SUCCESS in 39m 09s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d721fb9715a5477c8c5c709664150ad7 : SUCCESS in 27m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c77aa388635744cb94c9d1f628c79632 : SUCCESS in 18m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6b4e1f52c9b648248b925f459e678860 : SUCCESS in 31m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dc9657da90bd405fbe8fb88e448e3307 : SUCCESS in 50m 54s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ea8721486ce47e3b90109d961893c8c : SUCCESS in 1h 09m 43s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/e7057c79bdb9439f8c04c01e94d4e196 : SUCCESS in 1h 30m 44s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d4906fc75b224530bb061bd5cb8846ed : SUCCESS in 2h 25m 44s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/68c1a9c795564d06b61f0289acb2995a : FAILURE in 22m 36s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/3d1c4e10d0aa46b1a8a1d2600e19de7e : SUCCESS in 37m 54s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/bb8d428dd7f44bf3846a7cb920daa4ce : SUCCESS in 20m 40s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7d88f8cf319242b5b37e17f9d97d9f59 : SUCCESS in 52m 55s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9a9fdd422dab432fa1ae92cdd68b951b : SUCCESS in 56m 26s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/70f6164fa6214b548248e189676324f3 : SUCCESS in 36m 02s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/09cdae7e4fed48b1846f4bc48c136b7d : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b5bdb1a8d88c08a321b1bd4e4ecf07de282b2b2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-13 08:00:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e088bb9b3ad3beab6ab7fca76b69ff459196cf0c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 10:26:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"beebdd4723e52e9a4ead5db5dbd078f7089fb5a0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 10:26:56.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"d18de9483836a8444db74aa0e9c8a25887ce7bc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 10:27:18.000000000","message":"Patch Set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9b4acd4e883200c2ee42890098da02185dadd834","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-13 11:42:04.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0384c3498be0418b7d73df7480b45fd3f710a695","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 11:51:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aac5300bfff64e44819952c949b9e357\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/685e17ba6431440eb8a6aa53a39e11cf : SUCCESS in 41m 02s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/adf9840668424e808f5557b6e688e46e : SUCCESS in 1h 22m 48s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2509d7d7f9e444a5a4997142fc5bc99f : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/12fe300683dd4c0fa692bc71d9444861 : SUCCESS in 4m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c545ccd9e084833cc7b1426bc3e58b1d3344f4df","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 11:52:16.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"19f0164848ec06608bc9cd5b72ca6b4474378fed","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 11:52:35.000000000","message":"Topic set to kolla-ubuntu-noble-stable/2024.1","accounts_in_message":[],"_revision_number":2},{"id":"3645592c42c6ba0ccee60f4ff9a0eda92fa39f84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 13:40:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/100d7f0675e54b92bb4dab185c9899c4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a8c65a50f5aa40ac87993508e2d4784a : SUCCESS in 18m 15s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a761d866395743638af4a6d1e5cf83b7 : SUCCESS in 57m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/dc9c6510ee9f48cf8436311d3f703354 : SUCCESS in 14m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/58729226ac0140aabf0a063e63476676 : SUCCESS in 54m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8213f0cb4ee141f29c7423deaf134107 : SUCCESS in 29m 41s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/11acbcef798544e68f018a93a8246ec8 : SUCCESS in 2h 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f845ba1ba5784f2e81025eb4dbd088a9 : SUCCESS in 1m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/165bed14101d48aea3c458dff5f4854f : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/61cbfb35044d4b17aff517ecffee64c4 : SUCCESS in 3m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/169a315040f44795b6e95e93d4cdfdf8 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03a16abfe1e455487a9477194ec0fda : SUCCESS in 3m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6a20e5526c854971bcd1564ab916d57c : SUCCESS in 3m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1e721d2fa3c483fa8a34937a638627b : SUCCESS in 20m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3a9872bff2594114bb6159e2cdd3f6c1 : SUCCESS in 13m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4a88e93244e7440891df974441bfa1d1 : SUCCESS in 1h 18m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/173623bd58334de5951cc7bb1aa94368 : SUCCESS in 1h 33m 25s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/b5b12934e29445ef8094e5c6ba2adab9 : FAILURE in 1h 52m 23s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/87bfa2e4bdf548ef862088ced7e18b01 : FAILURE in 17m 47s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7798d32b676b4cec86ee1dfd8d8d0e25 : FAILURE in 26m 21s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/da5d729c94684d6a83965b4e039ecd41 : SUCCESS in 1h 26m 37s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/11218c7f5e494972959dc8c7ba2fab37 : SUCCESS in 26m 00s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/1366e0fb8bec43beb897ea54e1ed317c : SUCCESS in 28m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d7f581aa8a214933866b12e704bdf57b : SUCCESS in 1h 03m 52s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b1b7b3a01505484388323a4ffc405ef0 : SUCCESS in 1h 24m 43s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c5a5f8e3bfa94300bf8dce5b3052c945 : SUCCESS in 34m 22s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/daf0f084c51c465f9a9d2acd1b4ff036 : SUCCESS in 18m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fecce7215e824fe9bdb1af4ee7fde8bc : SUCCESS in 1h 08m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c859bcc51b534879a3d5606509eda32c : SUCCESS in 46m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0fd066f26b324f28b4ebe601efb5696c : SUCCESS in 1h 18m 18s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/892ff69e07e6446b9bf4146cc82cca1c : SUCCESS in 1h 21m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/02dbd9656eb24c0c84a9640552f63bc4 : TIMED_OUT in 3h 03m 05s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/847ea4fa3c1b4c0385f987603c645f17 : FAILURE in 1h 00m 21s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/fed82f9cb251460aab4b5cf8be4201c0 : SUCCESS in 55m 31s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ad70ac3a7e9e44429ad8ef7dd7768687 : SUCCESS in 24m 20s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/a5a47725e20247b8aa16b4468ed9a8ed : SUCCESS in 1h 24m 44s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ee3347c53d2d4f9e867528a397106f6c : SUCCESS in 48m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/73aed5fc4cbb410eb7ef175147092ece : SUCCESS in 1h 13m 44s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/c9d8437b684649788d57f0564580b64f : SUCCESS in 43m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"83abcbcde8d71dc56bf396b676818469fd407940","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-31 10:22:04.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"585a755e5f54fc7222e18d3e2d43cea811afce5c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-31 21:24:22.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5cb492d0bd365e976c8fc284bcad3207f43452c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 21:24:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6338c6ea4acd9efa8ab25cd4d891bc96ac108eb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 22:55:11.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5b13c6f885940ebb969515ce833f98c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf052c25da3f47cc8e8c97e73167c707 : SUCCESS in 19m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a743bd045da947b0a8f4154254c5bead : SUCCESS in 14m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5da4b5b8db5a4fada7f1dbdc9f14a0b9 : SUCCESS in 1h 17m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5b0dad066301416f96459ce49d84994b : SUCCESS in 55m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/986507d690e8431ab72987e2069e1823 : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ab8a4ca1a044ce683ef3e104a4502b1 : SUCCESS in 1m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/365b1e8a1c224f83aeaba86f42ffaae3 : SUCCESS in 2m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/533fa8635b77468cbbcbbad321de4ef6 : SUCCESS in 4m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/53bb24aca1b940c289533053e11e660e : SUCCESS in 1m 29s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6326d7a7d4874dc69393297e5da43def : SUCCESS in 13m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/30df7827aef2477e8e6e65b0fdb7d876 : SUCCESS in 22m 05s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/775c2a82102a4afa90949f846124b6a1 : SUCCESS in 30m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/44abbe8feaee4796b6abd14240820e77 : SUCCESS in 1h 26m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9c8e539045514ce7a3d95a61ec80f44a : SUCCESS in 16m 21s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/20095c4c4f844855bad054a8d4c295ed : SUCCESS in 56m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/88af78d480ef40e9928fe1c61ae90cbf : SUCCESS in 49m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2f061076e934416bb7ddd2fcb4704d8a : SUCCESS in 1h 03m 11s","accounts_in_message":[],"_revision_number":2},{"id":"a6934ac855316cb9c99a52da41f7237d66a3b96c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 22:55:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1a5a8dbfb837e3def289d38d91d68109348120d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 22:56:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68244d7a2fee4f9991bb345e1b6df98e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ee0f192f9704bfc9f3c9052ab1ba4c5 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d3cec115326c50e4f73c40805a58a585baf9d56","revisions":{"848811a192d5d1e830396c6d639517c44ec5d9ed":{"kind":"REWORK","_number":1,"created":"2025-03-12 10:17:41.000000000","uploader":{"_account_id":34076,"name":"Jakub Darmach","email":"jakub@stackhpc.com","username":"darmach"},"ref":"refs/changes/00/944100/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/944100/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/944100/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/944100/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/944100/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/944100/1"}}},"commit":{"parents":[{"commit":"9ac6af71cfba2a2f15c2eec4cb8dca712162e999","subject":"Prometheus update all exporters to the latest versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ac6af71cfba2a2f15c2eec4cb8dca712162e999"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-30 15:49:26.000000000","tz":120},"committer":{"name":"Jakub Darmach","email":"jakub@stackhpc.com","date":"2025-03-12 10:17:41.000000000","tz":0},"subject":"Added systemd-standalone-sysusers package for Debuntu","message":"Added systemd-standalone-sysusers package for Debuntu\n\nThis package provides a standalone sysusers binary with\nminimal dependencies, designed for use in non-systemd\nenvironments.\n\nThis package simply ensures that all packages that want\nto install systemd as dependency will have to settle for\nthis one because the systemd package breaks\nsystemd-standalone-sysusers, as it is an alternative.\n\nThis backport is necessary to fix failing Opensearch builds while backporting Noble support to 2024.1.\n\n[1] https://packages.debian.org/bookworm/systemd-standalone-sysusers\n[2] https://packages.ubuntu.com/noble/systemd-standalone-sysusers\n\nNeeded-by: Ie12c43dc1de122c203edc28181366457ccd67834\nChange-Id: I3ba85069c265844bd2f8b5e6d2b14f32e4898cfd\n(cherry picked from commit 55168f909d3c95b6dfc386d3fbee25fd20a7cae8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/848811a192d5d1e830396c6d639517c44ec5d9ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/848811a192d5d1e830396c6d639517c44ec5d9ed"}]},"branch":"refs/heads/stable/2024.1"},"0d3cec115326c50e4f73c40805a58a585baf9d56":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-13 10:27:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/00/944100/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/944100/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/944100/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/944100/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/944100/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/944100/2"}}},"commit":{"parents":[{"commit":"9ac6af71cfba2a2f15c2eec4cb8dca712162e999","subject":"Prometheus update all exporters to the latest versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ac6af71cfba2a2f15c2eec4cb8dca712162e999"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2024-09-30 15:49:26.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-13 10:27:18.000000000","tz":0},"subject":"Added systemd-standalone-sysusers package for Debuntu","message":"Added systemd-standalone-sysusers package for Debuntu\n\nThis package provides a standalone sysusers binary with\nminimal dependencies, designed for use in non-systemd\nenvironments.\n\nThis package simply ensures that all packages that want\nto install systemd as dependency will have to settle for\nthis one because the systemd package breaks\nsystemd-standalone-sysusers, as it is an alternative.\n\nThis backport is necessary to fix failing Opensearch\nbuilds while backporting Noble support to 2024.1.\n\n[1] https://packages.debian.org/bookworm/systemd-standalone-sysusers\n[2] https://packages.ubuntu.com/noble/systemd-standalone-sysusers\n\nNeeded-by: Ie12c43dc1de122c203edc28181366457ccd67834\nChange-Id: I3ba85069c265844bd2f8b5e6d2b14f32e4898cfd\n(cherry picked from commit 55168f909d3c95b6dfc386d3fbee25fd20a7cae8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d3cec115326c50e4f73c40805a58a585baf9d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d3cec115326c50e4f73c40805a58a585baf9d56"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
