)]}'
{"id":"openstack%2Fkolla~944273","triplet_id":"openstack%2Fkolla~master~Ic342d9d96dfd8ea03f89cdf04ff28e4a7070a54c","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-19 12:48:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic342d9d96dfd8ea03f89cdf04ff28e4a7070a54c","subject":"CI: Add qemu-user-static back","status":"MERGED","created":"2025-03-13 15:01:11.000000000","updated":"2025-03-19 12:49:39.000000000","submitted":"2025-03-19 12:48:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944273","meta_rev_id":"616f2020a3fa881415f7a697776e8fdeabbcb84b","_number":944273,"virtual_id_number":944273,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 12:48:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-14 08:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-13 22:41:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-14 08:30:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-03-14 08:30:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-14 08:30:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-13 16:45:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-13 18:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 22:41:15.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-14 08:30:49.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"70395000e0347d6819d6b83720cbcf6116c76956","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 15:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efa442f4ed52aea4feaad7b4b23f18f22a95e85d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 16:45:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e5a8d98381343e58eb6e8b9d5444ac9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/4dc435db27994e01bc09cb482d0f6b76 : FAILURE in 1h 43m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fc3b77d9df44889b9f3b7525a379262 : FAILURE in 1h 17m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b80069124245454e902296a830eeff8d : SUCCESS in 57m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1241fcf3411a4ba59e19f599f8c3aa1d : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dc4ceaddd27d47459ca664a39b4e9bee : SUCCESS in 3m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3f9adbbf100b436084728c18098d3f17 : FAILURE in 1h 28m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/26df57ec9b60489097828228fc9aba88 : FAILURE in 25m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b099af79291ba1477de409e53c2acbf9c70424e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 18:08:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cf09437772d4600930e636cd80d7fc4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9262ba2acfa345d49a9893e41ca31295 : SUCCESS in 28m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d3f9f7a1806d419b8b657d61cf4e33ef : SUCCESS in 1h 15m 25s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a8b86fef79f4452d8b9942265326e764 : SUCCESS in 19m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f34a4e2a63c14cada8c14cea2a44167e : SUCCESS in 30m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4959cb97ac034683822ec4a50071ec4b : SUCCESS in 1h 42m 12s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/21487ade83dc46e98e9e55948edc2900 : SUCCESS in 1h 49m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e0728e2d2134c8588ea5913f576354b : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5232d4e527a48efb1c72d63f882670d : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ffb103bcb77348d6a6a3b7604342cd35 : SUCCESS in 2m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c47e07b7a534e31af64ff2e8a37e713 : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7e8591a684d4013915be8d435c276e6 : SUCCESS in 3m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0912484d52e94b6e98d84c86c3b8efe9 : SUCCESS in 3m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b96aec8ca47c499f92a0286237fd5957 : SUCCESS in 27m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8962bdf728824438b300163dcb1a6862 : SUCCESS in 1h 30m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c07c68f331ac428ea72ec2f95094c81d : SUCCESS in 1h 08m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/28e3821255d84903ab23df32058907f1 : SUCCESS in 1h 34m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3d3e3006b84745e3a8b6ada19329757e : SUCCESS in 28m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d8120bab07c74ee3b312e9b654822526 : SUCCESS in 26m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3a591e1498c1484d8e35cce2ce23156a : SUCCESS in 1h 14m 59s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5afd186c13584ac2804a378f8f609c08 : TIMED_OUT in 3h 01m 16s","accounts_in_message":[],"_revision_number":1},{"id":"303f341e1a20e99ff4934a35a7d206081725d181","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-13 22:41:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"912f3d28f8e74454fa1267afdb4f70fb6f55af42","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-13 23:31:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2c215ee9a6428d20a9aa84e71260d9dccb496be6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 01:15:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338e110b4feb48e282ed3f9cd888ef00\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/266e181a0e9d43478b75a1983261c3ec : SUCCESS in 22m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5900dc5dc90742eaad5a067a57e11cf0 : SUCCESS in 52m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4766e54752674771898cf4610c6af48d : SUCCESS in 15m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2fd345c8fe4547e29a3aec4d4ade998f : SUCCESS in 25m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/75d3d34c56dd4bdea8c4706b2b72431d : SUCCESS in 57m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2b471797426e42ff84955ed5942370d8 : SUCCESS in 1h 43m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e4df0b03a744e19748d80e79716076 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0c621d2f1474e8d8fde8c30058823b3 : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7c5e6bfac20043d0ad26b47d29f4672f : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6744c47837e749e49c348a5f4badb0db : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b88ca1cfa7244382ac3282148baac2f2 : SUCCESS in 9m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3ea0e953923f4386be84d917787cbb70 : SUCCESS in 2m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9c2eacd760454135981a562bb4014171 : SUCCESS in 15m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fffef63018b04a2795ac935f43ca44bf : SUCCESS in 44m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9d64d2fd8df44aa48b12eac286485419 : SUCCESS in 36m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/76315b003e854902b762f89937b73f20 : SUCCESS in 46m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cdcd7598fb6041e196dbef2c72acbb27 : SUCCESS in 23m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/080789ede83442ac80ec1a80174b12d6 : SUCCESS in 31m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0b0f8c4b5156416696cc1b3561678175 : SUCCESS in 1h 09m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d6766132372a4407b72ef2fa01cf867c : SUCCESS in 1h 24m 36s","accounts_in_message":[],"_revision_number":1},{"id":"fc84627bccc476dcc94a1180594e663a42617f4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 02:17:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63121a0a146742cd88dc0d964d039198\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2659df64f0e940c38d306382f0884a5a : FAILURE in 1h 18m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fe0aee5040a3476c8e0826f3f35d81bc : FAILURE in 1h 06m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bf6b8c17adef4d5d88a2920de2e44124 : SUCCESS in 51m 41s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a5bc2c42508a4130a889676129f814c8 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/906483b1cc654751a34f8ffbde82ca82 : SUCCESS in 5m 20s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2a9fd4a092f94dd3b1651628c1da9db5 : FAILURE in 1h 05m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/da8019b633d749a281a616df91afde58 : FAILURE in 2h 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"83ebe1d032886d2d82281e15213e01f7b7a66e20","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-14 08:30:49.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"435486ed161b05157c9e06010f3d44a8ccdbc049","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 08:31:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f8b4ad69eb8831e15cad9abdf57436ec448b041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 10:02:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cd2e862bfdf4e6aaf9f09e9d94041bf\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3ef8b45a0f264286a295a956b148e462 : SUCCESS in 21m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0498be2d090e4f4b8bf9c47c51defd48 : SUCCESS in 32m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dfe62f5aa5514412be8a7667f73247a6 : FAILURE in 9m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf1c117fb0c3438683e71ae38a48f580 : SUCCESS in 1h 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0912726c3eb148e8ae1aee50fdae74f9 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97aba59b030747549b9ddfec505f8a03 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/70a8152a4ceb44408c90c4ebe86233ed : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f686ca3c449d4fe5812257a085f8e8e2 : SUCCESS in 5m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2c702a7357784589b17cde641667d7e3 : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/fc31f8afbd9f414da2797f6efda534b4 : SUCCESS in 20m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/63bc9f4e2c1840cbb3e3f4782f6715a9 : SUCCESS in 1h 25m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e468077ce63240a0878b2c3f630562ea : SUCCESS in 1h 03m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0d972cfd7cd140e7a06680d088c04dc3 : SUCCESS in 1h 22m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/99f9b674441c4c239316e1388052ac4f : SUCCESS in 23m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bfa2fc2f2f7d4714977b97e4096d6bd5 : SUCCESS in 24m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f3adbc1682184f79836b79e8cd55786a : SUCCESS in 58m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6bc4a6abf46a4e528625a2a7a766c930 : SUCCESS in 1h 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"90df25183040f6f46320a948cfa82c90fb500449","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-14 10:11:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"072a29a5bedcae27c628b352227ad7e53ead7dcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 11:59:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73d319cd62014bfbaf3463e298102e29\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d01f3de8965d462595426d77fdba6a92 : SUCCESS in 25m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8af27ae3e68d462bac12a596fa2b0ea3 : SUCCESS in 1h 10m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8fdd2dd2f7d34d5fbbbe9cbfbbad2f5d : SUCCESS in 10m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0f8b0364f8234f0e8d6741b0ef639c16 : SUCCESS in 30m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e7b5a15410d54aa68fdcdcf841b417c8 : SUCCESS in 57m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8edc539940f741c698dbd6627f833955 : SUCCESS in 1h 42m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97747b6f81d04af68143d293b95c4cd3 : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b01dfbb48c354f679a7cbc838bfb2678 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b17f85aac5e40028337d8ccd677d88d : SUCCESS in 1m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfdefe61110e48daaf66e30add42a0ba : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e7298e5a9f446e6902f4ce8a0dbd034 : SUCCESS in 5m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0998654ec03a44cf91bab3e5819e7838 : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7b391dc696b3419fac0cb967af492d03 : SUCCESS in 14m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5772a6e567f547b5b4c19c48104c7de1 : SUCCESS in 1h 08m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2b23a39406b746558f01e1c1dba17071 : SUCCESS in 45m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8f8b348a486b4878a009977c85f1bcd5 : SUCCESS in 1h 34m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f1b856234f742ddbae70f4f30d33d69 : SUCCESS in 23m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bf613daa767d48b6b3224859676685b2 : SUCCESS in 31m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f905269ae0b94b6d9c20039bb081efbc : SUCCESS in 1h 11m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/23f37cc0063048cbbc816b6e69ed8623 : SUCCESS in 1h 30m 03s","accounts_in_message":[],"_revision_number":1},{"id":"758b2ee5a2b23b6f8a0059a45a3b3aa6670fda76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 11:59:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d86014a2470458b9f38ed74be25f626c34a010c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 13:17:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2afd56e4e63d40e3b59d5401000bb3a5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/25ee41fe25bb42508e988d1b5bf42240 : TIMED_OUT in 3h 01m 20s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1727e645945d46e3bf88c3d5ce0e195f : FAILURE in 1h 24m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dbc32d6332da41eab26dd8cc81bcb622 : SUCCESS in 58m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a9e0fc12ad1a4709988aad38fd9df1cb : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e59989028ecb4b5f8d2d6f81b7d519df : SUCCESS in 3m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/08f337b8745f4969a30957d25027b053 : FAILURE in 1h 56m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/61b9391ac2e6423ba284dc838f88b412 : TIMED_OUT in 3h 01m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e3f2873ec38c34df379884f542f6b6e579ca0b28","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 13:49:20.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/af662dc4674d42d6bf53b5a3f7f780ad\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d033e236a6794977b883c8c740cdd594 : SUCCESS in 20m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a4e0e16df078483d8dec97a553d0f877 : SUCCESS in 24m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47dbf5af334342eabd3307c150ebf009 : SUCCESS in 38m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f303e89895ed4687b3b980dce112f928 : SUCCESS in 1h 29m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/875143ac81db47db9375381caf314a74 : SUCCESS in 3m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/43822cd6c12b4f20ba3bc4d73e4e4689 : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ecb41c8a2124db7ab974742b5f7f883 : SUCCESS in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aceb923eb40647a5b53ad1d6b12cb343 : SUCCESS in 5m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cb04746eeb7142c3871bbcc0f95e275c : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aa55a178fe5f4e50b82cea74ea0b323f : SUCCESS in 13m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2e2e2aced689409eaeb7dfbb83228812 : SUCCESS in 1h 38m 03s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f7470c7f96774f5bb90be6927a501e10 : SUCCESS in 1h 02m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/47560a8f813742d9acecce99d98e4161 : SUCCESS in 46m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b481962cc2ef40a6afdf4264fca1001d : SUCCESS in 21m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/83385030d305403b84200672eeac5ebb : SUCCESS in 29m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0ae23b1b3caa44d182d3359e21603693 : SUCCESS in 1h 02m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c25df32a367946feb4bc5cdc9363805d : FAILURE in 58m 27s","accounts_in_message":[],"_revision_number":1},{"id":"12be7d661fae6a3fce3865bffdf988333642be22","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-14 15:39:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"31cc2d19971180f3185ee8f549ce7d1471e457f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 18:11:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51e260f0c9da4f1cb629ac3daab58146\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/efee2c3fe826411d93931b99756f68a0 : SUCCESS in 25m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b46bdddbd26541a0943abe97ac2461e7 : SUCCESS in 1h 18m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9e18232e3f4540deb500e314bd2135f2 : SUCCESS in 23m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e42db17fcd2d4f5b9a81b1285da425ef : SUCCESS in 31m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e971a87c32644a35adfe4f14b33952f6 : SUCCESS in 1h 04m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8a0d095300a44377b50fef79399a9ca6 : SUCCESS in 1h 25m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb6e366a21ba4c4aa152ec6ed971e1a0 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6901b57b79554b24acf39b5d0346bd36 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0805ee30035448c39cc05b6d9171e8a8 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7c1679a85d743caa754a99e7942bf36 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a3c222003474382a39b1158cd4695f0 : SUCCESS in 4m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4a61b229ba34401cb9abef3b099c0bcb : SUCCESS in 2m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c2a627b06f524c49b345a5ed892f72e0 : SUCCESS in 24m 36s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/57713ae2db8748919acf65de7658a2bf : SUCCESS in 1h 24m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6dede83372ec46cf90a4d836bc605155 : SUCCESS in 1h 06m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/94791bc269644b039a49ef65cc350c29 : SUCCESS in 1h 34m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/38bfd22315c54a2d9d6a8820d4e8880e : SUCCESS in 11m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1190fef2234f465685736bc76bccbf04 : SUCCESS in 27m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f4dc079b48524039a8d282b228d9cd31 : SUCCESS in 32m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ba62dc8c53cf43098663c75a8b9aa601 : SUCCESS in 2h 12m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5cb5c286a62e522942f216a1fd5580ccfc47c109","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 18:11:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc19a676f595e024407be4a3b9d841056c358002","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 18:47:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63c5f7d7214a4439b8b88c9cff64a817\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ac0e36d542754658b3edd79675395a54 : TIMED_OUT in 3h 02m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ba83e5d5035e488998c46ba2a4fd8843 : FAILURE in 1h 47m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7fc12a7107c6471daea5f06649006abd : SUCCESS in 55m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/790bbc39b25e4188a49781e4d96f4de7 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f125ed03e1941ac9111fb094a57dc9e : SUCCESS in 3m 44s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cb51749f388747889a136c4d9159718f : TIMED_OUT in 3h 01m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9f451dac03a34097a41b91ce5fd39b76 : FAILURE in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e70eb07e55abfb07c0ef798106983b8d9d8a36e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 19:57:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/380e739b29264b118c9c375e27f7be3e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aead7446928b437ca1d053d9d975037e : SUCCESS in 17m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e921c6a6228942a19c6a8e74e9023579 : SUCCESS in 42m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fe6af4ba71c3419bb44ab57719aa8798 : FAILURE in 1h 08m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a8ec5b892083457091683a556e7208be : FAILURE in 1h 30m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d6ed79c58445b2977e8a0fa19b15d7 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a2cb1d737924d948f94df45558163f1 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad1c12c97d004af8bc5c82ec42207ca3 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1140b8a5bc4473c9ce970c07c8a814b : SUCCESS in 6m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3157401d467040399a6268304fa2e10c : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7a5c11c23ae8409080b272872d576aad : SUCCESS in 33m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/28d8195b5ad44e6aa1e628890df18d81 : SUCCESS in 1h 37m 09s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bac606af410d421cae0405a67879a211 : FAILURE in 57m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/67e720e5f1fa4cfa82f0c59e17fe8a15 : FAILURE in 1h 28m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4bae1dca3f14561af1b3a37dc4bca47 : SUCCESS in 20m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7741784175724182b22bf36ab80dea7f : SUCCESS in 27m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5e7350fe7a5b4260895679cee2c3210f : FAILURE in 24m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4a9a0802cfad4c729c436cb6a910b03b : FAILURE in 1h 19m 05s","accounts_in_message":[],"_revision_number":1},{"id":"4374fea8ba873cf7275b739aa13e6c75f6e575ae","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-19 08:55:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af217edac66e476bad73556164d51e2dc4ac36c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 10:47:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a3f6d48322e43828b9edcd5b50d862b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/319797275ef540eb89593df779266603 : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8c496cc98ee641bebf651e7b1d961b72 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/34a50f2006ae434b9517c8210d19377e : SUCCESS in 20m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ba8358377faa48cc993c2de11c377dcd : SUCCESS in 30m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/07ce82caeb4b4560b158e625bd5ac747 : SUCCESS in 1h 08m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d24978ea43534cd580fb1b67bd55aa19 : SUCCESS in 1h 50m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15fb671432594557bc96fa864ed054ff : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee4a8084affa492dac44f0416a8f640f : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f59f78798c3a4390b099925139cb59bf : SUCCESS in 3m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe0ab018dea2430e8b9b2c56ec661d53 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ffad933c2a48caa02cb09dd5963f08 : SUCCESS in 6m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5a11d15dfc344841a3985a1973c19a49 : SUCCESS in 3m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6ff2de98bfdd4929899abb1a7ab9826c : SUCCESS in 17m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8c0aee152b4c42c8be39fed9a5020794 : SUCCESS in 1h 13m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2e39b8d4547f41a78db609ce7ae6c40c : SUCCESS in 1h 03m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/aeb5f31b334b4366806913a77ae55b1b : SUCCESS in 1h 44m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/afc31ef64ac642e4ad7fd84833bcf1fb : SUCCESS in 12m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/810cf198080b4e93853760226279597c : SUCCESS in 15m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/557d80425b04449e81e05702a7a3f309 : SUCCESS in 30m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8dcc6a8cc7934f7dbf0e22127fbf5c2d : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":1},{"id":"873573830c5e8d0acee44cc0914224fb1f1bf038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 10:48:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"313320d1206c7e59f91625cb9b92799737e92b44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 11:45:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf2dbed7875d4d75858d3ad2897b8ef9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d5933ebf74fb4dbca7bca9dc03b7f1ad : FAILURE in 2h 48m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66eb927baa154bb3b1e920f1e13ef4c1 : FAILURE in 1h 23m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c61b1c3486b84b2580d006bdd0e6dfb6 : SUCCESS in 1h 22m 48s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/42e38543874741e0bcf4ac8374319ce4 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14bafde8d99b421cac838e3c695c117d : SUCCESS in 3m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c33aed878f65400096a85afe0bba4080 : FAILURE in 2h 35m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0e4dc158d716475faa6bca42b03e0a29 : FAILURE in 27m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"877ac67d535e261d1e818c1b70d84efe9cf455ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 12:48:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f184ae21517046fab1b0426cbaf2f490\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/34d6f05ff2af4851a9c010165a2a53ee : SUCCESS in 23m 44s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c1435b83ade445148f45454ddfc96011 : SUCCESS in 30m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/255b58b40c114942b67f607825f83720 : SUCCESS in 1h 24m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f6ef6566d5a045f4beaa6573bb8d7271 : SUCCESS in 1h 23m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7385060f98524078bd2d22bae3efbbcd : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f132904fdc094c2b947d45ff5e9cbf08 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/590e006ea36a444b9d7203326a17c30a : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7845f33d0710498d96b017492ec879aa : SUCCESS in 5m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f931ffdbd8c8474bbf9596e8c6db9ec8 : SUCCESS in 1m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cc71d51302cd4d13b43bc4ebd10a4daa : SUCCESS in 33m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/85c0116d2101465abf78dfdddb00bea9 : SUCCESS in 1h 16m 40s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2d48df3ebc7f4d63837190f1df44a274 : SUCCESS in 1h 23m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/794bbdb8f30d4bc6baadcca3883f42be : SUCCESS in 1h 55m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/45947738417c4aadb0b3d127d2790378 : SUCCESS in 21m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3af97071cb0e4f24af866d9ce13cf8a9 : SUCCESS in 29m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ad6c30f6e244d1f9ac44329206a357a : SUCCESS in 1h 07m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1c480ddc0ca94eb8a046edb5a104a983 : SUCCESS in 1h 19m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3c2fdffb7e837ba163e57c2ca6e017fc3c27af98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 12:48:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"616f2020a3fa881415f7a697776e8fdeabbcb84b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 12:49:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75afef63520f4cf688e9959ccc6aceb2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5349d0159ef3488ca215cca725bc2801 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4c9113f964b11fd4c648802cfa7a2b3ddb7cd93c","revisions":{"4c9113f964b11fd4c648802cfa7a2b3ddb7cd93c":{"kind":"REWORK","_number":1,"created":"2025-03-13 15:01:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/73/944273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/73/944273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/944273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/944273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/73/944273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/73/944273/1"}}},"commit":{"parents":[{"commit":"9efdcc01ca5a66304568a6582a182e4171e3e625","subject":"[release] Use UCA Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-13 15:00:55.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-13 15:01:09.000000000","tz":60},"subject":"CI: Add qemu-user-static back","message":"CI: Add qemu-user-static back\n\nIt was misssed when moving kolla-build-deps to a role\n\nChange-Id: Ic342d9d96dfd8ea03f89cdf04ff28e4a7070a54c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c9113f964b11fd4c648802cfa7a2b3ddb7cd93c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4c9113f964b11fd4c648802cfa7a2b3ddb7cd93c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
