)]}'
{"id":"openstack%2Fkolla~944316","triplet_id":"openstack%2Fkolla~master~I954106dc32b4c08604d5c40e00fe53e5bb18b266","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-03-17 10:23:00.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-19 20:32:39.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-03-19 20:12:20.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-03-19 20:32:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I954106dc32b4c08604d5c40e00fe53e5bb18b266","subject":"debian: Use Dalmatian extrepo release","status":"MERGED","created":"2025-03-14 08:07:02.000000000","updated":"2025-03-19 20:33:39.000000000","submitted":"2025-03-19 20:32:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944316","meta_rev_id":"35845797374169fcc98c5ebff45af16d7af45c34","_number":944316,"virtual_id_number":944316,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-19 20:32:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-19 08:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-19 20:12:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-19 18:51:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-19 18:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-03-19 13:45:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-03-19 08:55:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-14 08:59:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-14 10:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-14 12:22:49.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-14 14:28:32.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2025-03-15 06:48:31.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-19 08:55:03.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-03-19 13:45:15.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-03-19 18:51:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"129d2b35e42ec9fa49640f095028d8c05070e4ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-14 08:07:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7e8803758a059a791ed3083cc8d09e93cdc77d77","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 08:59:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87aa00dc1e9e4e3bb91fd7167dfb355b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/04235f54cae4462cbd57792ffad3489b : FAILURE in 5m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b89fe6db84154d1d9ad7ff8dfd59e15e : FAILURE in 5m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1699eba3304b4e368e3fdce720a7acb4 : SUCCESS in 50m 28s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ef4f04c6fbf4e549dfc210b75120463 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/86b9413ce1bf4defbde05a34a32968de : SUCCESS in 3m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a9780238c36f47fdb9a91fded23245d2 : FAILURE in 5m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f1b1af2bfed34c269615ae6b951f44ef : FAILURE in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e5e1bc7aec9f63821f926e71bc0f6443fb1693cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-14 10:50:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/604b20b7ce3641659da96cc3d3eb91d0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ce54f4b466264e25bb381646288bf9c6 : SUCCESS in 29m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c75578bf30134df09e7c025bb8deffa6 : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/24a0a12db322412687ace945cda44c50 : SUCCESS in 15m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/adf2f12573d8437993cdbcd8234be98f : SUCCESS in 31m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/85f0cb4083d946b49a8ce7d938d8a553 : SUCCESS in 34m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/31fe1fdafe384481a703da32208f1715 : SUCCESS in 1h 22m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b79876be8507406a864a706e882394a2 : SUCCESS in 1m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9413027c88644fd38eac254ff3d0696c : SUCCESS in 3m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b0def6fc0064d6b9829aa7c405c21e4 : SUCCESS in 3m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a91c43ddb8af464395a4252d29268428 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3128c25ae394ddb974c751567dde063 : SUCCESS in 7m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5ab952224dc043afa762a0d203747304 : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/020ede5e8afe48ff8d93b5be5ecf912b : SUCCESS in 33m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/76c63ea5dda540a2bfa2b43c18fceeb6 : SUCCESS in 1h 25m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f2e099c6b3784ee6b7eecfb1874741a4 : SUCCESS in 1h 09m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9328c1314fff4c809440fdbd421a864f : SUCCESS in 1h 41m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/d5dd77e6975948f680f8674680a970f3 : SUCCESS in 1h 42m 04s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e507859fe700457ea5e8d574a3d82f4c : FAILURE in 1h 30m 03s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/3a02adf2d7a3413db259e9391da18a82 : SUCCESS in 1h 08m 10s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/c796a12452d54efea20fcbb78c5d27a6 : SUCCESS in 23m 52s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0adfd5e6e77f4123bf9bbd6c12c4cc50 : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/64dfaecf6c40431abeb8bae917a0d840 : SUCCESS in 1h 19m 04s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c6ff999589974ebc942b96c7c4cafec8 : SUCCESS in 1h 14m 16s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/67366e082f4b4be9bf7c718e1d874764 : SUCCESS in 35m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/354da9d1acfa48ab9133877715a13d5c : SUCCESS in 13m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/28736af7aa8c49ea802808259508d981 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6aaebda500584bc681bfce0644252c36 : SUCCESS in 1h 11m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0d1aa617503940b898d4b3af2c5681fe : SUCCESS in 1h 27m 29s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/8d41758dbefe408499754935b6b7188c : SUCCESS in 1h 59m 14s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a03ae232e99842e7876ab52d5ff222cb : SUCCESS in 2h 21m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/213685836ded458dad02b111dab4b206 : SUCCESS in 1h 14m 57s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e38bb4726387494d902a0d5ce8200b9f : SUCCESS in 16m 20s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3d1755e70301451481a88b9274900441 : SUCCESS in 26m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/344d7d5120b243dab2480285977fb39b : FAILURE in 1h 00m 42s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/55fe68df456e46b2b1a3ac0a77202dde : SUCCESS in 1h 12m 57s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/6a1ef133f3a84e9fb415932bcd630d89 : SUCCESS in 31m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"043b61c1f5b39f9c53377facaf9714a0efad2fcd","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-14 10:52:52.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"0822660529a3a38fe8b7925ab825a5df9bdb3333","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-14 12:22:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0b27c8c9377276fdd366cfb9c4fb2f3c36696659","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-03-14 14:28:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4ecc159e52211a84aec565f6f87c1696de624943","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-15 06:48:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d0c88cebb4717ec5ee97599efc4ec9c73874a7a","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-15 15:10:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ca0c3705e3a1adfb18690632a3e3965b44aeebc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-17 15:10:46.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Backport-Candidate+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"37013bbced37eb51b7c5971633f23c925517efee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 16:12:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e4651add759429a883d67c6566b5309\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9ef80b0ff1e94fa6950c40b5e488ed68 : FAILURE in 3m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/23bab4a3af8b4d32bfc5b181c339c37c : FAILURE in 3m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3d8030098e14c149ce5288bd616e9a1 : FAILURE in 59m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a92a4c27d36c4dedaefeabd78f4a7cdf : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ceeda48700b6476989f5d1899375ef62 : SUCCESS in 4m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8d9df886485b4ebaae1e7afdcca537e6 : FAILURE in 3m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/06aef502327c4e15aa57567d0280233f : FAILURE in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7e2cb53e8f7376efed2d6b95e3046f1bd0f1ab95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 17:34:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/177a2bf855f346399c4b64893ac1ae26\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5ca1da77491544e8b92f4f6d061ca04b : SUCCESS in 27m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/73c195c9064e47109b6da89d8a558a24 : FAILURE in 48m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d5752d5e6dbf4db8b7592805708d41a9 : SUCCESS in 20m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0ea5cf3e4b4246e680fce7fa55a28f17 : SUCCESS in 31m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/857524edfa7f4152a523ccb11d8a4ce1 : FAILURE in 54m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4a3f9049af73448eaacd7a0c2e85a6dc : FAILURE in 1h 33m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e9e6bb5f6b84f14b3e0e699f3157c81 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6dab3c8726d54f1fb566064aefd31661 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e5cfdb9e5344082bde3c02ba002e94d : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/579fffd98f6c4d0c83b3b5369a378125 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37faf422f80347d2901ad22e5de91b35 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/95811d7974274d42bd4bbda982e81504 : SUCCESS in 1m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/df9d8f7258f640c1be2724478c0cb61c : SUCCESS in 2m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aef530661a294fdd866787071d1e73f5 : SUCCESS in 16m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/88352520094e44ecb66d5261e6e552bd : SUCCESS in 1h 25m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8662941397094272888a0e0ad47fa14c : FAILURE in 48m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2eee125763194bec882eff9b83e6680d : FAILURE in 1h 24m 12s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/9c1559991e134d2ba412f3c369755643 : FAILURE in 1h 18m 10s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ff45f0503ab1469db206e330f4b06084 : FAILURE in 1h 38m 41s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/e02b3929e4094c6f9679c6e521e8b96f : FAILURE in 1h 07m 25s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/93d0b50df9cc42acabe3cfce4bc890f9 : SUCCESS in 15m 06s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/e3b2963ab2a74dd0835aac56ca79460b : SUCCESS in 1h 08m 06s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d373490aa09440e4b37dad412febb8a3 : FAILURE in 1h 03m 21s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/132aef30036d4af69dd32ae670065c48 : SUCCESS in 1h 38m 02s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/78980c3d04c846b6b6d96458371f3a1d : SUCCESS in 34m 20s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e531450671b44a1681860e6502b565d3 : SUCCESS in 22m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f88bab8059a8417c9dbc553373d991e6 : SUCCESS in 32m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f9fbd7d6aa9d423fa33df5e8dea31fef : FAILURE in 51m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/35c988a75c5343c095db86e7617032a5 : POST_FAILURE in 1h 27m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4c2912a4a1d747dc92795d5ab74a040e : FAILURE in 1h 20m 22s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0a7c35d41c694db192a5b607a265f867 : FAILURE in 2h 19m 54s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/bfd7c7e0376d41e1a0faf9f2d9425b1e : FAILURE in 1h 10m 07s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/88e80f5d29974f40bb66be9017aaa5d8 : SUCCESS in 28m 39s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/9cf08efe12824299b22ee62aee40a6e9 : SUCCESS in 1h 02m 26s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2a7a52994a404537a5ac3be8103b183a : FAILURE in 44m 48s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a527b9fa29ab4914beb748586a886216 : SUCCESS in 38m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/019cdfcd0e944feda0c8b0e7827a8449 : SUCCESS in 32m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"11e141324d7b420e40bdbb38cb763274c3741196","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-19 08:55:03.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7a84be840b0d6da7c1b9db41956fef9ad67c160a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 10:46:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d022568aa39c4dd7bb4a58a134d393ac\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1b6aafc61c2b464e8c0ebbdc1f9e03ef : FAILURE in 5m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aabb6b53fbf84c03ba7f1ceb6b7e04d7 : FAILURE in 5m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/698dcfdc6b234506bc831a75db6a3d02 : SUCCESS in 1h 27m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ee00c9288db488a97212840e6d388d2 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c653547dc4b447f0bf382e7eb96b0af8 : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/facc139985e5416a8137f047122f88d2 : FAILURE in 4m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b8b37d7af45b4047b3b8963f6079ba7f : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a97b2dbb2ece57646084ef8093aa4b177336d3be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 12:08:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8cf9ee85bed4a2189220f8a928536c1\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c31dbdcda141413d9b41149f82f4faac : SUCCESS in 35m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f3a24b4f7ffe45d18a91606ef56edace : SUCCESS in 31m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b430d452b86a46a5ad51f1b14b0586a0 : SUCCESS in 21m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a33f7b6e94ea4f4da4aeee463381e8ad : SUCCESS in 1h 15m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/02ab585b1f084854957855264739d06c : SUCCESS in 1h 00m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a437fa6a5f33497182c40f8fa0b6aa1c : SUCCESS in 1h 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2354b7bda0ce46b58c77fdb50baf66d2 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d0a4016f9274861a84ba8e8be5b921c : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7734ac7b1d854bd2b1525efd2c537381 : SUCCESS in 3m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a677ee7373a43c69aa2850498cb98c2 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc6ea90b13554e7ba8a04a0c7789c240 : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff654784be144c2382245db85115c840 : SUCCESS in 3m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/75858161d04941c3882ee5ea210dd373 : SUCCESS in 12m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dde5dff2e1124deca49f91ee4c7e73b3 : SUCCESS in 24m 14s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c5df05ec7eff4f9da4a0c8ef0453949c : SUCCESS in 1h 01m 44s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/08c85b77c11346b2a5ea431b22d61d6c : SUCCESS in 1h 05m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/55d07b58440441498b8b9b986052a55f : TIMED_OUT in 3h 01m 32s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/36812f7efc344298b37a24606d543074 : SUCCESS in 1h 37m 10s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b3ab9dbd2ae44b10b10ceb2b6c822187 : FAILURE in 1h 41m 44s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/138c490a159545acb4245a4e933c4bed : SUCCESS in 1h 19m 38s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/950389389f06400f8ecb735ae1a116f4 : SUCCESS in 43m 01s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/4ec9db9661e645d5ba815baf3b6a550a : SUCCESS in 1h 02m 05s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ea9509b9dab44c318efc23f15ac5c549 : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8fced7c9dc4e4be5b1504179084e99d2 : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/435414ddc01f41a3bd995807271481bb : SUCCESS in 41m 57s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/15d0066d286e4719b27178bcccba54ba : SUCCESS in 34m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/58a4dffb137b473680ee11d429718fd1 : SUCCESS in 34m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2f9c75401ed74fe0b9305b33f4984aa4 : SUCCESS in 1h 52m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c6bf8567f47647fa993e9fc990029753 : SUCCESS in 1h 26m 53s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5fa82ce5561641c1bc32b0b839e4c6ae : SUCCESS in 1h 02m 56s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4cfb8bcfb488496abdf8fa9423744daa : SUCCESS in 2h 49m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/68d7b25ed7da425b9840c5bbd06927d6 : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b96cdc405b5c45cab681a2d478904b60 : FAILURE in 23m 29s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/e2f22b9778d049518b205bfc97e0873d : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/478ee12355624ce68daee303755e1055 : FAILURE in 56m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ea49591adba447fc81bc1e22763a7c04 : SUCCESS in 1h 25m 46s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/384092bac1d24db4a2dbe39ab086abb1 : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"776de658de468e2cfaa90d4a7da06e088d799298","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-19 13:45:10.000000000","message":"Patch Set 2: Backport-Candidate+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3c0e4d7dda3d78b82de9f84feba22c5781eb1d36","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-19 13:45:15.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1b404344ff2852161f3507fd38a20a7194eb0192","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-19 15:15:09.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ab954de51706d27158c5677056eba6874f3a7642","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 16:35:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab5c75abd9754f0c888141d70a153cd7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/34bde5fd8c774356a40a4c69b0f5eac1 : SUCCESS in 30m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8e8a389a6b484e3f968c7b239a3325ea : SUCCESS in 1h 14m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9bee9b1266624aab8e10db6b02d92c3f : SUCCESS in 13m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9dd51eec22db4388963034a9e5d77108 : SUCCESS in 19m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/37a7a45d634d42ec971621a6214e4846 : SUCCESS in 57m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b283319de4284db998ad6b21da33fbdd : SUCCESS in 56m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/114cb68fe0df4c0ba4bcac22c9ee6185 : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ba3a95d2cae40e5a0bdffaf398c3bde : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e70df4e3f626427f9585c934d6f0821c : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6fd1d2028074d2aa90550cbb56642c8 : SUCCESS in 3m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa27b175e05c48958a764b2affd14760 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbd5135d8d784eddae5da6645fb88243 : SUCCESS in 3m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/69cb465dfd724005a98f6a7b4df6a441 : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/71d1add8802242f99da22b1489df11f1 : SUCCESS in 28m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8839d6e828874ef0b1e6dda5df0c8aa4 : SUCCESS in 1h 15m 22s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/090f45c4d8e742f48ebe0fd41a0b7ba0 : SUCCESS in 1h 10m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f70578d2749642a38c5e565e368f08a2 : SUCCESS in 1h 37m 53s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e630c8aa53544839ac62d9890662e8f4 : SUCCESS in 1h 36m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/59f6a634a3054dcc865630bd3b557767 : FAILURE in 1h 11m 58s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8847cc3c83bc4628aa79618690cca232 : SUCCESS in 1h 17m 35s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/52fa2c7916e049ada654b133f1ec1bff : SUCCESS in 32m 24s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/a23c2fa14dbb4d0bbb9334ee6397e1e6 : SUCCESS in 57m 04s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/33a9c844ed0c483ebd7ff86d5d9d1ad7 : SUCCESS in 40m 09s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c2fc55f52ba84398a2d610ace1d4ba7b : SUCCESS in 1h 23m 15s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5183c68284164c8685b276e24cf11fdb : SUCCESS in 17m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/32d9dede024e47cc8553fc13c9bbc1be : SUCCESS in 12m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/fa06de3982764a2f93aa9d871c0a98bc : SUCCESS in 16m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c88d7aa2b9e441969b71be9bfe4dcbd9 : SUCCESS in 1h 09m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/16ab021552fa4318bb268991dd2e84e3 : SUCCESS in 1h 35m 33s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2ec436ad5df8438d8b8be7a40547d263 : SUCCESS in 2h 04m 41s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f07d71d8490d440d8571763674e89e7b : SUCCESS in 2h 38m 32s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9489986234784357897f2d9cc86a0aaf : SUCCESS in 1h 17m 45s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/3787c5a8ed17490085e69061db984780 : SUCCESS in 28m 04s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6ef44c63deb3473bb0b6d43e3abe5e2c : SUCCESS in 34m 51s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/db613f57fb4246d2ae704a13c26e3e35 : FAILURE in 52m 30s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/277f27d2718d4f788fc842a21c54f5e6 : FAILURE in 1h 28m 03s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/80c1cb2f72b94108bc7f235e27b263c7 : SUCCESS in 32m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5290ef10be401244a93a1cf9bc8c192191b7c11e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 16:50:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48521255fe634ffa817167d2613f6741\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d1d17fb9702245a88110401139db33e4 : TIMED_OUT in 3h 01m 38s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/367aae0e537e437e856a95184fe86ffe : FAILURE in 1h 59m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a98c9481fe714cceab35c0694be98a2c : SUCCESS in 54m 12s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1f9be4b0165945b3a4ad1c55b1c0e796 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0e3dce7989c64cbeb0600f72f522150d : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/40bd4a4398614a1ab0c32fd5820d1384 : FAILURE in 1h 55m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a1d3563eff424bb59280cba82114bce5 : TIMED_OUT in 3h 01m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"270d60d9085f55c78cd7aaac2c0c338cd25e9d1e","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-19 18:51:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b5dd809449b537c08e9c9380253782d36ef302a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 18:52:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d118469f10da0880de5d50c26d9bb5fe4b397569","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-19 20:12:20.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"24f8a53d04e15ff9dae36e76b1ef8071f1b3690d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:32:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/795aae922e114a98853ead97d66fc103\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b53bf051e88744b0882721f6d1cfff77 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fb6e042d4feb4e17a80eeef88cc2a048 : SUCCESS in 27m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/dd6312ce9cb74fcd80bfd8945abe01e3 : SUCCESS in 52m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7735bdf1e4244b72a4646bddb7a7ef43 : SUCCESS in 1h 23m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da1eb38f837740a1a5cb95a4ea9c453f : SUCCESS in 1m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e32a8528efe34249be968d8b246f77ba : SUCCESS in 5m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6e9001daf8f429abf1db7c35662928f : SUCCESS in 2m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84462ced78514695aa8fd4358fa89c29 : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c4a4a49f48449a695cb07153cf3ef45 : SUCCESS in 2m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4e2ebfedb7294acd83c318cbd293907d : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4b37c78e578e43b7b0898aa895270198 : SUCCESS in 13m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2f0985299d234143ad73299eff207939 : SUCCESS in 1h 22m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/83c21b63c99648aaab97c8a76e4c2ecc : SUCCESS in 1h 05m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fa95e01c2b49488fba67553a0a3bbe92 : SUCCESS in 1h 35m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/79df2c51fbf548a0a8dbdcc561619308 : SUCCESS in 24m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a4cb9e28ed0a456e84fae8f091058cc2 : SUCCESS in 28m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/798e20eac94b4fcca0740ef3b5c279c2 : SUCCESS in 1h 08m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8aa26a42a55d44728bf9225f50431dda : SUCCESS in 1h 27m 50s","accounts_in_message":[],"_revision_number":2},{"id":"67c2602f1f738a4038f881f3833f20ca183f31bc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:32:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"35845797374169fcc98c5ebff45af16d7af45c34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 20:33:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfb5bc950d7942bb848a6495e6e12f4f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/983067cd2f29403482508e8eb50a0366 : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/214311f79ca24e409e88e8d91859b2ea : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c6700a40450cf0ae4044a0ea6dc4c331a46eeea2","revisions":{"98a373b0f942d493d1aaff7efe1f53a528809390":{"kind":"REWORK","_number":1,"created":"2025-03-14 08:07:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/944316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/944316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/944316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/944316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/944316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/944316/1"}}},"commit":{"parents":[{"commit":"9efdcc01ca5a66304568a6582a182e4171e3e625","subject":"[release] Use UCA Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-14 08:06:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-14 08:07:00.000000000","tz":60},"subject":"debian: Use Dalmatian extrepo release","message":"debian: Use Dalmatian extrepo release\n\nChange-Id: I954106dc32b4c08604d5c40e00fe53e5bb18b266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98a373b0f942d493d1aaff7efe1f53a528809390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98a373b0f942d493d1aaff7efe1f53a528809390"}]},"branch":"refs/heads/master"},"c6700a40450cf0ae4044a0ea6dc4c331a46eeea2":{"kind":"REWORK","_number":2,"created":"2025-03-17 15:10:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/16/944316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/16/944316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/944316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/944316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/16/944316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/16/944316/2"}}},"commit":{"parents":[{"commit":"9efdcc01ca5a66304568a6582a182e4171e3e625","subject":"[release] Use UCA Epoxy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9efdcc01ca5a66304568a6582a182e4171e3e625"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-14 08:06:56.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-17 15:10:39.000000000","tz":60},"subject":"debian: Use Dalmatian extrepo release","message":"debian: Use Dalmatian extrepo release\n\nChange-Id: I954106dc32b4c08604d5c40e00fe53e5bb18b266\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6700a40450cf0ae4044a0ea6dc4c331a46eeea2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c6700a40450cf0ae4044a0ea6dc4c331a46eeea2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
