)]}'
{"id":"openstack%2Fkolla~944869","triplet_id":"openstack%2Fkolla~master~If5610148fc80acf13d4eb79fef78349764f08a17","project":"openstack/kolla","branch":"master","topic":"ironic-aarch64","attention_set":{},"removed_from_attention_set":{"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2025-03-31 11:42:45.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-30 21:54:49.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-03-31 11:42:45.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-03-31 10:23:58.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"If5610148fc80acf13d4eb79fef78349764f08a17","subject":"Add support for aarch64 ipxe to ironic-pxe image","status":"MERGED","created":"2025-03-18 13:23:48.000000000","updated":"2025-03-31 11:43:51.000000000","submitted":"2025-03-31 11:42:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944869-ironic-aarch64","meta_rev_id":"3aca8b2e8791f7ab084bce93e076d93ded1e9e96","_number":944869,"virtual_id_number":944869,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-31 11:42:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-30 21:54:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-03-31 10:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-31 10:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-31 10:23:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-03-31 10:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-18 14:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-18 14:52:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-18 15:07:42.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-03-19 19:48:41.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-25 11:09:08.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-03-25 16:14:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"d82048ce12cec317624fba8088cf112be6c2728c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-18 13:23:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5e5b55a06c37dccb31d8ded1d733c6262d96ebe8","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-18 13:24:07.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"4c5e4579afe66f66dfbde9f844f35d759f9f5e48","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-18 13:24:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"bcf3ac0b983e7d4d4a00944079e5d6154e53d1e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 14:38:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2bbd1bd0ab36407d8bf19d713d64e559\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6027e6e58e234f5fb745292f05f0fa95 : FAILURE in 2m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/66e4755f5b9d4b68be46128b15aab5fa : FAILURE in 2m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c1632efe9a244f785d8464a7af13515 : FAILURE in 1h 11m 35s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2a794f15653402cada7e3ae5ebf7188 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e2d13e940fc4e89a9813f9b1450b0cc : SUCCESS in 4m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c20fc2a88cdb4de5bc26a6137d89dbde : FAILURE in 3m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/54215cbd6f3b43f0acd63498f65fe783 : FAILURE in 5m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"da4ccbccdd867dfe3291213602d029f2fc9e816f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-18 14:52:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f250057ffdc499291ca3cc195f74b37\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d4d8ab105a6d40e6919731b5323a51ee : SUCCESS in 32m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f668f0600334413ebf707cf19f53fef0 : FAILURE in 50m 30s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b85a64b1c73e4b569f9f8a0882bbd3ee : SUCCESS in 19m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1ff66768f8eb418b88d57e58c3155ee0 : SUCCESS in 30m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d7f9b0bc2c284116abab69822640a9d2 : FAILURE in 44m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0950fa3f893c4c57a4fe8c35363b005d : FAILURE in 1h 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5916dc85a884450a820b633238a4dc8 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db3085cec4134f5797328c050e5f2434 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62c75a58acb045e2a27de4f2d145cd4a : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/940a2179ce7d4a4f97d2ec1f8b00cb8a : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/122a48e9b24642be8276c0eed3e80bf0 : SUCCESS in 4m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/38d4cde6763344078b8c8190e9b34c75 : SUCCESS in 4m 20s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/84425e51e7594f61b960d245d2b00299 : SUCCESS in 13m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/70de883194b24e28b9028a02e8619e79 : SUCCESS in 45m 00s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2e14222e40564f4c8aec17d5a49640a2 : FAILURE in 29m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d17eec161e6448c98fb3a49661161b43 : FAILURE in 46m 17s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6eb08a2746cf4999a036bc31eac22650 : FAILURE in 1h 13m 14s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/564e900c9c064c81845ffc0125f6d924 : SUCCESS in 24m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/48bd7b43fdcb4360b1446ff7c43cd7ce : SUCCESS in 33m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/291d4f7ab9d64f69bd1fd8bff5c8ad99 : FAILURE in 54m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b765f4dbb3de4c29b91075f8c3c8e1e5 : FAILURE in 1h 25m 11s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/48271d492b604e67b61a41ee4230923e : FAILURE in 1h 06m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9fa2af06fa7f9f7e2d5841a26246152a8d737420","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-19 08:49:54.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"605dc0c3a0b2e47d789ec12a00e8b53001a2560d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 10:38:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d5b629f519c4a54b71efb951dae1af2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ab788468154c4168b5b73fe899e5af08 : FAILURE in 7m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d0ace2aa09c42ceba05d092cd57465c : FAILURE in 5m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d1992dcc70f4937828a5c25df83ba0d : SUCCESS in 1h 28m 07s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca18f51cef1f4fbdb4791c4d1982cd84 : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbfb5bafdb9e4adbbdb97427c04f99e9 : SUCCESS in 4m 04s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b243fdc62ad1430e892859546f26bbe9 : FAILURE in 5m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68461cf62b0d49b38c703d5696e672a1 : FAILURE in 4m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"47f6b467adfb6c8d1eb034a152660d310d15e3a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-19 10:45:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edbc90a0350f4e3db1c35bc1943c8cb8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7830a97220b146b797736c5b53da188e : SUCCESS in 28m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6b2552cf8a484cf0bee924d5a534051d : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bc5628962531440087032209329e7723 : SUCCESS in 18m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/358fa1b7c4024ff0a58ba808757e5086 : SUCCESS in 33m 03s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d778f989ad0e419a98fc70379b587588 : SUCCESS in 1h 51m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f83f0b6ed98e4660ba45e873cf9238a3 : SUCCESS in 1h 28m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/528a5d611aba475c8133d0e36ca311ec : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2517409d66ac46d38e28f0aab2e61532 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9de64d5922784120b17f00f5154ebd67 : SUCCESS in 6m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c10d2276aca43a78b8ac3ba75c3e395 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/251b4142631a412bab602bd7a505147a : SUCCESS in 27m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8f57648f849d42248a29460d80c93627 : SUCCESS in 3m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c4adf9483b7b446faf82cda71b003406 : SUCCESS in 22m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9e9a48d77f5a45e1a45ff324cc64bb79 : SUCCESS in 1h 12m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2ed439f3fc5a4cb89729414b6e4b13ec : SUCCESS in 1h 07m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4e78b0273aa646be8be87785b7619245 : SUCCESS in 1h 21m 48s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/9b1cdfcb59f245bda4c0583035f79b49 : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aa6a094137ba4cc99a730ba7f62adc48 : SUCCESS in 1h 18m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ff04e12fe72d4271bd88f7d32d92bf7e : SUCCESS in 33m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/14944fd3f0b34483bf47b8ee0f7707bc : SUCCESS in 1h 03m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ed908a5da8c49d4a32c289b8ec7e146 : SUCCESS in 1h 32m 28s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3352705ba4bd4eb8aee601bd7ef4eda0 : SUCCESS in 1h 11m 28s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"184b468a2ef014db73ca9b5be307f01184243caf","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-24 17:21:12.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"8cc7b30a388c99866493b857cb346476b13f777c","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-24 17:22:16.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"ac25e89c6294c585b03aa63e088686fd17648fdb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-24 17:26:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a23e8d16674fb9d73c1eaa62021f87adbddd2368","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-24 17:43:06.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"720d61b748136b3cd02f7fa5603d8a18e952ce3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:24:35.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ec27fafcb364bd6995cd477f90b3ea5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b19ead6c9d4d4865bd80cd5b41537fa9 : FAILURE in 28m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/bd06dd03765a410a9df9e746111bd2c2 : FAILURE in 4m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5bcd890c92244d8d98a120a579e126e8 : FAILURE in 23m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/98d5663bf7f34c6fb9261d4a297b1876 : FAILURE in 23m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/19b18011527249658aa920ed45628076 : FAILURE in 4m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/562543a4d04548c2aa769569e7199307 : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6aa3e506f594e62af2d35c5820a58bf : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9da9ace0ca124dc9bcb254fc86158eb7 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/259042bffdfb40afabc823053f8c8945 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe343388948d45f6b771f1b39ab7a7a6 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/741c064c227248a88a05a4a1a32cce45 : SUCCESS in 4m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6f543452dcd34237833de309125a1cb8 : SUCCESS in 1m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b7a479426f074ef5a00fe030233ee015 : FAILURE in 15m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1286ae01474c4c779aaf870d73ccd926 : FAILURE in 40m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3940b9aee6794e6db45eda5111218089 : FAILURE in 3m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da4aa5e9f3a644e4aea58722f4a9a152 : FAILURE in 3m 49s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b2197df953764b97a84242e7de46f7b7 : FAILURE in 3m 51s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e745ad7dcc574b2e9109f4db17a3455c : FAILURE in 22m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bf9b655360544165a062e04f1c903944 : FAILURE in 22m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eeccf3eed0fd4f67a5ef99a698678171 : FAILURE in 2m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/efc71925642b4e6d9b81278fcacaf22c : FAILURE in 3m 07s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/2c63d991a0d44e0d9833b46e5501e211 : FAILURE in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"1fce7670ce56ce320ab4abc110201a4cefe9153b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:33:00.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e0a45dfa4564a34b6e310be96a18ff8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/425c6fd0195c4f26a46f1caa514f1f76 : FAILURE in 43m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e30f0365efd7496da1d0071d064945b2 : FAILURE in 47m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6b30bd2912224269b3b13906676f7cd1 : FAILURE in 5m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/291b52b2cc1144328e7783494ff15983 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3296880325f4aadada486146c7fe29d : SUCCESS in 3m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d3ef9fdb8cf04453b814002446aaf959 : FAILURE in 41m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0f9b7f577c974b3ba097c83a5717306f : FAILURE in 40m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"90b801c2de9e195626d21a6bef976bb26dd5daa4","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-25 11:08:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"e3a6a8a314cce774b9b6e5b178bd1ed6ac22c880","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-25 11:09:08.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"ed0ce80d2321e6b75349dfbd0ecc58580057d143","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-25 15:48:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"745813884b81a5d736d2e5914600fd61662eec00","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-25 16:14:02.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"77d3eed04bee1cf5536ea42cb02484b2a4af4b6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-26 18:23:30.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"4ef665f6a82a28dec0512a5848b499d8eec20c24","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-26 18:29:35.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"80a943520f370266481d4d69bc5876b531fd8183","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 19:58:34.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/affbc3e305b3489f963af1205058ce3e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e4f79f38ae7a4fe29427097d5a7f6e05 : SUCCESS in 23m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1dfa5c1aed514178bc31b84a35a9acbe : SUCCESS in 1h 11m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1d650605d3f449488d70cd12249a9850 : SUCCESS in 12m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4d31e41546594fd19dbfa19fc613efd4 : SUCCESS in 24m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9d1236a194ca41bcacf415db92cb1c18 : SUCCESS in 1h 02m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3dcb362ce42a4f679cd23d076735bb0d : SUCCESS in 56m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44f95f1ee54c437d819f796c75c76557 : FAILURE in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b964ccf8a8bb4d3db7e4e0d46eb3083a : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c06d4f02be1a465e8aec22dfc3669917 : SUCCESS in 3m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ffdb9a25ca34a9c9100ea1cb0f5d8f7 : SUCCESS in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a072d3a3f614af3893b78df85d46a80 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/781952ca9d0b465bad5cd5e371720d4d : SUCCESS in 2m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/021d48e8713e4389bd7c24a75bcb11e0 : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d63c1eafa47b4275a6001b293c289149 : SUCCESS in 23m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8c30a239f77f4edb8c1ce75d7af4d5dc : SUCCESS in 1h 00m 42s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/85800a78f0534ee8b173c6d255756c4c : SUCCESS in 57m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c9970f881d5b4358ae804823d3fdf31a : SUCCESS in 1h 28m 57s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f3570621373d49b299205882cc38ef1e : SUCCESS in 1h 17m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7a665dc57464463fbc6d2ce98040d710 : SUCCESS in 22m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9872d01c241f43d4acccc48f8b409d66 : SUCCESS in 39m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0d176a334b734ee08a4de036e5e9872f : SUCCESS in 57m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7ba46983525947b9a9dd969f851c7bf5 : SUCCESS in 1h 24m 54s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7f4192dcd2154e70ab21be6f4bf477c0 : SUCCESS in 1h 11m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"89916e694a8c3fadcd2eb09f0a2daf95383544d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-26 20:10:22.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"3dc99972ce14d928d5f61ae354fc9e7f92acfeac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 22:13:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d217e7316f614afe852bded523918205\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3d460ba46e6c43c1bde9640eefc01abf : SUCCESS in 26m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7b2c1f7690d643a8a586d33c57fa26d0 : SUCCESS in 1h 06m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/54db6b506eac496a9b96424a59207993 : SUCCESS in 19m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6f14be6af46b43fbab27be188c8cce2d : SUCCESS in 24m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f3823a18b8874edc9e7eba247c483ee3 : SUCCESS in 56m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c6bfa59e43b84a2798d07c0332121e63 : SUCCESS in 1h 18m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40827357d244a20b9ffd6dd35a2462f : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd97291f581943f6ae9916ef58e2c881 : SUCCESS in 2m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6851f2e1e0814ba38be93285de28219f : SUCCESS in 2m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7d849154d0642bb940f3acc2d04a71a : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029039e2604c45c2b13d0a936bab2aa0 : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0bc91fc08a64f0fbca05b13b4b5d18b : SUCCESS in 4m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/06804120489a45a79dadc6af59b55a68 : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c2bd138828d84b88a433629d465b14ba : SUCCESS in 23m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ed51e2bbf1c48818da9ce6b9704d88f : SUCCESS in 53m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/976c539774014456b49765d3444b2a96 : SUCCESS in 1h 10m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f2831efada5042ac80f71872b11eee43 : SUCCESS in 1h 28m 31s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/90aa93b591d54887a780277ce7a3007c : SUCCESS in 1h 18m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a51f566564144ac587152c1bec37bd84 : SUCCESS in 19m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/03b0ad0da06f4a0ea74f819c1f922a98 : SUCCESS in 29m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3201a27c24944bada196174f252ff65c : SUCCESS in 57m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8fa20e17136740bf83dfec857c204d26 : SUCCESS in 1h 20m 57s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/7fdc0a75db134442ba7950ec661d1f7a : SUCCESS in 1h 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"142e1d739c6c82c52b96d4848a8c609c1c8d388c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 22:43:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24a0de2532f44bb48f926c9c49268c74\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/54fd4338ed78401b86e0969d72193013 : SUCCESS in 32m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f071a57852774a01939ca84f0460e18f : SUCCESS in 38m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5705ef593eec439d938977d5b63fe26c : SUCCESS in 1h 29m 16s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/97152be6a370428e996e5c4004095871 : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a8e51ad4689346128f5ee53d7f320d1c : SUCCESS in 3m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/26eee12b70924259a1f51cdda05a1b4f : SUCCESS in 44m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/16a13c41901c404f92b5b07d16a89db6 : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96e4727291ecc7c8c1d3f67ce3acc335ca79e3a5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-27 10:00:59.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"71a05630fc40aa3060fe826bca8f020d13def240","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-27 11:33:34.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2b146bb0bd91dff4dc37fae97327c4a73174378f","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-03-27 11:34:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bd006066a0e9c09b5953020a781a3703f1485365","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 12:20:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6ded38b4d54437c8cbcd6ef2cb7aefd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/dd5ab88aae0643a384593f2eb5ec27f8 : SUCCESS in 35m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/637a6363ba8946ccbe9e17d647a305cb : SUCCESS in 33m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4b5ebebbc3084b479cabe764ce6d37d1 : FAILURE in 5m 42s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e066d461264c45b4bba04526e8a57041 : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d19d388777e4153863ad310bec2c9ba : SUCCESS in 3m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c9326f13e42346c7b5f81e7ea4ea5733 : SUCCESS in 43m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5795ddf8077b4d238003bdfae00bf97b : SUCCESS in 39m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e488a0bccd656d836ebf0b7ec71aaedc160c2720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 13:17:08.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0502a4adc1340d7a57b31f37a066c48\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/15ea7b1e3167483f8f7b2f1801f8bb63 : SUCCESS in 26m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/35d96f355e1b4da6a91c5aac853ccb31 : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ef6c4474c7aa4ee1a461bc5aff34e8bf : SUCCESS in 18m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c5e0af50818e4371bf4a6228749f1af3 : SUCCESS in 22m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2d388a24f4c34132bf581c87a3134cae : SUCCESS in 1h 08m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/be7b98c2fbcd497e9331ab1fa3b240a3 : SUCCESS in 1h 03m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/477761a7879a4fb08e2696404da86bc8 : SUCCESS in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c65850c3fe9e452eb2ca35329040f0d7 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d8bc595421c4f6dbf8c5a51a1b4606e : SUCCESS in 3m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3c2f8209cd49ca93b9ef7ef16dbcea : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62519184a2d0430ca1179d513d2227d6 : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33b4030c4b004a22ba783ed74021973a : SUCCESS in 1m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/09a1b84ea55145759e6cf7e227022c81 : SUCCESS in 1m 42s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/08c700ccce7242378c26d15e9573b455 : SUCCESS in 13m 10s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/39e0332cb65048828d68096529841316 : SUCCESS in 42m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/651580da342e4daab710ed1bc2701a70 : SUCCESS in 33m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/87edb554b11746818e452e0c2f3e1a7b : SUCCESS in 47m 49s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f38a6ae0c7c94736b0cb064f9ce14cbf : SUCCESS in 1h 17m 58s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3346b332e4224952a550494be8d3a3d2 : SUCCESS in 19m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9ec7312e8717452a88a3476e7426ced9 : SUCCESS in 28m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d016efc134b74d518d2d555a15046de1 : SUCCESS in 1h 15m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d85b1dec3dce45d48bc1cd4bfc317021 : SUCCESS in 1h 37m 35s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/6779990a2fd0406f8917c6d64c5c1fc0 : SUCCESS in 1h 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"dbc69a25fc09f134321708a8c354dbb584955221","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-30 21:54:49.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dffe91d5b357ae6b55a3375f0ed637fd129a0bd6","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-31 10:23:58.000000000","message":"Patch Set 7: Backport-Candidate+1 Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"882b30765b53ad3b41553b5c1a8fe2e724f20ffb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 10:24:31.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"899e650ab88924f9d02a5e0c9cd5424584330a91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 11:42:45.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20f46252cb6e46c9bcc2f9a6c6e4e611\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f4184b16a28149079f3e78fd4c2e9245 : SUCCESS in 10m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/de52a221b6fe437c8200ea6f5f3d691d : SUCCESS in 14m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4c063ec99d3c445cb0eed6dc3507c6c5 : SUCCESS in 30m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/457e774d64bb4d3a9fdf1acae23580a8 : SUCCESS in 47m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c1fea6d21234efe95f3a176ece55b1d : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e3cc39c0a45140d4957655525c234d12 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b1d1ab0eb52e434a8a771d444084e1d9 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd4bd3b8b3a4424b61c19e6bebb1ad8 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c57883aa687846768569892a182afb6b : SUCCESS in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e754eaa33edc45d4a7c0c6606c64a90b : SUCCESS in 4m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d4632c99166042a6a955a7a920fa1e28 : SUCCESS in 22m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bb3025c52c2945dc82e3f961bbb0bdfe : SUCCESS in 1h 10m 02s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/06290a04923049efae93c1f778a61c7b : SUCCESS in 53m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d5b573e93c8d4103a0a875a20b685af3 : SUCCESS in 1h 17m 10s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/113573ae6acc41c99b052dcd42542ed2 : SUCCESS in 17m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c5c37fbfecbb4694a1fcc6aa09cd2a74 : SUCCESS in 23m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/48abc72ee34746cf8ad9f1848de60386 : SUCCESS in 56m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/66361eed2f144f8ca266d977335ce91a : SUCCESS in 1h 17m 42s","accounts_in_message":[],"_revision_number":7},{"id":"875ac261ae7ef9ba199cbb0cae440723bda74498","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 11:42:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"3aca8b2e8791f7ab084bce93e076d93ded1e9e96","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 11:43:51.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f7de51377a4eae939ddb8287eb5e3d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d525f3276a9744a9b792ca8e669ed473 : SUCCESS in 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c256f33b61b40689f96307254ed3529 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"16b3223d8c1e236edf498695786c960aef121db1","revisions":{"8cefc52b80ab8c4f025e5090f2d3c7200d4ceb61":{"kind":"REWORK","_number":1,"created":"2025-03-18 13:23:48.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/1"}}},"commit":{"parents":[{"commit":"64f171ccfa02bfeb9c92a641e186bd987f0099ae","subject":"[release] Use RDO\u0027s Epoxy packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64f171ccfa02bfeb9c92a641e186bd987f0099ae"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:23:10.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cefc52b80ab8c4f025e5090f2d3c7200d4ceb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cefc52b80ab8c4f025e5090f2d3c7200d4ceb61"}]},"branch":"refs/heads/master"},"113c75ac872ce228d8b37fbc1134e3b840788625":{"kind":"REWORK","_number":2,"created":"2025-03-18 13:24:43.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/2"}}},"commit":{"parents":[{"commit":"64f171ccfa02bfeb9c92a641e186bd987f0099ae","subject":"[release] Use RDO\u0027s Epoxy packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64f171ccfa02bfeb9c92a641e186bd987f0099ae"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:24:36.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/113c75ac872ce228d8b37fbc1134e3b840788625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/113c75ac872ce228d8b37fbc1134e3b840788625"}]},"branch":"refs/heads/master"},"f9426b256f917aa342edf09c43273379d7414381":{"kind":"REWORK","_number":3,"created":"2025-03-24 17:26:56.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/3"}}},"commit":{"parents":[{"commit":"7667e2493116534e87ab511b5ff0ba6770f2ee92","subject":"Merge \"Remove swift container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7667e2493116534e87ab511b5ff0ba6770f2ee92"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-24 17:26:28.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9426b256f917aa342edf09c43273379d7414381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f9426b256f917aa342edf09c43273379d7414381"}]},"branch":"refs/heads/master"},"12ddfb7dc4eca50d1df6339116bbb18bcf3bc3cc":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-03-24 17:43:06.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/4"}}},"commit":{"parents":[{"commit":"7667e2493116534e87ab511b5ff0ba6770f2ee92","subject":"Merge \"Remove swift container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7667e2493116534e87ab511b5ff0ba6770f2ee92"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-24 17:38:12.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nSupport aarch64 ipxe only when using Ubuntu and Rocky\nLinux base images, as the Debian ipxe package does not\nprovide a suitable aarch64 ipxe binary.\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12ddfb7dc4eca50d1df6339116bbb18bcf3bc3cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/12ddfb7dc4eca50d1df6339116bbb18bcf3bc3cc"}]},"branch":"refs/heads/master"},"8be759489cc3f5c2f36fd36bb6f38ffe41950663":{"kind":"REWORK","_number":5,"created":"2025-03-26 18:23:30.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/5"}}},"commit":{"parents":[{"commit":"7667e2493116534e87ab511b5ff0ba6770f2ee92","subject":"Merge \"Remove swift container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7667e2493116534e87ab511b5ff0ba6770f2ee92"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-26 18:22:18.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nSupport aarch64 ipxe only when using Ubuntu and Rocky\nLinux base images, as the Debian ipxe package does not\nprovide a suitable aarch64 ipxe binary.\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8be759489cc3f5c2f36fd36bb6f38ffe41950663"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8be759489cc3f5c2f36fd36bb6f38ffe41950663"}]},"branch":"refs/heads/master"},"c36068af4b444b35d114f925586f4735732a6dec":{"kind":"REWORK","_number":6,"created":"2025-03-26 20:10:22.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/6"}}},"commit":{"parents":[{"commit":"7667e2493116534e87ab511b5ff0ba6770f2ee92","subject":"Merge \"Remove swift container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7667e2493116534e87ab511b5ff0ba6770f2ee92"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-26 20:10:11.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nSupport aarch64 ipxe only when using Ubuntu and Rocky\nLinux base images, as the Debian ipxe package does not\nprovide a suitable aarch64 ipxe binary.\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c36068af4b444b35d114f925586f4735732a6dec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c36068af4b444b35d114f925586f4735732a6dec"}]},"branch":"refs/heads/master"},"16b3223d8c1e236edf498695786c960aef121db1":{"kind":"REWORK","_number":7,"created":"2025-03-27 11:33:34.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/69/944869/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/69/944869/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/69/944869/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/69/944869/7"}}},"commit":{"parents":[{"commit":"7667e2493116534e87ab511b5ff0ba6770f2ee92","subject":"Merge \"Remove swift container images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7667e2493116534e87ab511b5ff0ba6770f2ee92"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-18 13:21:46.000000000","tz":0},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-03-27 11:33:13.000000000","tz":0},"subject":"Add support for aarch64 ipxe to ironic-pxe image","message":"Add support for aarch64 ipxe to ironic-pxe image\n\nSupport aarch64 ipxe only when using Ubuntu and Rocky\nLinux base images, as the Debian ipxe package does not\nprovide a suitable aarch64 ipxe binary.\n\nChange-Id: If5610148fc80acf13d4eb79fef78349764f08a17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16b3223d8c1e236edf498695786c960aef121db1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/16b3223d8c1e236edf498695786c960aef121db1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
