)]}'
{"id":"openstack%2Fkolla~945070","triplet_id":"openstack%2Fkolla~master~I0bdef4252d4f2850d2b1a40feac3c373e5c453e7","project":"openstack/kolla","branch":"master","topic":"ansible_2_18","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-03-25 16:38:58.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-20 12:33:42.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I0bdef4252d4f2850d2b1a40feac3c373e5c453e7","subject":"kolla-toolbox: Bump ansible-core to 2.18","status":"MERGED","created":"2025-03-20 09:23:50.000000000","updated":"2025-03-25 16:40:08.000000000","submitted":"2025-03-25 16:38:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945070-ansible_2_18","meta_rev_id":"dad95bcbd9e3d6a50f1054f4beb278a43fd2a304","_number":945070,"virtual_id_number":945070,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-25 16:38:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-20 12:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-21 19:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-25 10:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-25 10:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-03-25 10:03:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-20 12:33:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 09:27:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-20 11:00:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 19:29:16.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-25 10:03:44.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"2e87517f717648403cd1a921964044d42ebaf247","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-20 09:23:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"25f1ab1fade75a2549129955a4d4fcce35e9c514","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 09:27:03.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"42a45b539caae1e007752fa9fc48f0f5584455b1","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 09:27:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9994ce39c18816a3637bdd639835d2ecaf9e85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 11:00:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e32bff15dd4a96a520ff9d5c7fed2c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6280a77908be4d07904133e55ab7864a : SUCCESS in 29m 08s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5c443f56ec2640b7a0348f720adf5ba7 : SUCCESS in 1h 04m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5089fa48eb3e41118af2df37342f400e : SUCCESS in 10m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6b26c5e3c7234115a6ed7182a28fe20a : SUCCESS in 37m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f5bee7f4426f47448cc1c12ece0e2c64 : SUCCESS in 1h 03m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/52bfa9606bb94cb3adc94fadae667fce : SUCCESS in 1h 27m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/310d0f5e8579414cb5ee71b96c9eedef : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad39bb187c494b6a9e8bf6a56288c608 : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd06610073704ee5875873cbf9ff5011 : SUCCESS in 3m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7958ab9ddec641f8ab658365d5728a4e : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa896d4469584ec7921655e83534b417 : SUCCESS in 3m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d199c7a801914c21a2a401b50ba74046 : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e7b9ccfb231444a5b5eed6032f295e31 : SUCCESS in 31m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/144069a5c4ad42d0b11b433344f85e99 : SUCCESS in 1h 06m 26s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d23c47ce4504b888313d19432718f0e : SUCCESS in 1h 03m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c4fdef00001a44b9b37855d2ea4e88c8 : SUCCESS in 49m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d9c99aa51d8f4b96aabe3265f24e0f15 : SUCCESS in 24m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8c1e65ac1f794c649afc60bde9ee812c : SUCCESS in 28m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0c4f972e40b74224914ade1bec9aec14 : SUCCESS in 1h 07m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/eadbbebd69b84387b418e505ebe45d4f : SUCCESS in 1h 28m 36s","accounts_in_message":[],"_revision_number":1},{"id":"876d91e795f88b007efe0599e1cb23fff1e898e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-20 11:55:44.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d927211be33361a9f269c7005d0bddb33e94038c","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-20 11:56:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89d03574e24d22929851a4c015d865d42f9b0ef8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-20 11:56:41.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"a90becabfc104746d549c5b56c94d96a932ab709","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 12:33:42.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c671629ad1eb67985cfe691b2529f694e79b072e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 14:20:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a396c68cef4c2c8b060b925f5af44f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8500f78cfb2046cdb01ae58f7d4d7cc4 : SUCCESS in 25m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f4260ea5674f4e51802fa02280799722 : SUCCESS in 1h 11m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cf2b5e08b6ea4fa8a7ced297bc839566 : SUCCESS in 23m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/21685682d0aa414d812d68b8f71941db : SUCCESS in 35m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/511a9348786e4d458558769612771326 : SUCCESS in 1h 00m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b5d379b499b245d39f268a44981f3f89 : SUCCESS in 1h 58m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f100f5ca8748463ca7e96be0c91519d0 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d31fc2465de74648b3b82fe87c5a6aa3 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8668a5d7045484fa02fe45a9c449475 : SUCCESS in 2m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d52c44d0903464386a1a6cd7d271f4a : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd60bc93c2ee41f28b526537cff27259 : SUCCESS in 3m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38ba0948f805413c8cdd6005b17e9ca5 : SUCCESS in 2m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/df441471ea714805bf4f46729c3b141e : SUCCESS in 1m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c5a8b7d0822a4e7cb044870b5ea1b535 : SUCCESS in 19m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9cd3d29e1bd648ffa0393de31f539e6d : SUCCESS in 43m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/541dbda699d6498589d49596a4b12c83 : SUCCESS in 35m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d9e9292c1bd34e4b9e8c3d0dd436f09f : SUCCESS in 53m 00s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/168b946703fd417e8726e86802ef8285 : SUCCESS in 23m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/7d46b685826749548d22a32f63d8aa7b : SUCCESS in 20m 41s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/94f8a88a602f45b5b5992135947f09b6 : SUCCESS in 1h 10m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/86a8ce19f10246e4b367397691c7dfdb : SUCCESS in 1h 34m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1883c6509962bde926d170d91ba736fb76ed8356","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 14:58:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76675a8dce844091b80dfbfc6ad32aa1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7995cedd90b44980a2601dcddad19dc6 : TIMED_OUT in 3h 01m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ab480259f5614197a1749a4b7d8271bf : FAILURE in 1h 28m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/adba749a73b94c93a5fd691f53129447 : SUCCESS in 59m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c143956bfa60424a87b073fba0d94b35 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/52da74a738a84119bd44ef33586a17d4 : SUCCESS in 4m 00s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/121c1325fad049569f3a32ac3c6a1f87 : FAILURE in 1h 05m 34s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a3d5d81e9d1445bd8f1eff2168ca563c : FAILURE in 23m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"412893347ff99c78c6428d79f5eacb0c4a7cae7a","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-20 19:24:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"61172725ddfa9eab6fa638256d26c1870873211d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-21 19:29:16.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"dc786afd7a9ea2ad6b3de0bfe9738ccc08035be7","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-25 10:03:44.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4beec2ea15e2797f2faa620489555b2580eba662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 10:04:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0ad63682da8dbb59a5aae494c11d90aba3da9cea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 13:09:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/368e143a57354d4fab906c652e6bf230\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/39de6f970a5a4c88b593704a488eacf6 : SUCCESS in 20m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/910ece909d5448f99e72c8d32f85a1b2 : SUCCESS in 30m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/382d82020130404a80236a2e6736cf30 : SUCCESS in 1h 02m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/61c0afeffe97454fb1685c34aa88a1c9 : TIMED_OUT in 3h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c68d4708ad214716868fad9160d1283a : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c30e7ffac7843f6883f32a2b0a8985d : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/edb3d45cf4144633a7be0f7b2191059b : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85ba01dd522c4a7c94c1fa9e7eae85f7 : TIMED_OUT in 30m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f2fe3d572324d139f537367449e805b : SUCCESS in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b28d10eebef54899abf7f0d6b9775cd8 : SUCCESS in 3m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/e4a896a556fe4036b267d01d37354df6 : SUCCESS in 28m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0f68427182a54fe5b1f94bcf81567641 : SUCCESS in 56m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3b37e83ee4f24a8d8f1cd267afcef6ee : SUCCESS in 1h 02m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f620bb4bd3a94ddd8ba57752683c39a6 : SUCCESS in 1h 21m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cea80e62f68540e0a8f8539c7d268d58 : SUCCESS in 20m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/93a4618aa2ad4bfcba150c47b598b267 : SUCCESS in 35m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0e1bd15bf7344a73817b663bd69160df : SUCCESS in 2h 09m 51s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3e36a902d9ab43369fd731428c50bf08 : FAILURE in 47m 13s","accounts_in_message":[],"_revision_number":3},{"id":"71b2881a65432f71a4ad5cc6a34ac41fcef9f562","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-25 13:11:22.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8a2c926a3307e069d36b1e432309b471d94fcf8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 14:20:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00fb3e95aa64bb4a84fb4a2d791800a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5387a63e8551475486491d82c6bf5bc8 : SUCCESS in 54m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d3df358601d74e508550ae6647e74300 : SUCCESS in 1h 03m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/939af780779e4da3a39cb1a741cd33e3 : SUCCESS in 1h 05m 03s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fb3c003dfef8402dab6c55e5f14825e5 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ea74d8fc6e8b426686d2e8c9730344a4 : SUCCESS in 4m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1c163eb9b7984b29a4728e27a6d49508 : SUCCESS in 54m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0b82e5ea86e244019c662855e26ed1d2 : SUCCESS in 55m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fb849b4ccb8d32bbf417bcb87345613e271668d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 14:57:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b09ff958f5984b968038522761a18e84\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/2edb14c421c249648a76def0cb1042f0 : SUCCESS in 1h 11m 54s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/65aecfc8bfa149a299de685e63b51f97 : SUCCESS in 35m 11s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9aabe6cdf33e47d389cc23819d282a6e : SUCCESS in 39m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8d65e818815b438aa71b79656de105fb : SUCCESS in 49m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/092209c4a8f644faaf978705801ca15a : SUCCESS in 1h 40m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/87b907ccbdeb44c2961d77064c52a667 : SUCCESS in 1h 36m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/083b1d0f042541c8aec1e37d5059837a : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/330a465e65bc441d8f508eb99989ddec : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/785312a992944ce2bfaaa63a516103e5 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ecf54a07fea4c5eb4ac0acc1a593821 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9fd254b02344d268e1884135fc33cd4 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/825112438d194810add7d3ffeb7e8cd7 : SUCCESS in 3m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4792ff0e39eb4dfb8ebc580761f479b4 : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2bf05388fe0943df9262b4975001fd85 : SUCCESS in 23m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/98d41cefba984a98a992b34dd84ced77 : SUCCESS in 1h 21m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b318c1721899449d8387e042670f1d65 : SUCCESS in 1h 18m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/80220bc14b024117a8b7354398ecaae9 : SUCCESS in 1h 31m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/545938e6a48a4ab7bf85fdd1aad79b7e : SUCCESS in 24m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1c7cccdf768e4b378c1704a3c9457302 : SUCCESS in 14m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/243b082f37c34c79aef7cca2813b3903 : SUCCESS in 1h 00m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/677c92fda7804ff5b2db819b234913ee : SUCCESS in 1h 26m 29s","accounts_in_message":[],"_revision_number":3},{"id":"7166c5eb012a65667f6dc2d809f400a44a991723","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 14:57:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"4d3c80ab0146e9809753698be05331db987df213","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 16:38:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/602a332f01b84612a2a804a566091b9b\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8277abf0326f479fa656f240857a7747 : SUCCESS in 32m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bc4b7fc326104c16a749076c63c70a6c : SUCCESS in 31m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ee2c3c7c755041b18b921e8429cbadb4 : SUCCESS in 1h 09m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2b1a4b45f79e4a4987e515ae8a2fb778 : SUCCESS in 1h 39m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a24e520c69643d18de764e9d70ecba3 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9acc956c659741bf8fb5b756e6f9d1be : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7c1d212d9934eb9aa65d0d1631e5d83 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd03866bd25b43a19adc68e2c503f440 : SUCCESS in 7m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0218ed76f6e947a9925bd1c530e68215 : SUCCESS in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/55dfd956cf804b2ea405cc439f13f30d : SUCCESS in 2m 50s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ee59601757ff457db1ccfc287161d4b6 : SUCCESS in 14m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0651dba73c384178ad01bbf6c04616ea : SUCCESS in 1h 10m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0bb75b5a2be246d0ad822496007757a9 : SUCCESS in 35m 53s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4ab0ed56dea84151885f6260b06f236e : SUCCESS in 1h 35m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7e417e3444474c54adc1d563895b4a2a : SUCCESS in 35m 44s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cb679433f8f14c788ecdf4c6668dd9b7 : SUCCESS in 32m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec04d70351a34dbfbce65ce937a4bb3a : SUCCESS in 35m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5416e20d015d4a718e42b4737b9bfbde : SUCCESS in 1h 30m 14s","accounts_in_message":[],"_revision_number":3},{"id":"173ae27a218273e17b8f98ff1185c6874837ef28","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 16:38:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dad95bcbd9e3d6a50f1054f4beb278a43fd2a304","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 16:40:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0032493ab73240bbaf7c271749c150ec\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b328acb51aa948c49afde79318df701f : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9062b1689c264b6f94fa72813a522390 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e1fff6252581ee221c1972e671abd80f24393fae","revisions":{"e0d7a37d89f3959e1b8dd6b2938b7b1f30f1f86a":{"kind":"REWORK","_number":1,"created":"2025-03-20 09:23:50.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/70/945070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/945070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/945070/1"}}},"commit":{"parents":[{"commit":"d25e9a7a701f2c0d47c91ce4d829d4d67b4058c1","subject":"Merge \"debian: Use Dalmatian extrepo release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d25e9a7a701f2c0d47c91ce4d829d4d67b4058c1"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 09:23:16.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 09:23:16.000000000","tz":60},"subject":"kolla-toolbox: Bump ansible-core to 2.18","message":"kolla-toolbox: Bump ansible-core to 2.18\n\nChange-Id: I0bdef4252d4f2850d2b1a40feac3c373e5c453e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d7a37d89f3959e1b8dd6b2938b7b1f30f1f86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e0d7a37d89f3959e1b8dd6b2938b7b1f30f1f86a"}]},"branch":"refs/heads/master"},"209a6ca05810b1a3a139ec51325bc45c73c21bfe":{"kind":"NO_CHANGE","_number":2,"created":"2025-03-20 11:55:44.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/70/945070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/945070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/945070/2"}}},"commit":{"parents":[{"commit":"d25e9a7a701f2c0d47c91ce4d829d4d67b4058c1","subject":"Merge \"debian: Use Dalmatian extrepo release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d25e9a7a701f2c0d47c91ce4d829d4d67b4058c1"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 09:23:16.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 11:55:38.000000000","tz":60},"subject":"kolla-toolbox: Bump ansible-core to 2.18","message":"kolla-toolbox: Bump ansible-core to 2.18\n\nChange-Id: I0bdef4252d4f2850d2b1a40feac3c373e5c453e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/209a6ca05810b1a3a139ec51325bc45c73c21bfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/209a6ca05810b1a3a139ec51325bc45c73c21bfe"}]},"branch":"refs/heads/master"},"e1fff6252581ee221c1972e671abd80f24393fae":{"kind":"REWORK","_number":3,"created":"2025-03-20 11:56:41.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/70/945070/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/70/945070/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/70/945070/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/70/945070/3"}}},"commit":{"parents":[{"commit":"d25e9a7a701f2c0d47c91ce4d829d4d67b4058c1","subject":"Merge \"debian: Use Dalmatian extrepo release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d25e9a7a701f2c0d47c91ce4d829d4d67b4058c1"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 09:23:16.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 11:56:36.000000000","tz":60},"subject":"kolla-toolbox: Bump ansible-core to 2.18","message":"kolla-toolbox: Bump ansible-core to 2.18\n\nChange-Id: I0bdef4252d4f2850d2b1a40feac3c373e5c453e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1fff6252581ee221c1972e671abd80f24393fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e1fff6252581ee221c1972e671abd80f24393fae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
