)]}'
{"id":"openstack%2Fkolla~945114","triplet_id":"openstack%2Fkolla~master~I05a96f27b47a0f87bcf435872dd08ec0927232e3","project":"openstack/kolla","branch":"master","topic":"kolla_remove_swift","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-03-24 14:07:09.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-24 14:07:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I05a96f27b47a0f87bcf435872dd08ec0927232e3","subject":"Remove swift container images","status":"MERGED","created":"2025-03-20 14:56:27.000000000","updated":"2025-03-24 14:08:07.000000000","submitted":"2025-03-24 14:07:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945114-kolla_remove_swift","meta_rev_id":"84edf7d2e58b9021dfe05a7859fd4352479f5b15","_number":945114,"virtual_id_number":945114,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 14:07:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-03-21 21:30:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-24 12:28:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-24 12:28:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-03-21 21:30:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-03-21 16:30:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 16:36:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 21:30:01.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-03-24 12:28:52.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"499a752c6dc74ded7d6819f6636d20b575b64d57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 14:56:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"021feab75aca5fb94fd3ebba8e2f53d63f6a2e6a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 14:56:55.000000000","message":"Topic set to kolla_remove_swift","accounts_in_message":[],"_revision_number":1},{"id":"f69974208bbb866b252dc083fbe5c99659ac588d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 16:36:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b4cedd457ff423dab37e85621976137\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/314600b8b88244ad908ba11cdcf58c7c : SUCCESS in 2m 32s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d050097ac4b641d4878b5b6dda8a7959 : FAILURE in 32m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dbda10651ff34d1384f82e7232ac9e09 : FAILURE in 22m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7e103ee7960411db2e29d2cddedd533 : FAILURE in 10m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5e7703dbf1b34234bcef9dbf408b7fa7 : FAILURE in 31m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/60a02b355c864b579ff20c58cea2ae60 : FAILURE in 21m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/72eeadbe803b4f50b223771510bdf73f : FAILURE in 16m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10ae9b03fb644561b5fe1e8521869857 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14756975ce4140949d10e8a1ed0bb644 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52d6e37520cc443a81973749c3485aba : SUCCESS in 1m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35d07883dc284c548592c0a8603e854d : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d5ccff3d0664fb6a07be1e314c9e258 : SUCCESS in 3m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9e755e64adf7467e854d692bd62fc7f3 : SUCCESS in 3m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cad8f0be0fbb428a8654dfaa4ef7ba99 : FAILURE in 23m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e93522bb01cb40518f6cb775f47ec92d : FAILURE in 1h 39m 33s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d90dc08069b6447ea03b4520443e1c8d : FAILURE in 24m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/29366b175d754a6cbb1e5f2dc6e6f6d2 : FAILURE in 23m 34s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/038d69f0738e4901a21441c6d3003a20 : FAILURE in 22m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c0c0af7dd1a34785b81a5c286a7d812b : FAILURE in 29m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/9d16b136953a433e9dc79c5a0a65d813 : FAILURE in 20m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d78ad7c36c794a3db606a188c74e5111 : FAILURE in 18m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d840ab2b7bf3fa2b5c590474ee27888f745f7573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-20 16:41:29.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"930531a26ed9c2246b0ff4117bf112cad2138cc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 18:11:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/52ccafb864024b11b88c1b59ff747982\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbbcc96c13c043dd9ba6e152f83c0b86 : SUCCESS in 2m 43s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ddee05914a4a42eaa9e182c50782ebe0 : FAILURE in 34m 22s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/01b7838b93ad4294b993093f4d83580b : FAILURE in 1h 26m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1b3700d406574fa0935a446acf49566e : FAILURE in 22m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e6c0f30346d6404cb5ad5b2eadf57c79 : FAILURE in 1h 17m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b3a4055431384d0da663c945f038fb2e : FAILURE in 17m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2a7fb43c64774bf7a58c29e01450b294 : FAILURE in 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378790c24e514f47b634d30e77b21a29 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7b00b8a8521f4aff9240766cbcfef987 : SUCCESS in 5m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5173d782755448cb8dbb6c75626b32d : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de86d0022a7a4e0c883f0ad71168c4dd : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fed74e5a26f461eadfe2aa0aac5a104 : SUCCESS in 7m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/361c4ab0a5c34d57aa4e2800e04ca406 : SUCCESS in 2m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/153407a89ed64b15963b8ac45ef443e6 : FAILURE in 24m 27s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/91828fcdebef4b278a14cf1da0a41530 : FAILURE in 1h 25m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f78bfd8fd178496f8c13888fc15e73c5 : FAILURE in 23m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e744c02a8dec448eafc7bc95d2e1fcc8 : FAILURE in 13m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f6245405d6024c519d4611b63bff913c : FAILURE in 32m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/961d6431f52542a085d1bf9cc5c9821a : FAILURE in 23m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fe8fd8b46caa4989bf47b85a81ab51f2 : FAILURE in 20m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ab7a110f47be47de92fb03f8fa23ad0c : FAILURE in 23m 56s","accounts_in_message":[],"_revision_number":2},{"id":"694783c827ba81fd9e9771f4c05cd43454697e05","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 19:43:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e28a6c9b25244134bb0db44bc03920f2\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c668bc083df04f57b5861faf9ce23456 : FAILURE in 2h 18m 16s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/44eaa23d90964157a9d754e30a33c4e7 : FAILURE in 1h 31m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7ab0daeecd6e4c608e162cf7f5f08a8d : FAILURE in 30m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2237a101751145f59154fc954a3de19f : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2546c70ed07a42d989f08352bb72cd87 : SUCCESS in 3m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e1a62d9a88ee4cec89af61629f0fdc9d : FAILURE in 2h 39m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7d99909819eb4941a9135261beb185ac : TIMED_OUT in 3h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"17686e5c923b1de2ee6b9a96cc73c95ba51804c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-21 09:37:17.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"f8ac1d8d45e40bb11bbf71cb7f22cb53edcac268","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:05:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13ded32c510c46f1b9ce607788aa74fb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3e8595fe68b463aa513d883ce1f1ede : SUCCESS in 2m 30s\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7dac409fb7734a98bb81dac6748ca936 : SUCCESS in 29m 09s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c8838e4b02894de29fbd8d39799cf9f9 : SUCCESS in 1h 01m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1b7a4d84eedf4b319967567587abe08d : SUCCESS in 20m 41s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a47de266172947ef9c489595c0c11246 : SUCCESS in 15m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ca1be27c2f724251846c6ecb015219d6 : SUCCESS in 1h 01m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cf448c207d1747138819c17400bcefcc : SUCCESS in 2h 21m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/928dfe8f4340432f8474b6d87b89c06d : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b400151ced247f5b98e437f621e032c : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7faee4be885444f2a3d981e75affa2ad : SUCCESS in 3m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/767e249d8fd7401fb40444e06a432f5e : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f0790b88c849fc9adc596f71484585 : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d51b4b1cd7304eeaa1e71ee14f1f0dd9 : SUCCESS in 3m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/97f63cc0206f451daa4c023dd3a24bb4 : SUCCESS in 22m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/720ef04e5e1b4d7b8a62299ad960958f : SUCCESS in 1h 24m 36s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/106bc7e7332848569fdbdd4b1829bb45 : SUCCESS in 1h 09m 43s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9c6e8eda807c4b47be6c903a89e5b44a : SUCCESS in 1h 27m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f722d8f7a2964cdf9d2bc87b20b59166 : SUCCESS in 32m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/28474dd0ee6c434d9b13f2e12a9cd890 : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6f172b6a0e6942daa82861ea159f1720 : SUCCESS in 1h 11m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a2cb79a885ad400aa71a92e3d87dcb62 : SUCCESS in 1h 36m 16s","accounts_in_message":[],"_revision_number":3},{"id":"199e1b4f1248b571c49bae531662051fbb0e25ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 12:50:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67e8f85b853a478fb0ae88e08aaa3b06\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7daa2f55a78e4f86a0ee670578347794 : TIMED_OUT in 3h 01m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/871c564c1dba4d17867187efceac721d : TIMED_OUT in 3h 01m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f645d855d3c14d658a5ac1ca14d82153 : SUCCESS in 1h 04m 59s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25a71dd4022f4f868dd45a316f741797 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bcb7587290b1445e8088224494a5c496 : SUCCESS in 4m 10s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2f9d81d620ef46afb28843d97810254f : FAILURE in 1h 04m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ed2eb95d3a064a818b59e4fb7ff83e4d : TIMED_OUT in 3h 01m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f124cc3cd1d6ac548fbdc8ae7eda28d4d18b5fed","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-21 16:30:32.000000000","message":"Patch Set 3: Review-Priority+2","accounts_in_message":[],"_revision_number":3},{"id":"5562feddfc9b63416c49b333543df7e2ed8df229","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-21 21:30:01.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"5b8b968647bf058b86a7f1300b498972b499f32f","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 12:28:52.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2c11e020491db56e9e4b33760797c082d5b96e55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:29:37.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"85352ec964656d9fc87d3f159d08b4d5c9e3fa93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:07:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9994c98aa4b64c89a05498bcd8ab6b05\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/821ab0e7e82a4dfba01b0fb0fdf7f177 : SUCCESS in 2m 24s\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c886b5bd08514988849e1ba809c53055 : SUCCESS in 22m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cdda1a55f06849beaadfc97814de7434 : SUCCESS in 37m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0359e0ccadda4b57853c5da60ae35f5f : SUCCESS in 1h 10m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8ab9ff356acc4a619ee02ceb8a4d229b : SUCCESS in 1h 31m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e7b6aa6d16946b5a1bd4e0b7f15c619 : SUCCESS in 2m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9e6a451e13d24af0a908557e51fdea1c : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77eb05c63cf64d76a55e70d98c9a9ae2 : SUCCESS in 2m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3195157fa334e789dd55f329d053d8f : SUCCESS in 3m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ebb72c4848a84e9591b07ce1c38f0059 : SUCCESS in 2m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/93855f39966c4d3ab0d5669568570856 : SUCCESS in 14m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1a687d442fa043458e3a46c60f4a327f : SUCCESS in 1h 13m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/22d31947ae5043039badbc704da90c90 : SUCCESS in 1h 05m 56s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ee24e460ab594a67acc0f3693ee12d7f : SUCCESS in 55m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/515cf28853d44b77b92d1f2cc83173c9 : SUCCESS in 11m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b2fcf210c6c7417d80a42eabdaf3a355 : SUCCESS in 27m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6f05d48768c540a8b9edf5ae9f68b8ce : SUCCESS in 1h 08m 26s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/283093e5135145688408d9fc1fc6f4ef : SUCCESS in 1h 30m 09s","accounts_in_message":[],"_revision_number":3},{"id":"a43cdbe324b2fc9a457820b33aa4878cb9c62e7c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:07:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"84edf7d2e58b9021dfe05a7859fd4352479f5b15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:08:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0420ec78b904460abd38137106934f86\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a91755072d5411482bd68cb2eb67621 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f2fba3842cd520161f5c136297d806e9fe4d7521","revisions":{"e6794934b84fc0d2b6ec138af674eac1a052e263":{"kind":"REWORK","_number":1,"created":"2025-03-20 14:56:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/14/945114/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/945114/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/945114/1"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-20 14:56:14.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-20 14:56:26.000000000","tz":60},"subject":"Remove swift container images","message":"Remove swift container images\n\nChange-Id: I05a96f27b47a0f87bcf435872dd08ec0927232e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6794934b84fc0d2b6ec138af674eac1a052e263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e6794934b84fc0d2b6ec138af674eac1a052e263"}]},"branch":"refs/heads/master"},"a9dad13bdc086735479f713d83b0fc2fab141c2c":{"kind":"REWORK","_number":2,"created":"2025-03-20 16:41:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/14/945114/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/945114/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/945114/2"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-20 14:56:14.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-20 16:41:29.000000000","tz":0},"subject":"Remove swift container images","message":"Remove swift container images\n\nChange-Id: I05a96f27b47a0f87bcf435872dd08ec0927232e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9dad13bdc086735479f713d83b0fc2fab141c2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9dad13bdc086735479f713d83b0fc2fab141c2c"}]},"branch":"refs/heads/master"},"f2fba3842cd520161f5c136297d806e9fe4d7521":{"kind":"REWORK","_number":3,"created":"2025-03-21 09:37:17.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/14/945114/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/14/945114/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/14/945114/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/14/945114/3"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-20 14:56:14.000000000","tz":60},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2025-03-21 09:37:17.000000000","tz":0},"subject":"Remove swift container images","message":"Remove swift container images\n\nChange-Id: I05a96f27b47a0f87bcf435872dd08ec0927232e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2fba3842cd520161f5c136297d806e9fe4d7521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f2fba3842cd520161f5c136297d806e9fe4d7521"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
