)]}'
{"id":"openstack%2Fkolla~945157","triplet_id":"openstack%2Fkolla~master~I0e7ce418508f6684a20539c6b6129a134f91c71c","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-04-04 01:06:22.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-04-03 23:23:59.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-04 01:06:22.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-04-02 10:07:06.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-04-04 01:06:22.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0e7ce418508f6684a20539c6b6129a134f91c71c","subject":"bifrost: bump Ansible version to 11","status":"MERGED","created":"2025-03-20 19:21:39.000000000","updated":"2025-05-16 19:37:46.000000000","submitted":"2025-04-04 01:06:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945157","meta_rev_id":"3f3caab2fbd75bf931b8db58546702e9add84d5d","_number":945157,"virtual_id_number":945157,"owner":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-04 01:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-04-03 23:23:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":-1,"date":"2025-04-02 12:53:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-02 10:07:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-04-03 23:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-04-03 23:23:59.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-03 09:29:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-20 22:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-20 22:26:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-21 19:53:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-25 11:54:23.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-25 15:30:51.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-03-31 08:59:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"e508d82a0f5840e2e9251a3fb5a8f5ecd1d872fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-20 19:21:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a6e605f828f386b4a2a73677e6cabbce9546444","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:23:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff33cc8979d04a4a87d06cea7ecdaaae\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e5af8e5abe7443faaaad2d2922baf52a : TIMED_OUT in 3h 01m 13s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/50a2ab58015846839482be386f13c792 : FAILURE in 1h 50m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/68ba9532d95842d0b930b72c86898b6c : SUCCESS in 55m 29s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/32abe7ecf5d245eca1f546503c1cc8f5 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5987b30a9f464456bd1e376f347f6112 : SUCCESS in 3m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/19e6e8d5829a46ada0c502e993af4268 : FAILURE in 1h 09m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d125f212022e4802a8481292b92e9191 : FAILURE in 24m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"013e3275bf1f1df7c48a1389d9374211d5e7dea2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 22:26:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa8eefb74fbd43d9a29cad18cbbc8f8f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/00d9f0fc4d144910bee7c20841f70b2b : FAILURE in 1h 05m 45s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/76f1c7a9a5d14a3da52abdc289a450d6 : SUCCESS in 1h 48m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1fec193ba4e4410080cbaf69f82d267e : SUCCESS in 10m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ef9941eced7a431fa17373f041ed263e : SUCCESS in 32m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8dc31819f0994f1ca601fbf1c128ebc0 : SUCCESS in 35m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fa12d953de374d76ae0d7829a09038a7 : TIMED_OUT in 3h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83bb150fba14cbfb9194d190a240da0 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e07f147e4fc64d0589418a7b5b22e709 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d3933d14e454e1ba3174aca4a194d35 : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da17d6e48eb348bbae7cc36dc13a86e9 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef8263ee6084dfd8972885d6660ec4a : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f7249eb0e864b6699d9000597604f0e : SUCCESS in 5m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9597c1dbdcc54adc857b9af57e8f735c : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2427659345064ac383c4245b32669243 : FAILURE in 21m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6ea9f134fb06458a94fab86cd4a699f1 : FAILURE in 1h 06m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8239617300094ebf9402891c1d1f10a5 : SUCCESS in 1h 02m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f17b8dac03a54eac9855895c82529b1f : SUCCESS in 1h 36m 02s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/4d0021e58e3f4da196d8094368a28c2f : FAILURE in 27m 06s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/86ae578ddc6b413bb0b49c2fdcea24ed : SUCCESS in 27m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6da8e3b90e1b462b9d7f07c700574e70 : SUCCESS in 29m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/06a2fc653b634d2c855f02f0e2052c9c : SUCCESS in 1h 11m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7a23a00f12674c7fb96016a872a5ea95 : SUCCESS in 1h 40m 14s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/0f8c34867d9944faa20363c2dcb12d26 : SUCCESS in 41m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d47b7764993cd9afe899a7a60fc72e05443ece49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 11:04:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3bc5aa9094f8d15623df69bee4f31655ed56f527","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-25 11:54:23.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7774e8ff6c5a87a1aa819ee438f28cded952e9fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 12:17:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"ccd0fcbed0be3fa5e997734911b0314f909f8b3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 12:40:51.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a9d09dd19076f6cd3520ebd381a98791f15b18ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 13:13:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"de3b1c52c730ee1349a49a2537949e7830e344ed","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 13:14:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"512c8f84cd30ea6576a815f381d12580fe368f27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 15:04:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca6001581e584537aa83d220e4c9a96f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/57b9fccd0eb042e0b418506e04a8031f : SUCCESS in 29m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a0a332ba04fe48fba7bb98360f4b3621 : SUCCESS in 33m 47s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3717aff7696543288c67e49f77501398 : SUCCESS in 18m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6a54170031644cd6bb0a17308e7b7988 : SUCCESS in 29m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f0bc37cf48c84a809a06025fb0137b52 : SUCCESS in 1h 10m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/deb34ef53b2c41abbe9e8ce6b7ca9e34 : SUCCESS in 1h 36m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4caa3f07344e65be8e77e778bff645 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/183c6d15ed244612893cf771882fedbd : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e0b0f147c1184809a64125dec0c43ac3 : SUCCESS in 4m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e31b58dc0464c79933fbd8a866905f0 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ae2558b96a84675a2dc199172f00cd3 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e5d7a5dc26645f79560490605ce8904 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2fb4d58e768a4fa0a8a33e3f47a9c99b : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/285cec40a3b5481d92e2ec8ec5f21bfd : SUCCESS in 24m 33s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9e08ce21e43c49f28109d7e011ceb44a : SUCCESS in 1h 36m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/46709c55be524f639d2f07ac1ff77ad3 : SUCCESS in 1h 06m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6385d8dad1a34bfa90d38221ee7348b5 : SUCCESS in 1h 35m 18s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/ee821b3e795342a6b5bae3fd3149dc07 : FAILURE in 23m 54s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9443001c0f014a3d8a21397418fbf87e : SUCCESS in 24m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/84fa5ea08b3a4afcb3758fe9560a3870 : SUCCESS in 28m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f184cd828d41426c8f4b4b98030ce905 : SUCCESS in 1h 03m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/175844f5bd2c4d6893f7736c0abe1b9d : SUCCESS in 1h 25m 27s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/9e849e79682b431f857fba8ba64fff8d : SUCCESS in 30m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67bb85867cf7185cb136e5ad90de6fb0e12faa0b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-25 15:11:43.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3bf95ad93608fa66235bed50f0a825fa206f1d07","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-25 15:30:51.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"96cb0fa936402900160cd7d6ced19c2b9208cea2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-25 15:30:57.000000000","message":"Patch Set 5: -Code-Review","accounts_in_message":[],"_revision_number":5},{"id":"2bac26a97bf6408b319aa8527f3470d3ebdc68c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 15:33:39.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0a303aa95174f11bd6c96494d0bfd66\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1a14281076d8404f97e8857b270c8301 : SUCCESS in 1h 04m 00s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dd2062ec17c04e07b21b57c854eba9a7 : FAILURE in 50m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9fc1ba88585c48b78e5da81cd65909e7 : SUCCESS in 1h 03m 55s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/11d5810f421740ada6a1a185e349b6e7 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2f453a26d554cd6b9ccfcaa540150a2 : SUCCESS in 3m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2d8817c32ed34ea9a407025ae616ea56 : SUCCESS in 1h 02m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/beb2a9b079354eb8a9cd7df5ca28abed : SUCCESS in 52m 48s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3696ba109f45ea2da43ba06347efe73ba633906b","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 15:51:22.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b4a60c36acb048189dd25696792f54986e964c98","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-25 15:53:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a06982e1b5123f465610f998d611d9dcf123dd0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 17:52:31.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d94b27668da61d4c436245524868673f01234df8","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-25 17:53:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c3ff0cf54f438486fa107d1840daac710cae5068","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 19:54:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/122979ac6ce44689a718a9445704d6e5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/169a1d962d2b4bbab3e318a653552619 : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f89f91af91de40548753188251aafccf : SUCCESS in 58m 36s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e7d70194f0c04f188dd42d8b27a245c5 : SUCCESS in 1h 28m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/de78cae8214e4bb39cc40a7725119883 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50ec881782734cbfb2b6b010e49a8259 : SUCCESS in 4m 05s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/b7205a3f330443a0a0af9f74b8e3a490 : SUCCESS in 53m 13s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/95e801f86fa24bba8087a286a974659f : SUCCESS in 59m 30s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9b21a9dfd1380d52aa332791f0f1b7bf0537cf5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 20:04:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7829262f9e4d43718490f700260a45fe\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/bb1fcba5b78c42e1bf0aed11e0d4634d : SUCCESS in 15m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7051f82d40b54d4491136fa9ed3a396f : SUCCESS in 2h 03m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d2695455e69a4afebdf57c11008125d2 : SUCCESS in 20m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a3a4f96405f44c7c87aff141d3992847 : SUCCESS in 30m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d01e835ef4464aad9cb319523dcbffc6 : SUCCESS in 52m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/30bb0550791443ccabf139572b2a1d49 : SUCCESS in 1h 16m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/339ed55b44ce49eba3adf01751531579 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6786ee7d2cf64e15a1899931a5eca39a : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d1f1da5925146a190f2443ce312639d : SUCCESS in 2m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17b9f319bb14480e9b748c8a88020100 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89c54597e9784f688781b5609bab891d : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86e555ac554f4f8f8906975e5a900c11 : SUCCESS in 3m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4822a887ce6d4341ae826cee4c1afa84 : SUCCESS in 4m 16s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/542b8dd7daac4d219950586def3cacbc : SUCCESS in 43m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/43ad9ad2b91142f09e14c472f0ee0ba4 : SUCCESS in 1h 35m 19s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76641629ce89428d985503a3baa742d5 : SUCCESS in 1h 05m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5fa4804cb2854b6c8f4b2fc4670c0b61 : SUCCESS in 1h 30m 58s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/5884289ab33e4db9b858c02ffdcc976d : FAILURE in 25m 27s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a175f5d7a53c419b95f83fe55cd4baea : SUCCESS in 23m 16s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/27114ebcfd284e06aecaec6514fab1cf : SUCCESS in 27m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/934876c5f84042358b14425b99ca01b8 : SUCCESS in 31m 58s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d4b52e32a85942a09f24a945c2d3a670 : SUCCESS in 1h 21m 25s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/a169433f6058467b8e68edc815beceea : SUCCESS in 27m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f2784b2728e4b6a5eb5927500e1c2073a2a867ac","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-26 09:16:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"6ece9b967e7b583cf43e6a4783ef0b78811776dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 10:32:26.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/945498 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"d1e150ac7f36a696a34229e79951482b72e37496","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 10:32:29.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/945498 is needed.","accounts_in_message":[],"_revision_number":6},{"id":"1767b9b1cf277c480c1f565a7f5696b4786b289d","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-26 10:32:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"828c4c49b7f083a40b243ce280ff36c3afc06999","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 12:23:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1aef06428004ee7b2568b319cef3056\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f911010de62e4a25829c0ae83125d4b2 : SUCCESS in 31m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cc99ed1770464f08a2a7aa05ceb27d11 : SUCCESS in 32m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5cf96ebe01624746b6fe4153e17efe03 : SUCCESS in 1h 33m 06s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/361696762bb94b6891cda0fb7e87d4d7 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4ab971eb0a44db48cfb5d4d6117f8f9 : SUCCESS in 3m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5f5bc188732f450a8b4d159afbf29fe9 : SUCCESS in 28m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9f9c3925cd3a401c95207ddfdf55f0f1 : SUCCESS in 29m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8a1465b288beac1f68a0e4cc6c20e0f6fdcf9917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 12:29:54.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ac2beb280b446d7b9a3b359ca9912e0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a6a23e5db4344b8287eaeeaf6c9b1907 : SUCCESS in 14m 24s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ee473ca2f9b34481849440d01b812a03 : SUCCESS in 33m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/13bf491cd03146828f59c301bbf6f3cd : SUCCESS in 39m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3635da40da194b8ebb70d413424183e7 : SUCCESS in 28m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f330e563903d4055b9b1ef8d2c1bed86 : SUCCESS in 1h 06m 01s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1ef43713a95d4c97a5d8cbe4c56f2254 : SUCCESS in 1h 26m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7dac1d7b3ac4878a282e008c542bcd9 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/445df562d4f64f8bae28728999289d0d : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3469eca3b4944c7886adb058766c5ef0 : SUCCESS in 1m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa934525667f4a568f3a48de61609bef : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de965a4d4937449693ef80ebb50668d6 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44a4ee15390f46ae84850bfb9bfec0f1 : SUCCESS in 3m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d36cdfabae1646d2acdc55f61c4db1b8 : SUCCESS in 3m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dc5dbcf2f47d4eb5acd55c6be4fea545 : SUCCESS in 38m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ed43755856514b4eb658e3a023323b6b : SUCCESS in 1h 04m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b8732754da5e400a95f40e9a1896a8cb : SUCCESS in 1h 07m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f53a747b000844d2be54791c910adea8 : SUCCESS in 1h 22m 59s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/38520aaf536a4bb69ace2a59e8d444d7 : SUCCESS in 29m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/31bae7eb9a6149508a7ec656e03c9584 : SUCCESS in 49m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/68b73a03e27c4879b733289909d0e786 : SUCCESS in 29m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3e36dcc6e6f44bf68551079fbc9d75c6 : FAILURE in 1h 41m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e8bac0bcae404a1b9f02447a4fb236d2 : SUCCESS in 1h 28m 32s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/66432872703c4a658aa630f6e2f6501e : SUCCESS in 26m 41s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"95d02760cc1983a2f5c718b68a505a7b96621827","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-26 15:20:08.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ca2c23aeeb4c12e70b059f743d9abc4f454e9d3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 18:22:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b994e0520944af68786b27fe824680f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/cb537b57036247ac895b40cc50b941fd : SUCCESS in 39m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/90c35f2f5fcb4afda9846b9e021c2dfe : SUCCESS in 26m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/65217061c86841da983d3d2837c1e3e7 : SUCCESS in 59m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f3a2d7e02944bc0ad38bcf53b9642ba : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c6caaba5b804c6b9b2a972d5519855c : SUCCESS in 4m 11s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0b19ca93c57a4536b89f302fa5b2dd41 : SUCCESS in 29m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/69e142f3d320423fbc3c00729495f3c6 : SUCCESS in 27m 50s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"88848cb954c8237a57205fafb991ba817f3e3558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 18:33:53.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/453468d3dd9d4061a21fe2511ab5baf3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b6bb775be7f44be797d4c4026ec34c13 : SUCCESS in 25m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/313abf58bf144a8fa6ac7fbcb9005b7b : SUCCESS in 1h 11m 57s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1c939da8da84fc995255b1dc93cf0b9 : SUCCESS in 18m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6aa8178067b94ea7bbe46f17ba0b9bbb : SUCCESS in 28m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b6f5183db95a419fbbabb33170b3be1b : SUCCESS in 1h 09m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8c4e8044abe04e29b0281b53410762a8 : SUCCESS in 1h 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e92ce2441b2e4db79eebda000c6b91cf : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/237dfbb369d44e7f94f31ac3b70fc0db : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52e71e2a624c4b969a47ecc56a7fb455 : SUCCESS in 10m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ede9993b2ea742068cafccc57b67782a : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/560b1d642fc44eeaa739f39bd36e5384 : SUCCESS in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eb98213b53034e6cb737f8e7c7213213 : SUCCESS in 2m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/664c2b8c5ead4ddb8363819da3588254 : SUCCESS in 1m 53s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36550eba669f49f485b71f0b1fe997db : SUCCESS in 48m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/337b56d8c487485a82210a8c4d7c4d4d : SUCCESS in 1h 13m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/67667b2926a1452f87be63b4958387f0 : SUCCESS in 1h 05m 46s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b7173663e67a432098e66bec847ae080 : SUCCESS in 1h 33m 35s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/5ece7eb7c67f481a892a2a3f753fd1ef : SUCCESS in 29m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/47ab6a8d4a734101989dd8f1a5f2cabd : SUCCESS in 22m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/daa5582f8398429885392aff8346e512 : SUCCESS in 1h 18m 29s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/91f948206abf416ba0ecaae30509440a : SUCCESS in 1h 11m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e79d095207fb48e28345d52f27c623ce : TIMED_OUT in 3h 01m 18s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/da3f1967c6b1469abd80ecaa14e2d4be : SUCCESS in 33m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b78876d6cfc5f8da2dff29cb813dc7dfc5f2e983","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-27 08:42:48.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ca711cc58edc14b2f7f8f0a6b343d55f95e7b951","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 10:28:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/822c4f5f6a0a433aae225f8a8c35be24\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f807f584948c48c993b9471002795481 : SUCCESS in 29m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f8544d4d8a5b4fc6bedc861e27e3a40f : SUCCESS in 1h 06m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9531a73904be421b995aa794358a84da : SUCCESS in 23m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5d8600da468f433b97d00cdca5bc573f : SUCCESS in 24m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5f24e47828fb415393bf1ac32ff69e40 : SUCCESS in 1h 05m 50s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/34d415c939af487f9b8f6ee16c14ecd5 : SUCCESS in 1h 22m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8386800db384326a8bb611a0499f790 : SUCCESS in 3m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/490e7842d9864dda97d4d9af3bed92ef : SUCCESS in 2m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6160d9b4a7174b63a08589775b51ed6f : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/692d2b0708654de6883272da3dd3a68c : SUCCESS in 1m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59e3f99d3f847cbafd5e3d770d754f7 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6a015947205b428ea04db1d8e0f17b46 : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/00f21fee74294992935b645724638bd0 : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/36f4e085e9024fd5bad825c67b8985b4 : SUCCESS in 21m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b69b6353caba402d833dcd576e4ed0ce : SUCCESS in 1h 13m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/638cd5d5c1634df8ad44c59d42e7b79c : SUCCESS in 1h 07m 37s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/676b835182ce45318014f098cce5b97d : SUCCESS in 1h 23m 14s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/e153d184c85f430a8f9142c6cedea19d : SUCCESS in 31m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/30f86342e05543a3a0eaba13b3510abb : SUCCESS in 24m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/44813e9308484d2e9a98df204f6c66e2 : SUCCESS in 28m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ae5c3517e2bf484b816e5b8647a428f5 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cf2c9d7332f94fc4af36f2d5dd1b0d48 : SUCCESS in 1h 25m 30s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/60e58818a1d84313afb06a936a26f871 : SUCCESS in 31m 05s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96e867d4df3436b84ea31c6e1887fbfb0ef86000","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 11:00:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7770ecb22cd944aab4271fdb225cc4a4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c4fb6046b4074d518471c8c004ac59b6 : SUCCESS in 36m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42b9345f489d42ee8cc60309efa9f79a : SUCCESS in 34m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0909be26e2084252b6ee9ed17a48064c : FAILURE in 9m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/94a947510d524026ac0dc6cb8ad503d7 : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94bca22af37f4d198fb6d0220e3ec203 : SUCCESS in 4m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ed365e2088954b409729ef6a60dcbc13 : SUCCESS in 41m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ba4bb2ecd74425baa6178d27938cb53 : SUCCESS in 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"566102546a6e45df0b22b698b3e9746868dd7e17","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-27 11:00:28.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b1bcf314e5aee5a2a18a3d2df0c371140a74512e","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-03-27 11:31:55.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"dd2366d2d49dd20d33250ad6e88746ef6c89682c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-03-31 08:59:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"cc3f1b8b58c0cc82462406e9bf27dcc9326bf8ba","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-01 12:50:43.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"2e1489d4057326f2743f97c5e9eed0c996177828","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-01 12:51:11.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7c93713150fd36343928dc6d470ad9f4738e4b9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 13:52:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2cae59041d4afcabee3185bec7b265\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/78ae64acdc5e4e41adb15378817f2e19 : SUCCESS in 30m 57s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f4a13ea9f5314652b8147ac1098a7ec1 : SUCCESS in 27m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5213f8344bf6491b88073b2d30ad758f : SUCCESS in 59m 38s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffa8912dc7ca49f8b0a46bfd5f9a6d02 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/01cda8c2008b482db5ae81e80e2fa4db : SUCCESS in 3m 23s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3ecb711053f74c0a8cd700e10fc533a7 : SUCCESS in 34m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fb78e0bc69b94473a4d2b3553b48a4b6 : SUCCESS in 36m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"b601484b65e32bdaf2277b2b985fe52fdd7d7b33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 14:41:39.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b14da0c866b408b8a21d7da417033dc\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0e15dbd436914e5c983860899422ed7f : SUCCESS in 14m 23s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e6d08e44921a4496a82c0e7c5d9a42d7 : SUCCESS in 1h 12m 42s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ffc1498d24c34491844dbe4c542625c3 : SUCCESS in 23m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9591ffd901594c669a28e2f9859e9ee2 : SUCCESS in 29m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/705a3b00cef5415194337123820e461b : SUCCESS in 1h 05m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a0e5b4b82f274aabaa9abcdb0615530b : SUCCESS in 1h 35m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd94d80505034524a4996805c9bc408a : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c00245edf04642fbbe2b1b138b5c24ea : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7652b03240e145108702423a3a92a7b4 : SUCCESS in 3m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ee1ec1635454c98808fde1c0daa4645 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e561e71221b84a3a99069b6a87b5e50a : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9149677246af486faec2ce56bdce7d64 : SUCCESS in 3m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d114bda25ca64341b497c2c06d34999a : SUCCESS in 3m 46s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/815cffa9fac941b89e2ed0d8900e7d51 : SUCCESS in 12m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0a125303368b491b80cb012f1b72b801 : SUCCESS in 1h 16m 01s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b1676383dfc840b3be6cc33d6508c97f : SUCCESS in 1h 08m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/91ff4be934414ebbbe56a78ccb6cdf28 : SUCCESS in 1h 34m 16s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/da50b14fd71a4087aa6160d68ef1711e : SUCCESS in 32m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8713bd61c4494a8bbe2f10766a3e4cf5 : SUCCESS in 18m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f56ddf9492c0450297d7b9ab1f58f709 : SUCCESS in 26m 31s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/14903db6ed9440118d8b2a9375e19b07 : SUCCESS in 1h 00m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/55c1bfde594347df9afbf4f7299870e7 : SUCCESS in 1h 45m 31s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/c62bb7fa911c405d8c00e4e12887c156 : SUCCESS in 33m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f3bd62da082acdf2230e613b815e79a8bebc2437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-02 08:30:48.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"6ec16c026cb88a2e9a498650f76a3ac2b1089b10","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-02 08:31:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"74f4035de9ef422cc0a27f7bcdf0edfbdf45b93f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 09:57:00.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81cc11f8f15749aa9157a479bc2bd0be\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e2a81752925940e6a738bc3e9f2b50ed : SUCCESS in 45m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d71c7029190e48a894d1965c3713182e : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a1ca6610da7f4424a95b81550e6711ea : SUCCESS in 1h 24m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/198d13e9d13a4951812c4f74931dff4c : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94d6f6d666ed45bbb23dfe9f0c2b80f1 : SUCCESS in 4m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/fcdbca7025ab40e8a73481333260a34f : SUCCESS in 38m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/88bb05dae5254d2ab2241e571beb0afc : SUCCESS in 45m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"fd516662688083d39e2c48a430a5ffc8e7439075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 10:04:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea932602721b411abca3832b438eaec4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/682a72cbf48642f68a0150ddf9996e2b : SUCCESS in 25m 18s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f35a63832a414292a4e44c8da9340a1e : SUCCESS in 1h 08m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5651bbf613034403a3b2f92e02f1ade3 : SUCCESS in 21m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/af32bb2ee3b843f3935301729458e5bb : SUCCESS in 29m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cf893eb831814d469008c7d8e1b542e1 : SUCCESS in 30m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f9fd989452b944ef833bb79194cb0e72 : SUCCESS in 1h 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/125719b5f25f49d396bcac077a2daf91 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d6a0029ac34458ebde6d9a279f777f1 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c45867a16244d41a9600bfae9a03d91 : SUCCESS in 1m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb99d6b1c60b4f768a1e5608fb704450 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d86d5b8265647338eee7bc16193bbe9 : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a852d2a60ffc447994551d1ee09d81c3 : SUCCESS in 2m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b9820ad0530d45249da7ff0076d84a2c : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cda4af796dee4c11a1c4eceb6d3ff56d : SUCCESS in 27m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ace20b31a47c43b39978c42c2781ad59 : SUCCESS in 1h 17m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d01bd31029044aaba071c8ca3a929760 : SUCCESS in 1h 06m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a23ba459340842fc983f7ae10fa52a70 : SUCCESS in 1h 29m 03s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/6b5be1e2d7654b328cba47ecdc4e4912 : SUCCESS in 28m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5cc6e0d1c858460abbbe696dc4de3c4c : SUCCESS in 23m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/edb0acc8acd0423dafe04f032d93b53d : SUCCESS in 21m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/52984318414f4c8785584526f211f3f1 : SUCCESS in 1h 00m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8e88f82b3ad54fe9ba542947dfe1abe3 : SUCCESS in 1h 31m 26s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/58c3131db35d4429ac3dfd10fae1db6a : SUCCESS in 25m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c01b60bada19734463f7830c6f903b17e774ae21","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-02 10:07:06.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"c75923dd55563ebc4cf98d7fe69c92ad1f405c35","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-02 12:53:52.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8af69b707a5a500256441b5fcbd887b0470fddfa","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-03 09:29:22.000000000","message":"Patch Set 7: Review-Priority+2","accounts_in_message":[],"_revision_number":7},{"id":"6929f171c56deb9f32675cb0d82afbade59495b3","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-03 13:51:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"83c79d4dbff82a50e0d0d04b74f9a368dd2ed683","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-03 23:23:59.000000000","message":"Patch Set 7: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e287476015841308ca21c0ffe38b55ddf4dcf583","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 23:24:32.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"cbc562a34823cc9a374c1d62a7bc90a53781b517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 01:06:22.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb7ffbf282244dc88163cb1bbc6a1685\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86d2f9250d4b472d908ce6d96ac68ba7 : SUCCESS in 21m 38s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8eed1309ebf64b278bf425d4241caccb : SUCCESS in 20m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1132741b79a94b44a508ed0521e2f764 : SUCCESS in 1h 04m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/110d3b36f1f04f2ba5551fd0bca27ab8 : SUCCESS in 54m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1140a5f98ed4908a5bd3be1b58bd4b8 : SUCCESS in 6m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/09260f367e224227bfbd4d90c3d1e0dc : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d027199e5bc943d4860934756421ee96 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ed8b8a829a44afaab67c5a3e207194 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b894b604d95472b8cf2d4fb1dc0ad0b : SUCCESS in 3m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/37b56163341a4a98a3524ec638c285a2 : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ea9b1261834247189f7c3d97e8d47cc2 : SUCCESS in 15m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2b9c9f964f5f4682896f65cfe7b6419b : SUCCESS in 1h 26m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c3e9e57ddbff464aa7048dd65ef2057a : SUCCESS in 1h 03m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/897d9e231c0c41498175ff2ce1a6035c : SUCCESS in 1h 31m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a8abc5712c234bd2b865fb89aee30c3f : SUCCESS in 20m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/794e172df83540439c71a257aa7594aa : SUCCESS in 26m 42s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ade45e061de34cb39c8794e9ea9cfe57 : SUCCESS in 57m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d12ad0c74fed405fba0ca0c783a86d97 : SUCCESS in 1h 37m 30s","accounts_in_message":[],"_revision_number":7},{"id":"549141d2a064ae7e61463f1e6c4501bf7b83d65f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 01:06:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"354b66c14bf9fbcbac7765d4dddfa1eba8eedc3b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 01:07:26.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/859c2cee32e14a91af8debf04e9fc6c6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b10e459425f7458da66018f8b64d675f : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/77ba342f822242a3ae02e9ac797abe75 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":7},{"id":"3f3caab2fbd75bf931b8db58546702e9add84d5d","tag":"autogenerated:gerrit:revert","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-16 19:37:46.000000000","message":"Created a revert of this change as I69227a0548037411ab81890072aa779708cb85a0","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"c8fdb5550b15f98ed8f331f8d4b88ce150ccf3a5","revisions":{"f344be7947cf4bc119a6b8b50126f24c5acd5f78":{"kind":"REWORK","_number":1,"created":"2025-03-20 19:21:39.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/1"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f344be7947cf4bc119a6b8b50126f24c5acd5f78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/f344be7947cf4bc119a6b8b50126f24c5acd5f78"}]},"branch":"refs/heads/master"},"3451bb1f0750f3795c685aeeed79c22a8d78a38f":{"kind":"REWORK","_number":2,"created":"2025-03-25 11:04:33.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/2"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 11:03:51.000000000","tz":60},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3451bb1f0750f3795c685aeeed79c22a8d78a38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3451bb1f0750f3795c685aeeed79c22a8d78a38f"}]},"branch":"refs/heads/master"},"51315a384ba017b7ba9d565a869447ab57eb807f":{"kind":"REWORK","_number":3,"created":"2025-03-25 12:17:47.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/3"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 12:17:33.000000000","tz":60},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51315a384ba017b7ba9d565a869447ab57eb807f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/51315a384ba017b7ba9d565a869447ab57eb807f"}]},"branch":"refs/heads/master"},"7ba5c73876b2a82d7d8a77ae39a50d408fac8d83":{"kind":"REWORK","_number":4,"created":"2025-03-25 12:40:51.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/4"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 12:40:44.000000000","tz":60},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba5c73876b2a82d7d8a77ae39a50d408fac8d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7ba5c73876b2a82d7d8a77ae39a50d408fac8d83"}]},"branch":"refs/heads/master"},"d707f96c2b7ae15f7bfb083838695b33d35653e8":{"kind":"REWORK","_number":5,"created":"2025-03-25 13:13:37.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/5"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 13:13:28.000000000","tz":60},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d707f96c2b7ae15f7bfb083838695b33d35653e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d707f96c2b7ae15f7bfb083838695b33d35653e8"}]},"branch":"refs/heads/master"},"03b5b073c98a72144e5638d21b7e5731fbb207fc":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-03-25 17:52:31.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/6"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-25 17:52:31.000000000","tz":0},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/945498\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03b5b073c98a72144e5638d21b7e5731fbb207fc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/03b5b073c98a72144e5638d21b7e5731fbb207fc"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c8fdb5550b15f98ed8f331f8d4b88ce150ccf3a5":{"kind":"REWORK","_number":7,"created":"2025-04-02 08:30:48.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/57/945157/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/57/945157/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/57/945157/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/57/945157/7"}}},"commit":{"parents":[{"commit":"830a10922565c9e482d799e7feca47ffe16624b0","subject":"Merge \"Ironic: Add ESP image for UEFI virtual media\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/830a10922565c9e482d799e7feca47ffe16624b0"}]}],"author":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-03-20 19:20:54.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-04-02 08:30:46.000000000","tz":120},"subject":"bifrost: bump Ansible version to 11","message":"bifrost: bump Ansible version to 11\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/945498\nChange-Id: I0e7ce418508f6684a20539c6b6129a134f91c71c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8fdb5550b15f98ed8f331f8d4b88ce150ccf3a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c8fdb5550b15f98ed8f331f8d4b88ce150ccf3a5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
