)]}'
{"id":"openstack%2Fkolla~945296","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I11b7e19ac3a12147138a98fc1e030767f70b44c9","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-03-24 13:31:00.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I11b7e19ac3a12147138a98fc1e030767f70b44c9","subject":"fix sources.py based cross-arch builds","status":"MERGED","created":"2025-03-24 11:02:00.000000000","updated":"2025-03-24 15:00:32.000000000","submitted":"2025-03-24 14:59:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945296","cherry_pick_of_change":938791,"cherry_pick_of_patch_set":19,"meta_rev_id":"28c5ea19e916e8e153e11651d3446cdd9bd07a59","_number":945296,"virtual_id_number":945296,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 14:59:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-24 11:11:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-03-24 13:31:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-24 13:31:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 11:11:14.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-24 12:35:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8abab8fb79ded739c507f02f53cbd5421c7febdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:02:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f323f25d7d4cdb6a1efda06aa189cb88938f1ade","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-24 11:11:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"def197afe61030cfcf6c4228af79248c596118fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:35:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8373585ccd4cad910f9d4d13e23e3d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/19922b4a6c2e4c119f326b46aa1b52ef : SUCCESS in 21m 04s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/93aad8c3decb45d4bb9edbe6ce42ca86 : SUCCESS in 1h 03m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0a661f18299047428e8e5a746414b794 : SUCCESS in 22m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2178a836b683480da72cc1a3d464e36f : SUCCESS in 31m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/47faf36207f84ee791865d824a255676 : SUCCESS in 59m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/56a9764b9f184754a58027b7396b385c : SUCCESS in 1h 25m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6f73b6657aa4be78e011abb7fe76c69 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/663ed108e58d4b0a9d4e0c63bc6473b8 : SUCCESS in 3m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9bd825f98f554cf6b977dfe93ce76f7b : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4081bec4d0494f5e84b870d800b547e6 : SUCCESS in 2m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfa3948435fb427789936da2855d388b : SUCCESS in 2m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e62aee8ded2b45fa835ba93d9b849e32 : SUCCESS in 4m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/755d7f8a87ae4d0dbd468e412a03755a : SUCCESS in 3m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3452928198114d09860ed4bfe62814c8 : SUCCESS in 20m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d3a943d6a32247d28f97710d7f5b79f6 : SUCCESS in 13m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/92319fc49ae1420987446b0c7de72ec0 : SUCCESS in 34m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/03825f4a52fc457bb0c6d5dc45bb8ebd : SUCCESS in 1h 28m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/58cc02acf7cb4777b73149fec000fa2d : SUCCESS in 21m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9dbf08a6f6254289809d2e9008a52ba8 : SUCCESS in 26m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/24baa1bc26e448e1b3c0bcb099a607aa : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1176ead9ccc74ea186a2a4eb261ca485 : SUCCESS in 1h 32m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6b6e2c0c486404d3c9d73bdef04205715e6b39eb","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 13:31:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7d44c7adc704d7b2b38c7083989fed7a94af8923","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 13:31:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"26d4e85582cc58af08cd8ce283bffb4204098d6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:04:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/678056b611b149c4a358b671be9a4da4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b1ae1718ba1b40d9b47fab2baafd450f : TIMED_OUT in 3h 01m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96fa7a30f4394baf8542b85cc458259d : SUCCESS in 1h 20m 03s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2862b45491534ffb96051e7dd9d69891 : SUCCESS in 58m 02s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0487be54fdd4699937773b9a4a460c8 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2da4cf6c0cc4cbc9aa9a274fb48e187 : SUCCESS in 3m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/41ad747a0d824fadab78e0cafaa5a235 : SUCCESS in 56m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ef602fa6c13e4e7687d86d7091257e1b : FAILURE in 23m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"224abde0349bc11f515d585ce5347062065efe5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:59:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89085185b6c0474aa7b48a48a2aba5b7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ccd8fe8367fe4c75a9dda6dd91205a4f : SUCCESS in 20m 16s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e9ce0a86ebe84013b5315a51d4f3c975 : SUCCESS in 31m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0e3cf206eea241a7836f63d5fcdea38a : SUCCESS in 1h 00m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/669bf35884784f5b9cc3ad12cb3ac8ff : SUCCESS in 1h 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9caaf8c1b8ed4b419d9b23413deca6ca : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf580f3ffac841d787278e20f9f3ee85 : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56e0b75bf4334a8095fe3f618c2f0435 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc992456285142a297a3ca66b2de440d : SUCCESS in 4m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6d8bb2ec8e6a4d8fb1c97c0255c1b892 : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cae84930c16b4363b878ef708397a803 : SUCCESS in 19m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ef44aa05ce81488980d377cfb739a59a : SUCCESS in 29m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6449c373846c40f89b9bc5c892924102 : SUCCESS in 36m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f73e1ea7cc784f4184e85125016237e6 : SUCCESS in 1h 20m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e89f7ec6a68149d4b26aad9113254e24 : SUCCESS in 14m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b660bc65f1d2457283b8e41fb3d13841 : SUCCESS in 28m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/124952e72db64e889dd12ea724e1a03e : SUCCESS in 58m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0a137ef551ec4bf185c6609b89934418 : SUCCESS in 1h 15m 51s","accounts_in_message":[],"_revision_number":1},{"id":"520c3d755826d1fe995429e5674ad64f7ace5490","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:59:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"28c5ea19e916e8e153e11651d3446cdd9bd07a59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:00:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4808d18e42364f28badf7bcdb71efadf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8499315bba884c6c8bc484b5c9f13e96 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1ce31adf79e648b5d960868f000186eed5cffe5","revisions":{"c1ce31adf79e648b5d960868f000186eed5cffe5":{"kind":"REWORK","_number":1,"created":"2025-03-24 11:02:00.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/96/945296/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/96/945296/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/945296/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/945296/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/96/945296/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/96/945296/1"}}},"commit":{"parents":[{"commit":"fa7b8906110d5c5199930fd926f3e581bede39a9","subject":"Merge \"Adds missing nvme package in cinder-backup\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa7b8906110d5c5199930fd926f3e581bede39a9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-09 11:49:25.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:02:00.000000000","tz":0},"subject":"fix sources.py based cross-arch builds","message":"fix sources.py based cross-arch builds\n\nDerive debian_arch from base_arch by default when not set by the user.\nThis ensures that cross-architecture builds correctly download\nthe appropriate binaries, while preserving any explicit user override.\n\nCloses-Bug: #2103810\n\nChange-Id: I11b7e19ac3a12147138a98fc1e030767f70b44c9\n(cherry picked from commit 915189d10be83136d54f4216b81310b3dd16b82d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ce31adf79e648b5d960868f000186eed5cffe5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ce31adf79e648b5d960868f000186eed5cffe5"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
