)]}'
{"id":"openstack%2Fkolla~945300","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-03-24 20:21:12.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05","subject":"tests: Add cross-arch test for etcd","status":"MERGED","created":"2025-03-24 11:13:55.000000000","updated":"2025-03-24 23:31:10.000000000","submitted":"2025-03-24 23:28:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945300","cherry_pick_of_change":940442,"cherry_pick_of_patch_set":4,"meta_rev_id":"be1d6149dd2c1ff56cee4871d229f47b245539d1","_number":945300,"virtual_id_number":945300,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 23:28:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-03-24 13:31:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2025-03-24 13:31:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 13:03:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25400dd07c828c1a5effa5c20d16aa41a0ac64de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:13:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"00d4574bfbe9460d745d3a7bfea0898990b43943","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:22:31.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"362be7cae8bb0f745a341e458722e8626423e89c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 13:03:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ade038d532b4975bf78dd4052785244\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7955b026f5d043cfab6c8872694fda04 : SUCCESS in 19m 20s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ef63206d5d844ae1bea83e981b8061b7 : SUCCESS in 1h 07m 59s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/686d51528d62480dbe1d638a1f674867 : SUCCESS in 18m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dccb506e6d9141b3b2f1a4b6efabd7cc : SUCCESS in 29m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/88e533995eaa4e20a245dd22c4f14b70 : SUCCESS in 1h 07m 47s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7725f02c655246d38ade5c5670b90b4a : SUCCESS in 1h 38m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f7ebceb0901418dbcffb29a43ecdc27 : SUCCESS in 2m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a79dcf28b3264f0188edc28a335346b3 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7a74d5b9cca849e79d40541f44227883 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b5761b34ad748098144def47d766c79 : SUCCESS in 2m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ad367aaa9a04e4b8d636d2e4719acac : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bc3cbea32fe455cbfe1d0fe54d765d5 : SUCCESS in 4m 30s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6c280754959546f592dece8b72e2907a : SUCCESS in 2m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/25ed437e0ba64eac9bf918d7534f95c3 : SUCCESS in 23m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/592400a4064b4b769aba787ee6de3a46 : SUCCESS in 13m 48s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e8b59860b25b4a0ca921bd6cb843693d : SUCCESS in 36m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3be71584f21d42139beb966ec130b4d7 : SUCCESS in 1h 24m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/785c60cea9f24314afa88d1d6fd83af6 : SUCCESS in 23m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ac2432f26acc41e59f7797ba5b3413bf : SUCCESS in 24m 27s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/15e78a9fef0a49e993b2cf5115b332c2 : SUCCESS in 55m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/df0b0e72fd45473aa7f34c5e98324ced : SUCCESS in 1h 29m 08s","accounts_in_message":[],"_revision_number":2},{"id":"a645f281c9ff637bf8c4a9ff9690fd5a2e3880db","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 13:31:13.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0e914ab6020d54dc476d943e4e2160bd749add7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 13:31:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8a34effe194cdec01d9925d4b0c56b70d0298805","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 14:28:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed829e216ef34751b7e0634f98b2c05c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/df76007399c849c6a9a480b629444fd4 : TIMED_OUT in 3h 01m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8a026d1f3844423bbc09ecce5300b09b : TIMED_OUT in 3h 01m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/701d950f2f564ccd840907bad838c8a8 : SUCCESS in 1h 30m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3b7c2c516ea431da44b04832bd157bc : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6549a19af07e419d896d03b2ef219433 : SUCCESS in 4m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/f17b5e603c8742158a9e41dbba04aec3 : SUCCESS in 2h 26m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2de565d8fb0342c38e298629da5ecc27 : TIMED_OUT in 3h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8055dd7bbcbf9c1e05e96ad6bdd3c083fbad5c1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:06:36.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aef405c89e4c4ec9a3a8be92f500bbd2\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e7297349581048708c044b3a712efcf2 : SUCCESS in 17m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b3556555b4ca4353999779538432b752 : SUCCESS in 33m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fc3d6aa60e3c42b0945807f3017783a0 : SUCCESS in 36m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fd4e979cdde14bc7a3a2cc2037048afc : SUCCESS in 1h 22m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d77d609b896b43279b47d7968e653524 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffd814a3db414594bd33d5e150536d97 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4fea8e551abd42f7811d2804bbad077e : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585a8015fd9f40f28b8bcf69d911aac6 : SUCCESS in 2m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/31f17955c85a4b35bedc71980df3518f : SUCCESS in 3m 01s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f26a8f8dd92c4056ba0b2cbee4373c89 : SUCCESS in 19m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5317fff66c334cc79c26fd53e1d51d5c : SUCCESS in 27m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6aac87520a744f97836b673fff863261 : SUCCESS in 1h 08m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2e1c42d0a34d40faa063a1993d298d45 : FAILURE in 4m 46s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5cdc88318ecd4cdeb0176a456171706f : SUCCESS in 19m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/01462980ae2e47508f50c7df3b768fb2 : SUCCESS in 24m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/02db8e65db5d4115afdf38106cf97c22 : SUCCESS in 59m 37s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f008525c624141d2a9ad2635bbcd69c0 : SUCCESS in 48m 57s","accounts_in_message":[],"_revision_number":2},{"id":"bb3a80f070b0b5ef39993380d40e8fc20ce50463","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 15:16:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"077972e85d20b39ed264b514d4044d8e97136260","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:48:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/21c586fc3ac8428ca470eea1f9679bed\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c11683b1c7df46dcaeda1d98cc3ff0df : SUCCESS in 13m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/82c6a89de88c4131ae87dc916a16efac : FAILURE in 2m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a1e079bc81254d87afe3d3e6cf480961 : SUCCESS in 13m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/06289d6f42da48b2b5e1bc6bd6aa0c06 : SUCCESS in 24m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/87683dc552b643c48b581de4b0462c19 : FAILURE in 3m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/87d6d66667e7434e8921e77a55b10166 : FAILURE in 2m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7366cfe44412402485ff96348852eead : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f0612bae3c14275a1acf69f07e8b1bb : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d8373884b8a4436a94be8d3a6c5458f : SUCCESS in 1m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26f7ea2c41cd48ae97dcd81f369ea25a : SUCCESS in 2m 18s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7a4f9af6a654dcc8c02612cb5815b39 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dd63bc00c7e481f8d57873c8032c26a : SUCCESS in 2m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fb6722e8a96848c69fe465e24bb3079f : SUCCESS in 1m 59s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a18efb99d71f48df9d9c51d948ea74f2 : SUCCESS in 23m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/368cb5c376124894a72ae26c47866a1c : SUCCESS in 21m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/dfbdd49a83594ab1adde850baa319e1c : FAILURE in 3m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7225694637bb47b88685d648f6cd49cf : FAILURE in 2m 48s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bf819e38e47548b7a6354d837456c39d : SUCCESS in 25m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5bc24b5720564a7ba5e2573d58f96372 : SUCCESS in 20m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ec06c8ef5d0a40638b7cce4bf3c2b591 : FAILURE in 3m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/50914bbf57184cd99e3354e12ba155f6 : FAILURE in 2m 15s","accounts_in_message":[],"_revision_number":2},{"id":"0cc26ef67b0825d290301891c0a89febf31cd232","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 18:19:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d51b5c0c0b41048c800a42d4d198f3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/50c808fe3a354c219e5732a9e985dde9 : SUCCESS in 1h 57m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1af4f7dd20fc4d9ea33cc671221b29bf : SUCCESS in 1h 38m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2446c6b3be884177a4d5a9bcf87203c6 : FAILURE in 4m 10s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e9a34640cd754feab0261ba932e13945 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/872e87a9850e47bdad531a7c2e95a605 : SUCCESS in 3m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/87076b3de9a449c1962b82fda75084e8 : SUCCESS in 1h 44m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5669d49fa00f4eb98e9f80cfcfd88940 : TIMED_OUT in 3h 01m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c3938c49c4db71ae854363963fd5087aedfbd986","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 20:21:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c98834822a1cb7cdd30b092c3d0e827cbd1005b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 21:56:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/382f755a85ab464e825e46ad78baae6d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/416d8166bdff46b083f1a7d9221f56f7 : RETRY_LIMIT in 1m 27s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/168c370dedcf4731919801ead1e78d7a : RETRY_LIMIT in 1m 24s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b04b34f8712a476c925d76f8803b117b : SUCCESS in 20m 20s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/23859ba8ad984fbfabd4114caec73bd5 : SUCCESS in 21m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/497a1ffcacfa437d969dce7c5f2a89e1 : SUCCESS in 1h 06m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/85046e262032455aa47b229cad2de969 : SUCCESS in 1h 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/437f5d2b50cf41a388949760c5cfbcee : SUCCESS in 2m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e006734a64d84760a495f329bc36033f : SUCCESS in 2m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b18d647f59cd4d43aa0c7bd58c5897e5 : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d07f49578c4f4573879de81245280cad : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43d5f09a06884aa69749c9554d3604e4 : SUCCESS in 2m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d594f15fbd914932864cabe852e18cd8 : SUCCESS in 4m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0a299ae4adf64f96a7c23dba43cca104 : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4ba3ae8e39f74715a48c3dedb1739a64 : SUCCESS in 13m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/70e703e99fe746dea75330e40106e50e : SUCCESS in 25m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0db54a0ee90540bc9dff1f330339594b : SUCCESS in 1h 05m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/86bd0f62ecde484396f11c96bb5f64a7 : SUCCESS in 1h 29m 50s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ce42619fd1634ce3a9fe4a3fffdfbe9b : SUCCESS in 16m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c494d29004d04417a24c9d79c2ab87b7 : SUCCESS in 27m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dbd99551210c41718d26e06b1e1870ca : SUCCESS in 31m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6e32de18b51542008cb8fb747eab7f09 : SUCCESS in 1h 19m 40s","accounts_in_message":[],"_revision_number":2},{"id":"97b8dfb3f49ae9d030c06330fd78c03c3ed282cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 21:56:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5834b96a47b5258aaa939e9daa93942fa89f720a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:23:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/856aac1620654c52a044856f8094a59f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2524449af893451dac100e8573cff004 : RETRY_LIMIT in 1m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bc3880e324dd44fea4c603ac908b1567 : FAILURE in 2h 13m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/324f0c0520ff422ebc83e9277749ce70 : SUCCESS in 1h 09m 25s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/190d0da627444024874ce61efb130a44 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7c21361ca33843cc93493c3bcf798be5 : SUCCESS in 3m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0d0de2043981435f986bfb43d7ebd61b : SUCCESS in 1h 27m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ac2567ece0443e99b88b780c732df95 : TIMED_OUT in 3h 01m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b072cc11977c23d82580a86999dd66d94e179a31","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:28:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbfd4f5f40814e97be3fd039ba273787\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9424e36c4f634bbbb92934445f5de889 : SUCCESS in 18m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b9fb2103c8424a1eb01ceecf84cd56a9 : SUCCESS in 28m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2d453bebe2d14a7cba2f1aefdab0c97b : SUCCESS in 59m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d1848b98fc854d87b75559a14d9d722d : SUCCESS in 1h 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b85e30f1844ae7a63f36f7ca140d4f : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46a07d9072d84d369eb1bff4423075e9 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3d2d20ecc698408cb7ba495bc48176c5 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39ca41a9a6747bd801d170da1d444bd : SUCCESS in 4m 36s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/785bdbbfd91f4994b3a970309fb53fb9 : SUCCESS in 3m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/cbd3931f180745ecb468f5f7889fd6b3 : SUCCESS in 20m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fb477ffdf9dd4df3b3e69432035afa68 : SUCCESS in 24m 56s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6ebc0c2fa8374361b6979adf25e0b21c : SUCCESS in 1h 07m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/21b399dc72884c51bd74c38093e91f5b : SUCCESS in 1h 26m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b7850cc97563413caddc1ce36d7a0c21 : SUCCESS in 11m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6485ad62c886444389bd65a01908646d : SUCCESS in 15m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e1e3d7a6f3614d3ca03beb8ac760a89f : SUCCESS in 1h 06m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fbee1305d1c34f5e9d922a32921ea954 : SUCCESS in 1h 04m 13s","accounts_in_message":[],"_revision_number":2},{"id":"f1182e4f253e42e5e29dc6373b9f3fd8ed5b6668","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:28:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"be1d6149dd2c1ff56cee4871d229f47b245539d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:31:10.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e08f27ffa3b74d7193128be565ec309d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c819c6bd39af4ec5821d2971e816eca6 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"54208d858d4f7e740e399a1d4c4f58fc6ef648b2","revisions":{"6e7c36aaef1347b8fdf9ec0e05edf5ebdbb8928e":{"kind":"REWORK","_number":1,"created":"2025-03-24 11:13:55.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/00/945300/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/945300/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/945300/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/945300/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/945300/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/945300/1"}}},"commit":{"parents":[{"commit":"fa7b8906110d5c5199930fd926f3e581bede39a9","subject":"Merge \"Adds missing nvme package in cinder-backup\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa7b8906110d5c5199930fd926f3e581bede39a9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:13:55.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nRelated-Bug: #2103810\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n(cherry picked from commit 9c6f6ae2d9b1d412bde2821c48ff1f414066f566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e7c36aaef1347b8fdf9ec0e05edf5ebdbb8928e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6e7c36aaef1347b8fdf9ec0e05edf5ebdbb8928e"}]},"branch":"refs/heads/stable/2024.2"},"54208d858d4f7e740e399a1d4c4f58fc6ef648b2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-24 11:22:31.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/00/945300/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/00/945300/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/945300/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/945300/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/00/945300/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/00/945300/2"}}},"commit":{"parents":[{"commit":"c1ce31adf79e648b5d960868f000186eed5cffe5","subject":"fix sources.py based cross-arch builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c1ce31adf79e648b5d960868f000186eed5cffe5"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:22:31.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nRelated-Bug: #2103810\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n(cherry picked from commit 9c6f6ae2d9b1d412bde2821c48ff1f414066f566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54208d858d4f7e740e399a1d4c4f58fc6ef648b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54208d858d4f7e740e399a1d4c4f58fc6ef648b2"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
