)]}'
{"id":"openstack%2Fkolla~945301","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-03-24 13:31:50.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05","subject":"tests: Add cross-arch test for etcd","status":"MERGED","created":"2025-03-24 11:14:13.000000000","updated":"2025-03-24 15:20:44.000000000","submitted":"2025-03-24 15:19:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945301","cherry_pick_of_change":940442,"cherry_pick_of_patch_set":4,"meta_rev_id":"032314d0cb7b099258956e2c737c0c8deff8fca3","_number":945301,"virtual_id_number":945301,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 15:19:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-24 12:48:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2025-03-24 13:31:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-24 13:31:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 12:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-24 12:48:22.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-24 13:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f7642a3bc0b8d65d59b9df2fd667996518dab1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:14:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"2fe50415cb2d683f86d4ba1727caf35e3a100e0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:24:18.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"39b0865b8b8497305263c50a60e25010c3671228","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:38:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6230403b9bf54b8c9f98939aeb8b3881\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3b357c7ae054ae5a4f13e6d801a0c0b : SUCCESS in 56m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4ab1974cb7ee40c89345a10140e1a251 : SUCCESS in 1h 09m 47s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e704f5145d65474c890125ed3f907fc9 : SUCCESS in 3m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/160dd8859e4c4bef883b532e11a2b47c : SUCCESS in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3a9295e0347821b3e1c275255a4ab107848b1794","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-24 12:48:22.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a4a1d26ab82ba9abca6ddf10584ee48ebd578fe3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 13:12:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05837fa2703e4711a9ba17f9d3fb9362\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e8d2f82f22b34c1596268bb878d1c3db : SUCCESS in 16m 13s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0b0d6cc4bb264e1b98aee3465c8c7d77 : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f30891522a80427db8cae40fe628d158 : SUCCESS in 15m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cf9118b7a7be43ce91569f4fde93ab13 : SUCCESS in 29m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/00f63f603513465b971350177b3b8d02 : SUCCESS in 50m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/09cc97d5277f40ea9f0f245a732c6b79 : SUCCESS in 1h 42m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14db8004eb0442feb0434b558cf2fb1d : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37b2b1ed8b92413fa8c3faac724e516f : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d614c4144a1b4444b86ec6ba78f002cd : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e6fb8cd58824f49ab3d7e6745e5c4c1 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03f916284e8047ca9db361a2181ffd19 : SUCCESS in 5m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/84d6450f2a5243e19c013edf2b9d5fb2 : SUCCESS in 2m 13s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/18e6cd305ca942df94508d9260dea9d3 : SUCCESS in 22m 31s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b9caf69dbbed4920ae28dd3f2779220f : SUCCESS in 26m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d957a33b8e5a41d7b8d2150b1625d703 : SUCCESS in 51m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c267dc91d46f4900923944859fba5cb0 : SUCCESS in 1h 26m 43s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ec6a879cc9e24eee8b4f4d5a93506754 : SUCCESS in 19m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e9e237957a3e439eb5c2c50107eb83f2 : SUCCESS in 1h 01m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/815724deaa9d49efb21d8a3254fde5c8 : SUCCESS in 45m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/194fb0a47242470f9d37899810287ef4 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":2},{"id":"756771f44909f2a103e01982034afa5c2451257c","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 13:31:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aeca216bb2fde98df66d8c370fba99e0b0ccef02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 13:32:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2bfbdb27b7022745e2a3d43d6233940f307fa7aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:19:44.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bd8e58e9e0f480987b74758a5bedeea\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/73e7946eee0b455ca67433ca6cacf93a : SUCCESS in 19m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bfeb1fde3612449fb03bc5bfc07f8dad : SUCCESS in 30m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a2d532555eb947cbb6742e9a427e6ec0 : SUCCESS in 1h 08m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8e2ecdde1e3549caaed88c1b5e64e1cf : SUCCESS in 1h 37m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce64425d7a96414fb48996b6d922dec7 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e33ac42882794ec6b099199a02c2c2d5 : SUCCESS in 3m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c6dd0c60f6943758133a6deb8129b26 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/579243cbbf8e4cffaaad57b780380445 : SUCCESS in 6m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/80da2f7e9e1343d29fec3f5bef706175 : SUCCESS in 3m 19s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/55fa81d8455241d79b3d2891dcc096fd : SUCCESS in 19m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/92a84bcb76874f8aa5fce032059df009 : SUCCESS in 31m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76d2db0d01d04127a9f4d0faee2ab469 : SUCCESS in 29m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5776dd5d2ff24f92a7a2f84fc10ce682 : SUCCESS in 1h 24m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/70e32010fbae4f3d8dc1cee42bf5d340 : SUCCESS in 16m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6481054182d84a63beab6ecf3e5e51dc : SUCCESS in 40m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/952f096b752946f4b69eb467779ee0b9 : SUCCESS in 52m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/44eb1a12484346ecbf3cb2b2535f2b3d : SUCCESS in 1h 19m 16s","accounts_in_message":[],"_revision_number":2},{"id":"dfdbddde0ca5bc9846e288b83e363dd17cdec330","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:19:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"032314d0cb7b099258956e2c737c0c8deff8fca3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:20:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9dc89d5ed9945f688a7fcdd0b2d72b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/643d06c634f342ada1a6ca9a1809f073 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ef6688a994319024cc4e9e68546e6e2753ffd43","revisions":{"8ddc79e06a0e35b08c65943f43f625f0b7090aca":{"kind":"REWORK","_number":1,"created":"2025-03-24 11:14:13.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/01/945301/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/945301/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/945301/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/945301/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/945301/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/945301/1"}}},"commit":{"parents":[{"commit":"34d537a90ed3e3828178cf5dfb29d67ff0eb58bd","subject":"Merge \"Adds missing nvme package in cinder-backup\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/34d537a90ed3e3828178cf5dfb29d67ff0eb58bd"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:14:13.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nRelated-Bug: #2103810\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n(cherry picked from commit 9c6f6ae2d9b1d412bde2821c48ff1f414066f566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ddc79e06a0e35b08c65943f43f625f0b7090aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ddc79e06a0e35b08c65943f43f625f0b7090aca"}]},"branch":"refs/heads/stable/2024.1"},"9ef6688a994319024cc4e9e68546e6e2753ffd43":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-24 11:24:18.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/01/945301/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/01/945301/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/945301/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/945301/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/01/945301/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/01/945301/2"}}},"commit":{"parents":[{"commit":"6053fefb34e5c7ad51b90ea8f6d9e428e0a944e8","subject":"fix sources.py based cross-arch builds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6053fefb34e5c7ad51b90ea8f6d9e428e0a944e8"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-01-30 12:30:06.000000000","tz":0},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:24:18.000000000","tz":0},"subject":"tests: Add cross-arch test for etcd","message":"tests: Add cross-arch test for etcd\n\nEnsure that when building etcd with --base-arch aarch64, Kolla\nuses the arm64 tarball instead of amd64.\n\nRelated-Bug: #2103810\n\nChange-Id: I6248cb6646b2b0c7af1c29f9a0a244cef1c4cf05\n(cherry picked from commit 9c6f6ae2d9b1d412bde2821c48ff1f414066f566)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ef6688a994319024cc4e9e68546e6e2753ffd43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9ef6688a994319024cc4e9e68546e6e2753ffd43"}]},"branch":"refs/heads/stable/2024.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
