)]}'
{"id":"openstack%2Fkolla~945303","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I18cea367e94ff2433346c1b388d606e41b1831ad","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-03-24 15:18:12.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"I18cea367e94ff2433346c1b388d606e41b1831ad","subject":"CI: Switch aarch64 builds to native platform","status":"MERGED","created":"2025-03-24 11:16:53.000000000","updated":"2025-03-24 23:32:07.000000000","submitted":"2025-03-24 23:28:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945303","cherry_pick_of_change":945292,"cherry_pick_of_patch_set":2,"meta_rev_id":"d264aa87874d2b5db48e83aa7571809a81d4b53f","_number":945303,"virtual_id_number":945303,"owner":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-24 23:28:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":2,"date":"2025-03-24 13:31:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":1,"date":"2025-03-24 13:31:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-24 11:38:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"834739e0d6fc8393b37478fe2c48831f93b93a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:16:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3f3bec8a2f1b30b51dd94ae7741a34776db519ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 11:22:59.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"b94ae9a981bc5518a76411a191cc5fac38d26c51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 11:38:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/783849eaaff24d7f9bcba6b96a8e4eec\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c52b325b9f43f8942a14da1ab7aa2f : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a12f5d7494f746d2a5eae3d71fa4d103 : SUCCESS in 2m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/858563b53bf045daacaf553b682aa804 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0aeb23a834924e4e9c9c38eee0c44d51 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b3c5f07fe2e45ed87a73eeed8916383 : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0db206a1c0545b8bbb10d209d33ccc6 : SUCCESS in 6m 32s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1cd73a00c6e84f2db21face3b061f8a3 : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":2},{"id":"936acec6f2def9fd670982c61f1beba29ce3a415","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 12:23:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1532bdb14cf4a62a9b1cb034b7e8e4d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7f0ebe67e3094136a7f924cd0201f6b0 : SUCCESS in 47m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f1bee22ec5d14bf4b89ef9b369e8bcef : SUCCESS in 51m 23s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85ca7bd86ef7438fb0c82a92eb43c02a : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e462c1f692ad4307961275f9cda7752c : SUCCESS in 2m 53s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3ca373c2e5804146925f6213e9078bb2 : SUCCESS in 48m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8054bece71bf4d2298ba4da1af3d998c : FAILURE in 57m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5173f40571050dd7f9693dab0f0115e7a15cf1d4","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 13:31:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1bfa678cd53acf712a4f20793836a0a178c36701","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 13:31:59.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"14e67a65bb7f3f4c268150e9194ea4b2db319621","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:06:37.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/945300 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"bfbd0942cc7431990b80c087747a54921ad1db83","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-24 15:18:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0be5dbcf41a81a8b384640bca645bf2eb811c13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 15:27:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f980bb74c9b040a0bfa359e597115e64\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/665f2cafdbfc4e43aacf78cd43000d89 : SUCCESS in 2m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46580e6e74dd4e21908bc4fd7f94bfe7 : SUCCESS in 2m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7c8c88308f34870813bca5d58eb80a9 : SUCCESS in 2m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/827b9118ad01407b8324869b38d300d3 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/073fa42b9b654d1ab0fa6db7eb927df6 : SUCCESS in 2m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8461e444930440d58d3c32452febdd24 : SUCCESS in 4m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/7bd25cbf87c54477a47f5eafbaf0bf41 : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":2},{"id":"99532f58e0004c1d97255bcfbe3ba1611124c441","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 16:11:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c719d92c4ef4403c85cabc41edc4deb5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/85a62df90cc1466f95bab384c4f70249 : SUCCESS in 51m 08s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8929b37d439a44ea8d05ac6c86c9d402 : SUCCESS in 48m 46s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0319319335cd48da8441bc86a764a27c : SUCCESS in 3m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/487152adb01b4e4bac3d200a7b9ac34c : SUCCESS in 3m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/95d9a91bab6044b3870e8660d474f349 : SUCCESS in 40m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/85e5a2860b9e4d76816c9013b3700703 : FAILURE in 45m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a6361d5577e754dac2f49aa1cf079c766f49de7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 21:56:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"575edb43b1c1a52c637d61984f72de4fd1eae993","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:28:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25b9f8f1ca2c48489192f9990f908464\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016ff54e626f477aa6169c5cb6df044b : SUCCESS in 1m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b567e922bf9d48e6a988a522a264663f : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7b56770fd5d4dc18c47505d431db6f7 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80012edde2bb4bdd8369ae4faac5f50a : SUCCESS in 4m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/548447ac54ca478a930272a3bf130dc0 : SUCCESS in 2m 24s","accounts_in_message":[],"_revision_number":2},{"id":"75f1e36d363a9c3d6551f51cc8c646ac29abdfb4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:28:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d264aa87874d2b5db48e83aa7571809a81d4b53f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-24 23:32:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e545ba6437849c194a551f0e48e5b93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccd097e3581648b89fc6da1df0d8f6ff : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8577c6cd16b5f1a7e559424f9c9ee85e66b748c6","revisions":{"900a8d2a83fbf459ef2157d2d4a4dbe00a37f23d":{"kind":"REWORK","_number":1,"created":"2025-03-24 11:16:53.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/03/945303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/945303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/945303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/945303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/945303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/945303/1"}}},"commit":{"parents":[{"commit":"fa7b8906110d5c5199930fd926f3e581bede39a9","subject":"Merge \"Adds missing nvme package in cinder-backup\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fa7b8906110d5c5199930fd926f3e581bede39a9"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 08:18:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:16:53.000000000","tz":0},"subject":"CI: Switch aarch64 builds to native platform","message":"CI: Switch aarch64 builds to native platform\n\nChange-Id: I18cea367e94ff2433346c1b388d606e41b1831ad\n(cherry picked from commit dc0fbd20a1b7e0bb12fd2e14a2f27cd283b61967)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/900a8d2a83fbf459ef2157d2d4a4dbe00a37f23d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/900a8d2a83fbf459ef2157d2d4a4dbe00a37f23d"}]},"branch":"refs/heads/stable/2024.2"},"8577c6cd16b5f1a7e559424f9c9ee85e66b748c6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-24 11:22:59.000000000","uploader":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"ref":"refs/changes/03/945303/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/945303/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/945303/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/945303/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/945303/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/945303/2"}}},"commit":{"parents":[{"commit":"54208d858d4f7e740e399a1d4c4f58fc6ef648b2","subject":"tests: Add cross-arch test for etcd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/54208d858d4f7e740e399a1d4c4f58fc6ef648b2"}]}],"author":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 08:18:30.000000000","tz":60},"committer":{"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","date":"2025-03-24 11:22:59.000000000","tz":0},"subject":"CI: Switch aarch64 builds to native platform","message":"CI: Switch aarch64 builds to native platform\n\nChange-Id: I18cea367e94ff2433346c1b388d606e41b1831ad\n(cherry picked from commit dc0fbd20a1b7e0bb12fd2e14a2f27cd283b61967)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8577c6cd16b5f1a7e559424f9c9ee85e66b748c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8577c6cd16b5f1a7e559424f9c9ee85e66b748c6"}]},"branch":"refs/heads/stable/2024.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
