)]}'
{"id":"openstack%2Fkolla~945537","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I7d2367d4c277287f4db2885f5720a35d934f94b7","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-03-31 13:28:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7d2367d4c277287f4db2885f5720a35d934f94b7","subject":"Opensearch and opensearch-dashboard image size reduction","status":"MERGED","created":"2025-03-26 06:21:17.000000000","updated":"2025-03-31 13:29:01.000000000","submitted":"2025-03-31 13:28:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945537","cherry_pick_of_change":941384,"cherry_pick_of_patch_set":2,"meta_rev_id":"f88d285c38aab07281a0e82e80ea4d40ee9a17bf","_number":945537,"virtual_id_number":945537,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-31 13:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-03-26 08:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-03-26 15:47:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-03-30 21:56:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 06:21:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"state":"CC"},{"updated":"2025-03-26 07:38:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 07:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-26 15:47:45.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-03-31 10:18:18.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-31 11:58:02.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"}],"messages":[{"id":"142d54cb45d9e12025076def4904f4562697de66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-26 06:21:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"79e6193f5069a0721a1a4cbdba705e357d618092","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 07:38:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59014e6be43143029c4719619344955f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/50546a1b0d3e4da1855406014ca01027 : SUCCESS in 35m 39s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/514307625b414240875f2b6067fe6ea8 : SUCCESS in 30m 53s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/121fe5ec01d24fb29c8b2a0f586bd2b0 : SUCCESS in 1h 15m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/04bb688e53cf4ff98beccd86987353e0 : SUCCESS in 3m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bab7ea39bcbd45a5a3f043fdf06a1456 : SUCCESS in 3m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5f26b6a866f1487198917eb0679204c9 : SUCCESS in 39m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6ab21a74c76c49ea8472b2c5fd4ee545 : FAILURE in 38m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd0b116c3a35dc95de42eae30bf08cc3db0cd7eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 07:56:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88e9e58c3844aa8b2c95157946cee0c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a0c99fc907874be5b2396518d8a4c4bb : SUCCESS in 11m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e711affde27143ae99232dad2264998a : SUCCESS in 1h 11m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8902bb1ef9324ee2a33ce8ebb1027b58 : SUCCESS in 19m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/39c10e2f90ff426b956a5cce716dcb6b : SUCCESS in 23m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5f928efd666d48a383cd02d92377e809 : SUCCESS in 34m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fdd0e8ac6e1f4c8ba71cf6bf52e3a709 : SUCCESS in 52m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0049c73318dd4e759ba654b7fe3c568f : SUCCESS in 1m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a84a8182769548d1ab27c0cf02525245 : SUCCESS in 3m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edbc0fe676df46e3a6a24fbccbfc28ed : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f40c178ea0747a282f2e19bbbaa8dc4 : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/545d792f80b04d7286d6733bc4d02bf9 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95690ed1502e41e69a1d2caae2d71fb7 : SUCCESS in 5m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/679ce3a95b314880b3f247fe37cd6297 : SUCCESS in 3m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/700253904f2b4161ba2bff5aa7dc6630 : SUCCESS in 20m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ccff4527ae0d467999163214ec444803 : SUCCESS in 23m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d9d3d9aa922448e98e463eb1a43466a6 : SUCCESS in 50m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/bbc82472849e4d4092d03a4cb7bcd2c2 : SUCCESS in 1h 29m 53s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/99b7f6ae60b4408799ed4894460ef950 : SUCCESS in 23m 25s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4909727b2fac4058ac91ef90518bc480 : SUCCESS in 19m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a4d3b2539875400bb2571997d66c73ce : SUCCESS in 13m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1799b66aba8c451493ebe04ba45f853d : SUCCESS in 1h 09m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2c96da09109b4a21b29bb8588d5f80fd : SUCCESS in 1h 13m 28s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/714ea2f09a954576a58a084c92718c82 : SUCCESS in 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"df5bef05aaf1bd56a48bd25668a6c163fb36cf96","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-26 08:28:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e361a83f33dd246cca514e21929bd35c4e2617ee","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-26 15:47:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8b73c7196371a153aa06be1f49c11b420dfa347d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-30 21:56:03.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9ae7232bb027963e01bb82bc6ccf95ecb29b8328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-30 21:56:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f41e4e4cabdf000fdcc31fb699e5d94bca64c58a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-30 23:35:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9454828b02d84058a6a8ac034df81a78\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8f61527aba394301b016a60a226a282e : SUCCESS in 13m 50s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cd999600e52548c0a80a954187733332 : SUCCESS in 31m 43s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23bedf93f1214805af8fb3d5f86fb6ad : SUCCESS in 52m 16s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0540362b6824baaa47d828565131cb0 : SUCCESS in 1h 08m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/189913e893fd490ebf07c1c95498bcc0 : SUCCESS in 1m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d9715d166df4c828c4e696aacf54869 : SUCCESS in 1m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9258e13dea5c4c46a6e3ddef2c4d8cf0 : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de822633ac4b404ba9a54495b6176892 : SUCCESS in 5m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e4b9acc6bb6f4f118c4115019e744133 : SUCCESS in 2m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6e2b3028d9404f728f5dae2f07ec5f33 : SUCCESS in 28m 22s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1ef0b68088d24b22b228736f8a8d4ecb : SUCCESS in 31m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f2c0690aa7b2497eb08694c7bd43016a : POST_FAILURE in 29m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d71d5dccd13d480899a24d9fa3f0bc4f : SUCCESS in 1h 21m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/432fb2b00b214f7a9ce6f3da3a7e6ede : SUCCESS in 29m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ceb1196de37749b6842b87b392bde66e : SUCCESS in 33m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/92b1aa537459405a9f59a88902062722 : SUCCESS in 55m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7d0823d945ac4f1199d4690fa854cead : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"0cca4cc4f24c53c5e325ea24785c9e8603b9abb0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-31 10:18:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1dc592745e2435a4b13bdaa4a75a9b7ac792291","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 11:46:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e78070ae34a4d6baa3d014153ac57a0\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/840f68e7d329425caf1c17aadeb3f612 : SUCCESS in 37m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0c0bbd3934914a3e9648e38077957ee2 : SUCCESS in 37m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f9df5ddc2385436b978f1799337f648e : SUCCESS in 1h 26m 11s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4fa9f43c5cd945fb94f96f4751326a74 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8bc6492e6a104b5fb2b9349ee4abc216 : SUCCESS in 3m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/403d02303d1c40bebc25c681a6a065ff : SUCCESS in 28m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d179260a84954a9a95008e8e2340a4c1 : FAILURE in 32m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"85a9797869fa4a2abf2bcf68c2f54929d9cf9b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 12:00:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd989e3e9764069985971b7f71face7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/84717920b10341cd8f1206e000fd84ad : SUCCESS in 21m 16s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1cbcdf69a7854da49a533edc69a8f094 : SUCCESS in 56m 01s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f503dd8f9b0744c9b2c82d3dd3e77c6d : SUCCESS in 21m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c2e0fcf30a264054a7a7d2b96ae64d66 : SUCCESS in 22m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f65252cc14614449bc291489ba09aa1f : SUCCESS in 1h 08m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d4a5c906b4944302aa992eccaeecdf35 : SUCCESS in 1h 40m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c49a605820a4ee8b8989d940b48930d : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/311d5462734e4e4496221e3dc532caeb : SUCCESS in 3m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac5e469992cc4678a0f4f64cc69260ad : SUCCESS in 2m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/334ebbe8daaf4d749a5c76dfb43ae60e : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d21ba86461f64d5a89b7798cc9b040eb : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eee0b3bc660442b395e9c885196a376f : SUCCESS in 3m 59s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f9ce4f4ead7b4f6cb12ad43e5f7b799e : SUCCESS in 2m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f08cade19783469f8530a64b3ea3d99f : SUCCESS in 21m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/028bdc04d683447db18b3f3f42c2ba3b : SUCCESS in 16m 46s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b2ad7f0073dd434ba10036c0854ff4d5 : SUCCESS in 30m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f1e3aa1daff74bca9e5ab059815a4095 : SUCCESS in 1h 28m 13s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3742098aa04a404b9e6f4daffbf80ddb : SUCCESS in 30m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/076f324021944ad8a76d6cad4f0d8233 : SUCCESS in 21m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a8008e9037d94d3a9255cb038b275f60 : SUCCESS in 28m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6caf1a093464442c9056eec074a7dc6d : SUCCESS in 36m 57s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8f81ed6d7b1e4335aa808dacb3e0ae09 : SUCCESS in 41m 46s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5ff86c998f814537b9e45bda58e2b327 : SUCCESS in 26m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ee9ecbb0b83f5525de669cab11e2eb799604d8b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 12:00:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e6e62820f9e4e5c63a5d16327752b716839c5db1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:28:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1f5dbe7e784485baf5f5791c2331cc6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f1484580db1948d0be8c851d198bbb5e : SUCCESS in 15m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e940946d8cac4aa7ad38dafaed756f00 : SUCCESS in 22m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cae79962b57a44aea9fd5beb8d3de560 : SUCCESS in 29m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e786c8cc520c43e3b18ede283a94fe52 : SUCCESS in 45m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8276285f78ce4d9a9f62643db35645c1 : SUCCESS in 1m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/503c9669cdaa42ad94102184f65cae16 : SUCCESS in 3m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f7af3585d7f4e058ae252de89b76e70 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f42cbd5c6784061a5f4dfea71de3c68 : SUCCESS in 6m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b99e913835754a409104c9ab7d5ab0b1 : SUCCESS in 2m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/41e908b232384f748a662c89b5d21afb : SUCCESS in 18m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1d197e94a8d94da6a956cd50babc8d48 : SUCCESS in 23m 24s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/88ee6046e0b74d8ca8b05eb33164c807 : SUCCESS in 56m 47s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/dee5887a45984699bc44da93d47b2c5a : SUCCESS in 1h 22m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a6bf11aa81e9480e94e2ddb1801fe926 : SUCCESS in 16m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8a45a2de6c9c409ba55b4a8207b5ec7d : SUCCESS in 29m 18s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/86391ce9e8eb4962a09c347be8c87dfc : SUCCESS in 53m 46s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/10d100cc03fe40b78d8f4eb304c59a76 : SUCCESS in 1h 15m 20s","accounts_in_message":[],"_revision_number":1},{"id":"143b56173cd5afdeb37a9c0f6c20e983fc6aab45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:28:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f88d285c38aab07281a0e82e80ea4d40ee9a17bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 13:29:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dac973f88b024a4aaf15063b5ea5afdd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97ae2feca31f4bbe8cfa829227f66bc1 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9c24c0decebb57a48129cdcdeb32fe3393c0210","revisions":{"b9c24c0decebb57a48129cdcdeb32fe3393c0210":{"kind":"REWORK","_number":1,"created":"2025-03-26 06:21:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/37/945537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/37/945537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/945537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/945537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/37/945537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/37/945537/1"}}},"commit":{"parents":[{"commit":"8577c6cd16b5f1a7e559424f9c9ee85e66b748c6","subject":"CI: Switch aarch64 builds to native platform","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8577c6cd16b5f1a7e559424f9c9ee85e66b748c6"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-02-12 12:53:14.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-03-26 06:21:17.000000000","tz":0},"subject":"Opensearch and opensearch-dashboard image size reduction","message":"Opensearch and opensearch-dashboard image size reduction\n\nWhen building an image, the RUN command grants\npermissions which causes the previous layer\nto be copied.\n\nDuring the test, the image was reduced from 2.84GB to 1.57GB\n\nCloses-Bug: #2104007\n\nChange-Id: I7d2367d4c277287f4db2885f5720a35d934f94b7\n(cherry picked from commit 47a6ad020d7a52b14106938e03e27d65280999cb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9c24c0decebb57a48129cdcdeb32fe3393c0210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9c24c0decebb57a48129cdcdeb32fe3393c0210"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
