)]}'
{"id":"openstack%2Fkolla~945604","triplet_id":"openstack%2Fkolla~master~I4113976db743d3c5c4ff62f2876e224888f05a36","project":"openstack/kolla","branch":"master","topic":"prometheus-v3","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-04-24 08:36:06.000000000","reason":"Change was submitted"},"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2025-04-24 08:36:06.000000000","reason":"Change was submitted"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-24 06:42:55.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-04-24 06:44:29.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-22 17:44:46.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-04-24 08:36:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4113976db743d3c5c4ff62f2876e224888f05a36","subject":"Upgrade Prometheus to v3","status":"MERGED","created":"2025-03-26 14:35:18.000000000","updated":"2025-04-24 08:36:58.000000000","submitted":"2025-04-24 08:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945604-prometheus-v3","meta_rev_id":"8c3f17313fa889d4a6bf4355aa1737ac55f2c468","_number":945604,"virtual_id_number":945604,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-24 08:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-04-24 06:42:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-04-22 16:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-04-22 17:44:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-04-24 06:44:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2025-04-22 16:32:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-04-22 16:32:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2025-04-22 16:32:43.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-26 14:35:49.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-03-26 16:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-26 17:40:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 06:41:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-03-29 21:40:21.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-04-01 06:26:07.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-04-11 09:30:27.000000000","updated_by":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"CC"},{"updated":"2025-04-24 06:42:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"6586729d1ce4fb5cfedf7cbac1d2eac1d898cf45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-03-26 14:35:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2bc27c3d63b7f75575c3f99b0b025fc7dbceb363","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-26 14:35:49.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1fd239ce63376fdc9dab8eb8fb8dd05f86291707","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 16:36:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9e01173c5214aff987d007374fedfa9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/27bd50ea432647c0a95853427a633951 : SUCCESS in 46m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5b8d9a1b0d874194b70684f9984670e4 : SUCCESS in 37m 46s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/240d44c4090e44acb9d95749245099db : SUCCESS in 1h 03m 54s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a52be731001452285d77be605e0fb9d : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a89af6aa29074728b3dd8db0f281b09d : SUCCESS in 3m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a34e1a84d1344fff8fb57fa30a84c7d5 : SUCCESS in 41m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/cd50cece31db4e269a4ac7d78402605f : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7c18b93f2a38d6176d1ea0d993e86d5938eb0d56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 17:40:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2bc439e8f5204f0488e665354c9e7812\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e98ef188fade401a922e9ab6f27776e9 : SUCCESS in 1h 27m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/10f3a0649de44e7696a1bb8521ebb63f : SUCCESS in 2h 10m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0e775e07ab954f248825e9e299756b58 : SUCCESS in 59m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f475c0e457cd4f5aac0261d34d281231 : SUCCESS in 1h 07m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5a1e11ccfd5b411aa2e5bb41ced005ee : SUCCESS in 1h 48m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1ca4ef80d9604d8a88ca3807c9b9987d : TIMED_OUT in 3h 01m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e124ae16a21487ebf507234a5da049d : FAILURE in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99cfc6dfbdc04f12871fa42fd8b40dea : SUCCESS in 2m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5f3a58d1cd7b41058ab8674687f2b708 : SUCCESS in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/44d843298fce426db78fbbadcc65eef3 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3255541afc0b451ea9e7c133159e7b3e : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7384912954f042a1948a3cd794cbb5e1 : SUCCESS in 5m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/88d93687bcb64629821543d887e48efa : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9bd7d269ff3e4444882730ebc3524a0e : SUCCESS in 16m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3dfa9b08e3e94795bd11eb803187f13f : SUCCESS in 1h 21m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/aa79c2d283ec47149defe36630969431 : SUCCESS in 1h 25m 02s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/70d1bbab59314824a13f837a8291b4f2 : SUCCESS in 1h 33m 51s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2eaf3b8f6bd04b78be95533a2d41202f : FAILURE in 17m 21s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6f894011c6e844adbf6f9d65646eb327 : SUCCESS in 1h 11m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/23a30f2756d447fb8ee68dd25d7b52f3 : SUCCESS in 24m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d114b27dc1464339a1a22c7c290711c3 : SUCCESS in 1h 12m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fd9904a9773f4e3697083ed066fd6d61 : SUCCESS in 1h 39m 37s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/a5d7a5b858124dd1946ae83e00008cef : FAILURE in 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c7de037a6c35a99e81077f60a7df60e97ea5bf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-03-26 19:46:05.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e9315ef8af06c66a3d3fdfedcebe231747f2a16d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-03-26 19:47:57.000000000","message":"Topic set to prometheus-v3","accounts_in_message":[],"_revision_number":2},{"id":"8325ff2a3854262941dd6cef26a348b50ec98768","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 21:19:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb392b60ea214cad8a01e18c7de1edac\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c05f7147b8134454b22d2ec92edbbcc4 : SUCCESS in 32m 40s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c06c30a023f9401c89336886a5ce5feb : SUCCESS in 28m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6b7f57d10513414bbfcd0f57cc91826c : SUCCESS in 1h 01m 18s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f98941d2c3741b6a44556c5af9e7a62 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69784f7131114dfebc43aad92b8312de : SUCCESS in 4m 02s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/24d91f4e4b634aceada03bf9ec33b680 : SUCCESS in 41m 10s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bf9a260ee01a425db8b28cc0764abf87 : SUCCESS in 36m 36s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d22800551bf44d5390757864ce46fc0bfde6cedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 21:29:03.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad2acf61270c4b48a977bab7e3a5127a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a9f1c7898d524d309426b707bd6e452f : SUCCESS in 27m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/01dc2bc5d9bb4eb89858a3e9d01fec89 : SUCCESS in 40m 53s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/206d333c70ff471eab70decf15589a2f : SUCCESS in 21m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f0201d7e394f4208b73fe725d816fb19 : SUCCESS in 23m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/08f37405de57449ca82ac245eee71430 : SUCCESS in 54m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/496d260b287d4de282f3c60beaaeccb5 : SUCCESS in 1h 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc88b65c11e461e9b3aadd790b48cef : FAILURE in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dcc967165cb4a65bb3dee7c4ed8fc0b : SUCCESS in 3m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7edf12eb97fa4337bf95fe3514f66982 : SUCCESS in 4m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a88fa2bca708463cb55854e497014101 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74e7a2c24c7543dc868b15ec31fa764b : SUCCESS in 6m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/412cd6232eb44567accce4693dde81f1 : SUCCESS in 3m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/635d48ca93f946288bd5bef06cb3f488 : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ba49711fe8c1425d810cbb2b69ee80bc : SUCCESS in 23m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2d3b35aed17446a982690bb4400f0c99 : SUCCESS in 1h 15m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/5654c2721feb4b2fb96cf123ae27fcc4 : SUCCESS in 1h 08m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e711543e53884be8ade421eae7b349cd : SUCCESS in 1h 34m 50s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/3c04cd2afe9a43cfbac73f5eeacc4260 : FAILURE in 22m 05s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b3f25a80519a4a9099b9dc7ec08c5b0b : SUCCESS in 23m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6388b5c06bdc48e68eb0a5794f2daaf5 : SUCCESS in 28m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/684860a76f7d4a05aed09274c4abc96b : SUCCESS in 1h 13m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c58bae56b81a4ca283964572f401ae81 : SUCCESS in 1h 40m 12s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4aa1db993ca74c44ac7ed0d1f9cb0df0 : FAILURE in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c5c84a6269ba34f31f73287d56004fbd3d66fb95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-03-27 07:55:44.000000000","message":"Uploaded patch set 3.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"31edb7124b952f518fb802d88dd2819904ec37df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 09:00:35.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9698e024161241199cde8202f0215ba1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7b554ee8ce7f4a2abd19267d697f0b9d : SUCCESS in 36m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1a5c1598212b447c9b7a223d8cde64c8 : SUCCESS in 35m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/02d04448dc7c49f88d8ed13e8bae7f63 : SUCCESS in 1h 03m 00s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12a71548b8ae463b934b6de563aa2f1a : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b12091be7c2d4c0e92ebbc4bca52fcd3 : SUCCESS in 3m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9959ec601b2f4a8e9a6892d759b2e100 : SUCCESS in 38m 08s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e740c45dc9e4793a81c231214bb2a35 : SUCCESS in 38m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a6884c5cd5bdf3b494270a158d06968cef5bda0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-27 09:43:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bebf6284af74279a4e55226eb1253d7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/5606881190e7489a835fdac6f3bb4f2e : SUCCESS in 20m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/86ba62a806f843ef8f589d5c085c1f00 : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2b7d95854a7c435f9996e51843c78739 : SUCCESS in 22m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b541474f900d49f9b7ed0f4a4e5d5fdd : SUCCESS in 29m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a10f1f1d218b45a989966241b5aeedf7 : SUCCESS in 1h 14m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cf3ecafa5531453e81ec70cbb3ff3d67 : SUCCESS in 1h 42m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd3294a1ecec41e099b37d773da0d214 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d65db90e8a44971b39f2df5cd6514df : SUCCESS in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a43ff4dde0d4b80b7902c51c104a4f6 : SUCCESS in 3m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b7dc7b1577944aabf84bd3bd5d5fd13 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15b4acb1727741cf87b44354e916ae17 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/59502add99bf404ab43f5a807bc66dce : SUCCESS in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0a89d4641af94b0ab8ba188de49a4a75 : SUCCESS in 2m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/32f0ae3ef57b4c4a9c7a37ab0b74a09a : SUCCESS in 20m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/23173135c7b7463a8438075e53ac86e9 : SUCCESS in 1h 05m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/935b111bf32f40449620f927dcb8a5e3 : SUCCESS in 1h 05m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b51adfa5693e4224be479513ab680032 : SUCCESS in 1h 25m 15s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7cf0c5c7352c4bdeb3ecb3697a569fe0 : FAILURE in 20m 19s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1d49c4a275c3480a9abf3893df76cc8e : SUCCESS in 22m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2ff6aee01a5f4be99ff1fa9c27743ed6 : SUCCESS in 25m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c33764a732204a82a498bc5054048612 : SUCCESS in 1h 06m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/acc585f0a57048278f83229b61b47480 : SUCCESS in 1h 35m 54s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/08311af7c73148c5a8f2e46fe2ec8895 : FAILURE in 17m 39s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ace6450c8f28c41d03481eaf86c57d9cc3ad4fcb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-03-28 06:41:27.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cc6b1a39f05e2ff6e3f4756042989b192f751eb1","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-28 07:52:18.000000000","message":"Patch Set 3: Backport-Candidate-2 Code-Review+2 Review-Priority+1 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ba358be1046d825351829f07a7d2f96dde77840e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 07:52:40.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7f8ec5ca7531ea4e7188c520074c32c388f821d2","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-03-28 08:03:24.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"585b1e421df2c3358bb9a8ce0f37af2ad0dad905","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 08:03:38.000000000","message":"Patch Set 3: Verified-2\n\nThis change is unable to merge due to a missing merge requirement.","accounts_in_message":[],"_revision_number":3},{"id":"a2f11534fbc78e477837834d1f1704bef1ab8db2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-03-29 21:40:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1c10d8a96e60d4fe819117f015bc43a181a64f5b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-01 06:26:07.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01420e7ce1956f2e93c190ced84197470fe08f2d","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-02 13:00:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8f453e6078dc3ff8fd07088f30e1a87a31d3b50e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-11 09:28:47.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b1c485e1ce33c1313b0fea67516f3ee092aebb6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 09:29:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/07a7e12861eb4175b0b8c288b55c7921\n\n\nWarning:\n  Dependency cycle detected and project openstack/kolla-ansible doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"3f6daa46f979391fdd91811583bd0e0e387a03e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 09:29:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b69dc6daa0a42258958e25edd6e1ce1\n\n\nWarning:\n  Dependency cycle detected and project openstack/kolla-ansible doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"b1543ddf97743fe045b1d901b94fd735db730519","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-11 09:30:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"c2798259895ffbec6b791453b6f7a7e4f8d25ffd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 09:31:31.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c12b0ae39dc4ed0af16c2d914bb0298\n\n\nWarning:\n  Dependency cycle detected and project openstack/kolla-ansible doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"4bf5c66165414da2b7ba7aa88cdfe952e52042ae","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 09:31:33.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b95e68e59815475aa6e2b3ed760cf3e7\n\n\nWarning:\n  Dependency cycle detected and project openstack/kolla-ansible doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":4},{"id":"f66c58e613c76f13369992977f7586350c7e7952","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"date":"2025-04-11 09:33:33.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2bcf15582c7ea1182ea3e9d6c675fce5aa946773","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 10:54:20.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1b8b61999f54edfa13135ee68e82541\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/38753b994cf0424c843a586a8d1ff4c0 : SUCCESS in 56m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3bbf230321fe457eb36f89590534f125 : SUCCESS in 53m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/75a3285ac0d34bb9a7701adcd119c095 : SUCCESS in 1h 18m 39s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8aac2fcb29dd4039a28b64f367deb4a2 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5c950d6bffec4294abe84ccbd7b01136 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6eb4dda1b5034d4b8bea57d72c549fd1 : FAILURE in 9m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/35fabb269e3e4fe2a19dc8d9d876f6dc : SUCCESS in 49m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8c2c16e433a64a4e876206c70b4397ca : SUCCESS in 52m 22s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a705c5926fe348b8ba364751540d89e54fde5a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:20:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bd53fcc33fc416faeee24efeebd217c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a5a874e9dc064ee78d82482e0f363fa5 : SUCCESS in 21m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d2a768af2c4f43bba2f633604367ae2e : SUCCESS in 54m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fdc3c1952ce54ab7abf5dcb1db521a61 : SUCCESS in 15m 46s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/31ef9dd080dc4f089e76fff42d2dd022 : SUCCESS in 24m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/335c21884324458ba706dd7112d24ef5 : SUCCESS in 30m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3cb844de4c6e4db298906dfa5a6aa9a9 : SUCCESS in 49m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e3f874d58dd43a585980666211f286a : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/30bea62ef6934b229be379fa9bf92911 : SUCCESS in 2m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dbd4ba3104d4248818fdee0a97aff2c : SUCCESS in 1m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9065f7d3d5a245eb8f247dc1392305a1 : FAILURE in 4m 02s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b08d345ba26c488a97189f4ee39966e2 : SUCCESS in 1m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e250eb42b8a470cbce52243e5dd8614 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7782dca585c45ee88d670039faaccfa : SUCCESS in 2m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/61df890e5eb346c7a639fff8fd4b4ef0 : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/d9791287555b49eca9d460140345675b : SUCCESS in 12m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/887ce54d41d64dcf8b276ad6e61ab4f7 : SUCCESS in 1h 31m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6a562181230f4be6b89c59ece9801a8b : SUCCESS in 1h 03m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0a0c7bab3b904d2b8c2d85d7c0ccae08 : SUCCESS in 1h 37m 11s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/79f377e2d8a14030ac4cfa88027e8362 : FAILURE in 20m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/910f1a4e1e5f4364b404e6ede785de4a : SUCCESS in 21m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/14303e0f386a4383bd11a89081cb0fa1 : SUCCESS in 28m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d4c2c3036b0946f3b842da9f43c7600f : SUCCESS in 1h 07m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/13c0fbc08530412d90ddf9cc09a11d8a : SUCCESS in 1h 40m 58s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/669ab9556bc84c4e9365653b7b84ff7f : FAILURE in 17m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"34029aac2762a2108112a812b14d7e918fa867e0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-16 18:29:55.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"2437dfdff660264c426a6a1188b595a1b1e4fad4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-19 15:07:00.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"3cdf69059b73dde3fdc6dfffa944958eba2f2fd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 16:32:43.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1, Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4b440dabe35f474406e1e08fc0bbcf4001f0f509","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 16:32:48.000000000","message":"Patch Set 6: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b71cb9117738498d77ddc69490e156a90335b299","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-22 17:44:46.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2f1d688a391a73c1aa221af9edc9036e5bd9d7fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:18:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a63413307a742e4bdf8bcf13bcd08c6\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6426718c05b44ae48a9ed0d3777384c7 : SUCCESS in 31m 26s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b59360280e1b42049e071671db649169 : SUCCESS in 1h 05m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/73b923acafba431da802d5472dcc725e : SUCCESS in 15m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/49b2568025464e129e8c2e8190855175 : SUCCESS in 26m 38s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/97d1da6cffa6415fa315dbc95373c75c : SUCCESS in 48m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7adda31633014f6886f1c0e954e9b822 : SUCCESS in 1h 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14c8a4c5639c4f1282739ebfedb7b7fc : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b1fbd52fe5e4ee49ce10ddb36d62e26 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5067f43722bd4dd984abd93ec9f9b4c6 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6b17f2817f594abebeee0e3801f8ebc8 : FAILURE in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c67a6fd9ba75468fbc854201d449774a : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6eda646a857438dad0e5c87def66f78 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/87a85d17d2e64d718e2d34c428769b47 : SUCCESS in 3m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/873371f86c824a8f8e1ef9ad461ced1a : SUCCESS in 4m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5e15e13ca1b54ff2a6b949624da42780 : SUCCESS in 24m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/88942795bd3d4dd6a1743a8a00a7e31d : SUCCESS in 1h 12m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/09f92d570f174419bc2287189d9d1521 : SUCCESS in 49m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1bb0c0be50ab4bd8b4d2f8b212de0e6f : SUCCESS in 1h 34m 44s\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/79ff7e635a5f40be90ced1f7530ba2d9 : FAILURE in 20m 34s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/73d0a95aebce43d694f36fbfe9ccbdd4 : SUCCESS in 22m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/41ba813b30eb4760b686d66f1090e25b : SUCCESS in 26m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2ac8a75c19534b17a4721e38fd2898cd : SUCCESS in 1h 08m 07s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/44c054e09786435b9ee86e0b3fae9d81 : SUCCESS in 1h 36m 36s\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/d9ed6f1778d54059886a215972e15da4 : FAILURE in 17m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"cab0a2a3e8567f348e94cd9684334496fb37627f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:28:21.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d49fd9a663d43cf849c51cc6888495d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e1e3a504eb364d138a7839f35dd014df : SUCCESS in 54m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/20bd4837ac904159b7fd6b14522a556b : SUCCESS in 42m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4964d7cb2b8c4df79325662197023107 : SUCCESS in 1h 52m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8018f843eb764a3b843c9f05579253de : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f246a022ff43402187848ec97d4943c7 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/deed998ecfc8467c80fb77121320568a : FAILURE in 8m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/da4481e9584949cabf5f862e4775d3ea : SUCCESS in 31m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ac51455456974f93b4f3f3d72c4efe52 : SUCCESS in 52m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f2e513f2391095bd8e4fc3a1980ffe5d65e9d247","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-24 06:42:55.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"35de38b138abe01fe215e0fe0672b6cddeadeb2e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-24 06:44:29.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"56f483010d20fd0c7207a6451077a8c277c0cd3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 06:44:40.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3477139c1b3d4a8b4018e1a1a1d3fc3360607d4d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:36:06.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d857ad793d437194ac1bccaa1fd747\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/83ca330c8b8c4fa797f801e8fa3518d4 : SUCCESS in 21m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f892865de4424267b212e2f67c975db4 : SUCCESS in 30m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/716e06e201724e3cac091c9558fd0860 : SUCCESS in 1h 06m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/42217da322144c0aa3a8adb0884f3c21 : SUCCESS in 1h 35m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26599f03f0854824aca85f54787ff492 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/059a2dfa27734dd08e2bcdb70c76d3dd : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08ca222807a2429d827357f8fd60b022 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3c4b354854b4798b843f74393bb8568 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ebcf009871241e19bc119d78a2e3076 : SUCCESS in 1m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/861f2f803c8842728abaf78935290bb4 : SUCCESS in 2m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b04a72a6692f46969ef50fd8de03446e : SUCCESS in 13m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/88278170cdf1422cb3bb842bccf566d6 : SUCCESS in 41m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/75dbe7b3af4c490ba7a47537835e0034 : SUCCESS in 56m 59s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cd1182b96e4b4ab09f4306cead81435b : SUCCESS in 1h 38m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7cac619b78c4404ebe5c0988385eab82 : SUCCESS in 18m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f576a1358f944ed4a90246b6d1dc3d4e : SUCCESS in 26m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d7d0ee6743cd49bcbe7bfe79b5f733c6 : SUCCESS in 34m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c61bc35d19e64a038dd89a2d1b3bdbfc : SUCCESS in 1h 46m 26s","accounts_in_message":[],"_revision_number":6},{"id":"cb183a07b20bdb0b4a107e0f06fd16a38aae7875","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:36:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"8c3f17313fa889d4a6bf4355aa1737ac55f2c468","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 08:36:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25eab25310584d19bbf4ca2dd05ef2b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921cf2ff82064e10a2f7b499c899d87f : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffe5561f7b514c7fa88b7f4ec1d69d76 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ac79bc9b67823eab6d6e9a027a65261096cfcc19","revisions":{"1edd2cc02baea948fbd8e0032e8bd700c234003e":{"kind":"REWORK","_number":1,"created":"2025-03-26 14:35:18.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/04/945604/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/945604/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/945604/1"}}},"commit":{"parents":[{"commit":"e5e308f034d8d75e8298231c62bcf896273ec8f2","subject":"Merge \"Switch to 2025.1 sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e308f034d8d75e8298231c62bcf896273ec8f2"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"subject":"Upgrade Prometheus to v3","message":"Upgrade Prometheus to v3\n\nChange-Id: I4113976db743d3c5c4ff62f2876e224888f05a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edd2cc02baea948fbd8e0032e8bd700c234003e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1edd2cc02baea948fbd8e0032e8bd700c234003e"}]},"branch":"refs/heads/master"},"85882bd97ec5cb28d15f729090c4430d09374d6e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-03-26 19:46:05.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/04/945604/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/945604/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/945604/2"}}},"commit":{"parents":[{"commit":"e5e308f034d8d75e8298231c62bcf896273ec8f2","subject":"Merge \"Switch to 2025.1 sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e308f034d8d75e8298231c62bcf896273ec8f2"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 19:44:36.000000000","tz":60},"subject":"Upgrade Prometheus to v3","message":"Upgrade Prometheus to v3\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/945611\n\nChange-Id: I4113976db743d3c5c4ff62f2876e224888f05a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85882bd97ec5cb28d15f729090c4430d09374d6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/85882bd97ec5cb28d15f729090c4430d09374d6e"}]},"branch":"refs/heads/master"},"3db0ff84bd2591c148970ff7e2c89b5c61f4d851":{"kind":"REWORK","_number":3,"created":"2025-03-27 07:55:44.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/04/945604/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/945604/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/945604/3"}}},"commit":{"parents":[{"commit":"e5e308f034d8d75e8298231c62bcf896273ec8f2","subject":"Merge \"Switch to 2025.1 sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e308f034d8d75e8298231c62bcf896273ec8f2"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-27 07:54:55.000000000","tz":60},"subject":"Upgrade Prometheus to v3","message":"Upgrade Prometheus to v3\n\nNeeded-By: https://review.opendev.org/c/openstack/kolla-ansible/+/945611\n\nChange-Id: I4113976db743d3c5c4ff62f2876e224888f05a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db0ff84bd2591c148970ff7e2c89b5c61f4d851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3db0ff84bd2591c148970ff7e2c89b5c61f4d851"}]},"branch":"refs/heads/master"},"37808d4ba81a9174126a0afe1caf6b7b4cb0a9b8":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-11 09:28:47.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/04/945604/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/945604/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/945604/4"}}},"commit":{"parents":[{"commit":"e5e308f034d8d75e8298231c62bcf896273ec8f2","subject":"Merge \"Switch to 2025.1 sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e308f034d8d75e8298231c62bcf896273ec8f2"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-04-11 09:28:47.000000000","tz":0},"subject":"Upgrade Prometheus to v3","message":"Upgrade Prometheus to v3\n\nDepends-on: https://review.opendev.org/c/openstack/kolla-ansible/+/945611\n\nChange-Id: I4113976db743d3c5c4ff62f2876e224888f05a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37808d4ba81a9174126a0afe1caf6b7b4cb0a9b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/37808d4ba81a9174126a0afe1caf6b7b4cb0a9b8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7397116bb5928e2758403f88631daaa0c5578460":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-04-11 09:33:33.000000000","uploader":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"ref":"refs/changes/04/945604/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/945604/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/945604/5"}}},"commit":{"parents":[{"commit":"e5e308f034d8d75e8298231c62bcf896273ec8f2","subject":"Merge \"Switch to 2025.1 sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e308f034d8d75e8298231c62bcf896273ec8f2"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"committer":{"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","date":"2025-04-11 09:33:33.000000000","tz":0},"subject":"Upgrade Prometheus to v3","message":"Upgrade Prometheus to v3\n\nNeeded-by: https://review.opendev.org/c/openstack/kolla-ansible/+/945611\n\nChange-Id: I4113976db743d3c5c4ff62f2876e224888f05a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7397116bb5928e2758403f88631daaa0c5578460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7397116bb5928e2758403f88631daaa0c5578460"}]},"branch":"refs/heads/master","description":"Edit commit message"},"ac79bc9b67823eab6d6e9a027a65261096cfcc19":{"kind":"REWORK","_number":6,"created":"2025-04-22 16:32:43.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/04/945604/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/04/945604/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/04/945604/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/04/945604/6"}}},"commit":{"parents":[{"commit":"e5e308f034d8d75e8298231c62bcf896273ec8f2","subject":"Merge \"Switch to 2025.1 sources\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e5e308f034d8d75e8298231c62bcf896273ec8f2"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2025-03-26 14:34:21.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 16:32:38.000000000","tz":0},"subject":"Upgrade Prometheus to v3","message":"Upgrade Prometheus to v3\n\nNeeded-by: https://review.opendev.org/c/openstack/kolla-ansible/+/945611\n\nChange-Id: I4113976db743d3c5c4ff62f2876e224888f05a36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac79bc9b67823eab6d6e9a027a65261096cfcc19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ac79bc9b67823eab6d6e9a027a65261096cfcc19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
