)]}'
{"id":"openstack%2Fkolla~946215","triplet_id":"openstack%2Fkolla~master~I679ac66186d457ae53f89469fe22c5662089019a","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-05 17:18:13.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-05 13:28:02.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-05 17:18:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I679ac66186d457ae53f89469fe22c5662089019a","subject":"neutron: Add agents wrappers in the neutron-base image","status":"MERGED","created":"2025-04-02 21:43:50.000000000","updated":"2025-05-05 17:19:03.000000000","submitted":"2025-05-05 17:18:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946215","meta_rev_id":"fbbc360ae622fa1ae701ef2e3a601d0ddecabe58","_number":946215,"virtual_id_number":946215,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-05 17:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-05 11:45:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-05 11:45:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-05-05 07:30:05.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-05-05 11:45:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-05 07:30:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-02 23:20:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-02 23:41:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-03 06:29:33.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-04-07 20:16:59.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-02 09:25:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-03 09:36:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"51e98711359239745983e4c7d338431f4d84c369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-02 21:43:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf60a29ba20b5f0ed3802d8af346e95d1b4771f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 23:20:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e40b67291ac54050857c8343dd237ad9\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/bfbeaaa8d92f4eb0a7b7bf8b16615dda : SUCCESS in 54m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3f2c7104f5c44b98954c63baa6cc060 : SUCCESS in 43m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e1a29c1a48b7473ea7bf9661b7799cfa : SUCCESS in 1h 35m 11s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a724024057b349f1a28dd80e860d6198 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/396a499148024a6db2b37ca3e0fdda75 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3976821f6e94bffbddbbc81f69a64d2 : FAILURE in 5m 06s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/23b89dccc5534528947e517a4cb4bfa2 : SUCCESS in 52m 09s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6db2c27a26e74da983ca94ed3ea275bf : SUCCESS in 44m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eda26b3cfe63e5914c4e3e8250b40b6016cc6c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 23:41:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/571834f4ae3c48adbc39b8574f4df1ca\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0a5af1dac64a49678628e831c7e7ef55 : SUCCESS in 21m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/9293b7e3ee854fd6b54377b90aae1e4e : SUCCESS in 59m 54s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0078cdca66fb447aac9e0619781da2bf : SUCCESS in 15m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9f22c2de231e4fe1bc444c7d414a0e45 : SUCCESS in 22m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/88e597808fda40f48836faaa42c3c93b : SUCCESS in 57m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bd05ad3e71dc48509ea8d603ed69a111 : SUCCESS in 1h 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/849e09102be54d698d78813ff06dc210 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72d75bb91ab541a99d3804933c81e5a5 : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2c402499181e4dc0a33babdf1afa6cf9 : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f19e4884e8b145689337a993ac208969 : FAILURE in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/283d59f1760d4bbc8fad3b5d80ca850e : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ad00370f0e41ea88ded7b4201ec754 : SUCCESS in 6m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4945589ceb684a5dac76280486894a0d : SUCCESS in 2m 48s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3384ab0aaf2c48b2a42acb84a6d8ba39 : SUCCESS in 25m 35s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/15d2c265e89443ee8455c213fff0cf1d : SUCCESS in 1h 31m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f86e5204e55a4969ad29c54eb79c094c : SUCCESS in 31m 31s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/3262e83554434b739ea4d85bf460cd1f : SUCCESS in 1h 51m 54s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5b1531505e4f40509f83b425ad267559 : FAILURE in 49m 54s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4868b61a3dba42508aef423688d692ae : SUCCESS in 1h 18m 26s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/22a07286fcd0455e907b52d508726332 : SUCCESS in 11m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/aca201adb3364ada8208f9aa40801165 : SUCCESS in 27m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a8fc625998784c8899b554c09240e5fd : SUCCESS in 1h 07m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/672eb733a3354eee84c28bbb46eed37a : SUCCESS in 1h 24m 04s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/59dbd66706864ca68463dd5db1415f16 : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b82cacd41f324082b81ac89c754ac4f3 : SUCCESS in 1h 11m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41a8f13f4d7b2c9688a09d0bdd8d9a12d1497efd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-03 00:07:33.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7e7551bd193bdc559a4ba344c466cfb026b2c3ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 01:02:22.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/606d56746e1b4ea1ba0c69ec3d412006\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e1f5ce24f6b84ac5a18b59e758660a1a : SUCCESS in 30m 53s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fb97d6cca71f4d2da67d573bc1093383 : SUCCESS in 33m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7e035d41b6e04f75871284e93a6a9dc5 : FAILURE in 52m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/93d791c4587b4795ae64dad1c3f47e11 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c49e6248d94434bbdf6ade5731ba38a : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eac3d4fb9a394cf3a5edeccfc778135f : FAILURE in 6m 09s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/05a2ea7231074ae59b8a8bfb92be8651 : SUCCESS in 43m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/550137b3c1684f01aabfee760a4c32d4 : SUCCESS in 37m 23s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"77ab92003bfb3e8c779b10c2fb8793f7604883e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 01:45:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac144e5dd81f4ab680402243a4643904\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/17374261abd74d2fb4fc56f961d16d7c : SUCCESS in 24m 33s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5d74ff10322d4d0ea6684b5c22fc78f3 : FAILURE in 53m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c644f651c831467d92b6430dd89d108b : SUCCESS in 19m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d2fb5a5245574c08aa0649d69a92ca58 : SUCCESS in 29m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ce304fd7b4474d66bbe96e39ec123f48 : FAILURE in 47m 30s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/752cb6ef61904f758d406134847892ce : FAILURE in 1h 33m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781bf890d044434a896989d2e3bb1dab : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8659e79d27754ad486e81ab95489d409 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9939dfcfefaa48a5aa6d76df98288f68 : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f6fda11aae08424cbc583ec290e82c77 : FAILURE in 5m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad695fcf51b948c98ce5a2031dd3514d : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20f2ed5a03941238d47478f6048acb0 : SUCCESS in 5m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e7eb7fac6dad407d90052df994ce74f5 : SUCCESS in 2m 21s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1661c3cd192c4c6eb77215714070dd90 : SUCCESS in 16m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d61d44e89704dfe9dbd0b4ad5c0bb4c : SUCCESS in 59m 27s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/952903c1084f435db2a7ac51aaff7984 : FAILURE in 55m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d02232f1c3214f99bd66148e79bca8c9 : FAILURE in 1h 16m 41s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e782baa7440745ebbfa3149d0b708c8d : FAILURE in 33m 43s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/6ddca0a969934ac0b585931fcdbd99ff : FAILURE in 38m 01s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b23ba6fde08348459a1da94267bea2ca : SUCCESS in 16m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/829648ddd19e4932bd1357d60f98d69e : SUCCESS in 13m 21s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/35eeb1da896b4d56956169c8d068c8b9 : FAILURE in 50m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5f153141e1a04e88b75c2bfa700447f1 : FAILURE in 1h 15m 27s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/36799bbfa5a94146807cbd204d78de17 : FAILURE in 37m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e0c72b8b6d7443d5bda82d7b5e3d1dd6 : FAILURE in 1h 11m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"faad9f214fefdb8ca511ac7689883a4c10b907e3","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-03 06:29:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"69a7490f5c9527cc47d3296e2d68635dd9e2d72f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 07:33:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cfb218fd6924065b2c40c239ea9f7ca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3bb18e3865a74136bf6479ef94bbc676 : SUCCESS in 42m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/33c07593b5c542d68ee6d3ee783a24b0 : SUCCESS in 37m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e243fdc3ac5d42f99d0d5cfc433e85a8 : FAILURE in 1h 01m 46s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/216e3b883b934ef4b6495f1ae89d2b51 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0705eb9a2c0d437e8bcdcf17a447d384 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ecd2115846434fb3aef1ae2853326b6e : FAILURE in 7m 55s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/08de965f863e4ca9b91d38ad389060d5 : SUCCESS in 33m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/7c99c34387374fb7895582273df90440 : SUCCESS in 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a1d50e58a98d748e171f2c01f166f82840f2aa8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 08:10:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09e82cebbe1c41e0abd16072d6b91ec3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c1177d81f0964292b28b6d78ba160622 : SUCCESS in 25m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ff30241a8d2c49a5898066a731fcec0c : FAILURE in 49m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b44e2ff6415f40d38a71fe5dd030e8ba : SUCCESS in 17m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e7d11f56466442e59b2dc12ebaa0e8ad : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9280358c23624e538602bd5149b53d72 : FAILURE in 45m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7cace970847f4ac886c5b7c3d28c72a9 : FAILURE in 1h 14m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8116758995b64e35a864cd63619290fe : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06b2d35ecb474b52941139a2f8e5ed42 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6c5bf330571d4e17b0a53878d80fc15d : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fcf3fcb4236249d99044184ed6a78105 : FAILURE in 4m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88c0c9b1d580415a8c3f2c15ff68ab3a : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6437dc9d89444e24b73ee2ad61e08362 : SUCCESS in 5m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/da3f925bc089495cbae63b4f2df9133c : SUCCESS in 3m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8ebb2d48c50942db9a9358c2d054c48d : SUCCESS in 22m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a681f395b2bc4949b7085c185be3dfd3 : SUCCESS in 1h 07m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4d9e3d50dc6d452f8b3ffb1e84908243 : FAILURE in 6m 26s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2223819802d74a1eb8e5345a408e3ba3 : FAILURE in 1h 16m 43s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5b3da57f57734ca8894ba7793553f31e : FAILURE in 35m 21s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8f1344eed4ac4022b2f5944296ded4c1 : FAILURE in 1h 15m 13s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4982eaab07a4ae9af59d7254cd138fe : SUCCESS in 10m 37s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3f73d5bcdc454c1892abd8a32c095fda : SUCCESS in 28m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d09487021a3b423aafbd798ae16e972d : FAILURE in 49m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/94e120f291a64fd9849223819422502a : FAILURE in 1h 35m 24s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/709436971b95422db75136cb73c19da2 : FAILURE in 41m 35s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b63d526b5f4741ffbb08f1546cf1d1aa : FAILURE in 1h 04m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ab691232f3100bb6d2eb120d5487cefb870e3cd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-03 19:26:48.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"cdf1027776c51c60145dc26da501dec8f8e082ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 20:43:15.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6970e91806da42bd8edb6fe42da218fc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d35cf3cc39a84d3bb1728f9bdaa94ee1 : SUCCESS in 37m 41s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1403cd9c8436470581dd5b94c1a1de82 : SUCCESS in 44m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/00789b73241c4b8b80365c55d3355ef6 : FAILURE in 1h 14m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7b44416d51ad43e187d73fcdd940827b : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1d83f2899e55479580d725663253790e : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ae9fcf151dbd4eb989d26a66c7d4bc87 : FAILURE in 6m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/080b24711c6342a692c49df79fd1aa27 : SUCCESS in 28m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f0aa46b2a32e4cc2869599b048508b90 : SUCCESS in 38m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3269926ee82a3a2a09c9e207a3c7edb02e76a43b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 21:09:19.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8637b803a244ca789dcf9ea69ca3cf4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a96dc28cde464797bf2383af2bd44cd2 : SUCCESS in 28m 03s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2c3763539f6f4ea59ca2574b8f0387b5 : FAILURE in 51m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/45470487e2b44f8ba55e3b3901af33c0 : SUCCESS in 20m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0c89afd0b086400cb9769cdfa13d1617 : SUCCESS in 31m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d9154f796f194e4fa75c85f3bd056c80 : FAILURE in 49m 57s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9349b4a7de7145fead676665f034bee4 : FAILURE in 1h 19m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee5d0559e93c4eafaaf1fb3f8f049a2c : SUCCESS in 6m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7d17bc5d90741b48fd52de8ec80b9a9 : SUCCESS in 3m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/48bbd352f83a4d759e9038fbac9a70da : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa5fd80368974859826a5f778498d758 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6213753072984ef6958ec65ac60de288 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/382ad6ca432142fc9a4c0be07d8f4dc9 : SUCCESS in 4m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a27574ac566940c1b8e9388f222f439d : SUCCESS in 2m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f8e2bcac5e3e4050b9d100a040caed31 : SUCCESS in 20m 53s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3e1399c6394a442bb9f9b173f3932dcd : SUCCESS in 1h 14m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a1b0dfc6589748a1867cef0bc4678ef0 : FAILURE in 41m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/02e58b54d5054fe39c09acfea334b3e2 : FAILURE in 1h 26m 39s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/5934e4dc5eae425a97ae1eb3cddad60c : FAILURE in 36m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/335a80243f7444ebbd8843c6ecfefa39 : FAILURE in 1h 17m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/23b52b3e796d4a3488a6abd896773c5d : SUCCESS in 25m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/428d87ce10e6495ba5c7a2ac26154b49 : SUCCESS in 32m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/79c20c6209fc41f49fb5971dcc997f54 : FAILURE in 27m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ba85d18032964854be32cea482fa5dbd : FAILURE in 1h 40m 45s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/89ce749599c24678a96582e863f56529 : FAILURE in 49m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9c9010cb5f3d4ec89483a239c98ee74a : FAILURE in 1h 01m 04s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b37f1e40bfaf033ae0818414031db6009f66530a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-03 22:05:26.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"1a0aa3059d765187103af2eb26fa5102b3b0b216","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-03 22:05:30.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"0ff7ac5904907cf39b9e5e3e20e9d6c17d1f4782","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-03 23:13:50.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"d16446e3bcb76c980b55a090070a5c76eb531da8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 00:15:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce49fcaf952d449a87bab846f377e2a1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/569b04adaa8b49c1bd6d7525ffefd8e1 : SUCCESS in 42m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0a6bd73ff5f64cd0968ea51c6887060e : SUCCESS in 43m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ff3f53d172594cf4ab0463c3cda7d0fb : FAILURE in 59m 26s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/390b74ae71f54baeb786eb33f8f17fae : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9e46bf9940304a9a963bf2b7da1d014e : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ee8d08365c804e9bb404d8bb8bddc187 : FAILURE in 7m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/23f15bf027da435a91b8c5a6c4fd3a94 : SUCCESS in 47m 16s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bd7ee4982621403e844e2c81ac206ace : SUCCESS in 47m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"65d4a27727dbe4bed79ea13d91b8bd73d3712d7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 00:48:38.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16be9f4855c34211ae4ad751ad003e63\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/0b88e9b8265547acb2e1f2982274995a : SUCCESS in 21m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b9de438bfb8b407387b9a4900b49b723 : FAILURE in 43m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/31295f32a04040ad8342284ef0062d3a : SUCCESS in 14m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/03ecdafe63d647e7aafc7300b770e5e5 : SUCCESS in 21m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/39d943a1b1a446c4b70b60f071236934 : FAILURE in 28m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1b69eab63c4049d0a1abc6457d0f2e61 : FAILURE in 1h 15m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a65f9e68903450396aa8c89a98dc235 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c3813dad25324eaebd724de754fe6de2 : SUCCESS in 2m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ac41c001f408425993717929fc4bf000 : SUCCESS in 3m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1259d4f2680e4849870de51e1a2657d3 : FAILURE in 4m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d94d3d5c1b1746829cfa1a58c95cae62 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e813253158042b7943cca48894b8479 : SUCCESS in 6m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2cef01904a6c460998a74e92b8657311 : SUCCESS in 3m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4590879d78c141bab93bc4631742aa98 : SUCCESS in 25m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a2c006b621204e8fac67a87239747814 : SUCCESS in 1h 19m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a7d6b6c84d1e4e50b2483678de8d21a2 : FAILURE in 47m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/94661e1c7fae4b7c9db33e615e3bf094 : FAILURE in 1h 28m 55s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/adfa6545f4444541a34905c22832382c : FAILURE in 23m 46s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2c65bd7cf5714277a83d193e1205741b : FAILURE in 35m 26s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a55772aa28d0461c97a7ff2b178fd7e4 : SUCCESS in 17m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a09c54eff9e44876a9a8d6debae58ba4 : SUCCESS in 22m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e155f4198d1e4c7cb54d83316fbb3750 : FAILURE in 39m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9deb691e851c4c1298036c02f7be7779 : FAILURE in 1h 16m 37s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2ef559265cad41ba9340b31435d87322 : FAILURE in 43m 37s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/5b0edd8118e24fa49f7776da89586e0e : FAILURE in 58m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"af7d54bed2114e436d30cf141b53b7e1fe77cce9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-09 03:12:10.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"900c47ccf7ed973569d0d03d2db9c829adba11bd","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-09 03:13:31.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"5a3bb0e8555d1eecb424f466fb10e94787edb9da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 03:52:49.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7461b1ef25548299521c5f6eaf01892\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8e01ad747f45459c86272a3d07416c0d : FAILURE in 35m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/83373eca0a2542b6b5a535f4fe66f3eb : FAILURE in 33m 17s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4177909ac03a46be96f0aa03ef16d310 : FAILURE in 27m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30177bd4afb74a3ebdae2f68d5c33b20 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be7092b55e644fe19168d5f3c8f36a25 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e85946bb17d54ac3a3a91e349313a2c4 : FAILURE in 6m 52s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/01fc7938214148d1a56022b803c28648 : FAILURE in 34m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bf2d68e275ec4595a7dc43982c2dbd45 : FAILURE in 36m 51s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"c752a8265367d82ee2b13b0d96d6d64a883c1ab3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-09 04:21:04.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"ac644e650c7bab615bc9d4fb144cd5e601fd3fd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 05:02:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9b633ce8fe041c1927d981e27b880ed\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c70875e78d334f869f3856032572e725 : FAILURE in 30m 12s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4ea593810873414ab427b393103e6c84 : FAILURE in 32m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/23641c35c0704c9b91ee092ced39cdeb : FAILURE in 24m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/eec6e3f7b1c14add9fb4db13f3a61cb7 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d5e43f2877d45a892257b32e96af57e : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/95a7e598dffe4196b9b3f43b41c91268 : FAILURE in 5m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/05fc552a66984216b0cc7e248aac67b5 : FAILURE in 33m 00s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3d8303075393466f8d4c774e8566b834 : FAILURE in 39m 06s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"60e741c63c67fff5869b0855bb6715de36bdaf44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 05:11:09.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f78d04c8ea5944d1bb44594cb9709e64\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d80cdfef297945ee9c6250e7f4f3d647 : FAILURE in 14m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ab1b32d10c794d51a4a386b756fdfc13 : FAILURE in 10m 27s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fedea90d94d44341bdcd8d6a25dbbd98 : FAILURE in 10m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/90624d45b1ad4ee1a7af5dad9921255d : FAILURE in 21m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6ec7d39a359343fbb72619b2fc4cd081 : FAILURE in 8m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7b52de688b774d3d8e728e72cdf4ec67 : FAILURE in 17m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5daf3cb1ac3d4a3d8856d7560a29ce70 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77ffa09749e946f59d1bb5fbfe495f23 : SUCCESS in 1m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa6a2dc8bffa4daa9fd59ee4d9a7ada4 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e51ff27cfc1e4bf5a9aaa5582c3a78be : FAILURE in 4m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16f2172d23b74ef595bbec28a7e5ad9c : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c5b572d3b44e3f9cda286392b06b9d : SUCCESS in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8f90eaf6f8884d62bfef7a077574590d : SUCCESS in 2m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/35d7359ea29a4d87b1e7d1698586d555 : FAILURE in 17m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/22e23ba54305476db1640ca97f6b25ab : FAILURE in 48m 54s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2f7c734e67a44239e857918eea3861e : FAILURE in 21m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/495c6ed59e164fad871f5e0d99def4ba : FAILURE in 16m 04s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b913a64b45784f869f1b0be6c6025c34 : FAILURE in 19m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/96f13af1a0674b5f8e67bf47d87fd1e8 : FAILURE in 14m 07s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8030cbfe0c5f4cbcb7f7723eab4d99bd : FAILURE in 13m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8f808238622b4fb780c80e021d34016f : FAILURE in 17m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/59ab7d8d3c0548a7b5cca1f70c232ba5 : FAILURE in 12m 05s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/30a2439e7b7f48b59ebe5eae1d47b1b9 : FAILURE in 16m 55s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/731c8ac7e9b14491ac2c1769d70c5d0d : FAILURE in 17m 41s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4a5de0308413488d848e1353488f7d85 : FAILURE in 10m 14s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"340a29e1260d120e15663ec4fb27dd80c4158218","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 09:14:14.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"2ac5055866bca2f466457aa5275424f482dc23ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 10:55:17.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7150a4d36ac45fbb1f04565919c7ecd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/70538911fbf34b2487c1d6fda3a61240 : SUCCESS in 27m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6d5eb40e532341f3966a066f8f6caef3 : SUCCESS in 57m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2a4f33438b374066a1f74927dfc5b289 : SUCCESS in 21m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/86ad64c96d754197afaee5a3fc6d4dbd : SUCCESS in 15m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c68f204371c749e3a132ad8cd534adc0 : SUCCESS in 34m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4f5db526f9ef45ab8b498f63eacc5d60 : SUCCESS in 1h 27m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f58b01371fb647ddb53634c1e76936ad : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a19b8c585a234bf8b6d51a65149c599f : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4496c27bddc241b7bbb75a8a86b809f0 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/64fe0b87fa084484b56edb079bb32035 : FAILURE in 3m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5549974d4acc4664b6c47d925d7e7e7b : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e87f8110311041aba7bd2409489908f6 : SUCCESS in 6m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/1e95680268bf41f3b1f4fce816d3f6ba : SUCCESS in 4m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/81567d8fb26e413a8c4808c2a9e1daf2 : SUCCESS in 24m 21s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0c42c34a1ab54e10a5e3b98a47f132b0 : SUCCESS in 1h 11m 49s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4134488eea544a0499c8e00b07d580bd : SUCCESS in 1h 03m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/56f4b172c43445019be5f738d1df6298 : SUCCESS in 1h 36m 45s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/29e8db4ccb0d45b2a978b08cc7004a56 : FAILURE in 46m 01s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8064c8b80e604885a0ca861dba5d07a7 : SUCCESS in 1h 35m 05s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5e470ebb9c4841eba744fa1e274521e2 : SUCCESS in 26m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/83c4749f200a4854b932ec8666b3162e : SUCCESS in 31m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b76148fe2bb6405b899a037522a593fe : SUCCESS in 1h 01m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/306d4c77f755459a9a71224da3aa9af2 : SUCCESS in 1h 32m 29s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/6c21723ce24840b187f5256a45ae3ab0 : SUCCESS in 59m 40s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/71723eb2b27e45febe21139419ac4e84 : SUCCESS in 1h 19m 43s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"895a44e29a76d6eba9167327cc83c3fc4d129c93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 11:24:45.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/913e97804bba4206ba6b023435420473\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a6714bf2e70c406ab1a02c85d59d8add : SUCCESS in 40m 25s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9937266d55784bcf95a374484173c3e9 : SUCCESS in 48m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5c834bfab93c4f2ba972be0e77ee8966 : FAILURE in 2h 08m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4872ce078fa9459c8f3f951f2dda3414 : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4585a75509da46b0b396484ee4261376 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3ca85f8afbba4bfe94cb777092f2bd04 : FAILURE in 6m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3c896bf35d4247a89f87414576577a6f : SUCCESS in 1h 00m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fda69c3ebf3c46eab9cdbdc8239d7768 : SUCCESS in 54m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"58b7d71b1269865747ac955279baa50d9fdba91b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 11:49:17.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"1680bd60dc8acdc1642ff0b8f20dbf90e4af3a83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 13:12:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe14319b7e8043b3ad9dcd58ceb85af2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/172499af4daa4610a7598217d20136a5 : SUCCESS in 27m 12s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/c4fe18fa97054208a2bf0b2ba9a06b9c : SUCCESS in 1h 09m 46s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c2672eaede54463d9ba1e7d20c8b7c5d : SUCCESS in 11m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0dc299ec427043629b93724ac1018993 : SUCCESS in 32m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/1713a364d9ce4c86859783db11397787 : SUCCESS in 1h 01m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2c5b24c38651495590e26a9f121d7cec : SUCCESS in 1h 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7d86bee95444a89a48ee1860ffe3f40 : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f3357453c164bce8d5e9b46d5b226ee : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89af94884efc4ad5be1ad72777244f2b : SUCCESS in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20c8cbb7f2854af1a6f72b3b7f23262c : FAILURE in 5m 04s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45efc5f109b14c1d8bbe6b2dba6443a9 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f934a3b1070a4777855fdcfc3f9fa413 : SUCCESS in 3m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/bd1fe528363d4402ba18e574f68700a3 : SUCCESS in 1m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/34eb1e705cd34c2ca7c8d2e42e670797 : SUCCESS in 24m 03s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7549bc8f6c774e64b5ec35ca7407da26 : SUCCESS in 51m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a049ef33c0b648e8a35552d3b79482a6 : SUCCESS in 1h 06m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ab470efc0d484aa28719480291ea2927 : SUCCESS in 1h 20m 24s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/1d5dd5780668473d8dda433964757c49 : FAILURE in 49m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/0f699d3b90f047cd8ecb80492648cc8a : SUCCESS in 58m 41s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f3f8ec8d7dc44c33a81fd66a469507f8 : SUCCESS in 11m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/90562e9ac70646bbbab555e034d3cb2f : SUCCESS in 28m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cca950cd30f741fdb5c0f266ae126863 : SUCCESS in 52m 42s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f205c79e23f64bf98531f28a1ebc3531 : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c38e6d722ff644cbb0ae743fe6b8b912 : SUCCESS in 57m 02s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e54036d5c66748698fcea0a070bd8961 : SUCCESS in 1h 22m 32s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"cf2bdfc77ec1e08acb884a461d8b496aa5a15520","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 13:38:25.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b55d86be75fa4030bc5d358b8d9e7481\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ee0992bb23fd4171b798dc5a4c8e2e5a : SUCCESS in 55m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/52b696e25c5f41fb971ffce97b95a726 : SUCCESS in 53m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cd2b03879c6b47bea00f8c44739849c3 : SUCCESS in 1h 46m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ca6dc6e9aa7240f485cdcf2cabb03127 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/30a1efdf92de4dc4a35660fa0c0dd5f4 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ebeb3018ee3c4ebf945e64f6c7d7d15b : FAILURE in 10m 26s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/621ad97d099148f7b69e8b31d0a9b103 : SUCCESS in 54m 40s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6c5bae59f83d41ea9a4fc6ef5a89c360 : SUCCESS in 56m 41s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"e60855f48795b337e4339d742dd25702a118f0d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 15:05:20.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"fb62901dc242c36e5e67655ebf2563ec572fca6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:16:23.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bd2e1c5f2ab417c9f061659cfd892eb\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/75bf90b105b240ae8a163d9ad77f5004 : SUCCESS in 57m 32s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5f22a1fb45f447dd949708fc189b7d2b : SUCCESS in 46m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b237d0fe88144917975297ba506169c0 : FAILURE in 1h 09m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9eb24b543ebc4e9d9e56f3df68416215 : SUCCESS in 4m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b6dcf68cc28f4a4683cff89715fbb43c : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5f23f6a9a8ff431fbf5fd799c48de056 : FAILURE in 5m 32s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/82eea3023e674b05b5f95d4ac09cab24 : SUCCESS in 51m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/95b07caa419a442a8c53bcfde2ba7804 : FAILURE in 43m 22s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"f7bced47ee983e978ac8cbeedefe59f3ea1111c4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 16:21:23.000000000","message":"Patch Set 12: Published edit on patch set 11.","accounts_in_message":[],"_revision_number":12},{"id":"b51b3e0b7d1162c671994c8d9e9ff1075d2bfce3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 16:24:26.000000000","message":"Patch Set 13: Published edit on patch set 12.","accounts_in_message":[],"_revision_number":13},{"id":"3a8a6bc95b1ff60b17d3b05eb63c62e3810b9994","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:30:25.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf4487ef14ee437ab73369b26e097873\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/512f2475d87c445facd0014f0a24166d : SUCCESS in 57m 36s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/74d6a2653ef341d3949e3d3de5294f3a : FAILURE in 54m 44s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/42e73f6dfee74e47aa49426f890158fc : FAILURE in 38m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/637e3ab07c59408db4369069ea7fe9cd : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a366d2765103418da04f41193e58ee70 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c046dd65baf4887ad8830a87ef819b2 : FAILURE in 8m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4586b40a164740438162fd1208ed9629 : SUCCESS in 53m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/896ba73bde3c418a88174575ab56371c : FAILURE in 1h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"a2dd185e70fd335dba84fd80054249a77db7fc72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 17:34:01.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"0fdeb33d7a3a94b98ab9f5ab4f1a69f67bc94958","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:47:07.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/19a7af1fa38c4310a68c718266135e2a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/66b726dd87734281bdb9607cf577e913 : SUCCESS in 48m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/492d24f95d0443eb995add63ff520939 : SUCCESS in 44m 24s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a099733b93ea491b883aa2492df77696 : FAILURE in 1h 11m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/80bae518c1c6492d8aee6b9fd0a72d6e : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7bb47620b53e4536a121d9abae361dd1 : SUCCESS in 3m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a0dbaae67d8441d596082ff67b46bd7c : FAILURE in 5m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a82c4f4e7638409faaa3f4bf4bae70bb : SUCCESS in 31m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d3ac197dd6f04a17b261b4210a9056f1 : FAILURE in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"a0b7a2955fd03010a7a858b15c5ac718f9133999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:57:01.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac79a0c3bad44c2697cd3cf37a8f543a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/648d15e694814799aee41fc1d10f668f : SUCCESS in 16m 29s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/21e460ff2c344ad3b22526e392e3af38 : FAILURE in 36m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f69530cd1c3745d3835ed8d5c11d7e52 : SUCCESS in 20m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e3f67172ac72471f95046a1c4a3f79ca : SUCCESS in 26m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/95b46bd952ba4deaaaec9aa07ec13b5a : FAILURE in 29m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/876024fa6d974162a0e069cb22f17d43 : FAILURE in 1h 18m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bdfcdce74c44876a2905cc8832ef3f6 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77c0dc9bfe0f4c22b7f20e07008e55e6 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/724a44ac470745d8b2f6716e63d1b5bf : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3569fc159454c62a897207e48bccb06 : FAILURE in 5m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78500218ec2b471c9ebcba7035657163 : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb183ddde6414118bd6e9de44e5d1e30 : SUCCESS in 5m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/284887084a884ad596be693296a788fc : SUCCESS in 3m 04s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/20ab65d745fb4c91964d352cb0688540 : SUCCESS in 20m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/666e7fb04d174f98b646ebe1c791eeef : SUCCESS in 1h 00m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2953a5443fa247009c0fbff2924d3c27 : FAILURE in 33m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cf7af518395046aa89ee168552bfe27f : FAILURE in 1h 07m 00s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/87954afca27c4023b33382e19799ecf3 : FAILURE in 32m 29s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/180f8d3e0a994cfc9a0ceaf68553b33a : FAILURE in 52m 54s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/76a51bab9d4f44d0971274343aa338db : SUCCESS in 20m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/90bd2e4882f14735a175fffe650aa7a4 : SUCCESS in 24m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a95094d8a6c448469a38adbc0e6715c8 : FAILURE in 40m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c61f20a7ec564a5f85aac7b671c0c099 : FAILURE in 1h 12m 52s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b6c3fb59414c458fbcb6709fbbf61073 : FAILURE in 36m 12s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3a4c475884cf4696b2a620956af1f3b9 : FAILURE in 53m 21s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"3c8ea17f62ed4928322d695c16023e293e4eba6b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-23 09:07:09.000000000","message":"Patch Set 15: Published edit on patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"fbf7d77e99f8cc4fe2a90555652aca1cd79ba4f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 10:54:49.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd74054a7081468f9dbe9fb6b08f5267\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/b28bb9b393eb4e978f3b37c4ce18af2b : SUCCESS in 17m 31s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8dcec18ec4504143b645c3a625af80e3 : SUCCESS in 1h 09m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/60c7259aada44a8a9ee80572be2d2006 : SUCCESS in 20m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/af03de3f8917486286b50147df61ae39 : SUCCESS in 27m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/90c21563049a4e958475e6fa724493e8 : SUCCESS in 59m 51s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/cfe6bd7637d34b6b80a3498b568cfebd : SUCCESS in 1h 35m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec21f3511ab4e75a61499599e073099 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/087c076001904ec095089fef531aca3d : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b6ee4dcb1b34c28b68d5508895b9f44 : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/892fdd56284548bcaa2c6202a084a09d : FAILURE in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ae5e5003242421290891ea7c612351d : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f04c1fa0f5c4490ba9707ab484fdb7f : SUCCESS in 7m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe495acb18794699986cbcf226f3a285 : SUCCESS in 2m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f2974850c025483cb74225e0c78f20d5 : SUCCESS in 23m 45s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a8786170b2f041a2855dd3fbc28ac8b0 : SUCCESS in 1h 21m 08s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/efa180f0580f488684e33d9ffbbf041b : SUCCESS in 1h 09m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/445b15232e7d4a76b597ecabd130d880 : SUCCESS in 1h 40m 59s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/69903f794c744d9bbade471cb1248e63 : FAILURE in 42m 48s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/40e315e4a90c439f98b09ec0a33fb29b : SUCCESS in 1h 34m 49s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bb758c906e51457c8561ea7fd0cf8acb : SUCCESS in 21m 31s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f42cc4de54a54960a653eb35db523548 : SUCCESS in 23m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/524936d2057a4a2baf71af80c7b01f9f : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ccd8eb8a590c48cc9683f3b429082984 : SUCCESS in 1h 21m 24s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/bfe4038d07dd4fb9a16607c47e64ac30 : FAILURE in 1h 24m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4d4a52f3bbab4d7492523ed13e2f0cd1 : SUCCESS in 1h 30m 59s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"a15d8f6096cb55ea1d15d9e3addb397a192e90a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 11:04:07.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2394ef11805d432e97eefdf9c7e5d158\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/89c7fc4271e14c19a88431d32e2bc89e : SUCCESS in 49m 50s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6c13ee7cf218469c9680940e96e84f48 : SUCCESS in 47m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/72b8b7400f554911afc29d1b162f687a : SUCCESS in 1h 55m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1537cf2d37764995bac32358be6a3b75 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b809d4d65174e0ca2c5f7a26f90c950 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec71496270204a7c96ae3f221f8cf36b : FAILURE in 7m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d3e78f381064406abaa5ae4f1dc81260 : SUCCESS in 37m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2112d20cb4d646c8b4157d9f062ecfe6 : SUCCESS in 35m 25s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"794abb58953ee9d3af32c5d1f340984ab52d5aea","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-23 11:23:08.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"cb0bc889ef9e36c989f736b8301d7586c437622e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 12:37:27.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee5a36ce98ed410b88ff484614e4b364\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7067323bb0634890873c8783e1f8233b : SUCCESS in 54m 05s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/218d51fb5f054155bf63c40fab600d33 : SUCCESS in 47m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2a54789b15cd4e13b68ce906478cd45e : SUCCESS in 1h 12m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4f8ab88e310f43fdaa6e5668369963fc : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/93b26930e18d44a0b7e0d1721412586d : SUCCESS in 4m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec49efeb26064937a9398292583dcf0d : FAILURE in 5m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/02e974a79cb94a19847cc0a359a610d7 : SUCCESS in 39m 22s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/2cee69a6ea62416abde28395a63f7e6c : SUCCESS in 51m 46s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"361f075eb5c66c14cb6fa06db4c2d7791fe5fe13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 12:50:38.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/851f0cdf3e934cdb814421ac6b245ce4\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e2bbe9d1b81d40578a8f973ca9753905 : SUCCESS in 16m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/53a5bc20503643109db1250c1bf2400c : SUCCESS in 35m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3260252b054042c8b34c5de454045a48 : SUCCESS in 18m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/845fd48f23944b05930b486ca88140c3 : SUCCESS in 16m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/6955ffc234ca4bb3b473611a4d188b06 : SUCCESS in 50m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/96d0cd106c104c4084c8ba8898a19968 : SUCCESS in 51m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a075ffd2751411bb209ac1260391281 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91bd8613151241a5addaae4aace8b695 : SUCCESS in 1m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b340595ae6754db1b664c7f21d07e260 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a245e71b6249413d83367896d0f020ce : FAILURE in 4m 26s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/357f19b7e6324345a269d29098bb5199 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83ad49181fe94bd6952148f8c119178b : SUCCESS in 4m 34s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c35e6c8575624e08b3d2e93a17049565 : SUCCESS in 2m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/23369f30f232490a96d7a32e8e64d039 : SUCCESS in 21m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3512fc2f8cf543f98700b219884cd3b3 : SUCCESS in 58m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/00e2b2ce9e7a45d8b0e55804d08e854a : SUCCESS in 53m 32s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/7cb040b8cb2948a7bee4d120b1d81aa0 : SUCCESS in 1h 21m 11s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/8098176fe5a74aa3a3816c43ce33e930 : FAILURE in 42m 40s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/eaa4bcd2d4744e1387b0dd95fa235a1e : SUCCESS in 57m 12s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8803c97e516d4abc877fb9a070460935 : SUCCESS in 22m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ef4394fa8add454f8d2ec9616f334907 : SUCCESS in 17m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/50d5b33bd92940368ba6ac9c71a8b338 : SUCCESS in 59m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7b99f7f829494845941c07ee66f0fe76 : SUCCESS in 51m 05s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/2a780d2a2c424c769842b9d9323b6254 : FAILURE in 1h 20m 01s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3b8216a9a8e64b68af833caeac4ff35c : SUCCESS in 1h 21m 10s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"fe8a8505757502d8d3fb23224ff75994bd04e9a5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-23 17:40:39.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"a3e9ce6d8d930a7836f8824983f59feb07f3446b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-23 17:43:36.000000000","message":"Patch Set 18: Published edit on patch set 17.","accounts_in_message":[],"_revision_number":18},{"id":"d3f21117ccd489783103bc61af7b980b98f195e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 19:18:53.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ba5b997824e4c59b0ad3618fd367cfc\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/88175532ec964ddd9dfd3360937dad4e : SUCCESS in 53m 22s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/88d2525a0f114affa48a205f71be7184 : SUCCESS in 50m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c378766ba0de4a7881615ba38e6e14b1 : SUCCESS in 1h 33m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ce03ecd47dff4bf4a50c7025321462dc : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3dcfa570e76448abee9cc42e429c6ec : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c19ef0a9a7fe4637b0ae8fc7a7107203 : FAILURE in 8m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0079c4e99cf146f394e4a7281836de9a : SUCCESS in 46m 05s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eace544aeced42d398ac2215daadc851 : SUCCESS in 52m 30s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"beb12b52c4216147ecefab1cd9be38cb02ae9efc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 19:28:13.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35057ed294cf43c9931cac162de70676\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d6f56230d6f9471b97777b86bddd76eb : SUCCESS in 20m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/36c93f54ce944c35ae0db876d7efea12 : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/df9af927cedf45d0a3f9a4c4ff8b87e7 : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bf2d272e79a64caa85e288d29ef75bfe : SUCCESS in 25m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ab83d87b1f544de38c0e9465f5df82ae : SUCCESS in 1h 08m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9e02fd70329e4d1ca6ed9fcceb8ab20c : SUCCESS in 1h 12m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57e3ba24eb946799f2f0e87cca4209f : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c6246ca31ee41eaa0ad2e297f3f19ae : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c47defe7004b4cb98e08c1bcd8331964 : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8ee6298eaef2483c91bfcd9cf2c113ef : FAILURE in 4m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01fb9d1b9c3849288c58323260e7353c : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714ac5c897e048b088d18cccb46ba449 : SUCCESS in 5m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ef7ff5b9b9ff47648b38b5a2458f876e : SUCCESS in 1m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a06abcb6e24841d0b015f758d514a70d : SUCCESS in 14m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/71e2a80e83b94ef2ac5f576d89a56cde : SUCCESS in 1h 33m 15s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6a06b58af4d5491e87face83b34aaeed : SUCCESS in 54m 05s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/04a4c95823b4443d918de7d97c02aa91 : SUCCESS in 1h 37m 56s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c83c00070480425a91c8de50e75595fb : FAILURE in 34m 53s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b828d0bb34c14b02b7592a7801fc34a4 : SUCCESS in 1h 32m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0b17be070c1e43ad96c7fb7915668a6c : SUCCESS in 16m 35s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e8c00bb7f08742ad8534273cd41101bb : SUCCESS in 27m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/91a429b333304f548f2d74e5cd8459bf : SUCCESS in 52m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b2a50bfc4a1e4ea5b1e9b799f169e3e9 : SUCCESS in 1h 34m 52s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e997b44f7ddb4489a7388bc00353e6b9 : FAILURE in 57m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/08fc00cf80a7476b9a9de4c75b25a089 : SUCCESS in 1h 17m 33s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"5ca05890944344718745492bf50b24d5e7358892","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 08:48:29.000000000","message":"Patch Set 19: Published edit on patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"5dbfd34614445fd0b05fe0be5f03864169c24778","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 08:49:24.000000000","message":"Patch Set 20: Published edit on patch set 19.","accounts_in_message":[],"_revision_number":20},{"id":"f1cd440382aa6c637e36ecd5a863346cb5500aae","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 09:30:41.000000000","message":"Patch Set 21: Published edit on patch set 20.","accounts_in_message":[],"_revision_number":21},{"id":"df4fdcf773df57320bc5574c0a28e02583cc8e88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 11:10:13.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5bf7a344fe5402fa049e2ae404c8354\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/3e1dedb7683a4feb97e9ed1ce9cf7c6e : SUCCESS in 50m 59s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b5606ea8724f47b09b0fd56b5739924a : SUCCESS in 35m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bccec4d98357455197b0846b1c507688 : SUCCESS in 1h 37m 43s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f739bbd5f124987bdd2e6180d305f16 : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2fb5c933922f464092853505eeda777e : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9116e4a01aef452787ad568908d1e1f0 : FAILURE in 6m 14s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/118b2af276254958949fbff9e9be2a50 : SUCCESS in 36m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/97cc064d111d41d1b73b564e96564e45 : SUCCESS in 54m 22s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"79ba81afc9d2f13a51a7c5e96e34128af2168304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 11:19:51.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d75667e99c045fcafaff03613d45ea0\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c4373bcfa7884d8991ab26e885fcd1f0 : SUCCESS in 14m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/a631c2fdc9794b9bb49b43e674e52d8a : SUCCESS in 56m 18s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5b47bd976a834b47901df49c0ac3bc16 : SUCCESS in 20m 26s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5bef67a3788a45c9ae85fc3828433e50 : SUCCESS in 27m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a8945e2a70f743bc98fabdec542b4fad : SUCCESS in 55m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/85ad85e99330479496f8604b58fd446d : SUCCESS in 1h 45m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626b779abdee495b91e4aeb37f5607f5 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdaafcd741274e65b036ac44b8a8e23b : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0153e1362a2467ea1e06a0cf82b9770 : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/209c543300814f788d5e61376ee85803 : FAILURE in 4m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a19b8e16d5ce487eac80eaf47376fee9 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ca3767e7ad4929ba683351c705b556 : SUCCESS in 5m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/24a2ca7d6e0c45a9a198d29c84494370 : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9c6f6f3c01dc412f84c901f65054ee66 : SUCCESS in 23m 25s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7363c61c70e74228886e1332e9a84208 : SUCCESS in 1h 12m 57s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d1e040ffe0844c928e24f8909e6bc69f : SUCCESS in 56m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1ef9e1bb73ea4a16b068f002611da847 : SUCCESS in 1h 34m 03s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/28752c4894ab40ff8b504eaa8811bbdc : FAILURE in 46m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1d8ee4d9014040efb1506a73698e213f : SUCCESS in 1h 16m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/024810ca549648b5acf83d61ac11aa06 : SUCCESS in 22m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2871b5b445234abcab9cbc2eedac3e1a : SUCCESS in 25m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/454719fb71bb412ba65e3b168aadbfe8 : SUCCESS in 1h 00m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/72ba8dbe6506457584df704fcafd84c0 : SUCCESS in 1h 33m 25s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9c294e2aef2a4073a23c2fe76b843d57 : SUCCESS in 1h 06m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/09f01e379b73466d96e57ef943d72f95 : SUCCESS in 1h 22m 27s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"6084ba411e9d2c9d21681384cf78b7aa3cea226c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 11:42:11.000000000","message":"Patch Set 22: Published edit on patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"9f02b59238c4c4b0b4a83817edbf09e504ac32d9","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 12:51:00.000000000","message":"Patch Set 23: Published edit on patch set 22.","accounts_in_message":[],"_revision_number":23},{"id":"7cc19cd4021e447b3127091c2386321e727ae931","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 13:57:20.000000000","message":"Patch Set 24: Published edit on patch set 23.","accounts_in_message":[],"_revision_number":24},{"id":"ea32f4d10fcb0ce83a5d770a711602258d57385e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 16:17:28.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4642b6cbd15440286937818d2d31e50\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d0f97faefc38441e8643a7ad9894c9c8 : SUCCESS in 43m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fb47a2505978432dbe99bb44befcbd51 : SUCCESS in 38m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3f31e042fc9d436cb9bf56402a612695 : SUCCESS in 1h 08m 17s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c246e3b3bfc4abf9698f2181852b23e : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9d458d06c1fb4d7f80eee6164869ab29 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/25b1e1c87c354ef1a6314cca0a2b0bf3 : FAILURE in 9m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1f68c1bc33c342aab70ac04a8021e9ba : SUCCESS in 39m 27s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ab19088a07094b498b1e807c7be9bfe2 : SUCCESS in 44m 04s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"d492474053d25c62ad5f4adb38be6537f8201379","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 16:49:56.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3db3d51636e4e3ea5580339eeebb61c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/df42746f08534fb2807d531329185c25 : SUCCESS in 22m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6d7ec84a150a435eba60d0277afb6a32 : SUCCESS in 57m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/28e51a7ce2f640f48946fabd41db54be : SUCCESS in 19m 19s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c963302a244f47dc843a25ecfae03f81 : SUCCESS in 22m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/85433cce11664b32814e8ac58ae074ae : SUCCESS in 55m 28s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a4c370deec6a4e359efcc31f27f2b59d : SUCCESS in 1h 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee68a0f0557e4290a6c7ee1e83254a4a : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a94066f5396429b992f5eff1ea1ce46 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0683853318b84db488bbd0cd70774501 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5333f66f4ba46fda3b2ff23126699ea : FAILURE in 4m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1aa05519eea140f280ff5474e3866c3b : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b8c6c729f414598b5f172144e3a5fcf : SUCCESS in 6m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4baed10067bb442fa840a6e67b085cb0 : SUCCESS in 2m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3bfcbc4b54fa433984e39525e64386d7 : SUCCESS in 25m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/69c26faf1d144087b907388eef7cb7de : SUCCESS in 55m 50s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/217c40e3be9d4855a801fc08f9945a61 : SUCCESS in 1h 05m 04s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b964a1c06e93472083d84218b528e58d : SUCCESS in 1h 37m 47s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/6423f34db03e4a2f98b3f078a3e9deb0 : FAILURE in 23m 12s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/11d5b9064f014a57a84e47222f87751b : FAILURE in 1h 14m 44s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0029dfb6b208408981c2f693f4098402 : SUCCESS in 21m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d5e00d364d1144ceb415f48e28e91b9b : SUCCESS in 26m 15s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/46e3f491db334766ab693cbaac708d9e : SUCCESS in 1h 05m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f45ae1c61aab4f71bbcd24b8fc2dc23a : SUCCESS in 1h 52m 47s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fe00c9aee86e4e16beb6026f86676f93 : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/1650188cd64b47ffa638c9d9f70db219 : SUCCESS in 1h 26m 22s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"c8f3f727f30bd1567755f8c5d4017f0a4139f8d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 17:43:51.000000000","message":"Patch Set 25: Published edit on patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"f8b294cf5064b998aff016f046e56f4f37640635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 19:04:08.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb5566bd4c5445bd807730a4e6b8df27\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/f274aeb913754410b12aee65380c9e71 : SUCCESS in 39m 56s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a4680f84d381497eb9217a7d9b7927b2 : SUCCESS in 40m 51s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c8a28df782bd4855859c9f85f9135acb : SUCCESS in 1h 16m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c8cd1c23a8a491ba0ce981baed8080b : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c3de9cab5e146bcbe3dabf36acee41d : SUCCESS in 3m 34s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fae1ac90ec9c409a9326c269051c4da0 : FAILURE in 6m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0a69070b8a82473db28f0f7c93d83bae : SUCCESS in 44m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e09e09d251dc4fca96f99faa11b9accd : SUCCESS in 51m 35s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"ac59bf5066080a52f5c0201f140ce9fc54b64ac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 19:38:47.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3e5f618d82d4a0b8f74868edb7fd117\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/a273462c32a94e139dd964393e3081ea : SUCCESS in 32m 50s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/8b0faaba752740a7ba3ba66d7d6b4d37 : SUCCESS in 1h 06m 07s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cb7e108600c5475193fb10711521f67b : SUCCESS in 19m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3bc2f92898f8477a82debf319ab3ad95 : SUCCESS in 34m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5c6034bba6964908877de61bfc988b0b : SUCCESS in 59m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/48ecac1053d54dd0a2185b3495b4fd85 : SUCCESS in 1h 30m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/606c4e40019546bdacf21365c9c3a787 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f99e170f3ca04e87aa8594df59df29cf : SUCCESS in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41916e93c4e24f5bbb50467aac0fef99 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a0139ed899f348b7b1f53cac48b8ace3 : FAILURE in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/973269bcb06a40c191a63c599e4d3c88 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/443b7b41b08d4cbdafa390dce2800178 : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d14120c710ae4a0db7c8ac6a4c0b6436 : SUCCESS in 2m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dd3ca2cf858842958c6f1b4e6b60345c : SUCCESS in 24m 06s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aa5e86606f86458689d58077e69af7f5 : SUCCESS in 43m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d71bb7ade1b344919479fc14255a379d : SUCCESS in 1h 05m 08s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/02776a3a3cb64c8fbf12f2d86bacbcb8 : SUCCESS in 1h 18m 01s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/75c900d5cb284911a433f57bc3820803 : FAILURE in 23m 54s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f2a671c5c54641d6be889526978cac56 : SUCCESS in 1h 38m 54s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5de41418d1ca4fd286eb7e69f928a49e : SUCCESS in 17m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/58da845e4e8044cdbfd36b124cb4d6ad : SUCCESS in 14m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fdfb0f0208014ed59b75c80e7798a8ab : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d3816c0a079342be85075a8e0bb704fa : SUCCESS in 1h 26m 59s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9157c178b1d84039a76865921e751c95 : SUCCESS in 1h 04m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/def9b22d7fff43dca339b3268a9e6a50 : SUCCESS in 1h 28m 14s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"95673251060be2c2bc62bf703db41b12aff637e3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-29 07:09:57.000000000","message":"Patch Set 26: Published edit on patch set 25.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"be0ba099dcd0d288c596c1b2edba9051e54b7e04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 08:12:24.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d75b8c661949b6a68ea71f5a245330\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a4c60ee45768444f93c68426b5a84f77 : SUCCESS in 48m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f2fd117c9ba3455cb66a7540c18bfa77 : SUCCESS in 46m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3ff8fcfd539427e8577462339a5fa91 : SUCCESS in 1h 00m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4b7a6907f7ea4c70838978265f30ae42 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/16d064fff6b945c399118ddbc4e0058d : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d7cea8092fd2456da9d41e47dae95282 : FAILURE in 9m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/58f56b71b8c74865b61f5c06dd734982 : SUCCESS in 30m 42s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/382f0dfe5ce04924a5c7d677f80b7f5f : SUCCESS in 48m 41s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"72b79a09dcefd1869ef7056fe560841be004a6f7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-29 08:35:32.000000000","message":"Patch Set 27: Published edit on patch set 26.","accounts_in_message":[],"_revision_number":27},{"id":"e7a84c4b8c3d10b57438f7da993aef99ec0efb95","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-29 09:59:16.000000000","message":"Patch Set 28: Published edit on patch set 27.","accounts_in_message":[],"_revision_number":28},{"id":"59ddb748449d1c9a9f54f346f17ab79c1c212322","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 11:33:56.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b04b9c40a4d4ae8be6cea1104ce1573\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/50e9e40eb27c48fca30e9eab03be2df3 : SUCCESS in 37m 34s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/34e6dfd72dd34fee914811284d355f77 : SUCCESS in 1h 20m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/37bc46083f474d6789011fd4d3872764 : SUCCESS in 1h 00m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4844d7b6b4ac4836ad56d25634283341 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9b0e4b1eebe49358c33c28e1e1a8220 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d8e07e0fcfa44731b6f683578fb5fd42 : FAILURE in 5m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9255d311ddb04504b2067077f3c3a1bc : SUCCESS in 32m 11s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/553f1c5fb49b433f8046ed7896351d8f : SUCCESS in 1h 31m 48s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"05ac9f1e50f765b8156f177f2333227d198252d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 11:43:28.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e41ac34f741847e79eb06f574ebcd770\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/13d0c0bc20c74108aa98711b995986d1 : SUCCESS in 22m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/1499bc897d1a4f949201a02d298c63b6 : SUCCESS in 56m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b1383b4228a34d8ba3becdfd28306a16 : SUCCESS in 10m 28s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/22b4048a3b214060934ee7b576cf6f9e : SUCCESS in 13m 56s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8de5af3bbc9543e69ec795001a1ba62d : SUCCESS in 30m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d2d1710e250f42dd914a5796fff57f86 : SUCCESS in 49m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac47b06d431d4e80b8be513e25822550 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9076a316d974affb67b4b4b366f16e4 : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e255b5d0daf4025a1aef2f45982f9d0 : SUCCESS in 1m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7da7b64fe6104a2dadc5f817ecedd9e3 : FAILURE in 5m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab7d9348aea4d909dabd47fe4c04eb8 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cecd85fcba24d68a63fa8a5b56db777 : SUCCESS in 3m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cef90c32e67e47459b8960d77b5a2b37 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7ba1e9136e7940099f432e93a954c8e6 : SUCCESS in 30m 15s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/cdfa6fac01324176a434729e07f558fc : SUCCESS in 1h 17m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/370cb97e40594969be155cc09a8b245e : SUCCESS in 32m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/c633c2377c0745738ff67a220f7916e0 : SUCCESS in 1h 37m 32s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f5d74d6d38694b62b07139eb989e21ba : FAILURE in 46m 46s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b3cd5ffa3fc04850b722f4c37591fed1 : SUCCESS in 1h 32m 11s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1ad6e6066b234b03993c1aba7f1b6e94 : SUCCESS in 23m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6955daca5fe47d09d6803d2abf35bc6 : SUCCESS in 24m 09s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/96f7d3e466574f91b595c3be38fce02c : SUCCESS in 55m 45s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/20ef9429d53f4a27b6bf3d25a3104525 : SUCCESS in 1h 27m 52s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9e9d5be46fa04a8e9fe046803068a000 : SUCCESS in 1h 06m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/96ecde58f5554c9594887ebb8cd4a319 : SUCCESS in 1h 36m 02s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"47d27481b412cd09326f9a2390669328df98c563","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 10:05:13.000000000","message":"Patch Set 29: Published edit on patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"01912af70ae193e789a978622acb3523b0072947","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 10:05:17.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":29},{"id":"c63f24c505d1effbf0fe87cef35561aae45bcd41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 12:01:54.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4f2f30815f64c50a814be325ef12ca2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/efc6cecb67ab421b81235be59eb48d61 : SUCCESS in 14m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/be6e5739d95448dcbc074595d4140946 : SUCCESS in 32m 39s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a0a5760e521e45bf80e13c8be1604f64 : SUCCESS in 9m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a84a910323984735bc53e00081816097 : SUCCESS in 13m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/91f054ded40347abbe7ac1d97a1d5bf5 : SUCCESS in 30m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/40c8f3b3e8d64de18af6df05b89897fb : SUCCESS in 48m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc83f7604cd24376a599b352e2a31805 : FAILURE in 2m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/260d1601e573414c95b256e0d959022d : SUCCESS in 1m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e9fd4a1d32e144f58f16c05fa993bced : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9f42785bd1842dda891782896951c57 : FAILURE in 3m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f20d59e731847eeaabb3da8b8f63a9a : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/175292f7f8374cc8abb3ff3692031bea : SUCCESS in 5m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e2610e77b1c4449a93a19e9e12adb041 : SUCCESS in 2m 25s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b721e8b1c322475688fbbee70ed4a16f : SUCCESS in 20m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d6a1dfe45e2b4259a07368f58d469683 : SUCCESS in 1h 03m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/85722c2f3a5449a68fc5be4de1d3e3cc : SUCCESS in 1h 00m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/409d24983b18410f811991165e0655f7 : SUCCESS in 1h 15m 26s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/b92cc785db28489fab76a04a5d2af8c9 : FAILURE in 42m 07s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/655ac0c90681453e8e79aa4b89f2da90 : FAILURE in 1h 04m 08s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2209e46e4afb4c4ebd6ec47b368aad3c : SUCCESS in 19m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8798b43939664f248c3a93029309a81b : SUCCESS in 25m 54s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d064fc58a053458c8b6f9549f7d3ec21 : SUCCESS in 1h 05m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5fdb91321f504f53b1ec2902eef8be63 : SUCCESS in 1h 19m 45s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fab23b0ae71a46da92a02c615cf7ce8e : FAILURE in 52m 15s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b25afd88f8e344b4bdc62b596eb1a3ea : FAILURE in 1h 04m 53s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"c7aab0407e4be54d8a03d675a9ab4fb80e4a8848","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 12:47:43.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd28a7e92cbf4ea1ae9b18d3581cabf5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/11eb5b1bf148402b8f341b5dafb2da9c : SUCCESS in 45m 01s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/aae2534b59974cf1a5b829a66c156375 : SUCCESS in 33m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9b90800fbf114391b33f41a4f9951e80 : SUCCESS in 2h 15m 42s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a84d8bd74d5416e9f47fa97b61b69ec : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69e19a1e11244cf7be6366c854d7bad5 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9eb5392bb2be4b82a40e122e6e571403 : FAILURE in 9m 03s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/2c65682591314695b638116cdaa1a6f4 : SUCCESS in 1h 22m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ce3a944e0bc64b6493c102358548a190 : SUCCESS in 1h 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"5764f624642227843ce408963052f047f552db1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 13:01:53.000000000","message":"Patch Set 30: Published edit on patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":30},{"id":"2d4ed7a1b8906f70fd0bae4b681b716e9ff6c114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 13:32:15.000000000","message":"Patch Set 31: Published edit on patch set 30.","accounts_in_message":[],"_revision_number":31},{"id":"7bf6801ea9d70bbf38171ae39c29bd2da89bb91e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 13:32:28.000000000","message":"Patch Set 32: Published edit on patch set 31.","accounts_in_message":[],"_revision_number":32},{"id":"40c47762c96c2c4d89dcfc14edc6a221eff36090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 13:33:14.000000000","message":"Patch Set 33: Published edit on patch set 32.","accounts_in_message":[],"_revision_number":33},{"id":"d94cc7bbdfbe9027c8999261c45d8c5ae2e4bbff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 15:18:21.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5caceed3f324b3b962a5986d0b270c5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/31e02f60c5214fb4bf919defe05ad800 : SUCCESS in 26m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/eb3a01db5b154ca19e02e9e5a30897db : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf09644520de42a5843b9043b68b6fed : SUCCESS in 20m 01s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/df75a1690fcb43d881ebf244f176e4f0 : SUCCESS in 26m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e6378d58bde54aa19b3e841923d49e3f : SUCCESS in 53m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/43dbb4bb8f614291bf444e14d6f992f7 : SUCCESS in 1h 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94a034d419914d63892c697075a9445d : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a306692920f45fe8f7736c265504ad3 : SUCCESS in 2m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f34b3691f1a48e39cb6cf3ebc28b09d : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/11e96ee8ae9f46edbaacf5c3e39fe19a : FAILURE in 4m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f959f5fb34ec4d63b2a5f45bc9f17d25 : SUCCESS in 3m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20dcf00b48f41e9b0a0c1322762c2fe : SUCCESS in 7m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/79285a4044574bb3bd0dde1e804c2726 : SUCCESS in 2m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2ba7257a6dce468183342ddcde3c5a17 : SUCCESS in 21m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5c814e60740d48bd892a75917b7de7be : SUCCESS in 1h 10m 30s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ec6cdab8e10f449598f494dc30f84bed : SUCCESS in 1h 05m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f978bf46e91e4183a48ecef1fef2f5df : SUCCESS in 1h 39m 13s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e38b995596a14a6fae8738bca5b6a227 : FAILURE in 45m 31s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a3d4cd45c62d498681c28d3029250397 : FAILURE in 1h 09m 36s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4e13b678ce64acd9cb9b8aeab1416c7 : SUCCESS in 11m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b839dda5e3eb4b3e9a91e55d8b9e971f : SUCCESS in 24m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/a5de5623149c495c9f618e370da348df : SUCCESS in 30m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3d198dc0584d4a03ac4385684dc56857 : SUCCESS in 1h 31m 05s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/44761d30951a40989ce4f16ca079f43a : FAILURE in 46m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/005f6322baac411980973506f7d7d666 : FAILURE in 1h 05m 05s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"4938e5ece228b8278c639ba531270b716891de8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 15:36:07.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06a938ccf1a84d04b90d37b4dbdaf586\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/05ca39304d344807b3470c409f2bd743 : SUCCESS in 49m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/653599f13f7c4a5b9150882b0f2e68f6 : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/95e7f8ea7fbd42c99b5921a6c49f3b38 : SUCCESS in 1h 59m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/395a074ade1e4ef3a55b434250b1f602 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e06b7fedcd4b4a67a183d1fb8b3f32a1 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6eb121be68004f7aa3aaed1e96d1de5e : FAILURE in 10m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/187f2a8cd2d3483892beafee83ef0d04 : SUCCESS in 1h 01m 55s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5c71db7849a848d6887e9572bb8a0d42 : SUCCESS in 1h 03m 11s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"b3f7e5909d017614f1c901bb2f2a90b18ccb323a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 17:12:26.000000000","message":"Patch Set 34: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"78afa3990480b1329134507c08f1f97644ad8d61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-30 17:16:55.000000000","message":"Patch Set 35: Published edit on patch set 34.","accounts_in_message":[],"_revision_number":35},{"id":"d27fc000638959f0873a436d458c18011618c0d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:01:33.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4158079bb854554825dad396297ff58\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e67d4449f48948b7856f0a571ad18415 : SUCCESS in 26m 47s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/62dc6dfe499f4d3fb3cf2ff916400087 : SUCCESS in 1h 04m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d1d21bd299f44aa9a3ac829dce40a95 : SUCCESS in 18m 42s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2729c678f3934e13aef6392c45c7f27b : SUCCESS in 26m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b3c4e43dc627436fb96e5889a9c11dc6 : SUCCESS in 1h 08m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2bd16f18309346ce85a9c9c6245ca306 : SUCCESS in 1h 27m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58c573966ed548ecb3d4bf05730a2ce4 : SUCCESS in 1m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e0e32ce60eb46c499023dfd14585b87 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/509c17ad0a4f417c89261969b2e5bfb8 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dea84209cb64468db1ec0175b74bdd20 : FAILURE in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f9a85f2df484e4281ffc61756cab8cf : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/043b0985c25e4deeb5984d89dcd8f09c : SUCCESS in 6m 38s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/9be6a7dd99dc45ea89d86f47a76a843d : SUCCESS in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/68564655b3884b728bb64ac9307a76b9 : SUCCESS in 36m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1db55c5ee954a0097eb55c2b746771f : SUCCESS in 1h 11m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/92f0507e0f1c47e2b5ac75aff8a767d7 : SUCCESS in 1h 08m 20s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6d6a834029bc4687838345850405e92b : SUCCESS in 1h 25m 54s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/d52f8429b1714915a9e2cb95abde62ca : FAILURE in 47m 38s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/41466625b37845429495aaf3c4ac63a5 : FAILURE in 1h 09m 00s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0e3c51756a644301a4a3318f7173014c : SUCCESS in 23m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0a70e06778d64469a0fd2d672bdf0741 : SUCCESS in 28m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/740eabd76c4b44e7be3d59077a8be387 : SUCCESS in 1h 05m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9780673d10994dd1a49b663d63757203 : SUCCESS in 1h 28m 20s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/0f7dc85fafce47dd8fb771af7ef39dac : FAILURE in 1h 14m 21s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/4ef7a0f1c2144da0824a124988b3405f : FAILURE in 1h 09m 09s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"ba3653bf80b64114b5776fb5ed1abd5b56f604f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:23:02.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd23bf990da444129f1746cdadcdd85a\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/047084ddc81e453180fd62e6ac142710 : SUCCESS in 47m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6e8ccebcefe04ef0a501491a8c57d827 : SUCCESS in 44m 23s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1b5bd89b0844a21b967f5354c02f315 : SUCCESS in 2h 04m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/412c7f4f4553475d88d65cd320e1afce : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7090fd5e726e4bd2af72aad615dca4a8 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6fe229577132415eb9ad03826bfe15ae : FAILURE in 9m 38s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/23a047029b464408aa8d84d07dff7190 : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68a16baab7654ae7b9f095c5db313e23 : SUCCESS in 1h 12m 32s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"8009a3340ac1244e2870442b9cd37fc7da609901","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-02 09:25:52.000000000","message":"Patch Set 35:\n\n(4 comments)","accounts_in_message":[],"_revision_number":35},{"id":"168b6ea45fb4d18a6318954c26958cec76dc3f7a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-03 09:36:11.000000000","message":"Patch Set 35: Backport-Candidate-2 Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"782d201355f77eddd982513598b70250511730e9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 07:23:54.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"45abd73141c50b08c60ac75f79c20ca9af3f6a8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 07:30:05.000000000","message":"Patch Set 36: Published edit on patch set 35.\n\nCopied Votes:\n* Backport-Candidate-2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"5cf95945327562dfaf59cce76726b10047d4f13b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 07:31:57.000000000","message":"Patch Set 36:\n\n(2 comments)","accounts_in_message":[],"_revision_number":36},{"id":"d226dfd2a297602f5f3ce59f98bdb351b6a8bbb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 09:16:40.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f13d7cbd8924b29a324c86fde4e7497\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/701425209ba44b138dca7da8a8c75180 : SUCCESS in 24m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/026ed14ff0804662853b611b98de507d : SUCCESS in 1h 07m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e10aacbc7217426097ec3cfded869e75 : SUCCESS in 9m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2c96984febb64d36810b8a5fb59ab0ed : SUCCESS in 14m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c20cb61a7f6b404c80b7b776d9a39488 : SUCCESS in 30m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e845a693088e4f2ba29a90c7059c4836 : SUCCESS in 1h 29m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dc500eddc0248b8b59b272070bf3be8 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f67122fe2bfa4f5fb4af1af6ac5b9ecc : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77db084e3a7241598409b9c5d84d20e4 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/20d2eaa1e3284666b4bfa5b2163d790d : FAILURE in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f882e834bc7c473295dcffbf6a30e26b : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baac71cd205e467a9623c124835591a9 : SUCCESS in 6m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b1939fecbd0c4f129ccca45f42c8052e : SUCCESS in 2m 54s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/51f2628ece884c9e92b44770b226c343 : SUCCESS in 24m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/beedac13b88c47bb92484d377470b885 : SUCCESS in 1h 35m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/4ab82e9070144fda83d279013377240b : SUCCESS in 1h 05m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5850c99cfb674d7ea96e3b7c15b8f018 : SUCCESS in 1h 37m 58s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e55c9b395e4049b5bbe2c0396f5357cd : FAILURE in 52m 02s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/bd937d0da2464a809771e3d2f7324a7d : FAILURE in 1h 10m 42s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/561f790cb99940fa9fc1f363d7dcb2bf : SUCCESS in 24m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c8f20fb72ff14c82993ad2e3951f9301 : SUCCESS in 24m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0dc279b0680c44d3845764af7d79149f : SUCCESS in 1h 07m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4266eafad1574170aa04ab092fa7476a : SUCCESS in 1h 42m 24s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f13ff05987ec479e918894407a0a2338 : FAILURE in 51m 14s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/30360be02b2840e79fec140ad1ef8a15 : FAILURE in 1h 10m 49s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"64bc6b3ceb0ab8ee3b5a89800ab7d6133e57e34c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 09:51:09.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdca9b7ae9424f539e4a6d2132eaf014\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/34f812e3d9d44cedb2c33e98e9e85971 : SUCCESS in 1h 34m 45s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/18d810530a4f412399b745f06c05a28f : SUCCESS in 1h 18m 43s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9ad3745117ca44ada05ebe97057476d2 : SUCCESS in 2h 17m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5edf8f4f12d40efaf0e3e30e241fd46 : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fb62e4cfea6f42a3951c944b4bac57f9 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8f209588e0d64881a159dcc57c41c357 : FAILURE in 7m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/13a0e013072d45af8c80aab5b8032f2b : SUCCESS in 1h 26m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/d09bff93a8a04ac592a4d0d0b9e26227 : SUCCESS in 31m 58s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"45159442d1a98712863a1d7ace80412adef7a5c5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 11:45:49.000000000","message":"Patch Set 36: Backport-Candidate-2 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"da04e456af28e0dcb3db7e475555ccc15d185c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 11:45:59.000000000","message":"Patch Set 36: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":36},{"id":"a00007e5a6db8cb37d7c300605b3d1069413169d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:20:54.000000000","message":"Patch Set 36: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3cba97bb43c44c6e84387d50fe6b40a0\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ef9fdba248ae46eaa573032ff7dbe3f6 : SUCCESS in 15m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5c2bc04027b24cd6a7590497f3c1acdf : SUCCESS in 20m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/119aad714af94271bd87a34aab05a952 : SUCCESS in 1h 05m 06s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/543511d2c5e04cf7bd610519eb2fb209 : SUCCESS in 1h 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00a459568a4b4b1394b2aee491b2da69 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d695ba27bb6b48c09fa912387a97d0fa : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ce5a39b0fda460f9e5f84ac91703fa4 : SUCCESS in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ab1794ae3147ada785b527cd730a5d : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/05406eaa404d410dbb51d0307feb4981 : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3a77e76aa5814ca8ad9671c9d309ea3c : SUCCESS in 23m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c7fa2b5c2c9d4725ab5f4e8fde72ea97 : SUCCESS in 1h 15m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/3bb3cb90f5fa4f9193d7b8897093fb30 : SUCCESS in 1h 07m 11s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0890a99d854c46a3be05bfc9085dddab : SUCCESS in 1h 23m 40s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9c586758fc6d4bf4bca06db3fd3c7a84 : SUCCESS in 21m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2b4c95ab5f274620ab16f25599131be1 : SUCCESS in 29m 34s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f4f7b1175ebe451e95e5c3e2d9a74e45 : SUCCESS in 1h 07m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0d92e23a9cbc4c329320880c5ab010d8 : FAILURE in 54m 43s","accounts_in_message":[],"_revision_number":36},{"id":"d3204237589d9b8dc231d180039979037e10fe8d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 13:28:02.000000000","message":"Patch Set 36:\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"1e7df0048c6659fc81622512054785a01ffe0a8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:39:35.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3b7a677b564fbc9b8cd96576a6ec85\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/23183e6919504f9a8f04300784b4a9ff : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/136bf2542c6b482bb12f70e96b7f60ff : SUCCESS in 1h 01m 58s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7aeae6c3821e441d8ac5dfdec5e5ea2a : SUCCESS in 1h 09m 25s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3a50435c21c842028ae980c16bb8313b : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7f1ac9a293184148b15896f09fc3f491 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/15488b9741784e5484fc74bd3a40b029 : FAILURE in 7m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/aef20edaecaa4e24890655b3a610d167 : SUCCESS in 1h 02m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/08baa3b6456f4b63939b586c7964be1c : SUCCESS in 43m 32s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"166880f187e2b343b516e5baf3c67dfa80cdc283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:11:02.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f955949c65dd4515a5c3d4be38dcbe97\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9ddb16a61de74e61a8746d790e614270 : SUCCESS in 26m 59s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dbfc1c22207148a7834c1e2780e8f2ce : SUCCESS in 1h 07m 37s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a09f711013dd4695ae3f3627880298e6 : SUCCESS in 9m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cf3e452e29904c16899c2fc6da7dd418 : SUCCESS in 28m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b75ded5069af4f9393bed3604675edf4 : SUCCESS in 1h 18m 45s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7094461011ef4ef588dd3aa7f474bb7e : SUCCESS in 46m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c19824283d4748bca4138f0afe5b5292 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f26e13744a24f43a4a84ad98164ece6 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97d65969c0f84fe58a02d9b7c03de3e3 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f90406bce983462aad238b88a9985ef8 : FAILURE in 5m 06s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/405f08a9e13a4ee69404bfec5072f69f : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c41b6328ea2412d9a520e09ccb5ce13 : SUCCESS in 6m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/04b6c87278b5410480ec95694ddd9eda : SUCCESS in 3m 05s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/db994affed1f481aa9288bdfba0c83df : SUCCESS in 24m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/100ac16854fc40558d8a9b846009e527 : SUCCESS in 1h 17m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/861e598ee76140f08c291f8da08e76a7 : SUCCESS in 1h 06m 25s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e3548ea19f104b2491ce53292135cf0d : SUCCESS in 1h 28m 58s\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2f4e2e6d015b49cd9457ce07f05c58a9 : FAILURE in 45m 56s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/1f0acd84215745b1bd95e5bf6c244724 : FAILURE in 1h 06m 16s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a162b2ecfd4840a9992c093bf7a3dc22 : SUCCESS in 21m 39s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5c78dcb376b944ed8fd8095c3aca4e02 : SUCCESS in 26m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/0a97b54e44e146ef9fdb46d940835edc : SUCCESS in 1h 10m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7f6df54242504252a53c60e870b9e0a2 : SUCCESS in 1h 19m 02s\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/964724500af94203aee3cda55c32a34d : FAILURE in 54m 31s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/9819cfde36934dbb88f4fc073e480e11 : FAILURE in 1h 08m 05s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"ef16595a3331a6b78bd90c587aaa5d2c792fb40e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:11:13.000000000","message":"Patch Set 36: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":36},{"id":"f25593f7e64ae688a239b1543b1c726f7a409500","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:18:13.000000000","message":"Patch Set 36: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0bf6c0d34b44649a9665a3e2a9cef6\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/de4ca67e896c411286d95d68ccff6b70 : SUCCESS in 20m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a40e4cf3865a4d4bae2c91c64ec9a7a6 : SUCCESS in 28m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/21ddef66b3604b68ab476fee7fad4cd8 : SUCCESS in 1h 06m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4270c642cd5f4883affa4417260d1fd0 : SUCCESS in 1h 29m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36c15940a22c4ac9b7cb3baa617c7c1e : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/46f865d0b54a42208002a9946c58f090 : SUCCESS in 3m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/792c2eb380484ff2aa7d8638e169d152 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f9c8f6284dc4fd5ae03a2cbaf9bdc06 : SUCCESS in 6m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ac1cc205b85a4525aa883d9eb7e8dcad : SUCCESS in 2m 38s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4de1bd9b25574f1ab2476f8c91401ac1 : SUCCESS in 27m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/db052be88d4b40d78c6651491272b05d : SUCCESS in 1h 18m 20s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/702effdb1e714b78a865cbc77e1581ca : SUCCESS in 1h 00m 28s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ba8bd5cf4e3147e2a894ed08b055600f : SUCCESS in 1h 28m 32s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aef42c9b2bfe4d4691553405e65e2c66 : SUCCESS in 25m 42s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/946a344a16994b21883a6d3149ebe29a : SUCCESS in 32m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/672e8112ad1c4ce997a96912ea384e51 : SUCCESS in 1h 07m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/ad2a13fe1939451bb233d79e633ff6ec : SUCCESS in 1h 25m 45s","accounts_in_message":[],"_revision_number":36},{"id":"70155bec4abc545d69e7c760f1cc2345c4f2082e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:18:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":36},{"id":"fbbc360ae622fa1ae701ef2e3a601d0ddecabe58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:19:03.000000000","message":"Patch Set 36:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/704e7ba16ad74bb48ea5a5595fd49c19\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16a8e006f4b4a7493bdcd925e9d5045 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":36}],"current_revision_number":36,"current_revision":"a101add561b54869d97b6b308ba7d3195d372aae","revisions":{"d5a9ad3561e2abd33a5e1789fb1b2c323c7ae167":{"kind":"REWORK","_number":1,"created":"2025-04-02 21:43:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/1"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:48.000000000","tz":780},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5a9ad3561e2abd33a5e1789fb1b2c323c7ae167"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d5a9ad3561e2abd33a5e1789fb1b2c323c7ae167"}]},"branch":"refs/heads/master"},"1c6ef9431c6dc9c4f1a81878e1303d9a8118b334":{"kind":"REWORK","_number":2,"created":"2025-04-03 00:07:33.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/2"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-03 00:07:33.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c6ef9431c6dc9c4f1a81878e1303d9a8118b334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1c6ef9431c6dc9c4f1a81878e1303d9a8118b334"}]},"branch":"refs/heads/master"},"931e3c95882fbacd3cc765ece28f62c4cf00c45f":{"kind":"REWORK","_number":3,"created":"2025-04-03 19:26:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/3"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-03 19:26:36.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/931e3c95882fbacd3cc765ece28f62c4cf00c45f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/931e3c95882fbacd3cc765ece28f62c4cf00c45f"}]},"branch":"refs/heads/master"},"6a5810d0527ea787566eb923650932467aec985f":{"kind":"REWORK","_number":4,"created":"2025-04-03 22:05:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/4"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-03 22:05:16.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a5810d0527ea787566eb923650932467aec985f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6a5810d0527ea787566eb923650932467aec985f"}]},"branch":"refs/heads/master"},"0a3d779ad4138619cbdebec415063508d2283046":{"kind":"REWORK","_number":5,"created":"2025-04-03 23:13:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/5"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-03 23:13:42.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a3d779ad4138619cbdebec415063508d2283046"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0a3d779ad4138619cbdebec415063508d2283046"}]},"branch":"refs/heads/master"},"baa39b3a4e1f802f4489f8161a1d5d4336d3352a":{"kind":"REWORK","_number":6,"created":"2025-04-09 03:12:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/6"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-09 03:11:57.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa39b3a4e1f802f4489f8161a1d5d4336d3352a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/baa39b3a4e1f802f4489f8161a1d5d4336d3352a"}]},"branch":"refs/heads/master"},"4457c05b800d4af152978d6e5848852a78368e82":{"kind":"REWORK","_number":7,"created":"2025-04-09 03:13:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/7"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-09 03:13:31.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4457c05b800d4af152978d6e5848852a78368e82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4457c05b800d4af152978d6e5848852a78368e82"}]},"branch":"refs/heads/master"},"45b5d7dcb3ee7359276048f19cebf10a09e2dd7b":{"kind":"REWORK","_number":8,"created":"2025-04-09 04:21:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/8"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-09 04:21:03.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45b5d7dcb3ee7359276048f19cebf10a09e2dd7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/45b5d7dcb3ee7359276048f19cebf10a09e2dd7b"}]},"branch":"refs/heads/master"},"bbd399dc7f029bb666ce35ed6645f8832d01aaea":{"kind":"REWORK","_number":9,"created":"2025-04-22 09:14:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/9"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 09:14:14.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbd399dc7f029bb666ce35ed6645f8832d01aaea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbd399dc7f029bb666ce35ed6645f8832d01aaea"}]},"branch":"refs/heads/master"},"06173a42433e68ab0c37307deab1086c0c394e43":{"kind":"REWORK","_number":10,"created":"2025-04-22 11:49:17.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/10"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 11:49:16.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06173a42433e68ab0c37307deab1086c0c394e43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06173a42433e68ab0c37307deab1086c0c394e43"}]},"branch":"refs/heads/master"},"647192674031485264e907f721faa2e0e08470ce":{"kind":"REWORK","_number":11,"created":"2025-04-22 15:05:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/11"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 15:05:20.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/647192674031485264e907f721faa2e0e08470ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/647192674031485264e907f721faa2e0e08470ce"}]},"branch":"refs/heads/master"},"9a3e4fbe407efeee39918878cfd8046c2b637619":{"kind":"REWORK","_number":12,"created":"2025-04-22 16:21:23.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/12"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 16:21:20.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3e4fbe407efeee39918878cfd8046c2b637619"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a3e4fbe407efeee39918878cfd8046c2b637619"}]},"branch":"refs/heads/master"},"3a4e02c19daf76005fe374dbf90707644882926d":{"kind":"REWORK","_number":13,"created":"2025-04-22 16:24:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/13"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 16:24:22.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a4e02c19daf76005fe374dbf90707644882926d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3a4e02c19daf76005fe374dbf90707644882926d"}]},"branch":"refs/heads/master"},"bff0f48f1fd92a48ad58d67a9de2ea97bf6094b3":{"kind":"REWORK","_number":14,"created":"2025-04-22 17:34:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/14"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-22 17:34:01.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bff0f48f1fd92a48ad58d67a9de2ea97bf6094b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bff0f48f1fd92a48ad58d67a9de2ea97bf6094b3"}]},"branch":"refs/heads/master"},"7f15d2d99f483cf2af9c77b6950eacc62f63e702":{"kind":"REWORK","_number":15,"created":"2025-04-23 09:07:09.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/15"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-23 09:07:09.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f15d2d99f483cf2af9c77b6950eacc62f63e702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f15d2d99f483cf2af9c77b6950eacc62f63e702"}]},"branch":"refs/heads/master"},"22fea60762c038ebb5e03d4f6b9feef3a32d65ca":{"kind":"REWORK","_number":16,"created":"2025-04-23 11:23:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/16"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-23 11:23:08.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22fea60762c038ebb5e03d4f6b9feef3a32d65ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/22fea60762c038ebb5e03d4f6b9feef3a32d65ca"}]},"branch":"refs/heads/master"},"55c69b85257b5f1bf76294232d2c3d4b56b99e0f":{"kind":"REWORK","_number":17,"created":"2025-04-23 17:40:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/17"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-23 17:40:35.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55c69b85257b5f1bf76294232d2c3d4b56b99e0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/55c69b85257b5f1bf76294232d2c3d4b56b99e0f"}]},"branch":"refs/heads/master"},"eb94e3a7ad2296db590c6a0b0321a6f2c223bb61":{"kind":"REWORK","_number":18,"created":"2025-04-23 17:43:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/18"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-23 17:43:36.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb94e3a7ad2296db590c6a0b0321a6f2c223bb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb94e3a7ad2296db590c6a0b0321a6f2c223bb61"}]},"branch":"refs/heads/master"},"9834ad96a9d9e3d3b71191ce52060cb101276924":{"kind":"REWORK","_number":19,"created":"2025-04-28 08:48:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/19"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 08:48:29.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9834ad96a9d9e3d3b71191ce52060cb101276924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9834ad96a9d9e3d3b71191ce52060cb101276924"}]},"branch":"refs/heads/master"},"9b96136d8dc01fd94df37e80a8f1b48e706e8b6c":{"kind":"REWORK","_number":20,"created":"2025-04-28 08:49:24.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/20"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 08:49:24.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b96136d8dc01fd94df37e80a8f1b48e706e8b6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9b96136d8dc01fd94df37e80a8f1b48e706e8b6c"}]},"branch":"refs/heads/master"},"bbccc35c2bf47502907371c7c9fe553fb49b5ebb":{"kind":"REWORK","_number":21,"created":"2025-04-28 09:30:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/21"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 09:30:40.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbccc35c2bf47502907371c7c9fe553fb49b5ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bbccc35c2bf47502907371c7c9fe553fb49b5ebb"}]},"branch":"refs/heads/master"},"d3f67b079dbf19392d9864522ec1a4216ee60bc2":{"kind":"REWORK","_number":22,"created":"2025-04-28 11:42:11.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/22"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 11:42:11.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3f67b079dbf19392d9864522ec1a4216ee60bc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d3f67b079dbf19392d9864522ec1a4216ee60bc2"}]},"branch":"refs/heads/master"},"9a17852ae6535f75c7a6e79fa8bc3d7578c5df1d":{"kind":"REWORK","_number":23,"created":"2025-04-28 12:51:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/23"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 12:50:57.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a17852ae6535f75c7a6e79fa8bc3d7578c5df1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/9a17852ae6535f75c7a6e79fa8bc3d7578c5df1d"}]},"branch":"refs/heads/master"},"651ca575b52b4c29811ebf36556255fad409a7f6":{"kind":"REWORK","_number":24,"created":"2025-04-28 13:57:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/24"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 13:57:20.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/651ca575b52b4c29811ebf36556255fad409a7f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/651ca575b52b4c29811ebf36556255fad409a7f6"}]},"branch":"refs/heads/master"},"fe12c55f905f6e739e738ee6c2aa4607ec5d3e1a":{"kind":"REWORK","_number":25,"created":"2025-04-28 17:43:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/25"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-28 17:43:51.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe12c55f905f6e739e738ee6c2aa4607ec5d3e1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe12c55f905f6e739e738ee6c2aa4607ec5d3e1a"}]},"branch":"refs/heads/master"},"efb561f2803af0393e381b81ebfde86b5005b039":{"kind":"REWORK","_number":26,"created":"2025-04-29 07:09:57.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/26"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-29 07:09:57.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb561f2803af0393e381b81ebfde86b5005b039"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb561f2803af0393e381b81ebfde86b5005b039"}]},"branch":"refs/heads/master"},"76cdd378c44afe9a3b8a1b7f07b86dfc84def24a":{"kind":"REWORK","_number":27,"created":"2025-04-29 08:35:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/27"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-29 08:35:32.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76cdd378c44afe9a3b8a1b7f07b86dfc84def24a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/76cdd378c44afe9a3b8a1b7f07b86dfc84def24a"}]},"branch":"refs/heads/master"},"8c8a9457636da360214990ed0a555889369be206":{"kind":"REWORK","_number":28,"created":"2025-04-29 09:59:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/28"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-29 09:59:16.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c8a9457636da360214990ed0a555889369be206"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8c8a9457636da360214990ed0a555889369be206"}]},"branch":"refs/heads/master"},"3ec93cd4b41b33bb743466e85e84a0a8f231b30b":{"kind":"REWORK","_number":29,"created":"2025-04-30 10:05:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/29"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 10:05:13.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ec93cd4b41b33bb743466e85e84a0a8f231b30b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ec93cd4b41b33bb743466e85e84a0a8f231b30b"}]},"branch":"refs/heads/master"},"513e87821806fa102c7de208de7c9e34d7676719":{"kind":"REWORK","_number":30,"created":"2025-04-30 13:01:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/30"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 13:01:53.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/513e87821806fa102c7de208de7c9e34d7676719"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/513e87821806fa102c7de208de7c9e34d7676719"}]},"branch":"refs/heads/master"},"42b14827b2bfb9aadce560539fcb118870b7c2b1":{"kind":"REWORK","_number":31,"created":"2025-04-30 13:32:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/31"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 13:32:11.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42b14827b2bfb9aadce560539fcb118870b7c2b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42b14827b2bfb9aadce560539fcb118870b7c2b1"}]},"branch":"refs/heads/master"},"b116cf6810893b368c17a7620ec3fe0ca8114201":{"kind":"REWORK","_number":32,"created":"2025-04-30 13:32:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/32"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 13:32:25.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b116cf6810893b368c17a7620ec3fe0ca8114201"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b116cf6810893b368c17a7620ec3fe0ca8114201"}]},"branch":"refs/heads/master"},"87bafd809ce7c8f22cbe5710e02acb659a5cde09":{"kind":"REWORK","_number":33,"created":"2025-04-30 13:33:14.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/33"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 13:33:09.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87bafd809ce7c8f22cbe5710e02acb659a5cde09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87bafd809ce7c8f22cbe5710e02acb659a5cde09"}]},"branch":"refs/heads/master"},"efb11ec7960c882264fd3bb55786bb8a67f7bc62":{"kind":"NO_CODE_CHANGE","_number":34,"created":"2025-04-30 17:12:26.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/34"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 17:12:26.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nThese scripts have been based on TripleO version of the\nsame feature.\n\nUnfortunately due to the fact that Neutron/python is\nstripping env vars when it\u0027s running dnsmasq/haproxy/etc\nprocesses - we need to do the ,,config\u0027\u0027 handling in\ncopy-wrappers script.\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb11ec7960c882264fd3bb55786bb8a67f7bc62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/efb11ec7960c882264fd3bb55786bb8a67f7bc62"}]},"branch":"refs/heads/master","description":"Edit commit message"},"bad208fe35b5df05b85b1ef92aeda0f26530e088":{"kind":"REWORK","_number":35,"created":"2025-04-30 17:16:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/35"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-30 17:16:30.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nThese scripts have been based on TripleO version of the\nsame feature.\n\nUnfortunately due to the fact that Neutron/python is\nstripping env vars when it\u0027s running dnsmasq/haproxy/etc\nprocesses - we need to do the ,,config\u0027\u0027 handling in\ncopy-wrappers script.\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bad208fe35b5df05b85b1ef92aeda0f26530e088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bad208fe35b5df05b85b1ef92aeda0f26530e088"}]},"branch":"refs/heads/master"},"a101add561b54869d97b6b308ba7d3195d372aae":{"kind":"REWORK","_number":36,"created":"2025-05-05 07:30:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/15/946215/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/15/946215/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/15/946215/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/15/946215/36"}}},"commit":{"parents":[{"commit":"72e0d38bff339916d803ca58998fa869258c6a74","subject":"Merge \"Add libnetfilter-log to l3 agent container\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/72e0d38bff339916d803ca58998fa869258c6a74"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-04-02 21:43:44.000000000","tz":780},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-05 07:29:58.000000000","tz":0},"subject":"neutron: Add agents wrappers in the neutron-base image","message":"neutron: Add agents wrappers in the neutron-base image\n\nThese scripts have been based on TripleO version of the\nsame feature.\n\nUnfortunately due to the fact that Neutron/python is\nstripping env vars when it\u0027s running dnsmasq/haproxy/etc\nprocesses - we need to do the ,,config\u0027\u0027 handling in\ncopy-wrappers script.\n\nChange-Id: I679ac66186d457ae53f89469fe22c5662089019a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a101add561b54869d97b6b308ba7d3195d372aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a101add561b54869d97b6b308ba7d3195d372aae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
