)]}'
{"id":"openstack%2Fkolla~946659","triplet_id":"openstack%2Fkolla~master~Ie19fcb441a2e54a60762e5c8483487b713a29ddd","project":"openstack/kolla","branch":"master","topic":"el-aarch64-ipxe-snponly","attention_set":{},"removed_from_attention_set":{"34212":{"account":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"last_update":"2025-04-28 11:29:56.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-17 19:26:33.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-04-16 13:59:08.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"Ie19fcb441a2e54a60762e5c8483487b713a29ddd","subject":"Fix preparation of /tftpboot for ironic-pxe","status":"MERGED","created":"2025-04-08 16:28:03.000000000","updated":"2025-04-28 11:30:49.000000000","submitted":"2025-04-28 11:29:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946659-el-aarch64-ipxe-snponly","meta_rev_id":"62dffcd3c55ce9863a3071ae2470c149acb5f8b3","_number":946659,"virtual_id_number":946659,"owner":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-28 11:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-04-22 17:09:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-04-17 19:26:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-16 13:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-04-22 17:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-04-22 17:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-16 14:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 16:28:48.000000000","updated_by":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-04-08 17:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-08 18:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-08 19:35:29.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-04-22 17:09:39.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"dafedc8adcc95377dd98b49b25d7e8548c027469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-04-08 16:28:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4dffaf55f33cae377b83c560883c61ed267051f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 17:39:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ba6e0f1e9954b7384cd6385206c88dd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/fd8dc07a955a4d3fa96bb98257ef060f : SUCCESS in 28m 31s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/04825fa20d434286b637da19410cc71c : SUCCESS in 39m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/383687af274d41ebb14e7257c245c941 : SUCCESS in 1h 09m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/95328be87b5b4afdb77650b2ae821e5c : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb6f3425ce5043cea328a96506cd7d58 : SUCCESS in 4m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22d2ed74882d4a5cba8dfe3bb96e77ea : FAILURE in 7m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e87d1d9f4bc04d32b19a65d092471cc2 : SUCCESS in 36m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/590dccfc7a774d6d85fe24dbf0458905 : SUCCESS in 37m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"30933901f9668c032d57c63e6d6ad6205fc68102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 18:12:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66c0c8a6d16040ceae716be18774741c\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/e1a51eb7195f4c46a056f584dd77886a : SUCCESS in 26m 48s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3700d7cfa7f0476aab3302b70ba8a546 : SUCCESS in 44m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e13dbcd5dcc3477d908f0bee25c4ff17 : SUCCESS in 10m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/47d8bc322ecd497d90b1d73a27b44c4e : SUCCESS in 28m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/467f584a3c5c4a0d84f7d9bd88c9f9f8 : SUCCESS in 51m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6d108da3e25b4b9caf76a424988de078 : SUCCESS in 56m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/390ef7469d904d7e8d2336924b4ac4be : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86bbd80e72c7480a80599a9fa7b0b08b : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b31bc51ce1b34b4cb35f98b7db04d83d : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e01f58d59b4248778fb41e60d70c3d6e : FAILURE in 3m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6e181e5d7d544b49a07d8bb644cde78 : SUCCESS in 1m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acb5ae2099e4cfeab77ec2aa4998434 : SUCCESS in 5m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b4e99744e0645a79132e8bdc59bd375 : SUCCESS in 3m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b00159c897ca4b90a6197608a327db7b : SUCCESS in 2m 15s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9fe7ce37480e403486bf447ffb197f4d : SUCCESS in 12m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e7a7e6d7a6404858b5455a5027a46fec : SUCCESS in 58m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/05ea9ecd839940beb30c385689110b92 : SUCCESS in 52m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/5c2f3b67f23a4117885a6ae9231e8071 : SUCCESS in 1h 38m 43s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/b538a36b0e914269b61849038b452330 : SUCCESS in 58m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fb96517889a24a19b2c9415fe444636b : SUCCESS in 17m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/11c475df98c940b1a8e0b6300b7aabc1 : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f2f442c9cfad48c881be1d589a187096 : SUCCESS in 1h 04m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e0982b3d39034918a98266f6812c9077 : SUCCESS in 1h 28m 14s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/89b370487bef4785bba0339bba2f577c : SUCCESS in 1h 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e1d308e61b9833f2dd6ae1b3b91cf99510c53bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"date":"2025-04-08 20:04:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"434bafe880ae728e04402fd4f043bc1f11da7a84","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 21:17:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec893a3aeb8a46adb58ae83444d8af97\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/59782de5fa9e4083bb9c953f85b4069b : SUCCESS in 36m 07s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2f5c50235a0942339aefea562395bebb : SUCCESS in 34m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/acf80a19ae144de58d216e3036fc9ebc : SUCCESS in 1h 11m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/af238e2379104b7b87facdf5fdaf2810 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/411477c2c2cd49ec80476ad6a82bb5e3 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/aab4c2813a7f47a1904b6e7dde675c15 : FAILURE in 7m 30s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/15f16c91e7804097bbfc897648401ea5 : SUCCESS in 32m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4909a0bad3b14c73a794b16094e44b9a : SUCCESS in 36m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6f7de7d5d65414186e8baa7035dd172ee5f721ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 21:42:55.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7c0e3b36764fc7ac9798b6193651e5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3c26da1ade2b4d028c3afe447c182a45 : SUCCESS in 21m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/996a238375334ee2a3b42e9eae259707 : SUCCESS in 54m 50s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/0b2cc01cae05458d9ede171de9d2152a : SUCCESS in 16m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7a6385c39dd344ba96c6e97f8102b1ea : SUCCESS in 15m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b74ca7ea236b4805a192f230f245ce56 : SUCCESS in 51m 19s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/31ae15da351643408acd70465a6bbf18 : SUCCESS in 1h 24m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f45ce24f784eac9c924bd458e7d1c2 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01cbf58242d245bba90263c03fdbe404 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aa39383b82f4225b6cd85bc222dc386 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac2c27dfe9ef45289c0ec55ced60f385 : FAILURE in 6m 41s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2dba8b61bd0247e18c3f1124caece27b : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a49bdf1ec6d044c08f38398209177fb5 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46dc65063d504a808e2d64754cef760a : SUCCESS in 3m 06s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eb9be8a7eb0f49b2815aab0b5b6dcd1c : SUCCESS in 3m 26s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c652cfb4cf114711b3fd98771894452d : SUCCESS in 22m 40s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/0c7b81e3beae4905bef1c945ca1897d0 : SUCCESS in 1h 16m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/753e9619e1b14ac1b399b696c1549bd9 : SUCCESS in 1h 12m 55s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8a78b2e2ecf54c2089bd1f3747acb18e : SUCCESS in 1h 37m 55s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/8f974fdeb4b44cdc86563e95dc327c5e : SUCCESS in 1h 12m 18s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ff671e72d8234abba84965ef95c98c7b : SUCCESS in 18m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/22772556e4564c8090f2499b47e37111 : SUCCESS in 29m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aa976a425844406eb260ed9c1539cb2e : SUCCESS in 1h 05m 39s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2a1bf8a79e504b4583f44db1b299123d : SUCCESS in 49m 58s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/fb73dd93605b40bab8cc967021663087 : SUCCESS in 1h 03m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3b9b1a7a884965f09ada2da930d08035361ab548","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-16 13:59:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"67bcc3259422c13a0408a02f7401cf08d418ed2a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-16 14:50:46.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"801d5c8178b42739f863726da4953185071727f4","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-17 19:26:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2488024c266c39e4d1a59be0214ca2610cd6acec","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 17:09:39.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a9845ace922b3c86d402cbdac4ad13b8659e4b8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:09:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"81cbac103a315aa49bad89d78614c07ea8fe3763","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 18:59:31.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/610ab2b7343e48b0bcd7e0d9153c1275\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a7f78cef1e3b462db6d6ef39a0755f0f : SUCCESS in 23m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1a18d93f90e44d468357306aa91e217d : SUCCESS in 26m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/88c1bb20f1ec4d88bc2f3b31fdaed740 : SUCCESS in 53m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/17ad63080e4a4571b4c3cbd3551e2e55 : SUCCESS in 49m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/910dd2050c4d424abf128fc979a1c8a6 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27009aabd5794304a796465b7df9b7fe : SUCCESS in 1m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7b247cbcee394cc8ac7c66434552d407 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db0e13f44ae402ea0b24bbaa2e2ced6 : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e22ee5fdc816486a920b8c040d21b0a7 : SUCCESS in 2m 42s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5975d346235a4880ae28215a4b43b307 : SUCCESS in 1m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/06a2c546ac9e43919cc322e9d80ade1d : SUCCESS in 24m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/292b9447bb414a6086cb143e3538098e : SUCCESS in 1h 00m 45s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/25215649827e4ac29438553f31c3ab1f : SUCCESS in 33m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/42f4bc9afe9b4a1f8135a6e381cd9246 : SUCCESS in 1h 44m 16s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d41c666996d842109b6f14435eae9b92 : FAILURE in 21m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/de84bbf182ca450dbe974ae7204a1e96 : SUCCESS in 26m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6dc4e5d50b57405c904125375e5aadf1 : SUCCESS in 58m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e9b41891a48249c6954e521bf2d0ce74 : SUCCESS in 1h 36m 01s","accounts_in_message":[],"_revision_number":2},{"id":"90c0da2a6568dc9cc1545ea69e2aa09cd6d3a91a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 07:56:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"017676247ce340d0a42d60afbf095f69a54febf4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 08:58:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/974cc54b251545a9a362d383d2185764\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/7ed345cc7f9a421eb5b60082a8163445 : SUCCESS in 47m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0b891ea69ce240a58de51c173ceb3bf0 : SUCCESS in 26m 07s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/dddb92c807e1422c958a4b083193bd05 : SUCCESS in 59m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b1a78f6dda1848499849ca14f3e810d5 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7d4bcada008c4200b7072e42e0c6ddca : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3fcba22d805c4ce597fd794c97d79740 : FAILURE in 6m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/4281a8129eb549529757a99b789b1cf3 : SUCCESS in 40m 35s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e3085c9306d3413bb92458b39682aafc : SUCCESS in 44m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2551d85f521c06d3844674b0e2ef5023e3c4e83d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 09:41:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2982dd3b6fc545478400ff7a96c4ea57\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/42d6f10854d54480a8d4867a2a06bc50 : SUCCESS in 22m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ca936783da7f4792a357402c200b0147 : SUCCESS in 1h 05m 58s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b01dc4cc242d4d689d859454e16537d7 : SUCCESS in 18m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9f7e6264191044c09b1436b45b284d02 : SUCCESS in 26m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f081bc4a6a49497ebae042114eac7035 : SUCCESS in 1h 01m 48s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/783753b5b0f945698fddcfa500065b76 : SUCCESS in 1h 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a60e56de2cd14733b870c077447b725c : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c32b4b2a3f4044b6b7e6fa9859b6ce19 : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7ac6d7bcaf043719d653501f52cab9f : SUCCESS in 2m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2169aa4eae784737809182ec49624dcb : FAILURE in 4m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d98e1f72ddc74a2abc4dc77348c727f7 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c8caccd673b4c2186434467596acbc0 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1202e3cd6a94f44854efbfbaf4710e4 : SUCCESS in 2m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/71c29da9fa7049e78d2d5be88bda7ca2 : SUCCESS in 3m 11s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3155ef6a3c6e4de68c041c0d0e23fd38 : SUCCESS in 22m 41s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e165766e2eaa45c18a32c804ad231d8b : SUCCESS in 1h 22m 07s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2a0fd8d3fd414e188cf36f6ca80fb999 : SUCCESS in 1h 00m 07s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6fb6107346a040d6bfb9cfe2fdbd7093 : SUCCESS in 1h 34m 59s\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/7885c28b6d904205b300b56cbf8af7cf : SUCCESS in 1h 10m 31s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4ba9e21602344afabc46edd4a221e98 : SUCCESS in 23m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8c0f8d77910844f2b0048094914fb66a : SUCCESS in 26m 46s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f413908f69b342ca9ca3a3f958d78b79 : SUCCESS in 1h 04m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0cee9101920c44818182fe643fb94208 : SUCCESS in 1h 26m 57s\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/577fb798b7ba4e9a92d9a2fce13a79d4 : SUCCESS in 1h 20m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8a2bf790ba487381be4b6a5d1e56216079746a26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 09:41:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b9dc95d1f4668e13d9b76fd59b539148a45f55e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 11:29:56.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2988323b715b41d8af32395851a465ea\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/86fee29619b94a8c9a5f1ce1cfbb03aa : SUCCESS in 13m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/98326d7fefdd41dabfd989c7194f72a7 : SUCCESS in 21m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e99dac7928244eb484863f9e45823859 : SUCCESS in 50m 58s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8be63a3253ea4f4da34d66bca86b2539 : SUCCESS in 48m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00f5682599164c438cef598972984c62 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590e1cdd164349849972286c11d9b309 : SUCCESS in 1m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bcfc568313544189bf7ea9e0db3170f : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f157ade18e14514a9b7c76aca443a70 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784f9baad0db4afc85d61c1c50d21456 : SUCCESS in 3m 10s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/491ffe11d6d8472d840eb0a813a98ffe : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7362b5da0c5b47a98b928bdbb4d89fa2 : SUCCESS in 23m 17s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/9207a94d305c443abee223894046b251 : SUCCESS in 57m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a2f864cf0ccc415fb751331e9a393eef : SUCCESS in 1h 06m 45s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0b8692904a7949868e9b1b3eb25717c7 : SUCCESS in 1h 36m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a8d87b7808374954bd1157d95c8e4186 : SUCCESS in 21m 38s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bc4e119df1af412a863a6fc1527a2c13 : SUCCESS in 30m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b0b62f36162a4916ad0ebb9aca5de87a : SUCCESS in 1h 05m 24s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a99a652af15a4b0091ed6c8947af5d9f : SUCCESS in 1h 18m 00s","accounts_in_message":[],"_revision_number":2},{"id":"ecf10d1238cc67b85e18c26b531744feba964ab9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 11:29:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"62dffcd3c55ce9863a3071ae2470c149acb5f8b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 11:30:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1788ee877115412dbecc4a2e908660d2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8858cb2b1fcf4c89913aa8fe8364ee11 : SUCCESS in 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce4e3b4e632e4c7bbc98f7e53b90ddd7 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"80d931982a81cf3821249e077c21ed36d9d9b170","revisions":{"bb234c6ce8c2ff96691c9fc645ec78d2b26ceddd":{"kind":"REWORK","_number":1,"created":"2025-04-08 16:28:03.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/59/946659/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/946659/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/946659/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/946659/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/946659/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/946659/1"}}},"commit":{"parents":[{"commit":"e7391f65bdaa7f0cb75443a5a0139415dc82550a","subject":"Merge \"bifrost: bump Ansible version to 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7391f65bdaa7f0cb75443a5a0139415dc82550a"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-04-08 16:25:02.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-04-08 16:27:35.000000000","tz":60},"subject":"Fix preparation of /tftpboot for ironic-pxe","message":"Fix preparation of /tftpboot for ironic-pxe\n\nCopy aarch64 snponly.efi to /tftpboot during bootstrapping\nof Centos- and Rocky-based ironic-pxe.\n\nDon\u0027t use $KOLLA_BASE_ARCH when symlinking to\n/tftpboot/ipxe.efi, because an aarch64 variant of this\nbinary doesn\u0027t exist, just x86_64.\n\nChange-Id: Ie19fcb441a2e54a60762e5c8483487b713a29ddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb234c6ce8c2ff96691c9fc645ec78d2b26ceddd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb234c6ce8c2ff96691c9fc645ec78d2b26ceddd"}]},"branch":"refs/heads/master"},"80d931982a81cf3821249e077c21ed36d9d9b170":{"kind":"REWORK","_number":2,"created":"2025-04-08 20:04:21.000000000","uploader":{"_account_id":34212,"name":"Matt Anson","email":"matta@stackhpc.com","username":"m-anson","status":"StackHPC"},"ref":"refs/changes/59/946659/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/59/946659/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/946659/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/946659/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/59/946659/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/59/946659/2"}}},"commit":{"parents":[{"commit":"e7391f65bdaa7f0cb75443a5a0139415dc82550a","subject":"Merge \"bifrost: bump Ansible version to 11\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7391f65bdaa7f0cb75443a5a0139415dc82550a"}]}],"author":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-04-08 16:25:02.000000000","tz":60},"committer":{"name":"Matt Anson","email":"matta@stackhpc.com","date":"2025-04-08 20:04:04.000000000","tz":60},"subject":"Fix preparation of /tftpboot for ironic-pxe","message":"Fix preparation of /tftpboot for ironic-pxe\n\nCopy aarch64 snponly.efi to /tftpboot during bootstrapping\nof Centos- and Rocky-based ironic-pxe.\n\nDon\u0027t use $KOLLA_BASE_ARCH when symlinking to\n/tftpboot/ipxe.efi, because an aarch64 variant of this\nbinary doesn\u0027t exist, just x86_64.\n\nChange-Id: Ie19fcb441a2e54a60762e5c8483487b713a29ddd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80d931982a81cf3821249e077c21ed36d9d9b170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/80d931982a81cf3821249e077c21ed36d9d9b170"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
