)]}'
{"id":"openstack%2Fkolla~947239","triplet_id":"openstack%2Fkolla~master~Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3","project":"openstack/kolla","branch":"master","topic":"bump-alertmanager","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2025-04-17 22:12:50.000000000","reason":"\u003cGERRIT_ACCOUNT_35010\u003e replied on the change","reason_account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}},"32657":{"account":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"last_update":"2025-04-17 22:12:50.000000000","reason":"\u003cGERRIT_ACCOUNT_35010\u003e replied on the change","reason_account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}},"35010":{"account":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"last_update":"2025-04-22 19:08:04.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-19 14:58:15.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-04-18 08:22:02.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-04-18 06:50:34.000000000","reason":"\u003cGERRIT_ACCOUNT_15197\u003e replied on the change","reason_account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}},"hashtags":[],"change_id":"Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3","subject":"Update prometheus alertmanager to newest available version","status":"MERGED","created":"2025-04-15 13:00:17.000000000","updated":"2025-04-22 19:09:04.000000000","submitted":"2025-04-22 19:08:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947239-bump-alertmanager","meta_rev_id":"71ad63890625bdb2ed10b2d527c34b74fadea387","_number":947239,"virtual_id_number":947239,"owner":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-22 19:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":1,"date":"2025-04-17 22:18:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":2,"date":"2025-04-22 17:11:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-04-19 14:58:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-04-18 08:22:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-18 06:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-04-22 17:11:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":1,"date":"2025-04-22 17:11:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 13:01:45.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2025-04-15 13:02:41.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-04-15 14:14:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 14:47:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-15 15:20:53.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2025-04-16 05:45:05.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-04-17 14:59:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-04-18 06:50:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2025-04-22 17:11:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"81169a05c5777065122475355f1bd40e6e432ac4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-15 13:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2a8f20a7f829e6ae653161015b42b3ba530e635d","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-15 13:00:51.000000000","message":"Patch Set 1: Code-Review+1\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"98bc11de2028d3158cf7606d6c3bf1ec6a90ac4b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2025-04-15 13:01:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d4eb5f791781d5163f103e3893ed86f1df2b89df","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-15 13:02:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9b52009ec49ae16a79e8642b86f6ada67317f8fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:14:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f62f33eee7e941ada703f97d6fab021f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/b06b0308b6eb4a67b099e9f103158bbb : SUCCESS in 47m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/30b0a8f6a73c4b6db1ff1a63ac9cf1b9 : SUCCESS in 42m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6d8348e0a7c849f2bf50d37deccfc619 : SUCCESS in 1h 12m 20s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2cd9512ef99e497189e434ecaa4ca119 : SUCCESS in 3m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6fd45ae68cb461396958cbfa2e4b345 : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dcf38869d8864b16a27e116399540bc9 : FAILURE in 6m 36s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0a3a246d2055463ab2c4e1d5dbc2e230 : SUCCESS in 44m 21s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0c05f5ff75f94094824dd33c167cbaaf : SUCCESS in 48m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"224d2a6f8b0545c6ecd191c14ec7520376910da2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 14:47:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cfa4d9e5d9b49baa20068fbe0b9c712\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9e2419b2a04647fd98289eec62690ac8 : FAILURE in 23m 07s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f5c5d3197a65416680ff5cc59e3584da : SUCCESS in 1h 17m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/aee1cf5ffafa45608b167886b33c0343 : FAILURE in 19m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f150bfb5883c4bbea089c5451ef3affe : FAILURE in 20m 27s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0ed8c19b506d4e189dd6232cd48bd1d0 : SUCCESS in 1h 02m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ff73d587b0914ed0a0120f510e44aea1 : SUCCESS in 1h 21m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/761bb8cf5d98436b8878d0822a54a142 : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/669418b393b5415d86edbc7f7edbba56 : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f365bd1540b444685d6f95eb0105abd : SUCCESS in 2m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/93ec0790356a47aab439722805c936f0 : FAILURE in 8m 51s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca8270f3674440b5990af03d9f4561ab : SUCCESS in 2m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc417c00ab5449ce8c22f77f22116d0f : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/155f445ee77042148cbc681450c4042e : SUCCESS in 3m 12s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3e76fb3965fe49f6ac7aa9787ccca47c : SUCCESS in 2m 37s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/846ed93c354849cc85a5add0f53d0c96 : FAILURE in 24m 26s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f66f3e5cd81846e48e8206ffb32e4870 : FAILURE in 1h 11m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b496b4906a674e05a8263ab28f175c72 : SUCCESS in 1h 13m 48s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/68f8479c294f4fa1bf5a27d359f6faeb : SUCCESS in 1h 39m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e3eb6ef2aaa446a1bbb6685d1e57c274 : FAILURE in 1h 32m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/83cfd7ae9a7e40b4816e2301823937f6 : FAILURE in 25m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/988e800f708d4056a2894e7b000359d2 : SUCCESS in 1h 12m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b034a17052f745689e6849284f99f51b : SUCCESS in 1h 20m 18s","accounts_in_message":[],"_revision_number":1},{"id":"ce63fca56182dd8a0d27ece4536131cacf071bbe","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2025-04-15 15:20:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"315dc5be5fb9d9405be48558655e618dce80ad35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-15 15:56:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"4174624560dd5e21661e3591ef043660fb616e54","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 16:47:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/053f240fe9454ba8bec36a5820552dce\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ad152e8fa2ec45348024e6f01c6723e3 : SUCCESS in 38m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ef4e2d81870840ea8477352a38132eb3 : SUCCESS in 34m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/86b054db9bc345df88188ae03ab9aa60 : FAILURE in 42m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4bd2e79f875c433bb60be8e0afea600f : SUCCESS in 3m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c0dc2a0f894400e9feb3ced6eaa5c65 : SUCCESS in 3m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9ba513ea4fe24bbbbb961839ffb0dc09 : FAILURE in 4m 57s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c941555a75e64a568c27de9588c8545a : SUCCESS in 46m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/05e64d9b27224bf586c877b59f8414f2 : SUCCESS in 41m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d6d5801af891232ce527cfea7ec3a64813017d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 17:43:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d296b32131402c8583ded07211b35f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fcc60afa781045abb80fea7a84d0fdf6 : SUCCESS in 25m 01s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7df25a11ae7349e0852ae05038dcbcaf : SUCCESS in 59m 51s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7f7b1b92be4449dd858c2e92dcc9aaf2 : SUCCESS in 18m 43s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fcea2ee872434f8794f17ab94bc1c9d8 : SUCCESS in 26m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/42a4d05773de48e0a1bd6b98b7657798 : SUCCESS in 58m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ad06f51665274cfe9a60b66d655c9d8c : SUCCESS in 48m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce7fd424a52446639af5f86b98620a64 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3101c3c583ee4a44a3114a96d422b98b : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17538a3c5f9847b1951d9ac1f0751607 : SUCCESS in 3m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/332cbb855cf34f3abbd77e90c2421835 : FAILURE in 2m 58s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74ca50c3a0764c59b6fb6e77da5ed623 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55cdf3bd50304de0aec90e2759ecc5dd : SUCCESS in 7m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26667c6ac7014507aaa59c6f02f24c32 : SUCCESS in 3m 58s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5d9353ec7a4642af9e01576acf906618 : SUCCESS in 2m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ca01fb1f1b594ea1a00b64cc35a8aa3e : SUCCESS in 20m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1818c00fce13490e9bab272f5bb0f058 : SUCCESS in 1h 13m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/76b0241af5f04f708927d206543a54ac : SUCCESS in 1h 04m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/ca5d3d1ff2a4439d91c502cebc97615f : SUCCESS in 1h 17m 02s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/349556a5cbd04bd79822d1f38d90ce4f : SUCCESS in 20m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/14fac176cd7e496abc35b2fbcc0ac422 : SUCCESS in 22m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7285caa65878413baf479626d6c732df : SUCCESS in 30m 14s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f2f3ea958b104fe79759bc21f6df73a7 : SUCCESS in 1h 42m 18s","accounts_in_message":[],"_revision_number":2},{"id":"e22a2b6acdd3c2db60426242ecb964b125b0059b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-16 05:45:05.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1c4183b169879defc62a76dbd2f7453833f580a8","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-16 10:15:16.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c84f52ae53fd2b49c186a5fd09f4052260ff3d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-16 14:24:32.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6e552063f840133fca49c3803b777026449486bc","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-16 14:41:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1e5186e829ccf942b25734491d158e672e0e3589","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-16 15:12:53.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"174f1959fc0e7818b9219852c7aaadf00938bba5","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-16 15:26:40.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"eb94e8e19a953cc53cd58bad1eb9f840845b3591","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 16:13:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58950a654054e289a8da56b2edfaea6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/34fb44846a2546638c6928f4a8eb778e : SUCCESS in 33m 18s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e38ead0e018344fb81f05ea1b7b91a6f : SUCCESS in 29m 57s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b92c4c123baa4c0e90617ad25233bfd4 : SUCCESS in 59m 07s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/952e430bb4ec47e790900da042496f26 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf7f40ebd359456c96db8c22da1983aa : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a9ae9d1caf5f465781f9239b06855b35 : FAILURE in 4m 46s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/07d467d464cd4673884161f4e5806175 : SUCCESS in 32m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fa63f48dc8a54ee38804e5eb78464ce2 : SUCCESS in 41m 09s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3502567ddb57a4985d0510815044c2425499cca8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 17:06:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9d616de5e0f41ef8d1a0a04d4b500b7\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/21b6b669acb2469fa532ef5e607d4a5d : SUCCESS in 27m 28s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0f49ca8cd14442d3a9b2f6b328432248 : SUCCESS in 1h 01m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5fac8798b65a427398a5f554a86e4b71 : SUCCESS in 19m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5c67588c75f6420c962b9bcc2cb8ceb8 : SUCCESS in 29m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/71d5a35f29874d0e8acf03ab7256a6c3 : SUCCESS in 1h 04m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3e49e5165f7d4374bc1798a758d95ca7 : SUCCESS in 1h 30m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4e0222a0c440c1a53f4509f14dad4f : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6da58bc2278477eacaeebf8706d54d3 : SUCCESS in 2m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f881a67dd48c470084eb1bd817917a53 : SUCCESS in 3m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ad2048fdffea461d90dabca3b2c3ddc3 : FAILURE in 5m 24s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e7c9b015bc140f2be98c9527ab7fd62 : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0922d4957a2b43339e9cdc3f7853e109 : SUCCESS in 3m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7e5e17558fb4a70b85a2456c476fb72 : SUCCESS in 2m 40s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/53d5fdf08e4c4b1a8c043ac608fc86b0 : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0c82212b34464a75865c12af2c7ec298 : SUCCESS in 13m 47s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd51cfc4af814e6391b4082e74353adc : SUCCESS in 1h 37m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ff785645787444bd9f839b8f6c3c43bc : SUCCESS in 1h 05m 12s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e62f16aaf3aa41f8ab26759d226c704d : SUCCESS in 1h 34m 37s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b00879dcd3e847b4bdf31785c14946da : SUCCESS in 22m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d0c699dcc0754425b9ecdba14eec7ced : SUCCESS in 20m 39s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ac886a461c5d40458e4bb7b635ae5042 : SUCCESS in 1h 08m 10s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/289397dd82f84a1f96b6f88bff9398ca : FAILURE in 1h 41m 35s","accounts_in_message":[],"_revision_number":4},{"id":"d6157a57fd9352b94bc62d5ce686988af0db3094","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-17 14:59:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"fe01777122435a0d69605942fc86525e7293a538","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 16:31:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61cdf82708704fb389448c99293eb72b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8cd11278668f454092fc3da093befa6d : SUCCESS in 38m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/62d863366077415fac2ca92529def807 : SUCCESS in 41m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/220ba2ecdf0f4149bda3265075e17605 : SUCCESS in 1h 30m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a15710de2e4b44a5a187f04de75f2836 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9efaca63c68c46a4897e25421edb1a9e : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d743eeae87b84cc29f54e7a66dab77ed : FAILURE in 8m 37s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e096eed25e01426cb6a677b63dfaf84c : SUCCESS in 42m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/95d384b9377a417a99e2a1178809e829 : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"853ea8703ce2951de7799219f2d891f626037196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 16:38:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da2f485142664a4c98a396534c5551f2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8c619a8d9ce54ceeaaade99c0eedf3e1 : SUCCESS in 26m 41s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e84beb1c197048db8c958098638a3303 : SUCCESS in 1h 07m 32s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/05a2ceac7984472c8878215996a31b41 : SUCCESS in 22m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1282035d33144741ae03aaf42b84292f : SUCCESS in 31m 26s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/48b40cee69704a5b847713ee26d1c9d9 : SUCCESS in 1h 02m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/90fb26558f654f22853df9acb8921b81 : SUCCESS in 48m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/472435f4bb134bec86af431f33489ba8 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d154f1b47b18486c80f8f6b475848876 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4397ec66bd474e27aefbe07650ac6bde : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/170dcbcc2590439db1fa65f32d1f2a8e : FAILURE in 5m 03s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9732f09db6d347f6849e88310fdaf102 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e8b0db55d0448bd9967c0c5135be929 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7806dd8ee5f64d19bd5cc2d03ad139e1 : SUCCESS in 2m 56s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/746ea88c495d491880a01183339433c9 : SUCCESS in 3m 45s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/da16239fb41a4262bb0f8f1a1dd2a56c : SUCCESS in 19m 39s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/698dee4e2d0e4eae9e36e1a5a32d71e6 : SUCCESS in 58m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/71b579feec2e4e9593c7aa099e7eb08c : SUCCESS in 1h 07m 23s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e72b19a53048486f97363f612cc6f911 : SUCCESS in 1h 19m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4d2f5d0012c240a2977988c8a37ced81 : SUCCESS in 19m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/242d3e1c2b0b437b9416c02610255984 : SUCCESS in 30m 05s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/cd2e9431514f4b28bdfa275306bcdb3d : SUCCESS in 1h 03m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b39c9f33453445e4bcb62e02a682d0e6 : SUCCESS in 1h 34m 41s","accounts_in_message":[],"_revision_number":4},{"id":"f0e62b3ec98f6913686fdceb2bd569e855e36724","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-17 19:30:33.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"bee2b37cbc6e5a9116cdf42c132a8345ab1a5665","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-17 19:55:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0a454a4a8dd94066b60f28c24b48e345ae92eb08","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-17 22:12:50.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e4db9eca00f8acfe98e6ca1953eb9ab1c47c13f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-17 22:17:51.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"b9ee855e3f4c5d5e928bb2dc9db7fcce09452796","author":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"date":"2025-04-17 22:18:53.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"95ca98a955264757f44d7d698c0f14664f8127bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 23:19:40.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70befdf383c54d0cb731a00f785b09e5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/eb696fa6b9ba448093adb8ac8bb4ed59 : SUCCESS in 36m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/28a526652ff64ed5a7ff3b6dd003909c : SUCCESS in 29m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/76607820563149198af601508a645804 : SUCCESS in 1h 00m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/02d63d16ddd34548b33a9c6619c629db : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/df161b21b1ae4866bdf8848f19c6279d : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/59bd916261954575867a31614d128afc : FAILURE in 5m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/69c1edc3c57143a5af4f899e216a510f : SUCCESS in 32m 01s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/337ce3aec0574a8d856d51ae0b64f824 : SUCCESS in 33m 28s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"08ad341d68ef64958cc1dfec0c5111aa453aeeee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 00:03:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a61fc0acaf94343a15dd1d90da04be8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/47df524a4e434b749f6dafd67716554f : SUCCESS in 41m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/72e8aaff767f4b878f2446ea5a942ac5 : SUCCESS in 31m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6e48a211315c4e5e89ba8bb2d32871b2 : SUCCESS in 10m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/373279ae123f4158a462c483e2c7864d : SUCCESS in 14m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d4956d5d206141c5b63d30f64df60bf9 : SUCCESS in 1h 05m 03s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/398f4d27d8ed44d38935cd62f84d7bf4 : SUCCESS in 1h 32m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/433662b6e612498d9a49e42ec8379573 : SUCCESS in 2m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c924c5e6b97441baa05fa78309c0a60 : SUCCESS in 2m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3875c3615064a80a9db2201d85ba8ed : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ee308eeb4e254a4b816fee98414a681f : FAILURE in 5m 25s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78c6d89c57e74c468453bd4d8265c53d : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1fc10725d444d2fadee6f28bdd9cd8e : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bdeaf277b37b45aea2d3a42786915818 : SUCCESS in 3m 14s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fcdecd973121424cb403be53790d6d75 : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6fb7be32982e4d29a100db72cb31cc16 : SUCCESS in 23m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b7f2b503b83844f3b164f076490c5316 : SUCCESS in 53m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/10b2864329134cf2be3d971b6cb83335 : SUCCESS in 1h 07m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4b20e972588c439fb13dc8e33a346b1f : SUCCESS in 1h 10m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/66ebf0848dfb45b68d9d341d77e68136 : SUCCESS in 19m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/334dc4aac53242da87f5e795bf92a701 : SUCCESS in 20m 37s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8821b610b0284c92a4f8dc5c1891cca0 : SUCCESS in 1h 02m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/edf2a5e8eff94d138a8a11f27d35bb5d : SUCCESS in 1h 39m 47s","accounts_in_message":[],"_revision_number":5},{"id":"3fb26e1cc30b9b046bae6dd95beac57b57b642a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-04-18 06:50:34.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e57d3dfb3faa46f787218c7d9ff336453d9e5a9b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-04-18 08:22:02.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"f0b2299ff0436e7562fa71ff925e437213e1fdd9","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-19 14:58:15.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"581968b1dfa18aa01c46787a4c4989046dc738ea","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-22 17:11:41.000000000","message":"Patch Set 5: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"20fc31aba73847281b53ee86ac5f7af1f63b1cdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 17:11:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a110bc4a9f079429e031acceef61833bf7e6ef87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:08:03.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a750c9185aa24a63916f1b386b05aac7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fb91e33974f0408ab507ed734a91cd5f : SUCCESS in 10m 32s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8098a244f8db43cbb8cef744cd2b8abd : SUCCESS in 28m 51s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/14909aefa0594d4ea5ef48bfa5c0420e : SUCCESS in 33m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2acf31068dd74821a4bc71aea95b94af : SUCCESS in 1h 29m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4286ef6ade747f182e958869cc6f446 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/844bc14e8b9946d2bb6815611d0f9aef : SUCCESS in 2m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fed5fadea036469980b6c5dc767180ec : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5976147fbde24b51821d471812ab999c : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91f09a395fb54b40878d97bd68936950 : SUCCESS in 3m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f14ad27617f54f56855337037e47a7a8 : SUCCESS in 2m 28s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c3eaa97070c645b6a934acb1db37e8bf : SUCCESS in 20m 58s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f8b39ff92de24a76b07eb928ab6019db : SUCCESS in 1h 12m 04s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/07c40f8c3c8c4632a02fdf4ae085f919 : SUCCESS in 1h 08m 03s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/13d5913bfecd4187b237e5b1ea4f4a30 : SUCCESS in 1h 32m 41s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fccb44b907f84ca48e63cb464de793ab : SUCCESS in 23m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2ccf40d1c01a4279acf6e07d253525e8 : SUCCESS in 40m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7adfdc42d0f347539f3b896a76b841ae : SUCCESS in 1h 20m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e3fb7f817b7d4b748a0f710c539e24a2 : SUCCESS in 1h 37m 10s","accounts_in_message":[],"_revision_number":5},{"id":"2efe40cb1a7bbef6a50cb8a6127b5e0403bfa67c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:08:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"71ad63890625bdb2ed10b2d527c34b74fadea387","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 19:09:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3d19f3b1234cd4a3fbc9d36b99796c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4006d0cc0fcc4fbfbb154cdcbe40b22c : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0c2db98d11c4eac914ee231fc43a4eb : SUCCESS in 42s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"bd42e9d435ac219630e9196bc1a5c6c384b7cc3a","revisions":{"2c48c4d865fee01b37302f8d5bb7922fa032292f":{"kind":"REWORK","_number":1,"created":"2025-04-15 13:00:17.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/39/947239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/947239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/947239/1"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 12:17:49.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 12:17:49.000000000","tz":60},"subject":"Update prometheus alertmanager to newest available version","message":"Update prometheus alertmanager to newest available version\n\n* prometheus-alertmanager: 0.28.0 -\u003e 0.28.1\n\nChange-Id: Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c48c4d865fee01b37302f8d5bb7922fa032292f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2c48c4d865fee01b37302f8d5bb7922fa032292f"}]},"branch":"refs/heads/master"},"1f3276b1e500e0d52f0917e5b116ba3bb09c6294":{"kind":"REWORK","_number":2,"created":"2025-04-15 15:56:22.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/39/947239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/947239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/947239/2"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 12:17:49.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 15:55:54.000000000","tz":60},"subject":"Update prometheus alertmanager to newest available version","message":"Update prometheus alertmanager to newest available version\n\n* prometheus-alertmanager: 0.28.0 -\u003e 0.28.1\n\nChange-Id: Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f3276b1e500e0d52f0917e5b116ba3bb09c6294"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1f3276b1e500e0d52f0917e5b116ba3bb09c6294"}]},"branch":"refs/heads/master"},"2d5efee67d2986906381ea40a58dd4ff8cc99d47":{"kind":"REWORK","_number":3,"created":"2025-04-16 14:24:32.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/39/947239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/947239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/947239/3"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 12:17:49.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-16 14:24:18.000000000","tz":60},"subject":"Update prometheus alertmanager to newest available version","message":"Update prometheus alertmanager to newest available version\n\n* prometheus-alertmanager: 0.28.0 -\u003e 0.28.1\n\nChange-Id: Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d5efee67d2986906381ea40a58dd4ff8cc99d47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d5efee67d2986906381ea40a58dd4ff8cc99d47"}]},"branch":"refs/heads/master"},"1a7c5e318aed2f2e2f574feb90628c8f239c3bf2":{"kind":"REWORK","_number":4,"created":"2025-04-16 15:12:53.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/39/947239/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/947239/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/947239/4"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 12:17:49.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-16 15:12:47.000000000","tz":60},"subject":"Update prometheus alertmanager to newest available version","message":"Update prometheus alertmanager to newest available version\n\n* prometheus-alertmanager: 0.28.0 -\u003e 0.28.1\n\nChange-Id: Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a7c5e318aed2f2e2f574feb90628c8f239c3bf2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1a7c5e318aed2f2e2f574feb90628c8f239c3bf2"}]},"branch":"refs/heads/master"},"bd42e9d435ac219630e9196bc1a5c6c384b7cc3a":{"kind":"REWORK","_number":5,"created":"2025-04-17 22:17:51.000000000","uploader":{"_account_id":35010,"name":"Jack Hodgkiss","display_name":"jackhodgkiss","email":"jack@stackhpc.com","username":"jackhodgkiss"},"ref":"refs/changes/39/947239/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/39/947239/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/39/947239/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/39/947239/5"}}},"commit":{"parents":[{"commit":"a0af54161e1734ab150dd5d7aea22947eadf0b91","subject":"Merge \"rabbitmq: Bump to 4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a0af54161e1734ab150dd5d7aea22947eadf0b91"}]}],"author":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-15 12:17:49.000000000","tz":60},"committer":{"name":"Jack Hodgkiss","email":"identity@jackhodgkiss.uk","date":"2025-04-17 22:17:20.000000000","tz":60},"subject":"Update prometheus alertmanager to newest available version","message":"Update prometheus alertmanager to newest available version\n\n* prometheus-alertmanager: 0.28.0 -\u003e 0.28.1\n\nChange-Id: Iec5b936ae77f2d8298803fbfe50f55b2ed69f4b3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd42e9d435ac219630e9196bc1a5c6c384b7cc3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bd42e9d435ac219630e9196bc1a5c6c384b7cc3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
