)]}'
{"id":"openstack%2Fkolla~947724","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I7944c59d1bb83d42631c3d0ecc14aa8e25ce620e","project":"openstack/kolla","branch":"stable/2024.2","topic":"bug-2107733","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-04-29 05:59:17.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-04-29 07:35:20.000000000","reason":"Change was submitted"},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-04-29 07:35:20.000000000","reason":"Change was submitted"},"36765":{"account":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"last_update":"2025-04-29 07:35:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7944c59d1bb83d42631c3d0ecc14aa8e25ce620e","subject":"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation","status":"MERGED","created":"2025-04-20 20:17:58.000000000","updated":"2025-04-29 07:36:11.000000000","submitted":"2025-04-29 07:35:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947724-bug-2107733","meta_rev_id":"d52303871b54121871a28435a7a841bda5231ba9","_number":947724,"virtual_id_number":947724,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-29 07:35:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-04-29 05:59:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-04-29 05:59:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-20 20:20:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-04-20 20:20:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-04-20 20:20:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":36765,"name":"Seunghun Lee","display_name":"Seunghun Lee","email":"seunghun@stackhpc.com","username":"seunghun1ee","status":"Cloud Engineer at StackHPC"},"state":"REVIEWER"},{"updated":"2025-04-20 21:43:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-20 22:26:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c20db8b62d05db350483c98e29da5f69a04fb1cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-20 20:17:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c1344383f38f341c1b3f192d513029d709b39793","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-20 21:43:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc5005f07a2c4060a08fabd35e3a1d7e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/50aa2bb917ed4b3fbf6d512957f82cc2 : SUCCESS in 28m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ca4b895f6da54034b05474ce4142063c : SUCCESS in 31m 11s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/474de5575da34acf98904537ab77ef07 : SUCCESS in 1h 23m 22s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3a5005803e9e4d6ba60194881d7b9452 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8011978e401f434bbb95abddcda08598 : SUCCESS in 3m 39s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/e4a96d802ef44ff8afdd36f996c575bd : SUCCESS in 27m 24s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bc1c00efe8424bdba898c4e2ebbf40f9 : FAILURE in 28m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf6af042c5803514d86404c99c99eacd999900d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-20 22:26:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a405d503abc34517a76a91b921e84232\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/31fcb2047bca460fa6918f19cb2f51cc : SUCCESS in 11m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/79d909a181d54bc7936b81aabf08fc1c : SUCCESS in 31m 12s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/207e11b9e82a4088943844041e7521f9 : SUCCESS in 13m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f14ad9a92baa47e49376132bb3e0c569 : SUCCESS in 15m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/22b00564aa9643fe9d2502a824d01e5f : SUCCESS in 34m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2399d17fdefa4e20b866d7cb89057386 : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef4ebfde92b04722b7579dea2e27e194 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/722c31e47ede41959b8bb6489f7f662a : SUCCESS in 3m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa476ba0169543a689a19530a6078783 : SUCCESS in 3m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/204564e4e92f477fb6688546ac3b0e58 : SUCCESS in 2m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38cf9856e3704679acb4b5aeecbbf86f : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1300c5dcec4f442b86b4357a9b360497 : SUCCESS in 5m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f01b7e1aabd745c9932ee35d6466a5bd : SUCCESS in 2m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9444a37a494e401eb9410fb7bde64b2b : SUCCESS in 17m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ba549a3e2c249c7857b9d812a75202c : SUCCESS in 27m 13s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/79d1d241e3c64463bfcf2af9d676af46 : SUCCESS in 57m 14s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/e5699d98fd4b42b7a0a6e93a8d78ed9c : SUCCESS in 1h 12m 35s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/644cae0d646346d68349400c3fe6099b : SUCCESS in 1h 32m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eaafbfb4cc5d44d888e3820ef6891b5c : FAILURE in 2h 07m 34s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/58275d3d703f4b79ae6b21d31fa79430 : SUCCESS in 1h 05m 55s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a8563701b00c4e4995b5db40cd8e8fd3 : SUCCESS in 1h 09m 20s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/d7b4f161971b42338637e5212d49f8ce : SUCCESS in 48m 24s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/2283fb42b3ba4dac86fa709645af08f2 : SUCCESS in 53m 07s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9b9b44d54a32445485d36150b4f77395 : SUCCESS in 1h 09m 50s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e0215838b7ef41e49cccc9db6c2d40eb : SUCCESS in 15m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9015eda6cb75400f9a1743d282840d47 : SUCCESS in 19m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/38feacbaa11a476dad763f2498c055b5 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/61ba52cf34754a6b84a218cf1070a3ac : SUCCESS in 41m 56s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/c18136960b16484f8d6e4a840161f97d : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/11aeea44cd194996abb8050f2f8de3ef : SUCCESS in 1h 11m 49s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/efcc3826c475483cbec2c275165cd62f : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/ad6292724ba24f0ba5ca9c75acfefdb2 : SUCCESS in 1h 04m 01s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/72bd92940957495ba8f6bccea2bd33fe : SUCCESS in 55m 29s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fb67d023f29848cd9773715b824080fd : SUCCESS in 1h 08m 39s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8bb41235c8a347d58d2247826b9a0d37 : SUCCESS in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f22edb0061ab5e5ded8c246cbf312e94f2c22423","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-04-20 22:48:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6afe434ac24101579b9b2ff4f0c95f187fc354c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-20 23:55:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2ac2462a5845898380cbe353d41ec5\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1477c0378c3846348d3a6aa7aa83da8e : SUCCESS in 34m 15s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec924aa70c9f44e787945d0a1403fa79 : SUCCESS in 43m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b3591ae0d1164fd8a484e0fd9a0c428e : SUCCESS in 1h 05m 04s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a0196d5689c4e9ba41ccdd9abe614ca : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a72ffedb9d8e4aa8b0f06451d14cb462 : SUCCESS in 3m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/7e244c955fea4dbb9ea9a8a30358dbe0 : SUCCESS in 37m 58s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a7d182b75a1f485293c803c2dd6399b5 : FAILURE in 42m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b53cd5f4e17b75deeba3a938899873cc41f2dc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 00:41:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0acd9eee8ef541d5982ac2947d7cd8c3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ac52866e99ef4bd59099f2495fc4a945 : SUCCESS in 17m 34s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/24ddeb64a9f54c5082573d6ccbce5c58 : SUCCESS in 50m 14s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d0588084b0a540c49f934d41dba71bda : SUCCESS in 15m 51s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e4d0282e581b4b339e8adc359606dadd : SUCCESS in 22m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5c6c9dd6d1804041bad9625e9b83e90a : SUCCESS in 30m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2a4dcb13282f4fec800689f3d8dba317 : SUCCESS in 1h 09m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5dab33603cc43efabd1592ab7a9fffa : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cec6fca11ea49a39c0583f0d4c08d96 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8f4e9447cf2b4b639514f3df9a0915f8 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fe957a1038d64cf699bbab417d72a789 : SUCCESS in 3m 07s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02c524237651438886745aa22a1ee9d5 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19780b392ab64481b1ecf197631c1d7d : SUCCESS in 4m 25s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0d7d546b0ea74a5184033fbe9efcc7eb : SUCCESS in 2m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/dbf7cfb8419f44d2a137c350979d1b7e : SUCCESS in 10m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dd30ca8f8d724108aea51f66d7ea13f5 : SUCCESS in 13m 16s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/fd44837657f74bee88cccf0e7eef2768 : SUCCESS in 31m 35s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/af53cd38494d4f4e81f3f075a3a06df7 : SUCCESS in 49m 34s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/1648417f072041fbae32f730d8108f5c : SUCCESS in 52m 39s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/17e2fe32271b4342a2e1f41d86d0b08b : SUCCESS in 1h 23m 41s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/6db71d671feb4878b2dcd6bcf1168ec0 : SUCCESS in 1h 15m 29s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/acbf0add8a0c47d9b2cf23fb0cf293a4 : SUCCESS in 37m 38s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/031c647869324730a49e9831973be0a4 : SUCCESS in 1h 00m 04s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/f8e7deba5c914d94b0456ece8769187c : SUCCESS in 35m 19s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/4f5da1a8ab0a4c469c0bc9c919d2061b : SUCCESS in 1h 15m 05s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3b918f27d5cf4a71a357d4913e0a8856 : SUCCESS in 13m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b4c47631ad8c4ea2a0c0465038fb591e : SUCCESS in 13m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ef0fafb928c84c7aa63dcacb24aa458d : SUCCESS in 1h 00m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/130c217c21b045cf8ad6c7e83b503104 : SUCCESS in 1h 21m 48s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/ace8ba9b7cb7412c802fcb22d293fc98 : SUCCESS in 1h 43m 57s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6d252dffeb7943fd84757d03a74dc27a : SUCCESS in 1h 52m 49s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/8637681aa7504c40bd9abcc865c00897 : SUCCESS in 58m 46s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/cfc656caaaa6420d8e0f70e30b43d7a0 : SUCCESS in 1h 02m 52s (non-voting)\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3c5f97771b4e4238896544712b5c8b2b : SUCCESS in 44m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/bdddc7cc04764b9aaa27c9e5b592b9a2 : FAILURE in 42m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/8f84f05b19d8444a967c575b038fb560 : SUCCESS in 1h 03m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f440b9bef7d097fb2ccd1a56ed283e60c809f35e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-29 05:59:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5695d75585403bafaa9e67fc70866567b8dd8808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 05:59:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0b6e01eb0f72891058da7586c86a6f843d198a94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 07:35:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46e1e3642af4db2b7b5f35f7f302c11\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/767ada59a7684e8ea418d3a8219004d4 : SUCCESS in 19m 49s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2536f6c8b23b4f6886f7c858f3f04d9a : SUCCESS in 25m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d01c3f477a5d45128f200aa6d91970a3 : SUCCESS in 1h 02m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bda837a01c004cd1be1262dff443f672 : SUCCESS in 1h 31m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f25852670c4e508aec7ec63fc08cea : SUCCESS in 2m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/484bfed121a3469d8c14ce7ff2162545 : SUCCESS in 2m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e0f4cb673de4055ae6d5caab0089787 : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fae90677f00745c896c8ee272c9e7c5d : SUCCESS in 4m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e55e35a7c44648f7ab96ee47ab82d1bb : SUCCESS in 2m 33s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ad4b5817b2fc40669861472805443b9a : SUCCESS in 17m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c416cd54eb5749d0a2e3054c5620aa3a : SUCCESS in 25m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/887d45347cb04e77876f0201599c6f58 : SUCCESS in 1h 05m 38s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fee9a25a24714e94ab8f728a67be4869 : SUCCESS in 1h 34m 28s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e7492b6a9bad4fff93f7c86abf313425 : SUCCESS in 19m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4eb977512add49ae87e831217b215c46 : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8f3054f86525447f933cba5fb3e3247b : SUCCESS in 1h 00m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0185db26c4884a26af95525541d9cc78 : SUCCESS in 1h 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"4e567e4efdcd094e5e626f754c4fb7d04c4685a2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 07:35:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d52303871b54121871a28435a7a841bda5231ba9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 07:36:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e0e890d33d94a30b65622e164795e5d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9280aa6745d4ac3ac68d5b7f7eb7181 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3ef707286efa66b3286ab40ecf652fdb97bc5bfc","revisions":{"3ef707286efa66b3286ab40ecf652fdb97bc5bfc":{"kind":"REWORK","_number":1,"created":"2025-04-20 20:17:58.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/24/947724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/24/947724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/947724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/947724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/24/947724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/24/947724/1"}}},"commit":{"parents":[{"commit":"b9c24c0decebb57a48129cdcdeb32fe3393c0210","subject":"Opensearch and opensearch-dashboard image size reduction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b9c24c0decebb57a48129cdcdeb32fe3393c0210"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-20 10:15:09.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-04-20 20:17:07.000000000","tz":180},"subject":"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation","message":"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\n\nFixes the issue introduced in I5a9162a30b23cfabb64dabe5c12c226fa8627abd\nwhen \u0027openstack-base\u0027 container built with \u0027pycadf\u0027 plugin from git.\n\nCloses-Bug: #2107733\nRelated-Bug: #2074075\nChange-Id: I7944c59d1bb83d42631c3d0ecc14aa8e25ce620e\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ef707286efa66b3286ab40ecf652fdb97bc5bfc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3ef707286efa66b3286ab40ecf652fdb97bc5bfc"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
