)]}'
{"id":"openstack%2Fkolla~948084","triplet_id":"openstack%2Fkolla~master~Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac","project":"openstack/kolla","branch":"master","topic":"fix_luna_permissions","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-04-28 11:08:34.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"35125":{"account":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"last_update":"2025-05-06 07:45:46.000000000","reason":"Change was submitted"},"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2025-05-06 07:45:46.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-06 06:03:31.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-06 07:45:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac","subject":"Enabling support for Thales Luna user id and group id.","status":"MERGED","created":"2025-04-24 14:45:22.000000000","updated":"2025-05-06 07:46:32.000000000","submitted":"2025-05-06 07:45:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948084-fix_luna_permissions","meta_rev_id":"916fb9721f132c939565dc9c1474afda4a81143c","_number":948084,"virtual_id_number":948084,"owner":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 07:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2025-05-06 06:03:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-05 11:52:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2025-05-05 13:19:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2025-05-06 06:03:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-05-02 10:22:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":-1,"date":"2025-05-06 06:03:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":1,"date":"2025-05-06 06:03:31.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-24 15:23:04.000000000","updated_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2025-04-24 16:25:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-24 16:33:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-25 12:39:06.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-04-25 13:51:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-04-28 07:59:06.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-05 11:52:57.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-05-05 13:19:27.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"}],"messages":[{"id":"9408a4384c03949fb58c6fff89ee01f529eb408e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-24 14:45:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"825dcd0a2445a3e11c5a7c303d8ab46a393ba868","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-04-24 15:23:04.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a10c9b4f7cb5ae78a346b2e8ea43e65c01bb7bea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 16:25:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20c240efa12a4b79b93d60021bfed453\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/8df58d8f4617467d97e8c095e0cbfcb4 : FAILURE in 47m 11s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4279aba002104029b077bc932c7de02f : FAILURE in 43m 39s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2d2d63db2ccd45059f110c161d693dc2 : SUCCESS in 1h 37m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c231a1637a6435a8033485d3d616e93 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e35a94115d6445097b5266f418d1382 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f4758031261f447a8632a7478f792c9a : FAILURE in 8m 43s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/5f7871285ca7486980a145d3192680d6 : FAILURE in 50m 52s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4aa421a394b646309bf171d7a105a25a : POST_FAILURE in 50m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"460dbe95b7ef9fbdea523b48eab3a2da379eb0f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 16:33:12.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/238db75cbc874235bbb1d3b860b3eef3\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/18e14b75819644518f1224bd45ff1fb3 : FAILURE in 23m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/04d46403b11342e2b888a45cd28ceb99 : SUCCESS in 1h 00m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/92b54e062a334b139314976174872f1b : FAILURE in 16m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fd0d5cc373f74a498167b61ae85ec20e : FAILURE in 26m 52s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0975cf01681c4ca3833b2b7026813104 : SUCCESS in 30m 42s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/015488850a6d4b1bb6763ea8c4817ef3 : SUCCESS in 49m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5fc107b8ca9494fbc0f5da6e171058d : FAILURE in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe596938b0e6464daa272735de9cbc27 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54593c9f173e49f38087e50f348aeac5 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f43e5ba520fa41b2a679189e11cab96e : FAILURE in 5m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de7bcfc51dd14c66a89312f16d24ee47 : SUCCESS in 2m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4168dd61a65d4fabaa9da05b3d0c92c5 : SUCCESS in 7m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fe666987769743f288fbe358fd3ccf3b : SUCCESS in 2m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6a63875c535c44dbbe8c8dd17c9ee6de : FAILURE in 23m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/d86a5215f97b462a98cd41cd091a3e51 : FAILURE in 1h 02m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/bdee2050ef7a4157b5a2141670e28690 : SUCCESS in 1h 00m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/966f4db8c40747179220fc0fdc59718f : SUCCESS in 1h 14m 49s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72f8f0c4a37c4ac4980b4f6ea8d432fd : FAILURE in 20m 04s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ffba833b4bff4574985a4c40065ace4a : FAILURE in 27m 40s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7c3bdde2a3b24b789c699e112d3a3725 : SUCCESS in 1h 04m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3097fb601c6d466e931c3a708085cdc0 : SUCCESS in 1h 31m 41s","accounts_in_message":[],"_revision_number":1},{"id":"16f80580b62a76045baf0f5c0d0e4978a927977b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-25 11:47:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b763efb3501f2e7aee3bf5681c38657697b74623","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-25 11:49:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6bf00986e20643086bca22ce806a871c77a7b922","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 13:32:59.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40e78ddfa1c6457595312ae04b51e34d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3162d6b382d74494bec5c11c37be871a : FAILURE in 26m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6fa528ca6a764460928d852bb1b63c90 : SUCCESS in 1h 11m 10s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a024834040664327a758c065ecb8b3bc : FAILURE in 20m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/f05c823be90d4d88be00df40c79a303a : FAILURE in 14m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9da4bbafeef7421890642a70ccbd3f46 : SUCCESS in 1h 01m 32s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e0c5b85cdc6b4f1d85dfc50233d2a163 : SUCCESS in 49m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb0ed155b0b54dd7a016b41008591e11 : FAILURE in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbb1aaf899bf49ab9dc6681064ace043 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a62474f3e06c484ab1acf53036c777ee : SUCCESS in 3m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9f98f70e7b944eba920096c24cf885f3 : FAILURE in 5m 11s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7e056e457144378a5252103c86ce320 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19b87501e55a437da320e8eac70f256f : SUCCESS in 6m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fc91c3fb9fc548628c1cea87d90118f0 : SUCCESS in 2m 41s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/ec0a0f06f9e74266aab8e4f487612de1 : FAILURE in 25m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/aad82d33be2a4727a40107a07e3160d7 : FAILURE in 54m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8ce3328b0f9943c79874ce88a277e5da : SUCCESS in 50m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/683eb841edac456ea7d515b48bf7d4aa : SUCCESS in 1h 40m 18s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2be40d69f4f44287972bfdef6afbaf8f : FAILURE in 18m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/10c7c42b5d7a4ab7b9de985e3d2b88dc : FAILURE in 33m 48s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/93f17dab02ee4195b912102cf71f7e16 : SUCCESS in 58m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4a1b185dafe34cb3b802d3707ece1337 : SUCCESS in 1h 32m 30s","accounts_in_message":[],"_revision_number":2},{"id":"60c39f703c331e035480129723567f1605212097","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-25 13:41:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"734cea02f3caa9294dce9f991e1c77feed23a2de","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-04-25 13:42:43.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e0c35abdda6d850190407f2cdabe4896d52a71ad","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-25 13:51:41.000000000","message":"Patch Set 3: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72aa942b4e41383f5b9abf3e4a75f3696ef6c2e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-25 14:37:28.000000000","message":"Uploaded patch set 4.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"641d9cab9226a66affa9e725d9928ff420f953a2","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-04-25 14:53:52.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61c2b5a073d98949f4ec98125907da835b9ec7eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:16:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa4713ea2ff64509864e8a9438b999d9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/75179d13d173488b8b3506db5646c717 : SUCCESS in 14m 05s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/28f654792bef4dc28bb89d1d81ec6448 : SUCCESS in 33m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/da017c1d7bbc49638bf67b9f739952d7 : SUCCESS in 20m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/cfdc28ba3cd6426ab4370896c75eff6c : SUCCESS in 26m 16s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/de03da89e9df457c985787b159c2d3b2 : SUCCESS in 1h 10m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bdbf0123f8174b5ebdbe97b8dfc7640a : SUCCESS in 1h 17m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f00f17771da4c20843da8527f13aa46 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e19297b191864c2c870514e720840368 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40792f1ec93743c69acfc918c97e42f7 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac8a005b6ecf49b09b98ebb1bb2c570d : FAILURE in 4m 16s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b2566fe56944d11a94a7ecdc8d5b696 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f5d5cb79bf84b97ba50c48559814bde : SUCCESS in 3m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/95634b8540854cd6b8c1a664726d32c9 : SUCCESS in 3m 24s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9f5fe26ef7ca4af199ac6cbc1f23d816 : SUCCESS in 21m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1251f82c7ea34768961ee45e908b2c52 : SUCCESS in 1h 20m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/69415f54b5264b8e9c003e993d584b96 : SUCCESS in 1h 04m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4aa6c91b69cd479d8c4d4bf49f82f6f6 : SUCCESS in 1h 28m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1d8fad2a02e54627bbec2dd5ee75e493 : SUCCESS in 24m 33s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c4f8acf415b5464e95484632ddb3952e : SUCCESS in 27m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3ddcac9d79294266ba8caccbe7cf5f77 : SUCCESS in 31m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7c3aa80ea64d4472a2a47d7cde8039c9 : SUCCESS in 1h 34m 20s","accounts_in_message":[],"_revision_number":4},{"id":"9309f080e2c9133b376b971ea4962e4517f29fc1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 16:35:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccad1632837c47aa803e4057cac82b78\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e74cee9befdc4b7e8070d6acbb31b5e4 : SUCCESS in 38m 17s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/09fda225e8a94af0ac85776481ff56ee : SUCCESS in 34m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ec6fdc2612364dbe8963011712d1138c : SUCCESS in 1h 53m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/229e46348437495fb1527d3e99e54260 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fd09d7523924d729479ae6e111b8711 : SUCCESS in 3m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53e57ea1bed14dedaeaabaffaf7c4aff : FAILURE in 11m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0b05f720b79342dd8779929d0f6857ce : SUCCESS in 55m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/63c5e8fbf8254fb2a63fc4a0e0c4e1bc : SUCCESS in 27m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"8ee336090ef4a03418197c1db88968868b721697","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 07:59:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45ef484b9e777ced963cfd4d6991dc5264e1bc2c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-04-28 07:59:15.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"e5add80333b8bc650a34b8d638159db1a5f9f6bf","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-28 10:20:53.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5ca41384a768a21a8799479d4d61ccc9a917f981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-04-29 10:44:33.000000000","message":"Uploaded patch set 5.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"6926879983854d2a02ac74aed521b7fe1ae87269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 12:33:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c03b85987ac455ca8f93cca08f0db3d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6d694551a1fc4b3ba48041507495cf41 : SUCCESS in 30m 02s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/0f73c5d3f34a407ca9af58d458a3fef1 : SUCCESS in 1h 04m 13s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d1eda64a7ae946f39917ec6ea5a93ae6 : SUCCESS in 20m 14s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2d7ba161f3734d8a9ac8fa622058f82c : SUCCESS in 27m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4ad58513869244d881b173438d56d933 : SUCCESS in 1h 08m 34s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/98cb6f3640af4beb91be8392bf5ab44f : SUCCESS in 1h 29m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7f5f9ec08504f7b9c9eb5ebef61dcbc : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1017d72794a04a9a9d20b16616c89fa9 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b8208d94ca844018fee9772b74a15e7 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/930ee4f0347f4ccda86191fd46c6636b : FAILURE in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eab80d8242814a4093c731d42f4c1262 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63f36d7120b4bc08837e3bf0ff3a90a : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/429bfc498fe34a2f9d5a32fd93b619d7 : FAILURE in 3m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ce6fff4afc3d43abb99ae830e97cd9d2 : SUCCESS in 4m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b0cb934f721c4edea5f2d3161e655181 : SUCCESS in 23m 51s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5b69eade1b8e48a4a641358faebdc148 : SUCCESS in 1h 18m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f50dbc628df14f9c998033d60ee87dcb : SUCCESS in 1h 08m 17s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/1d6238d6d541422caf4fcd614af505ed : SUCCESS in 1h 31m 24s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/aae4c06e84b446018a43d2b146eb7c5c : SUCCESS in 17m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/710922ae42f4480f8465239554bb4377 : SUCCESS in 32m 51s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fbb9f49afae44819be17416f4489685c : SUCCESS in 1h 04m 38s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6874398f8a3c4300804b69b9826ad4dd : SUCCESS in 1h 40m 07s","accounts_in_message":[],"_revision_number":5},{"id":"288201803b453f392b08038ead40249677a910f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 12:46:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4a8b1bf055647e6aa50e0f35834e924\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6d4dbbe82d3340899f2951ade34f041e : SUCCESS in 35m 48s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a385fda268c94db79c6ee49d7dbdaee7 : SUCCESS in 30m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2b5f84efe90f4ce58e6fd2471c825120 : SUCCESS in 1h 59m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d47299e0bd174074abc8b4e1d755a75f : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9b33a726f9564ae7b97a666b7f9c70ed : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c86d421a3e70416d9763e9a89e9aa0c6 : FAILURE in 11m 54s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/1f5796fd664843d8a71d1ab3a6d6dd10 : SUCCESS in 52m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/fdc3b481512c43e4aac30058fae38fcd : SUCCESS in 39m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"09e40bde293bd32dedc05690f5b4a20a0ee0abcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-05-02 10:22:10.000000000","message":"Uploaded patch set 6.\n\nCopied Votes:\n* Backport-Candidate-1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"6affa53e2370b0ce0bba5f7cbc0e91609566ea36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 11:53:37.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca8c806eca8043c7bf4c46c54475c23a\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7b83d262f54048ff951ae50a5f75facb : SUCCESS in 24m 57s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/3627b66c8b744b7892659f7ed410cbd2 : SUCCESS in 1h 09m 17s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4b30331471cc4558b79f66460178faac : SUCCESS in 23m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/93851a4cd74145bbb627c66b38944339 : SUCCESS in 22m 33s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/455f18a2bfae4608b35f8f6d94ea9df5 : FAILURE in 40m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4179708169f84f55b938af70d0897c00 : SUCCESS in 1h 26m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/713225076ad04a16a786717cac1c3c14 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82b18d75d5634a1594664d3422d63772 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/706d72576e0048cbb5b4102bb9379f9e : SUCCESS in 2m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f90589d3ae54e8289cfacdf244449ce : FAILURE in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/411800b446cd43dd911af1c5572b5551 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0794b2b4ec4406a7298df2b7d9df22 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ee82552c89c4d4298f760b7cbfd129a : SUCCESS in 2m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ec9a976a7ebe4f4b87d896af1f7c168f : SUCCESS in 2m 58s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a773688b4d4547b1b1e1903da0662967 : SUCCESS in 18m 16s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/a776a57a6e6d4b2d8964e4096d67150d : SUCCESS in 1h 18m 51s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/630d758dbab2437d974a88beaf65ec8e : SUCCESS in 1h 05m 54s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/fabafd3a39044623bd3c0db2fd3cdff5 : SUCCESS in 1h 15m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0b25b2697579445f8550699866dc0687 : SUCCESS in 17m 53s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/005f49a1b8e44456b85441e7da5e7155 : SUCCESS in 23m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f4eb37c6dcbe421aac588ab745b085d6 : SUCCESS in 51m 32s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3f50ec84c60145e9a56cf92f83813ce5 : SUCCESS in 1h 15m 54s","accounts_in_message":[],"_revision_number":6},{"id":"59bdeaaa4b15c0acd0f7d9041cd64bc388ff7f3b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 12:01:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/223454d7f26c4c9681cefd0fc0143f72\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ab4e3bf604a54deea04bfebb01010039 : SUCCESS in 49m 29s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d60180f71055436d9ea96cf0ded85acf : SUCCESS in 35m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/33700db114b5463aa3405b60412b4c94 : SUCCESS in 1h 37m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8b9018ed4e7b4771a3fcaf4a63b38f6b : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50d51061a3f24ce6afcc8d8fae74b354 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dad353069198406aa8f2c0e9d3aaac52 : FAILURE in 6m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ac46037f32d1411da2e7b788339eea37 : SUCCESS in 46m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/68ecc6c63dbd4fef92a9276f26bde1e7 : SUCCESS in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"62524674335fb1714df8c80c85a860cc3a82a967","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-05-02 13:58:19.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"eed384805d0a32d25a16ec5e5c5f045523481b04","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2025-05-02 18:31:02.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2073aa296d03f39a985abad42b26ac27e2ac2f35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 19:40:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2cbad39a2be4aa39da3a7c2973af2d4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/35911a5e22a3416a8e1e2c8b018fd13d : SUCCESS in 58m 47s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1f9070cbf6724209ba57310122b033d1 : SUCCESS in 48m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c3aee1f75d6341b785e0cf4c3752552b : SUCCESS in 1h 07m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9302eb490f7a4a6db7291aa27262a303 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/eda4234430be44dab6afe8202438e5c8 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1115fb771c824889b9040d8458a0e39e : FAILURE in 9m 24s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3e24d7632da9404bb4537799aa43e925 : SUCCESS in 55m 45s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5fe286a679664382863107bf6969ace3 : SUCCESS in 50m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d2bb30e0effd8cfd275f69233392ad5ffca3f5a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 20:06:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22331339b78d443597f22ce6ef3d0069\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/7da3f25680c843faa4803d9cfacb1518 : SUCCESS in 19m 58s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/fe9ecfb0ece34e77ac767b096a59136c : SUCCESS in 1h 09m 26s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f6a0bf4f509b49e88e317e367f6b83aa : SUCCESS in 18m 48s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3c31c9cbb59b409bbdda432c64b919b6 : SUCCESS in 20m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/73a31ff3ad4a4f0f9b86292bf1c55a2a : SUCCESS in 1h 02m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/76cbc6b9facd48d0b5f4d6222857e55a : SUCCESS in 1h 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1150774e2da411c9417a6a926e8dbc4 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c6279c8257184edeb6ee43e8249ae349 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aa5a6262561543fea7944f7ea3610aa0 : SUCCESS in 2m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da34c5dcad5f489cbe9552fcccec77ab : FAILURE in 5m 40s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91bf1f442f534eb0a522f8add4b7dea2 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05bcb7ec71164b17980c6e11c91abf1c : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50db2fae96464ddab5d38e35fd2ae711 : SUCCESS in 2m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d3c89ea5eb4e4bb4a82dd9ba4fcac242 : SUCCESS in 2m 30s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/aafd1876737e4b26b71401beb562bd2c : SUCCESS in 22m 19s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5d933536ac9f443497bbb08adb377b00 : SUCCESS in 1h 17m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/91836f60c2dc4077afbadfe6d3d46478 : SUCCESS in 1h 02m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/6f5ffa9a3b874df6bc6a27ab35a4f9c5 : SUCCESS in 1h 30m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2bef2f2fb39e4ee3b5428b0ab487b7b9 : SUCCESS in 28m 26s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/34c797c37a994a4bb139a51c5e99e752 : SUCCESS in 34m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/69626ba21cb5412b959c7640fda580d6 : SUCCESS in 1h 03m 52s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/032d91cfdb4446438a11aaec1585a9e1 : SUCCESS in 1h 30m 40s","accounts_in_message":[],"_revision_number":6},{"id":"2227c10c0979acde605c5dd8b7a7ab8c0f71afb5","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-05-05 11:52:57.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e879ba46ebd6d6f00f4d5ebbcef93ae7f7e04651","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2025-05-05 13:19:27.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"2c47280e088343f1a2f2d5f7d77ecaa4fefa1c52","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-06 06:03:31.000000000","message":"Patch Set 6: Backport-Candidate-1 Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d92a3a7fb2121e3bb0ebf4a78e480731b90d8984","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 06:04:04.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d44375c36297997ca9ef3c6c17baf1f3d74649ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:45:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e912f74ec3814742914844b99d778bcf\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f9c7605e192e4c15a8c0f0c4325c332d : SUCCESS in 9m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/de4475c1f97f46988f654e09a2a49891 : SUCCESS in 26m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/28809e0efbf7496b9739d632ce53412e : SUCCESS in 1h 03m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/11a69b70a8714886aaaa5a0adc777b4b : SUCCESS in 1h 36m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ba74af2821f4cc7b864688ab21d10c7 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/850733b84b3942bc88401243a9346683 : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3650e0ca43834a4494914e3af7038c1e : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45fb3c2b60d24c129d0816f2c75b95e8 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3357afbd4ec4eb8b22fcefcfb776e96 : SUCCESS in 3m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/766d22fe53be44dfb28a5324406bbfaa : SUCCESS in 2m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/9d3c59e59ead4b72be27a8e49f60cbca : SUCCESS in 21m 54s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6e12cc8c91c14d3c9d6e8d32511aedf7 : SUCCESS in 56m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8ede49f10978444c857ad37f3a17c98a : SUCCESS in 1h 08m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/496cb8807feb4f4f9994f3af7bb080a7 : SUCCESS in 1h 32m 39s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/cdcf94db918e4956bae5b8c60f7f2c90 : SUCCESS in 23m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cda25b4ed6ef45678f882664f619d8f5 : SUCCESS in 35m 01s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/822b7182dffb490688936f4e8a47bf33 : SUCCESS in 1h 01m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3b5d302691c543a8ac53885ecfca6738 : SUCCESS in 1h 34m 45s","accounts_in_message":[],"_revision_number":6},{"id":"fe79789cc094649af4832e182fe060ff7ab5345f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:45:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"916fb9721f132c939565dc9c1474afda4a81143c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:46:32.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac304a7e88db4ad49956c6672ccbf201\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc4ec8660dc24b60be0dc5f15933da25 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01fd19fbd901431c8a2978eaa7627013 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"d85bc282b97a477e8944524019b5bb886cee71da","revisions":{"a538f423cedd9699997c6548f0eacefd9fbcd953":{"kind":"REWORK","_number":1,"created":"2025-04-24 14:45:22.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/84/948084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/948084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/948084/1"}}},"commit":{"parents":[{"commit":"e9d31761441177ac3782182c1af0ea82f65d0f7c","subject":"Merge \"Upgrade Prometheus to v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9d31761441177ac3782182c1af0ea82f65d0f7c"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:42:38.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:43:20.000000000","tz":0},"subject":"Enabling support for Thales Luna user id and group id.","message":"Enabling support for Thales Luna user id and group id.\n\nThe new client software version requires the use of a specific group\ncalled \"hsmusers\", and for consistency reasons, we are specifying both,\nthe user id and the group id, and inserting the Barbican username inside\nof such a group.\n\nChange-Id: Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a538f423cedd9699997c6548f0eacefd9fbcd953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a538f423cedd9699997c6548f0eacefd9fbcd953"}]},"branch":"refs/heads/master"},"98634d2b7d9d333db3f5c8e1ac395c2fe9315bb7":{"kind":"REWORK","_number":2,"created":"2025-04-25 11:47:41.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/84/948084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/948084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/948084/2"}}},"commit":{"parents":[{"commit":"e9d31761441177ac3782182c1af0ea82f65d0f7c","subject":"Merge \"Upgrade Prometheus to v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9d31761441177ac3782182c1af0ea82f65d0f7c"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:42:38.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-25 11:45:19.000000000","tz":0},"subject":"Enabling support for Thales Luna user id and group id.","message":"Enabling support for Thales Luna user id and group id.\n\nThe new client software version requires the use of a specific group\ncalled \"hsmusers\", and for consistency reasons, we are specifying both,\nthe user id and the group id, and inserting the Barbican username inside\nof such a group.\n\nChange-Id: Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98634d2b7d9d333db3f5c8e1ac395c2fe9315bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/98634d2b7d9d333db3f5c8e1ac395c2fe9315bb7"}]},"branch":"refs/heads/master"},"a5ab84f7ea84f9bc992dd32ce19eef677d2d3a25":{"kind":"REWORK","_number":3,"created":"2025-04-25 13:41:21.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/84/948084/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/948084/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/948084/3"}}},"commit":{"parents":[{"commit":"e9d31761441177ac3782182c1af0ea82f65d0f7c","subject":"Merge \"Upgrade Prometheus to v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9d31761441177ac3782182c1af0ea82f65d0f7c"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:42:38.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-25 13:38:59.000000000","tz":0},"subject":"Enabling support for Thales Luna user id and group id.","message":"Enabling support for Thales Luna user id and group id.\n\nThe new client software version requires the use of a specific group\ncalled \"hsmusers\", and for consistency reasons, we are specifying both,\nthe user id and the group id, and inserting the Barbican username inside\nof such a group.\n\nChange-Id: Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5ab84f7ea84f9bc992dd32ce19eef677d2d3a25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a5ab84f7ea84f9bc992dd32ce19eef677d2d3a25"}]},"branch":"refs/heads/master"},"5bac399152c205d939c77483efb41c364d9fb5e5":{"kind":"REWORK","_number":4,"created":"2025-04-25 14:37:28.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/84/948084/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/948084/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/948084/4"}}},"commit":{"parents":[{"commit":"e9d31761441177ac3782182c1af0ea82f65d0f7c","subject":"Merge \"Upgrade Prometheus to v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9d31761441177ac3782182c1af0ea82f65d0f7c"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:42:38.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-25 14:34:59.000000000","tz":0},"subject":"Enabling support for Thales Luna user id and group id.","message":"Enabling support for Thales Luna user id and group id.\n\nThe new client software version requires the use of a specific group\ncalled \"hsmusers\", and for consistency reasons, we are specifying both,\nthe user id and the group id, and inserting the Barbican username inside\nof such a group.\n\nMore information can be found on\nhttps://www.thalesdocs.com/gphsm/luna/7/docs/pci/Content/install/client_install/linux_install.htm\n\nChange-Id: Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bac399152c205d939c77483efb41c364d9fb5e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5bac399152c205d939c77483efb41c364d9fb5e5"}]},"branch":"refs/heads/master"},"7cb6c9824934e6c8e22c10455ce104f825f360e7":{"kind":"REWORK","_number":5,"created":"2025-04-29 10:44:33.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/84/948084/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/948084/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/948084/5"}}},"commit":{"parents":[{"commit":"e9d31761441177ac3782182c1af0ea82f65d0f7c","subject":"Merge \"Upgrade Prometheus to v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9d31761441177ac3782182c1af0ea82f65d0f7c"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:42:38.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-29 10:42:25.000000000","tz":0},"subject":"Enabling support for Thales Luna user id and group id.","message":"Enabling support for Thales Luna user id and group id.\n\nThe new client software version requires the use of a specific group\ncalled \"hsmusers\", and for consistency reasons, we are specifying both,\nthe user id and the group id, and inserting the Barbican username inside\nof such a group.\n\nMore information can be found on\nhttps://www.thalesdocs.com/gphsm/luna/7/docs/pci/Content/install/client_install/linux_install.htm\n\nChange-Id: Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cb6c9824934e6c8e22c10455ce104f825f360e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7cb6c9824934e6c8e22c10455ce104f825f360e7"}]},"branch":"refs/heads/master"},"d85bc282b97a477e8944524019b5bb886cee71da":{"kind":"REWORK","_number":6,"created":"2025-05-02 10:22:10.000000000","uploader":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"ref":"refs/changes/84/948084/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/84/948084/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/84/948084/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/84/948084/6"}}},"commit":{"parents":[{"commit":"e9d31761441177ac3782182c1af0ea82f65d0f7c","subject":"Merge \"Upgrade Prometheus to v3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e9d31761441177ac3782182c1af0ea82f65d0f7c"}]}],"author":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-04-24 14:42:38.000000000","tz":0},"committer":{"name":"Mauricio Harley","email":"mharley@redhat.com","date":"2025-05-02 10:20:00.000000000","tz":0},"subject":"Enabling support for Thales Luna user id and group id.","message":"Enabling support for Thales Luna user id and group id.\n\nThe new client software version requires the use of a specific group\ncalled \"hsmusers\", and for consistency reasons, we are specifying both,\nthe user id and the group id, and inserting the Barbican username inside\nof such a group.\n\nMore information can be found on\nhttps://www.thalesdocs.com/gphsm/luna/7/docs/pci/Content/install/client_install/linux_install.htm\n\nChange-Id: Ib7d0c879b3741ca472cd0bbf3a9a88a0fee25cac\nSigned-off-by: Mauricio Harley \u003cmharley@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d85bc282b97a477e8944524019b5bb886cee71da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d85bc282b97a477e8944524019b5bb886cee71da"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
