)]}'
{"id":"openstack%2Fkolla~948723","triplet_id":"openstack%2Fkolla~master~I42dd2490d14070074fe1fa7ae9e0a4cbf095b06f","project":"openstack/kolla","branch":"master","topic":"bifrost-rpm-fix","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-05 09:54:22.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-06 07:54:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I42dd2490d14070074fe1fa7ae9e0a4cbf095b06f","subject":"bifrost: fix creation of python3 lib symlink","status":"MERGED","created":"2025-05-02 18:12:21.000000000","updated":"2025-05-06 07:54:59.000000000","submitted":"2025-05-06 07:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948723-bifrost-rpm-fix","meta_rev_id":"5f486293ac8ce86d28c7deb24015ed1529d9841e","_number":948723,"virtual_id_number":948723,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-06 07:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-05-06 06:04:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-05 09:54:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-05-06 06:04:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-05-06 06:04:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-05-05 07:33:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-02 19:42:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-02 19:55:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-03 21:14:32.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"},{"updated":"2025-05-05 07:33:18.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-05 09:54:22.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"518a18420961ace6c41c5c603363ef4eed75f1e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-02 18:12:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"98cb92a200ccd8b9e6acfa7514b95bc070559bfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 19:42:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e376ac40c11848e29c84b4efd0044828\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/5136e62239594312b8ef68bf2f315117 : SUCCESS in 54m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/330bab1aa3a244bbb8a0477b80a75d7c : SUCCESS in 26m 31s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/953c379b20c3401b91c3b1c852ff3686 : SUCCESS in 1h 28m 44s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/65e24c52aecf45aba8708b18b83bf7d6 : SUCCESS in 3m 28s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0023035250f9490cb1d9782df8f685e8 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/469dd61210664b35aa30709dabd7f132 : FAILURE in 8m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/60e930f5c06245bd9d18893fede5c878 : SUCCESS in 54m 20s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/bb3e935739b94895a032ee620100663a : SUCCESS in 50m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5bd5b184399ee8811544ecab9ce1c5f34d6b579a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 19:55:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d4278f69c4d46d097c51763868acf20\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fcbad8eb0939457d8bf2139c9808e0a8 : SUCCESS in 26m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/246cd35511154b4baa2a191fdf3e2803 : SUCCESS in 1h 06m 41s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6b76c68d913549f98f5b35d8102979fb : SUCCESS in 20m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b11c47e0eebc4d308fe7bcd7e807bcb9 : SUCCESS in 21m 54s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/a0ca909dbce642479c0c626ec7371cef : SUCCESS in 1h 02m 37s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/ddff1536be7f437181e62a92074e0ce8 : SUCCESS in 1h 34m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d1ee028ff044a7892e9474ef1177773 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af021d4eef5a469a9de1a0b5b6e61a99 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89db850d6e8d4d87ba27743b7bf9e278 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4668457567924f1582ebd5b7062ecc81 : FAILURE in 5m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3505bac938c74673b7564034884e47f5 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4c62d238944d7ebfcbd35e5874e6a1 : SUCCESS in 6m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0ed0072ac03a4edd97d0abed6cd76654 : SUCCESS in 2m 51s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a459cf9dcb7d42d09f150788ce82f33a : SUCCESS in 25m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/64c29993d2394507a1bace1eedb6e595 : SUCCESS in 1h 37m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/050ef59631f748d1a6a161864fc7c800 : SUCCESS in 1h 05m 06s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/14ddc16d28ba44a1a31fa86daeb0b0bf : SUCCESS in 1h 37m 38s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/fde11598fe1e4cbd9226ed78d8d5b0e8 : SUCCESS in 31m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a9831991080c4749acee41a1f5c0cdc1 : SUCCESS in 22m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dbcdd18e83a54dc2955916e60c83b086 : SUCCESS in 35m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f905762f6b30453b82ccf82d03744976 : SUCCESS in 59m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/05c0a443a51a4df78b3444e732d9638b : SUCCESS in 1h 34m 16s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/89354695f1104f48908d5749f6384b94 : SUCCESS in 26m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0fddf9d9aa717a0296ab1af3e17647773cef3563","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-03 21:14:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb9aff0f466826a3a379ba3ae0cb22217147bd9d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 07:33:18.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"6f6bab6a9926893bf92f0a18af62c5e61ff784ae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-05 07:35:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"37a883b94904b6cf2758bb8b0b147dc14b5e4803","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 09:54:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"690305cfb13dbadd59e6f65fb21a7ed98b5fd2b7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-06 06:04:02.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3d11d806ec6758f0c160f0371728cdb4fe6c8b30","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 06:04:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca80e93d87307cb75aa0652ae8836410a0e27dd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:54:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16050a4b062544f0bb1db26987ebf082\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/272a0746c9d14490ba89aa2b58ae243c : SUCCESS in 20m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c641f92b24c849e4a9263e7f6efc91d9 : SUCCESS in 27m 29s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/64a36eef122f4efaa9e5336be764559c : SUCCESS in 1h 04m 10s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9f9b83fdb8954b3697379b2b89aefd4c : SUCCESS in 1h 45m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c8dc43cbea4695b3047ce8379930c4 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f5b6d14d629a43389d5746fa76e0133f : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17b15950e16b4fb3bf1933d59ee7e161 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a118d4269769498aab128a6922104f27 : SUCCESS in 6m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a628ae1e6dcf485d8f4881b32f8ab823 : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/7e63f3edf0334fcbb98e107d74d82054 : SUCCESS in 38m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/f60920c163e942f5b29532f749076977 : SUCCESS in 1h 28m 52s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0a25b851bb454712be77309031334ee1 : SUCCESS in 1h 37m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0f539bfd6ec349f98798d1bb71bbbcdc : SUCCESS in 1h 35m 56s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/45b58af4c108455598d6b5677828557b : SUCCESS in 20m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/041d6a52a4394de5a4077571b7eae240 : SUCCESS in 27m 52s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e1a36d4d3ea84ad1a199c42b43719067 : SUCCESS in 1h 02m 30s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/897f58f434e34459894856c80340c388 : SUCCESS in 1h 35m 44s","accounts_in_message":[],"_revision_number":1},{"id":"00b4d331614a9b31c6c29575f2920f8b0e7d8668","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:54:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5f486293ac8ce86d28c7deb24015ed1529d9841e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:54:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90e3f8a2a79d48ceb134632b59fe6064\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623a8619bef74339981643f09091e6ba : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e33902c30cb7e07d48621d3c574594b2084f1bdc","revisions":{"e33902c30cb7e07d48621d3c574594b2084f1bdc":{"kind":"REWORK","_number":1,"created":"2025-05-02 18:12:21.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/948723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/23/948723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/948723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/948723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/23/948723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/23/948723/1"}}},"commit":{"parents":[{"commit":"aa44d7a9d1a57c8453cfa520b70110fad51e8464","subject":"Merge \"Fix preparation of /tftpboot for ironic-pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa44d7a9d1a57c8453cfa520b70110fad51e8464"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-02 18:11:38.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-02 18:11:38.000000000","tz":120},"subject":"bifrost: fix creation of python3 lib symlink","message":"bifrost: fix creation of python3 lib symlink\n\nWhen {{ venv_path }}/lib/python3 already exist and we run the ln\ncommand, we end up with:\n\n    /var/lib/kolla/venv/lib/\n    ├── python3 -\u003e python3.9\n    ├── python3.12\n    │   └── site-packages\n    └── python3.9\n        ├── python3.12 -\u003e python3.12\n        └── site-packages\n\nInstead, it should be:\n\n    /var/lib/kolla/venv/lib/\n    ├── python3 -\u003e python3.12\n    ├── python3.12\n    │   └── site-packages\n    └── python3.9\n        └── site-packages\n\nRemove the existing link first to ensure it is created properly.\n\nChange-Id: I42dd2490d14070074fe1fa7ae9e0a4cbf095b06f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e33902c30cb7e07d48621d3c574594b2084f1bdc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e33902c30cb7e07d48621d3c574594b2084f1bdc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
