)]}'
{"id":"openstack%2Fkolla~948762","triplet_id":"openstack%2Fkolla~master~I3cde698b93a5da17652b9d39c45b24b3e0d5a50c","project":"openstack/kolla","branch":"master","topic":"fix-easy-install-removed","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-05 18:27:42.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_14200\u003e by clicking the attention icon","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-05 18:26:33.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-05 18:27:44.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_14200\u003e by clicking the attention icon","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I3cde698b93a5da17652b9d39c45b24b3e0d5a50c","subject":"Fix issue with install kolla from source","status":"ABANDONED","created":"2025-05-04 14:49:24.000000000","updated":"2025-05-05 21:00:11.000000000","total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6d319659228d89857a25f3be081ae25d5b2c694","_number":948762,"virtual_id_number":948762,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-05 11:25:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-2,"date":"2025-05-05 17:55:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-05-05 18:26:33.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-04 15:38:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-04 15:39:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-05 07:34:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-05-05 17:55:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"90b364fbde28e30922e3ad606d15194de2bcfdb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-04 14:49:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"01a7a0bfcdf6a9ef3e2ae25db8ca6852d856e6cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 15:38:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7ebe8979d3d4b16bd142a9cf04f9a92\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aa848ad27d45429ca379505b8338b1a7 : FAILURE in 46m 21s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e7322b069b764e46989c15f6e6684e49 : FAILURE in 46m 48s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c43c9e99cf5d4582a0d18b1876ee8076 : FAILURE in 5m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c208bb2b0225479381b1bd2737e4b49f : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/966716aad10a4a228d56ebf712dcd9ea : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4600e5c54d4142fb8bf712164009e8ab : FAILURE in 8m 58s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/93a1e89ca1d54eef92728eb94e4b07c9 : FAILURE in 46m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c9699b0fb7da401a89cc3a193c290e7b : RETRY_LIMIT in 2m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"913067d6b38e2488bf5305bc6eab84414617a60a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 15:39:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09d20f3f32ac46f188393c3182d86d45\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ed096d136a02441eb940dcd91e5ddc01 : FAILURE in 15m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5df5dbdf190c405e8abe1da5819b9762 : FAILURE in 4m 38s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9c734cddee78426ebf94c1aca8dfb345 : FAILURE in 12m 09s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/1f0e7ccf9cb6443cb89d473945c98d4c : FAILURE in 16m 59s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8ff729717c1453fac082a3d500e03b4 : FAILURE in 2m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2f289a99f44442d9a59fe6bf1b3b77ce : FAILURE in 3m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9e6fc8b3e91240a6a13265f64fd9a70a : FAILURE in 1m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223e53627eb044f3ac040cff59b94533 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf3ba1308801438498279f62161826e6 : SUCCESS in 1m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/273b32d5e1c441b18ffaad8792d3787d : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd25f3a3d7534607877c509094a91f70 : FAILURE in 2m 53s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a3aec1df7dbb403ba026723e6e0f192b : SUCCESS in 2m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e639e9afb6c94a35b768644563b964e0 : SUCCESS in 4m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e24ae12359c343ad82bf9572223022e6 : SUCCESS in 3m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1898cabd39b745188ebdad496dfbc42f : FAILURE in 10m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/35c4cc5582df4123b2ee8e35561ae7a9 : FAILURE in 49m 37s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7eebd35db2044be4b2f666be258b986a : FAILURE in 3m 30s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/30e47e871a5140daa942000993fe5247 : FAILURE in 3m 42s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9d011b632e3449be9f517ee7bdec9579 : RETRY_LIMIT in 1m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2fa8cd32143e4195b62a1669611ffff8 : RETRY_LIMIT in 1m 10s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/648f4cca1f5b4b0e8009301b5fe7f139 : FAILURE in 3m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/47130345694c4dabb5657f07687157a4 : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e951fc1ad810c375569557174923faffc9b77e86","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-04 17:55:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45b76413ddd7347d2c3dac5c72b9dd6a1a0a2a65","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-05 07:34:32.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"06c300cc455e4b481a7d1d9ee3c087c5c6ff372a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 09:47:27.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7671d2eb3ff8b7947919c4ea8f762d9d7261e067","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 09:47:41.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d78ea7d7d6a7f3e0d8288955fbbad8f719cd47c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 10:53:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7dbdfc43fc7e4ce9a92fa86c2f25c31e\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1090ceb7734c46e59f55699d75550bdc : SUCCESS in 54m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a94ad08929ef4ac8921b6be8278d2908 : SUCCESS in 49m 28s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/41c5ab2cf3484908ab84d7d90d49461c : SUCCESS in 1h 04m 19s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/135c21f697424ab4adc0d93960afcb60 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a62b876fecc24e89ac2386b328280754 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b336f09315e64b1ea54d552b78bd8d24 : FAILURE in 8m 47s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8381e3e0ee654f508b031a0f036af3a3 : SUCCESS in 44m 19s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1a551155f7614b69ae4e9926fb80c55d : SUCCESS in 49m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"16e9e161f1f4db650b2749da093499a08708e6a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 11:25:04.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/645419ae980c47068ddbe4e937555280\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/d06c396c63274ba38bc97fc83b1be765 : SUCCESS in 24m 51s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/79105117965b4257bb9e9e7bda4cac69 : SUCCESS in 1h 06m 48s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f247ca21699241ccb0b667530d17b733 : SUCCESS in 18m 40s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b0fd4efc31e24833a9995debc0e36411 : SUCCESS in 13m 41s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/06195e3125d44eb0a86d8d0aa894ce89 : SUCCESS in 58m 27s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/abe0fe456b9947519a94b3cbb2ce30f0 : SUCCESS in 47m 49s\n- requirements-check https://zuul.opendev.org/t/openstack/build/01cc130971414a52afa704cf8b448bef : FAILURE in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db2d1af84bc6446fbadffa0ea980ad27 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f529241b0c11471b877d7ce92a6d2afa : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7feb781058847eb8c8b8c5bc22b7c7b : SUCCESS in 2m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3afd47d70cd74fd88b615b078fdf4d3c : FAILURE in 4m 27s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e967f3937d5449b8bf91497491b323b : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe0c90e8b8c44958e523938535f5d8d : SUCCESS in 5m 17s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0a6b9697ff934dc98ed1a4824e36fa88 : SUCCESS in 2m 18s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/890b1aaf81bc403a8c958a456868d542 : SUCCESS in 24m 57s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/ad8c53aa256e4984aead32b9d3b02703 : SUCCESS in 1h 17m 35s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/885a700fe34e4a5ca917eff579b0d47d : SUCCESS in 1h 05m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9e86c9f969d34d59a3d5aa0fb268d8d2 : SUCCESS in 47m 04s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/759a090745924d8db9cb1d06f915198f : SUCCESS in 21m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/82eb5cd17f7e476c853ea9a4bd30be0c : SUCCESS in 28m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2fef90c8edf54ffd8465475ac755acd1 : SUCCESS in 59m 20s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/148544942cce4be48b75300ea93abf4a : SUCCESS in 1h 32m 31s","accounts_in_message":[],"_revision_number":2},{"id":"827e8e465db409362b25d41a34d133046b8922c5","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 17:55:46.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c35ddab85a71cf3472ae5cf2ba7a31f5cbfeb10f","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 18:26:33.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6d319659228d89857a25f3be081ae25d5b2c694","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 21:00:11.000000000","message":"Abandoned\n\n80.3.0 has been yanked from pypi, 80.3.1 should have fixed the immediate issues","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"71d67c57854e218c53774201d909491e8cb681bd","revisions":{"205ecf7cb605efd2a99908948aa580e024bae8ca":{"kind":"REWORK","_number":1,"created":"2025-05-04 14:49:24.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/62/948762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/948762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/948762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/948762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/948762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/948762/1"}}},"commit":{"parents":[{"commit":"aa44d7a9d1a57c8453cfa520b70110fad51e8464","subject":"Merge \"Fix preparation of /tftpboot for ironic-pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa44d7a9d1a57c8453cfa520b70110fad51e8464"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-04 14:41:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-04 14:47:29.000000000","tz":180},"subject":"Fix issue with install Kolla from source","message":"Fix issue with install Kolla from source\n\nSince setuptools v80.3.0 [1] released, Kolla install from soure failed\nwith an error \u0027AttributeError: module \u0027setuptools.command.easy_install\u0027\nhas no attribute \u0027sys_executable\u0027 becasue \u0027easy_install\u0027 was removed.\nNow we need to pin setuptools version everywhere or migrate to the\npyproject.toml.\n\n1. https://setuptools.pypa.io/en/stable/history.html#v80-3-0\n\nChange-Id: I3cde698b93a5da17652b9d39c45b24b3e0d5a50c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/205ecf7cb605efd2a99908948aa580e024bae8ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/205ecf7cb605efd2a99908948aa580e024bae8ca"}]},"branch":"refs/heads/master"},"71d67c57854e218c53774201d909491e8cb681bd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-05 09:47:27.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/62/948762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/62/948762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/948762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/948762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/62/948762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/62/948762/2"}}},"commit":{"parents":[{"commit":"aa44d7a9d1a57c8453cfa520b70110fad51e8464","subject":"Merge \"Fix preparation of /tftpboot for ironic-pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa44d7a9d1a57c8453cfa520b70110fad51e8464"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-04 14:41:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-05 09:47:12.000000000","tz":180},"subject":"Fix issue with install kolla from source","message":"Fix issue with install kolla from source\n\nSince setuptools v80.3.0 [1] released, kolla install from source failed\nwith an error \u0027AttributeError: module \u0027setuptools.command.easy_install\u0027\nhas no attribute \u0027sys_executable\u0027 because \u0027easy_install\u0027 was removed.\nNow we need to pin setuptools version everywhere or migrate to\npyproject.toml.\n\n1. https://setuptools.pypa.io/en/stable/history.html#v80-3-0\n\nChange-Id: I3cde698b93a5da17652b9d39c45b24b3e0d5a50c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71d67c57854e218c53774201d909491e8cb681bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/71d67c57854e218c53774201d909491e8cb681bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
