)]}'
{"id":"openstack%2Fkolla~948764","triplet_id":"openstack%2Fkolla~stable%2F2024.1~I3cde698b93a5da17652b9d39c45b24b3e0d5a50c","project":"openstack/kolla","branch":"stable/2024.1","topic":"fix-easy-install-removed","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-05 09:49:35.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I3cde698b93a5da17652b9d39c45b24b3e0d5a50c","subject":"Fix issue with install Kolla from source","status":"ABANDONED","created":"2025-05-04 14:52:54.000000000","updated":"2025-05-05 21:00:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"fef8323c542ed2a8693c8e12819abad3a0e7c6a0","_number":948764,"virtual_id_number":948764,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-04 15:36:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-04 15:36:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0ed77f37a44f43edbf6527286c9189d45e214951","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-04 14:52:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"813383ce7788f4d7bb544310df24e34dcfc335f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 15:36:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b6d2bd7d4f1414fa8fabe45cce92901\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/80adbdb494ef421693120dd1c026424c : SUCCESS in 19m 40s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5e9b3ad888a9440093ad3fd2e8ea95a6 : FAILURE in 4m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4fe110885fe348788eda8e0d9d45bf38 : SUCCESS in 18m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/84c601ebebf749e1a4944c3838dff0ca : SUCCESS in 28m 25s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/eb1496cec77b418bb46802963c6ed17a : FAILURE in 4m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5aa8ba8022204db6be52bfc895a0f02e : FAILURE in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c14bc15958411cbdc243ce72f89267 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6f80fa6d3c14b6686a78abd2652099a : SUCCESS in 2m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1120b4ff69754034ae2e884435876cbc : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52bb8eb786cd40f29d99fff1bc665e3b : SUCCESS in 2m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0150fa6bfcf4fa88f43f13fb057a58d : SUCCESS in 2m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d2596b87a2264c7f9f218768875402ed : FAILURE in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35ba3759ac6043b8b7cbd30accde3da1 : SUCCESS in 4m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/29b8cf7ca8064b869fc4f4f5d1fb95cc : SUCCESS in 2m 09s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f9325208dd0648f2adf295ffbf4be050 : SUCCESS in 16m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/68e5791fed0242cf99bddec735636eb6 : SUCCESS in 26m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a26507d4cee342fcaf1bdc355291bc70 : FAILURE in 3m 42s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/89d1f6f6de4e4041983ed9db9b0979e2 : FAILURE in 3m 54s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/235cc5d075ed4fcb90422770c564b38f : SUCCESS in 14m 30s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2b1b87139f464807a60fc3702484ea80 : RETRY_LIMIT in 1m 11s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/6fb83d5e24aa4117a43e87805949b2b9 : SUCCESS in 42m 52s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/4ec5914669d84a5db8a4add9a0cc6840 : RETRY_LIMIT in 1m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b67dfd7e676316dd11c53dbfc04813da98d01f15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-04 15:36:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edcc581aaaf14d1490438bcd6920d5ab\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d163159b7d794af38fd5942d9c1e44b4 : SUCCESS in 40m 40s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01e3400c10ab4af09516ec39a6fcb9dd : FAILURE in 5m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89397585e3574942b59551aba1341756 : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2519038c3dc45dfa25f1c78b4381099 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d9ccb8195ea499297ed362afe73b1cc : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"066d6145cacd9a9d0d31abf2fde33f33b25d0f4f","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 09:49:35.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"fef8323c542ed2a8693c8e12819abad3a0e7c6a0","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-05 21:00:37.000000000","message":"Abandoned\n\n80.3.0 has been yanked from pypi, 80.3.1 should have fixed the immediate issues","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7d2161222683edba29a809af4d5537ff8f0773f6","revisions":{"7d2161222683edba29a809af4d5537ff8f0773f6":{"kind":"REWORK","_number":1,"created":"2025-05-04 14:52:54.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/64/948764/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/64/948764/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/948764/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/948764/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/64/948764/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/64/948764/1"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-04 14:41:30.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2025-05-04 14:52:25.000000000","tz":180},"subject":"Fix issue with install Kolla from source","message":"Fix issue with install Kolla from source\n\nSince setuptools v80.3.0 [1] released, Kolla install from soure failed\nwith an error \u0027AttributeError: module \u0027setuptools.command.easy_install\u0027\nhas no attribute \u0027sys_executable\u0027 becasue \u0027easy_install\u0027 was removed.\nNow we need to pin setuptools version everywhere or migrate to the\npyproject.toml.\n\n1. https://setuptools.pypa.io/en/stable/history.html#v80-3-0\n\nChange-Id: I3cde698b93a5da17652b9d39c45b24b3e0d5a50c\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n(cherry picked from commit 205ecf7cb605efd2a99908948aa580e024bae8ca)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d2161222683edba29a809af4d5537ff8f0773f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7d2161222683edba29a809af4d5537ff8f0773f6"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
