)]}'
{"id":"openstack%2Fkolla~948803","triplet_id":"openstack%2Fkolla~master~Ic75fdaedfa62816fae0cc0b7924425b574feb1ee","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-08 06:53:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"Ic75fdaedfa62816fae0cc0b7924425b574feb1ee","subject":"CI: Fix pip mirror multiline","status":"MERGED","created":"2025-05-05 13:45:27.000000000","updated":"2025-05-08 07:46:26.000000000","submitted":"2025-05-08 07:45:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948803","meta_rev_id":"28a3a6915c5eb7ad8c71902054d6276a7f9a6856","_number":948803,"virtual_id_number":948803,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-08 07:45:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-08 06:27:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-08 06:53:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-08 06:53:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-05-08 06:27:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-08 06:27:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 15:48:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-08 03:35:43.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-08 06:27:02.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"1dbc96130e3116e2e8b6e1d1b689e545f2e2044b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 13:45:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c623bd7d6f452661098b061d825a02fb74c70945","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:48:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843b355e52dc47b3894fdc9bb0dd6047\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fe72ab93c7c9477f808174f2c5b75cda : SUCCESS in 29m 36s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/615a11d644414234806eac39efea6c35 : SUCCESS in 1h 12m 52s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/56afd59d84104b76bc1b4042db3a9d73 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/012b7f3ba7294b9b9d9ec012e11c5513 : SUCCESS in 28m 15s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/193711c545754d2cb45fd3affdf9df60 : SUCCESS in 1h 10m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/d30d681e455e4b69a60e139924e03e58 : SUCCESS in 1h 41m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18f19fbfb7f94711829775d50de85b39 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0171144008f4ae098afae4d754d0bfa : SUCCESS in 2m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3dd41e553ca416bb32df6dd59810648 : SUCCESS in 2m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c1d50315893243cb8fad82e9efd1e698 : FAILURE in 5m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bea8220219624e13bd504657d8eab54e : SUCCESS in 2m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f7257628084aa7b728c3f58437fa7c : SUCCESS in 5m 48s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e06faaf371e843b0878790b6f8aa881d : SUCCESS in 2m 32s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/87ee3573d61e4e5897df7fc320b8a16d : SUCCESS in 25m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/3d4e30312e5a46cea4c897cfe228540a : SUCCESS in 1h 14m 41s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/50d9d2dcf7004d69a296d84d4cc88e6b : SUCCESS in 1h 08m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/da1f90db467145cdafa358c5a52b28a3 : SUCCESS in 1h 30m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/78381091eca149a3a259eafded3585e7 : SUCCESS in 24m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2df81fa18935450baf38a621dcc83686 : SUCCESS in 30m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/42d59702dba4415d9e8d07b735105abe : SUCCESS in 1h 07m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/e7ec1998a3dc402daa98ef4deb36b2af : SUCCESS in 1h 32m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a58ecee2dc08c5703d772d5a77b18be1e52c24bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-05 16:17:32.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"619c29d4a1aa1043999cddfd3c8c01ce6cae26ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 17:59:24.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb60c813f96f4c308f0cf6a8699f70a2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8dd75fa5af53492190105fba9a7f1856 : SUCCESS in 24m 14s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/6f5bb8bfe2cc4bdea17e5ad64cb2af41 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a5b1c9fa23004e6fb6a85ce13a235f2d : SUCCESS in 18m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6a4ab4cbaacc4824ad962d0b3eb83dea : SUCCESS in 31m 17s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8d354e68d91d41789de83551a7d972f6 : SUCCESS in 1h 07m 29s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/62236fbcd9924a678f15f436d74ea6e9 : SUCCESS in 1h 36m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7ebe736e124458a1b3444fcac6de7f : FAILURE in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f974a3d0974c49e59b41ed0c3fba7b49 : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f87a9e2295d4db4bf4fca2872967bf8 : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96542e12eadf4c3194b76eea752b843d : FAILURE in 4m 44s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5eba2d6e29f468eb7913f1461184764 : SUCCESS in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/babcea8cc8624913883ad18d65e3c368 : SUCCESS in 5m 46s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2bb35990992041b5b4058199bf12f047 : SUCCESS in 2m 52s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c6f1ecd251584c72ab0c617c1a9fddab : SUCCESS in 25m 52s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/24c0a2ec4452430db5eab2e183f00d98 : SUCCESS in 1h 07m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/353f03ca3bad406688bafa8e84f53545 : SUCCESS in 1h 09m 24s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f7976e5bcc3d45a1893b8f73115d915f : SUCCESS in 1h 31m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2ca141100aab4378989b91cfc9ba4fce : SUCCESS in 18m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a6fad6a643b44d2a89f262d2163fcceb : SUCCESS in 31m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/27c7a50968c44efebb53d1902965b62d : SUCCESS in 1h 08m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/57e13c8a8e904e8c8eaf2a9011917f3e : SUCCESS in 1h 17m 54s","accounts_in_message":[],"_revision_number":2},{"id":"c1ea3c21cc651b804ef1bb1a062759363646e161","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 18:13:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd7e056a2c5b42338278a185e6aeb871\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2be3382101814c6f8e3946e963437e64 : SUCCESS in 1h 08m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4cd1a993545141da8202f157aa911047 : SUCCESS in 40m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/38d66082798443b18cd08ed0b8caeaa3 : FAILURE in 1h 51m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/17ffcdaea2214457ab558c63cac07673 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d8bd28ee958d4e798b6d8d743b0691fe : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d105e36459354f578c4abc8ad1294e7d : FAILURE in 10m 49s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d9b419c8b5674a34af960818b45b4ce9 : SUCCESS in 36m 04s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e02beb1f031a4a79aa91e0dc91b3f66e : SUCCESS in 1h 08m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e06d4b59067a6efeba7ecb8ade8ef139192e0242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-06 06:06:59.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b44ca3fd46f73ce0f8512b5bfc9c546d3f79fd12","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 07:36:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1075d28cb0c347a0ac7281ac1daacfa3\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/e69ab9f525c84830b2698253886a2c9d : SUCCESS in 1h 20m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/418c222fe3c34c08aa6ea856e2eadd10 : SUCCESS in 1h 15m 13s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2067d2c433943a4af4a29f24ba2cf50 : SUCCESS in 1h 27m 30s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e62c1baced354268bfa1e912ea4b41a0 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/98a77cb3c4f84cca84392e03ea42918d : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/006de6a9f1e7431299cc6eca42bc1bc8 : FAILURE in 5m 19s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d6269ebdc8b04ce8839fed35e0f36f09 : SUCCESS in 24m 50s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/372f4f4f42ba4681a9ea9a70b47af6a3 : SUCCESS in 1h 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7ebb43fc299511f4d32e26794eb06cf17d55f9f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-06 09:13:30.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd7327db81d043f18dfda89408375ac2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/fd8867acbf2d47318f91a2c6c4d5f10b : SUCCESS in 24m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/df002cad34674c718acbdd15ff3460e4 : SUCCESS in 1h 11m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fe230459215a4c19b34c19ca4c4c9a27 : SUCCESS in 18m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b37be0840daa44db88b70ab661ecbe85 : SUCCESS in 31m 58s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c74d61a4d1c14d38b226ee0c8263587f : SUCCESS in 1h 03m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/80e2854488f349c7bdf9a3510f34f53e : SUCCESS in 1h 24m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1433d33704f94857836af5c56f664b81 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65b300be0c0f43e78e37e18fff25e08d : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c69ae2cafbd4ab5a3ec529172f48098 : SUCCESS in 3m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a9343443cb364f448352714d6a6c12fb : FAILURE in 5m 17s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84de9a2040e7431cac111d18e99e06a9 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c22ac68dcb14073b997e55710cc007a : SUCCESS in 6m 24s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/52dc9cbff83e493592aa83ed06489b69 : SUCCESS in 2m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/163bdb73453f453e8a2e761db41cabe4 : SUCCESS in 24m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c1e4b2f116d340748ae7fae763bb2eca : SUCCESS in 52m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0c7f91aa2d0b4782a3ca65b525dd982a : SUCCESS in 1h 02m 36s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a54d52593a23410f9a977206f628b933 : SUCCESS in 1h 29m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/18f19f34338a466f841714968b919055 : SUCCESS in 22m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/60c18722fecd44faad16600ecff11704 : TIMED_OUT in 3h 01m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/2dfbba478ec74dac98f7b444eb89b99b : SUCCESS in 56m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1d60fdbe906149b7812c36e5047da2d2 : SUCCESS in 1h 20m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3fcb4fd40f4c1a5dfc72025f632b1eaddff19d6d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-08 03:35:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"319f69dac972590ec406bb60790ba205b72fbe98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 04:42:49.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0def68e80a1f494e9573246892104b9f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d34de71f943a4dc39e881ee251b46baf : SUCCESS in 38m 58s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7340a1ee0902400f85dbd2d333ce6af0 : SUCCESS in 30m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/293e949bd0bd4ad7a52a46d10af5985c : SUCCESS in 1h 05m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f75feffed46c48dd8b42a1649235f3fb : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad8328106ec14cefab838de07e7bf31b : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5a8fb5ffd384a308331b501e8b54e3e : FAILURE in 7m 51s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/874573cecbdf4e419e6716b945afcdfd : SUCCESS in 32m 38s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b3733e6e85974dd79624dae6bb6fef66 : FAILURE in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4cb1bd9c12d6b62f3718721612184f5077f04bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 04:43:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59c28912b4c64c67bcd646aef734f484\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ce93d014d32842068d928a6c7c5f00ec : SUCCESS in 23m 11s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/16b2e053760a43ba93954edaf6d1050b : SUCCESS in 56m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9283957268a44ceab6f3565ad0b217a4 : SUCCESS in 20m 00s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a73bd269427b45a38a19849a08a2aa02 : SUCCESS in 28m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2c03ccf683604e4f9e08f2b740077a84 : SUCCESS in 57m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf6cabe29494796bf2758c6418f21c5 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab403b8076cd42a9a64cb9bcfc84bb78 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e0781e600664a2c8d17bdb2e8161c57 : SUCCESS in 1m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e7bee79c75d45089cfbc5d00915c231 : FAILURE in 4m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd1fc34c68924386875417dd23ed513f : SUCCESS in 2m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2498d5d79fd467a8cb18cea3a7ed20e : SUCCESS in 3m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/65550d1d6cb3468a937f7a609216cd68 : SUCCESS in 2m 35s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0483ea7e92a6426bb051c14bb9658cae : SUCCESS in 13m 18s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/fc68cc064e8642ec8641e34e20838868 : SUCCESS in 57m 47s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/c7b9ca062eac418e96d330f146e446c8 : SUCCESS in 31m 57s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a4d0be9bad4b48619495c0c316595db1 : SUCCESS in 20m 54s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ebdf116f6b80476691eb14e2b1eea192 : SUCCESS in 30m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/660fba1fe0904da4a8ab39106327c4db : SUCCESS in 1h 07m 57s","accounts_in_message":[],"_revision_number":3},{"id":"1513ebab23966a61a3a15d2430c0b6fab45271b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-08 06:27:02.000000000","message":"Patch Set 3: Backport-Candidate-1 Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"dcb998a9c2ac2d60a3a0a835c8c8757b9a46cfb5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-08 06:53:04.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4d0caa032bfbdd1164ce25703c83ba1d0d0274e1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 06:53:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fe3cb452d9336fc2f6feec9b34b65cc46b79b33d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 07:45:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a2481fbd4454467bc465d805bd45855\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7b61f7f8438e401dbf82722add3937b7 : SUCCESS in 14m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3ae5c733b2714e249169db2199f8359c : SUCCESS in 22m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/38cce3ffc13e40fdabeeaf9c00079170 : SUCCESS in 51m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/422bd7d295364873b392766d3cd3e6e4 : SUCCESS in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b70f8ab531c84161a8fc06f9cda40d79 : SUCCESS in 1m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5d5efe21acd44a891c07cc04b83687b : SUCCESS in 1m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9566526ea1a461795b9ac8eb069e9de : SUCCESS in 3m 05s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5c99cc63a0244c21ace418de81805fec : SUCCESS in 3m 06s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/213ce939f6414be8b83deebdc9e05be1 : SUCCESS in 13m 49s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8785158591ad4f7782d8b8b85644ca4c : SUCCESS in 45m 59s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b95d81ce89c646dfaa69b2ae5e38e1cf : SUCCESS in 34m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/79d6d9d6c16e4a6ca0a26eebda36b858 : SUCCESS in 23m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8164fda2deda4ea6be1497bb76270f8b : SUCCESS in 29m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/96094ac451854bb3942bbfd137acebe2 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9f1153e4fcbbfcf18081c30e0f08b034ff089006","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 07:45:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"28a3a6915c5eb7ad8c71902054d6276a7f9a6856","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 07:46:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27cd93562b644145849ef3076cb92a1e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/107685694d1e4ff6a45a4f2a25cba8ad : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"06ac69bd87a2ae97d87aea3daebce61ab25387a7","revisions":{"4bb4af568e95fdb761d535c9f788481d175b6b2f":{"kind":"REWORK","_number":1,"created":"2025-05-05 13:45:27.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/948803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/948803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/948803/1"}}},"commit":{"parents":[{"commit":"aa44d7a9d1a57c8453cfa520b70110fad51e8464","subject":"Merge \"Fix preparation of /tftpboot for ironic-pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa44d7a9d1a57c8453cfa520b70110fad51e8464"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-05 13:45:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-05 13:45:50.000000000","tz":120},"subject":"CI: Fix pip mirror multiline","message":"CI: Fix pip mirror multiline\n\nChange-Id: Ic75fdaedfa62816fae0cc0b7924425b574feb1ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bb4af568e95fdb761d535c9f788481d175b6b2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4bb4af568e95fdb761d535c9f788481d175b6b2f"}]},"branch":"refs/heads/master"},"b07c4c1c9369263575ea8071093eacc926a15e28":{"kind":"REWORK","_number":2,"created":"2025-05-05 16:17:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/948803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/948803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/948803/2"}}},"commit":{"parents":[{"commit":"aa44d7a9d1a57c8453cfa520b70110fad51e8464","subject":"Merge \"Fix preparation of /tftpboot for ironic-pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa44d7a9d1a57c8453cfa520b70110fad51e8464"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-05 13:45:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-05 16:17:32.000000000","tz":0},"subject":"CI: Fix pip mirror multiline","message":"CI: Fix pip mirror multiline\n\nChange-Id: Ic75fdaedfa62816fae0cc0b7924425b574feb1ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07c4c1c9369263575ea8071093eacc926a15e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b07c4c1c9369263575ea8071093eacc926a15e28"}]},"branch":"refs/heads/master"},"06ac69bd87a2ae97d87aea3daebce61ab25387a7":{"kind":"REWORK","_number":3,"created":"2025-05-06 06:06:59.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/03/948803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/03/948803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/03/948803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/03/948803/3"}}},"commit":{"parents":[{"commit":"aa44d7a9d1a57c8453cfa520b70110fad51e8464","subject":"Merge \"Fix preparation of /tftpboot for ironic-pxe\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/aa44d7a9d1a57c8453cfa520b70110fad51e8464"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-05 13:45:40.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-06 06:06:59.000000000","tz":0},"subject":"CI: Fix pip mirror multiline","message":"CI: Fix pip mirror multiline\n\nChange-Id: Ic75fdaedfa62816fae0cc0b7924425b574feb1ee\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ac69bd87a2ae97d87aea3daebce61ab25387a7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/06ac69bd87a2ae97d87aea3daebce61ab25387a7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
