)]}'
{"id":"openstack%2Fkolla~948997","triplet_id":"openstack%2Fkolla~master~If301b9ebd7d0dc38bea4d60467e8e80a7a49825d","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-07 22:16:11.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-07 19:21:50.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-07 22:16:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If301b9ebd7d0dc38bea4d60467e8e80a7a49825d","subject":"mariadb: pin to 10.11.11","status":"MERGED","created":"2025-05-07 08:45:46.000000000","updated":"2025-05-15 05:08:37.000000000","submitted":"2025-05-07 22:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948997","cherry_pick_of_change":949062,"cherry_pick_of_patch_set":1,"meta_rev_id":"8ebb6f9901ad375fa18be4fb2a67073048bce591","_number":948997,"virtual_id_number":948997,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 22:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-05-07 19:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-05-07 19:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-05-07 19:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-05-07 19:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 11:37:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-07 14:16:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 15:53:57.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-07 19:07:41.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"19bf37cb0c05d8b62c343c440474bc9604269a6a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 08:45:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7cc4c769fd46d5165e73041f261b84fe7265d8e5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 09:06:55.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"e033924d30229f855995d66fb44d88f38bb34a1a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 09:44:02.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3d4f34de05aa35ab9189e15f96465dc990a98068","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 10:03:35.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"c357967fdf1a23541e527913f969515d7cf11b3d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 10:33:28.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"83ff19d1b9bd53060c4e193f1d1ddd711a46d5a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:37:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b525c1b4f21408db90594ef89450f76\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/2f9bab7b69c249c19cddb8ce87196eb9 : SUCCESS in 42m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cd74e4f7a93f4351a1e1659f5b1c6d94 : SUCCESS in 40m 38s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/974e6ef08ce44808b7d1c30573a906de : FAILURE in 1h 01m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/61c9e159e32b4da4b71385cfa750c63b : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5ee36ae30664995bebe0990d476cfac : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb8655890fbd4265a27acccb31a6e93e : FAILURE in 7m 16s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cab87541c902427d9ffa034b6c3a6959 : SUCCESS in 42m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e07c8566ce1c4e38ae392d415bab0926 : SUCCESS in 44m 11s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"f07ab8352c6ca8952d9c8c39af05788e94e646fa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 11:39:08.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"85503e2924e87532a7d97846a6feeb16947daa28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:59:03.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27752d28a6174a9a8f12dcf5ae78994f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/ee872c35afc840bda5ba6b6d3a499049 : SUCCESS in 45m 04s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5a75412be8ba49069ba03a5cd5270088 : SUCCESS in 32m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/bb5744e08ee948e898f01c1b542a8011 : FAILURE in 1h 18m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b2851b87ea94027b4a45b3bd0c0133f : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad75a9491cd84a4c931bba41ead5a5e5 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bdc28204e9e542d193c5a3305bdb6579 : FAILURE in 5m 13s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/c9930f4bf92849ea9d8d2c17f1d59845 : FAILURE in 39m 32s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1415af9d20fa47209db0149b21180c25 : SUCCESS in 35m 01s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"d35fb0ade3c7d42e29abc660144bf00030464a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:16:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb6a5ea6c033488382a98412da44a815\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/47e1dc2eb2e2460d947b985440cc6cb2 : SUCCESS in 27m 25s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/18a56d0bdc164dbaa3c9f0c900873e31 : SUCCESS in 1h 09m 44s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4094dc6e6d184cc48b97f84cc71a9fc7 : SUCCESS in 20m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/276130a1ccd945cc94ce289fb08849f9 : SUCCESS in 28m 39s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/10ddc364541f4cc4beeeb2aa1c455a66 : FAILURE in 27m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/47730eb2ba044f25a99c35deb5c8fe69 : FAILURE in 53m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ee11b0672e4c51bff4e277404bfa37 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9cf9887bf1144b58bd10aece986a13da : SUCCESS in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ea8046022afb4fa8aec1c24f4fc40f9a : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2d25d7726e6540e191a535db96c90f50 : FAILURE in 7m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20110d8a6ce94235b68cef8e0b0af91e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79373c79a83e400a838379e58467dc79 : SUCCESS in 6m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a553432ae96c4797a03009e7cd47940c : SUCCESS in 3m 55s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1b200a02e95f4d2e8eedd360e5f9151e : SUCCESS in 20m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c6693ddf90094b63a58a0fe0525f316e : SUCCESS in 42m 38s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8e8b3f2042004caea38644c977387943 : SUCCESS in 1h 02m 18s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/edf91855672044ab8912746038ccc1b7 : FAILURE in 52m 46s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/60a8c331e09e4bd7a5e850fafc911379 : SUCCESS in 1h 51m 52s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c96425a044b849afac6d28dafdd93fec : FAILURE in 1h 28m 15s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/1096aaddb5a8426384844a6b3a7abda1 : SUCCESS in 43m 19s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4b8bc784daf34d1ea54c951a95de0bff : SUCCESS in 27m 48s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/2e856344b76742a0a8e2f53c4545b8ef : SUCCESS in 55m 58s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/7bb85bbdb4294a08a70b6e0e53e2dc91 : FAILURE in 40m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/2d7a6606af534f0688d17e99cde4ca63 : FAILURE in 1h 12m 28s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0c9ad8337540453bbd67e7caad49a5d1 : SUCCESS in 17m 32s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ae8f264dd72a497cba20c8cb79374552 : SUCCESS in 13m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2dec087a6f0743448bb2f4e58fb3f241 : SUCCESS in 30m 44s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b0eefc50d2174f3189e34e411ed8f51d : FAILURE in 40m 15s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/31fa16650cca45c38c244c6e80509405 : FAILURE in 45m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/42d22295233240b3a763c187774e01ed : FAILURE in 1h 09m 35s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b28a80f9d5ce4786a11daed6270dccda : FAILURE in 42m 26s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/58252874c1f848babc70cc8dfd7d8969 : FAILURE in 1h 37m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/4b85ad5e0ac641149ab31fbb14199cec : TIMED_OUT in 2h 31m 54s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b7096161bd4d46c09f33ac421fbfe6e4 : SUCCESS in 56m 37s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/94adbcf21a8e4b978ea746a6dd961f67 : FAILURE in 45m 28s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a0c02ec2f6974c1984094708a206ed37 : FAILURE in 1h 10m 53s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/96d2f701ea8a4d1fa61aed3641e3ea89 : SUCCESS in 26m 18s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"07059c2eeac81464e015bb1890ae19427d164d23","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-07 15:56:11.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"45d0c24275aff21548f86f6ec808776e886410c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 17:51:02.000000000","message":"Patch Set 7: Cherry Picked from branch stable/2024.2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"b4d72d99ba41819fc0f27eb34656c3395c653e38","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 18:00:19.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"689d5e3789a2a62e82198d3495f922be80e6091c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 18:00:45.000000000","message":"Patch Set 8: Backport-Candidate+1 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1809232b28deceb7d470ae76d932f5860e7ae6b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 18:11:50.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7be628c8a7eb98b6698775521443afaf13f66318","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 18:11:59.000000000","message":"Patch Set 9: Backport-Candidate+1 Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c550c9a89565e6d9da1c4fb36a6b3fa10be6adb8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-07 19:07:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"bbbd5f4d83611cb77fba154ebc83e2aa5d31e9b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 19:21:28.000000000","message":"Patch Set 10: Published edit on patch set 9.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":10},{"id":"e3089e7a0900da2a744a31ff57195e59d2ab68cc","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 19:21:50.000000000","message":"Patch Set 10: Backport-Candidate+1 Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bb9cbde1c177726036df45155349c22d969df312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 20:22:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f2160109ad4d6396f651449c66b61f\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/9fd8da92287c4dd49a04648819393ae9 : SUCCESS in 37m 23s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/84a6e3d629c54e8f993c8f2885f53c7d : SUCCESS in 29m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ebcc4c9caf43404da6db45ea0fba74fb : SUCCESS in 59m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b35a91a010a64ff58c84dcfc7a1ca1d2 : SUCCESS in 2m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e256e724cc574f93a2ba49fa8fff8611 : SUCCESS in 3m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ef8aba64594c4771a28c51e38a245c1a : FAILURE in 7m 29s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/49e4b206b6754ba49e1504ab783ca314 : SUCCESS in 40m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/37df25d42f864f2cb9b23474deb074a8 : SUCCESS in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"a119a8301c4a5cf0062ecea050a3fde2b0905a1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 21:03:18.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7ea813c405b40d2aed4d5b41f8693e2\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/da44da8cf7754eacafbd0f27d4efd84e : SUCCESS in 22m 56s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e0529aa52ce34384b1e5ee5d30ad63b4 : SUCCESS in 1h 10m 43s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b135525ee904952b6b74399b2e01311 : SUCCESS in 21m 15s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/10dba14f8c63436a9c1a7ff2ae7238a5 : SUCCESS in 14m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/29894dc02e974a3c9d6dba2eda972c14 : SUCCESS in 1h 01m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48ffcaa0f574e4f805b9b74931a68cd : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/337d86a9fab748e893ef609f8cbb0ff5 : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e489949b1f9641a29d2d6b1af26db7a9 : SUCCESS in 2m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/833b42b103f54159bd0dec92d93d8899 : FAILURE in 4m 36s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/304c78491d3e456f90fc8d192362a300 : SUCCESS in 2m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e482ca34e8d146f192a90818c0d133e2 : SUCCESS in 5m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/fc8ba47a526b44ffaa718c195f8ab820 : SUCCESS in 2m 44s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/6e2ea86a5c264324a9c11287f6b071e9 : SUCCESS in 17m 28s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/5c0ad17cedd14267848fb07d5cd756ae : SUCCESS in 59m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/58da01a1dfa14abf86922a6b046a1cb2 : SUCCESS in 55m 59s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/ad61e1d89bfc4424aa2cafc3dbec2986 : FAILURE in 1h 39m 15s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/f7852d55ddae4bf0913e4ca82946f943 : SUCCESS in 1h 07m 02s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/655c05beed864e5e9b5f63455c490348 : SUCCESS in 18m 31s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/62303e6d19954bf8a198e23ac691f6d3 : SUCCESS in 49m 15s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/c0ec648d3cc54d00bb13881f50b0702d : FAILURE in 46m 32s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/d9c95a5ae16d483ba45c87495b9c54b8 : FAILURE in 59m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/656dc4cc4c5048dbac636f3008306564 : SUCCESS in 15m 16s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c3419ca7b02b4d4a98e789a1a4005e53 : SUCCESS in 23m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ad023b22df0945eaa612905f6a258666 : SUCCESS in 28m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/421736938c0f4fd29611c8be6588a6c1 : SUCCESS in 58m 56s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/596a648054234f6db6dcd94f315cef0b : SUCCESS in 1h 40m 53s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/90c862df77de412b8ddbd24673daf6ea : SUCCESS in 50m 12s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b329c5accee5492196c9bf497e2328cb : SUCCESS in 23m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/d3261c923e3845118f52064cc1cd4ee3 : SUCCESS in 1h 01m 07s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/9f24749523da483d8e284d67c816cef0 : FAILURE in 52m 47s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a88c037dd93846dd8e67ecdb6bd3c88b : FAILURE in 1h 09m 53s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0ae4e8e586074b3ab66c85f5c015952e : SUCCESS in 13m 42s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"f172e516b71ca5c2a0151db53c2c2f40270505fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 21:03:35.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"5890d32a883054a99ccd62d2e03936aa1e3f19ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 22:16:11.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0272b76453f44eaf978075da6d11afea\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/27e1ca996b1a41d789baa6308f5777cc : SUCCESS in 17m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ef1774fad44c47168436dda67cf3da1b : SUCCESS in 22m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5f600e3fc1e64471bbd531b89ef38c2c : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e036614dccf423b8eb9b2670df52a3a : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84dfd0e968eb4bc29aff8ebf8d3229d2 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44cb975dca5b40539c46e3272c8c1d58 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e13f00195e6647adb7cbcb8c11b72c29 : SUCCESS in 3m 21s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/09dfea154f5648b893be7c617036b060 : SUCCESS in 3m 08s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a1dbcd33e9c94dac9f5969c4fa2deb34 : SUCCESS in 20m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/e597459192f54f90b3eb23b4dab756c4 : SUCCESS in 1h 07m 21s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7d5314d37bd94fd08dab9662a2dbdd6f : SUCCESS in 55m 20s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6e46356059c74458803c6f9b674ac72f : SUCCESS in 27m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5cab1a5179d245cbb7dffe1a4eee5e5a : SUCCESS in 13m 45s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4723434cc0254df2b15c98399ef91ff9 : SUCCESS in 1h 09m 23s","accounts_in_message":[],"_revision_number":10},{"id":"dc6c246bc4526ac0874133178df6f1a6624a3bc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 22:16:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"2172b2f51800c574e2edc8c152be9314763ee643","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 22:17:36.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b04ce20a1224b379b4b8ca581b5e872\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/790c62f4ef914fbb87425c5ea6a3c06f : SUCCESS in 43s","accounts_in_message":[],"_revision_number":10},{"id":"8ebb6f9901ad375fa18be4fb2a67073048bce591","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 05:08:37.000000000","message":"Created a revert of this change as I0a8cdc232e5b46cb47e4a3e1b0af7b2402103690","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"535b2f270faf24cfa30b9aa1a896a62c5090f1aa","revisions":{"c15d4dc88364d98249451ff4e2a6e7590338efd1":{"kind":"REWORK","_number":1,"created":"2025-05-07 08:45:46.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/1"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:46:09.000000000","tz":120},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c15d4dc88364d98249451ff4e2a6e7590338efd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c15d4dc88364d98249451ff4e2a6e7590338efd1"}]},"branch":"refs/heads/master"},"01f501c0ed12da69f4a45d9e382f502269854100":{"kind":"REWORK","_number":2,"created":"2025-05-07 09:06:55.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/2"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 09:06:55.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01f501c0ed12da69f4a45d9e382f502269854100"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/01f501c0ed12da69f4a45d9e382f502269854100"}]},"branch":"refs/heads/master"},"288aad26609c991ee06fcbfc6595ca2f57bd7ac5":{"kind":"REWORK","_number":3,"created":"2025-05-07 09:44:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/3"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 09:44:01.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/288aad26609c991ee06fcbfc6595ca2f57bd7ac5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/288aad26609c991ee06fcbfc6595ca2f57bd7ac5"}]},"branch":"refs/heads/master"},"d1786ed56e32b140326d4b145825b7fd8c4c1b04":{"kind":"REWORK","_number":4,"created":"2025-05-07 10:03:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/4"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 10:03:35.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1786ed56e32b140326d4b145825b7fd8c4c1b04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d1786ed56e32b140326d4b145825b7fd8c4c1b04"}]},"branch":"refs/heads/master"},"6f208e82333ddc5a6ed03ef611e6e095aa38033a":{"kind":"REWORK","_number":5,"created":"2025-05-07 10:33:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/5"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 10:33:06.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f208e82333ddc5a6ed03ef611e6e095aa38033a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6f208e82333ddc5a6ed03ef611e6e095aa38033a"}]},"branch":"refs/heads/master"},"27bfcb327a7b228320890283f9612566daba81f2":{"kind":"REWORK","_number":6,"created":"2025-05-07 11:39:08.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/6"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 11:39:07.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27bfcb327a7b228320890283f9612566daba81f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/27bfcb327a7b228320890283f9612566daba81f2"}]},"branch":"refs/heads/master"},"7f62830315113b9773006985eb9173c3bedc2aac":{"kind":"REWORK","_number":7,"created":"2025-05-07 17:51:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/7"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 17:51:02.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f62830315113b9773006985eb9173c3bedc2aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/7f62830315113b9773006985eb9173c3bedc2aac"}]},"branch":"refs/heads/master"},"8ee5d43999f42a279f96679012225386516c8038":{"kind":"REWORK","_number":8,"created":"2025-05-07 18:00:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/8"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 18:00:19.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nTemporarily disable upgrade jobs to get epoxy rc1\nbranched this week.\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee5d43999f42a279f96679012225386516c8038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8ee5d43999f42a279f96679012225386516c8038"}]},"branch":"refs/heads/master"},"bb5726d2890d5c6fbe78ec69ceb175754e258810":{"kind":"REWORK","_number":9,"created":"2025-05-07 18:11:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/9"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 18:11:50.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nTemporarily disable upgrade jobs to get epoxy rc1\nbranched this week.\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb5726d2890d5c6fbe78ec69ceb175754e258810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bb5726d2890d5c6fbe78ec69ceb175754e258810"}]},"branch":"refs/heads/master"},"535b2f270faf24cfa30b9aa1a896a62c5090f1aa":{"kind":"REWORK","_number":10,"created":"2025-05-07 19:21:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/97/948997/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/97/948997/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/97/948997/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/97/948997/10"}}},"commit":{"parents":[{"commit":"02c54efb42251235d421269e66ae7f5b30d7e4e9","subject":"Merge \"Allow specification of a key type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/02c54efb42251235d421269e66ae7f5b30d7e4e9"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 19:21:28.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nTemporarily disable upgrade jobs to get epoxy rc1\nbranched this week.\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535b2f270faf24cfa30b9aa1a896a62c5090f1aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535b2f270faf24cfa30b9aa1a896a62c5090f1aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d1"],"failing_atoms":["label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
