)]}'
{"id":"openstack%2Fkolla~949013","triplet_id":"openstack%2Fkolla~stable%2F2024.1~If301b9ebd7d0dc38bea4d60467e8e80a7a49825d","project":"openstack/kolla","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-08 05:14:59.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-08 08:35:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If301b9ebd7d0dc38bea4d60467e8e80a7a49825d","subject":"mariadb: pin to 10.11.11","status":"MERGED","created":"2025-05-07 11:39:53.000000000","updated":"2025-05-15 05:07:50.000000000","submitted":"2025-05-08 08:35:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949013","cherry_pick_of_change":949021,"cherry_pick_of_patch_set":8,"meta_rev_id":"69d850a690e4449777bc94feda3c9368f4bdb84c","_number":949013,"virtual_id_number":949013,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-08 08:35:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-05-08 07:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-05-08 07:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-05-08 05:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 13:11:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-07 13:38:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 15:54:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"c05a191288d1101dbdd17475e228c7c5f3aa8b0e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 11:39:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7097ffb25714aaf8ff933f51534595b5ce045314","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 11:47:40.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"5c87b46f5a692b2e776eecbd589bbd2d8fe2893d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 12:00:16.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"044c1ce72865f722044bdeb43587e99020d5c06a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:11:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea29567ff2b64b97aa05e976b85a6efa\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/91e681e124b14fe981092056174062e6 : SUCCESS in 36m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2ef4457aeffe4b078021bb8846917d63 : FAILURE in 1h 24m 32s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a5d869d6f57d495fa17e817984c00669 : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/74efd322107646cd8b8dee24f1957daa : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bdbae86379ed4e08a5333bf1a35bda18 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"8423c1e0dce2a80cc15e4634ee5af670f930dd6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:38:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86623e3682bf4f908f27b1d8da664ac8\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/f9ef232e336b46a3b96a1075e6495751 : SUCCESS in 22m 35s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/af481e6d92b341e1a001fe3eeaee5e62 : SUCCESS in 36m 19s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/cbfeceff5669497a994224213caa7563 : SUCCESS in 19m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/01a28b5d5dce415fae54f478246a90c7 : SUCCESS in 27m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/799caf3add94442e8e118ab664f8bd5b : FAILURE in 50m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/24112b5348384e94b515eac8396776f6 : FAILURE in 46m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/504a08972a2c4dd882cea067caf19037 : SUCCESS in 2m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f5bb8b541f849eabb4c425f258344ae : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fa9779c3df0c4d7595d83c89767f8540 : SUCCESS in 2m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce22c16d27fc46b3ba3ab5c7efd97b37 : SUCCESS in 2m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6bdecd519c14448dbafcfd3f463538da : SUCCESS in 1m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4284ff65eaee4f1885b135f5fa98cd44 : SUCCESS in 4m 37s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/2b5400fb8915405eb70cf9ba2bf35735 : SUCCESS in 2m 31s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/67289d9fb80749e0b19eae5ff977d7f1 : SUCCESS in 19m 20s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/b3153c36e2074bb2b8c200414c6ee582 : SUCCESS in 19m 32s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6fec9b2c9fa54e15bbfa16f7f3d7047c : SUCCESS in 33m 40s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2203e8ab39a84d4297996481c18ade3f : FAILURE in 27m 17s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/4671db5994ad4c6db4127c3dfaf190b5 : SUCCESS in 1h 51m 18s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c479b9b9b09a4827aa9b1693ea8a6972 : FAILURE in 12m 39s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/76555a0249314842b573ffa1bad26939 : SUCCESS in 1h 11m 21s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/29fae1cd6a014defa795a8ab809c4e53 : SUCCESS in 1h 29m 30s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/70f6b65a542c45b0959d2218b46c4f79 : SUCCESS in 27m 49s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/31d03ffa5c1442db88ecd2610140b85d : SUCCESS in 1h 00m 56s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/0c52895ba4604ded847ef3aad75744f6 : SUCCESS in 56m 42s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b241322e83da4d0aafac2057a4f11e17 : SUCCESS in 1h 39m 51s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/5fe486e5f451450b91f10866662a2cbf : SUCCESS in 27m 27s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b3eb9a286fa4ebbab21d5691940f006 : SUCCESS in 13m 36s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/d4ab291735944b7f834b3f71a3a31104 : SUCCESS in 18m 57s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/1b4d8ab101a049258d70a867cef5534a : SUCCESS in 1h 06m 16s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/d89e07d4b3bb42da8f8383b985c49c36 : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"3132f7edb1d892f7209afb5a282757dd4b742a22","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-07 15:54:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0530f4c526d38f9e99fe578d9b4a984faa11ec88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 16:37:39.000000000","message":"Patch Set 2: Cherry Picked from branch stable/2023.2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"27370eada25b866502b40296e9b1643d30814351","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-07 16:41:57.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a394d0c2900e2d0c6405f9825533dd626b68673","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 17:42:34.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":3},{"id":"7497fb7c4a3e8cde62e3235f0883671376db53bb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 17:43:12.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b2820d6e260cdbbf76dc0ebfe8501b445f1eedb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 17:44:05.000000000","message":"Patch Set 4: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"68add0072a3a3777a58a8a323b45987a5928b089","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 17:44:11.000000000","message":"Patch Set 4: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"4118a46ed9a24194d0340985e16a09485df55f19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 18:09:01.000000000","message":"Patch Set 5: Cherry Picked from branch stable/2023.2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"31466049a46345d05c7938672cff060b239c06a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 18:09:21.000000000","message":"Patch Set 6: Published edit on patch set 5.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":6},{"id":"76c27619f0273168e23ab4a70d69ca611c9818d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:22:23.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c658192ffc1c446b8be121d1edd2bf33\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1379e9a7dc684c89bf3d2e081ebc2d5f : SUCCESS in 37m 09s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7810dc5ec53d49d7beb83a21c40718b3 : SUCCESS in 1h 11m 04s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/340a8af56ad04229ab6c42e4d7cf2e0e : SUCCESS in 3m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/28839e5d74f24758852ed95372ea3bfe : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5f3718e4b6ae439c9e4c4c2acda39b4d : SUCCESS in 3m 48s","accounts_in_message":[],"_revision_number":6},{"id":"63464736057dfc39b5f63065284764e4b7db412a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 19:43:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e5858f894aa47e28be3d0649ee64a2b\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/4ca5660417b74380ad7f1967b7b23136 : SUCCESS in 14m 46s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/987d5968f344444b9233c7818eab7e92 : SUCCESS in 57m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/724533f08c73454d88828b6eebd17723 : SUCCESS in 19m 54s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/bae9f2654ac6416aa1ec2dc13c73a2cb : SUCCESS in 29m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f1ecef9252794d0f908df57e5dd82853 : SUCCESS in 57m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/934f5c6bef7c456fa1ad2a1c594e792c : FAILURE in 55m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00aebb11119143d99c1d2494a72fefc6 : SUCCESS in 2m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40c36974c73640ab96bbf23ef6190f8e : SUCCESS in 3m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/210283f2278a44b7b7ac87d98b9d4212 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f91a760e95fe471dbad0ce07751f3d5c : SUCCESS in 1m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f01e5b36036440b8a648dc0959ccbf2a : SUCCESS in 1m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7002f87f072f40bd86647582831f535a : SUCCESS in 2m 31s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c9709dd759f94de789292595d37c67d2 : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f7b3dc8c1eb0446a8c6ce95b9d227cb3 : SUCCESS in 16m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1db32bae005243678ec3b601b0d199e8 : SUCCESS in 23m 18s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f389d64d564046f4a38f7ffb012a0b3c : SUCCESS in 1h 00m 58s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/9e894c25a41140c7bd7bae577b7defbe : FAILURE in 49m 51s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/620b4d4a21fe4c6a995ab650d5aa38c1 : SUCCESS in 1h 25m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7b6198a22ef44132a3401fb4c42fd844 : FAILURE in 18m 52s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/2a62a26c1fc1473aa339c0b297ddf603 : SUCCESS in 1h 02m 04s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/b00fe11f888e4158a0d424f6039f853a : SUCCESS in 1h 24m 34s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/fe9ae92f011a439eaf6cf62035e984c0 : SUCCESS in 22m 38s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c51ae7f5d5724dfeb8604aa05962c3eb : SUCCESS in 59m 08s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/ca193fff41b54a8b96ae3394ef5f5ca1 : SUCCESS in 1h 00m 54s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/c13eaa4bffe34cbf801afe624b60a569 : SUCCESS in 1h 33m 16s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f4b4e8299a9a4510b8487153c6c06bf2 : SUCCESS in 16m 30s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64e001ade31d455faf24d5a53c3c3fa0 : SUCCESS in 14m 17s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/db4ac1ab814a4ac8935cbc24b3fc2028 : SUCCESS in 22m 17s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/de8f855aff7944d284b1ff0da18f069b : SUCCESS in 1h 00m 56s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2df6fadb80e84d0cbf6867df30c92c47 : SUCCESS in 25m 56s","accounts_in_message":[],"_revision_number":6},{"id":"b76316627be06a0e7e02120c12382d370611fa26","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-08 05:14:59.000000000","message":"Patch Set 6: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c4ff46525cb5c3c67ad5815594dd21373903c7a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 06:40:02.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c724cf90bc3d45ee9f3ce48671d74da3\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/acb06dddbb144cd9bc7b41efbed9da9f : SUCCESS in 27m 45s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/527937db9ca941f6b2cfbdfd04c52f0a : SUCCESS in 1h 23m 11s (non-voting)\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9b52c93327349e8930e2fb1f11cd49b : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff69598cafa64de99a148bb02bba7c88 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0efd0757a95e46fca9c5d62bf4b56bd0 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":6},{"id":"89d7ccdfc3f6770b89d6e42b9ac30e8e8057630c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 07:19:22.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4220ac0db4dd49769d4882181b95fbbd\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/63c76e9820674a46a6c3c3d8de2eefc4 : SUCCESS in 17m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/7443017587694b29a7bfcac671003206 : SUCCESS in 1h 05m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f2e92e571581448ab6a7bfd8235f065e : SUCCESS in 15m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4824808413b74dff8d98249558680ec7 : SUCCESS in 21m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d46c38b5cd734ae79b103510db62f2ae : SUCCESS in 1h 07m 17s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/49af02b6e00e484f9d4ac8ee36bed942 : SUCCESS in 1h 29m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922686509ca040b6b7e184612dabf0f4 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521666e2a8914bbe986eeafb969652f9 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea696454a90a4305a78da38b68666dfc : SUCCESS in 5m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f19b8d1e61742f3bd1493ba55c708d1 : SUCCESS in 4m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e36782f2ebf4122824a5c1e9b724016 : SUCCESS in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff71451dbcfe4e5bb05f26112d9973bb : SUCCESS in 5m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a13a9a2aa6984c0397ce7ab77e08aad5 : SUCCESS in 2m 22s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/0e8a9a3877f243d1863bd8eebe6ff78e : SUCCESS in 18m 48s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7894443d8d34407a92a8ed938631506f : SUCCESS in 28m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/1249d84d939f4a0caef272c44e16ff29 : SUCCESS in 1h 00m 21s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/886c8d63f7434f48ad59b0c6738d440d : SUCCESS in 1h 23m 53s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/e76dc6d1ae914bfb883790992e408e14 : SUCCESS in 1h 45m 33s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7b75ce95a19a40898997169ecf8fc8a8 : FAILURE in 10m 31s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/83776129be2e4411ae099da9cd0a629a : SUCCESS in 1h 21m 11s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/a12a401423b6404081664df733295dcb : SUCCESS in 1h 03m 59s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/113206f49c9e4886a0cf54a366d01744 : SUCCESS in 16m 48s (non-voting)\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/97926c7811ed400299f66bcc51d34216 : SUCCESS in 59m 44s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/be2a07f100464ad0a5e2e35c0dbb3119 : SUCCESS in 1h 14m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/cfe4468855a04a8081e87f87281a5f21 : SUCCESS in 1h 06m 25s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/909eb7be572649a8b485d7882bb33301 : SUCCESS in 14m 47s (non-voting)\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4376f2d9d54946f89b7ccd2d0af54633 : SUCCESS in 16m 42s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/36a8034125bb4bbb81c631b7fc9e45f5 : SUCCESS in 16m 44s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/d166715a31574aaaa44dd7c2afa30614 : SUCCESS in 31m 04s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/2e364d9cbf864c6bb45fa483abea5694 : SUCCESS in 16m 21s","accounts_in_message":[],"_revision_number":6},{"id":"667c7fb8632526228dc822e33541e6a54bc44c33","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-08 07:25:14.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"155993ddc56fbfc69e4cc272942df01929faaf1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 07:25:25.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"935717f0b8a07516f4272a6fd1cd54892a6dc9e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:35:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0d5abf463ab4a3581f92e07a0c94184\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/829ff5f6ce0e4cfb838a4b8f361f2931 : SUCCESS in 14m 39s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/96fb9eb497bf4f84b7909bdeee521764 : SUCCESS in 21m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/3436e1fad0474917a1dfe7d3f3d91c36 : SUCCESS in 49m 07s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a9ee188f258d494999b5518ea5b0d951 : SUCCESS in 1h 09m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f1a130c9262421e9b4fd2ae8c389951 : SUCCESS in 2m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1802d42061074e20b51237f215acae76 : SUCCESS in 2m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8ea5e55e6c23430ebcb42c5f40732e56 : SUCCESS in 2m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d7dcf4e24414ecc8827c0fa7096c63b : SUCCESS in 2m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/df78d81e956a4dee809d87e2c30ae883 : SUCCESS in 2m 12s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/07fec1e5f6e0471eb5deabfbb74b9c03 : SUCCESS in 10m 56s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/72c8d1d133de4f4b9391cd6e18d4ac2e : SUCCESS in 23m 25s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/e54ad91eccf846e1981718f1a7ce4ade : SUCCESS in 57m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/420223c74fa34b18b3bda03e13d884b1 : SUCCESS in 44m 14s\n- kolla-build-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1c019cf97cbb4cbbbcb6e209320a943a : SUCCESS in 16m 08s\n- kolla-build-ubuntu-noble https://zuul.opendev.org/t/openstack/build/17a9a393584045bd82cc1b2d713a1c14 : SUCCESS in 12m 53s\n- kolla-build-ubuntu-jammy-podman https://zuul.opendev.org/t/openstack/build/87debd723fe84c7184cbe1258dcdeee5 : SUCCESS in 57m 50s\n- kolla-build-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/845d201b204b4600ad83dd09c137de28 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":6},{"id":"9002fb82842de57d9db0d37de885bbc03863b00d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:35:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"6f391f8f4905892a25a56353970e9d47238410bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:36:36.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f5f42b0cae4d42bcf838be416c0742\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bb68d327164414a93daa8fc2ab0ca38 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":6},{"id":"69d850a690e4449777bc94feda3c9368f4bdb84c","tag":"autogenerated:gerrit:revert","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 05:07:50.000000000","message":"Created a revert of this change as I9d559990f96e6d293da6baf8d9b7680be58ff440","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"eb29ede7b3d5d82671cde525b9eaa7fcc4a31ab0","revisions":{"61e647cca92687c0c14ea1d992c6db9c32cb4867":{"kind":"REWORK","_number":1,"created":"2025-05-07 11:39:53.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/949013/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/949013/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/949013/1"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 11:39:53.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61e647cca92687c0c14ea1d992c6db9c32cb4867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/61e647cca92687c0c14ea1d992c6db9c32cb4867"}]},"branch":"refs/heads/stable/2024.1"},"6655329d664442a1b0cfff13be9336acd8497a89":{"kind":"REWORK","_number":2,"created":"2025-05-07 16:37:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/949013/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/949013/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/949013/2"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 16:37:39.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nMarking bifrost as unbuildable has been included in this\nchange.\nBifrost is failing because it\u0027s trying to fetch\nstable/2023.2 instead of an EOL tag.\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6655329d664442a1b0cfff13be9336acd8497a89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6655329d664442a1b0cfff13be9336acd8497a89"}]},"branch":"refs/heads/stable/2024.1"},"898f72b7c999e120a4e7c47dc01a9ef62bc7da09":{"kind":"REWORK","_number":3,"created":"2025-05-07 17:42:34.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/949013/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/949013/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/949013/3"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 16:37:47.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nMarking bifrost as unbuildable has been included in this\nchange.\nBifrost is failing because it\u0027s trying to fetch\nstable/2023.2 instead of an EOL tag.\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/898f72b7c999e120a4e7c47dc01a9ef62bc7da09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/898f72b7c999e120a4e7c47dc01a9ef62bc7da09"}]},"branch":"refs/heads/stable/2024.1"},"4b32e42320eb4462b69e9e4507c454842060dbc7":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-05-07 17:44:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/949013/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/949013/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/949013/4"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 17:44:05.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b32e42320eb4462b69e9e4507c454842060dbc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4b32e42320eb4462b69e9e4507c454842060dbc7"}]},"branch":"refs/heads/stable/2024.1","description":"Edit commit message"},"e7ca1c6bd5b62b43e60a31c14e29b5c44e4866aa":{"kind":"REWORK","_number":5,"created":"2025-05-07 18:09:01.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/949013/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/949013/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/949013/5"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 18:09:01.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nMarking bifrost as unbuildable has been included in this\nchange.\nBifrost is failing because it\u0027s trying to fetch\nstable/2023.2 instead of an EOL tag.\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n(cherry picked from commit b0cd0dfe6eb8ba133fcfc09e6a6b91edf5de6f96)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7ca1c6bd5b62b43e60a31c14e29b5c44e4866aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7ca1c6bd5b62b43e60a31c14e29b5c44e4866aa"}]},"branch":"refs/heads/stable/2024.1"},"eb29ede7b3d5d82671cde525b9eaa7fcc4a31ab0":{"kind":"REWORK","_number":6,"created":"2025-05-07 18:09:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/13/949013/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/13/949013/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/13/949013/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/13/949013/6"}}},"commit":{"parents":[{"commit":"1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5","subject":"Merge \"[2024.2 and 2024.1 only] Fix issue with \u0027pycadf\u0027 installation\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/1e2bcd5f26efbae0d80c683a6f0e56e6ee4accb5"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 08:44:39.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 18:09:18.000000000","tz":0},"subject":"mariadb: pin to 10.11.11","message":"mariadb: pin to 10.11.11\n\nCloses-Bug: #2110122\n\nChange-Id: If301b9ebd7d0dc38bea4d60467e8e80a7a49825d\n(cherry picked from commit b0cd0dfe6eb8ba133fcfc09e6a6b91edf5de6f96)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb29ede7b3d5d82671cde525b9eaa7fcc4a31ab0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/eb29ede7b3d5d82671cde525b9eaa7fcc4a31ab0"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
