)]}'
{"id":"openstack%2Fkolla~949022","triplet_id":"openstack%2Fkolla~stable%2F2023.2~I2facb38430ec9883c69f1a652e54116c383287fd","project":"openstack/kolla","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-07 13:53:35.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}},"hashtags":[],"change_id":"I2facb38430ec9883c69f1a652e54116c383287fd","subject":"CI: Drop upgrade jobs from 2023.1","status":"MERGED","created":"2025-05-07 12:10:52.000000000","updated":"2025-06-03 13:34:51.000000000","submitted":"2025-05-07 14:06:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949022","meta_rev_id":"f742a1388930a8ff9f7002878a25b82dbd326ca0","_number":949022,"virtual_id_number":949022,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 14:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-03 13:34:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-07 13:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-07 13:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-03 13:34:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 12:20:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 12:49:56.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"013e391d9ae7e4b0e670cb4e977aef3061ccfa03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 12:10:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9d9b7402583585064426d580bca46a1dd576c74","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:20:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea7c12432d124b2281e8d34fb296b414\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d237504ebc9497390260e0e327b94bd : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/034352d76ced4e5fb787745dab3ea5f3 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08149d588e664ab194e449d68fb3da3d : SUCCESS in 4m 45s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c00f6d81fb274390919ef425505bfd68 : SUCCESS in 2m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ac0f78550b3100344aa0af20aa39b4d0117cdb00","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-07 12:49:56.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"e6f7a5894d96c4ebeccb4af278f1f93d292d8970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 13:09:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"368cc87e7065b742417cd8ba9e54bf8393ffa057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:20:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9de68fe0f34f5da09caba8c83fc249\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/797ae3b1b89c499bad3e017121541187 : SUCCESS in 2m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b6523b1fa04455a8c8c621183a7ef95 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7a98bb359e94b808d455fff8b7cd49d : SUCCESS in 4m 53s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/26b6536134874f208ba51395991f5892 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":2},{"id":"f17d263cd0317c3be25a79ae7296c1d62fae9027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 13:53:28.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7a681bba560b1d7ad523202bdf03cc08aec8df18","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-07 13:53:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7d8072a6835135206cc4791582af26aa895df89b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:58:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f56cea9257344343b2716e40d648df41\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9047821d06ea49a0b3e82c72bb8ae72b : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4b7cc8a17614379af7789beb123591d : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6deda5df91b5405c8446f5c47bf495d2 : SUCCESS in 3m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4363d33091ac44b0ad4fd70ddf12b10b : SUCCESS in 1m 50s","accounts_in_message":[],"_revision_number":3},{"id":"42b91e4c1c89397646dd0faf023db09455bc9407","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:58:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37fc29d6519d7b2ad4f7997262b3b92ab150c4cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:06:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4245f297484d489ea65b342386d386a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01246d3850f64cc299256d50fa2f02dc : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c2312d90c4d4181bf72903ad346a284 : SUCCESS in 6m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/be4273f3b4e144749855fd6e07ae2fff : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":3},{"id":"132be3f4846e65625e91c5a9e19d544077cbac9a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:06:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2fc12468758476c9130eeb260a5463ea87dccd71","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:07:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6194c328dcf546289f0ffec5ee6d0e30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4381c32a1bb94bf3a4dd45be46e55d2a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3},{"id":"f742a1388930a8ff9f7002878a25b82dbd326ca0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-03 13:34:51.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fe9c55be40d38c0d4af9e21fd6c13b206f001e84","revisions":{"8314049554ab8b710f9d3a31f0fbac258f555476":{"kind":"REWORK","_number":1,"created":"2025-05-07 12:10:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/949022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/949022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/949022/1"}}},"commit":{"parents":[{"commit":"00d434e2c5cacf5549f326af8416e48a50bc8df8","subject":"Opensearch and opensearch-dashboard image size reduction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00d434e2c5cacf5549f326af8416e48a50bc8df8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 12:10:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 12:11:16.000000000","tz":120},"subject":"CI: Drop upgrade jobs from 2023.1","message":"CI: Drop upgrade jobs from 2023.1\n\n2023.1 is unmaintained\n\nChange-Id: I2facb38430ec9883c69f1a652e54116c383287fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8314049554ab8b710f9d3a31f0fbac258f555476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8314049554ab8b710f9d3a31f0fbac258f555476"}]},"branch":"refs/heads/stable/2023.2"},"cc3f9d8bacf1d39854379672adc4241387ae41cb":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-07 13:09:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/949022/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/949022/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/949022/2"}}},"commit":{"parents":[{"commit":"ce8641af1f77e53b7056bef1f10046d5df25f096","subject":"[2023.2-only] Mark bifrost as unbuildable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ce8641af1f77e53b7056bef1f10046d5df25f096"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 12:10:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 13:09:16.000000000","tz":0},"subject":"CI: Drop upgrade jobs from 2023.1","message":"CI: Drop upgrade jobs from 2023.1\n\n2023.1 is unmaintained\n\nChange-Id: I2facb38430ec9883c69f1a652e54116c383287fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc3f9d8bacf1d39854379672adc4241387ae41cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/cc3f9d8bacf1d39854379672adc4241387ae41cb"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"},"fe9c55be40d38c0d4af9e21fd6c13b206f001e84":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-07 13:53:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/22/949022/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/22/949022/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/22/949022/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/22/949022/3"}}},"commit":{"parents":[{"commit":"00d434e2c5cacf5549f326af8416e48a50bc8df8","subject":"Opensearch and opensearch-dashboard image size reduction","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/00d434e2c5cacf5549f326af8416e48a50bc8df8"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 12:10:53.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-07 13:53:28.000000000","tz":0},"subject":"CI: Drop upgrade jobs from 2023.1","message":"CI: Drop upgrade jobs from 2023.1\n\n2023.1 is unmaintained\n\nChange-Id: I2facb38430ec9883c69f1a652e54116c383287fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9c55be40d38c0d4af9e21fd6c13b206f001e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/fe9c55be40d38c0d4af9e21fd6c13b206f001e84"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
