)]}'
{"id":"openstack%2Fkolla~949230","triplet_id":"openstack%2Fkolla~master~I911462285ec44d38e719b7b61b67800fbae97245","project":"openstack/kolla","branch":"master","topic":"wip/2087865","attention_set":{},"removed_from_attention_set":{"35366":{"account":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"last_update":"2026-07-27 14:37:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I911462285ec44d38e719b7b61b67800fbae97245","subject":"WIP: docker buildx via python_on_whales library","status":"ABANDONED","created":"2025-05-08 16:51:03.000000000","updated":"2026-07-27 14:37:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"242479c0801a4ba4b0b4f3dae312540158189a60","_number":949230,"virtual_id_number":949230,"owner":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-08 17:25:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"all":[{"value":-1,"date":"2025-05-08 16:55:26.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 17:07:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-08 17:25:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-27 13:57:43.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-27 21:29:52.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"}],"messages":[{"id":"51abcb56806ed4aecaaf49488d0dd7a9463803e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2025-05-08 16:51:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67a0a321450eea445d6413baced044afdefc627d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2025-05-08 16:54:23.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5936cc981107002d872fa5ec6cdb47784227874c","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2025-05-08 16:55:26.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"fd95b8986e469fb06724cf946d1c2bbc32d165b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:07:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c2ac86acf14f5a99eebb542e8c32e1\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/d6107b4101ed42119e59f6ad7960d5a7 : FAILURE in 5m 55s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d7db67d4e4ac42b8a6e2ef730b8d7633 : FAILURE in 4m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eba4f450ac5f4b329bd7fc4832f5d9f0 : FAILURE in 11m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/40b611c4be8a4b458d5534cd7b54fdc8 : FAILURE in 3m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c542b67b28f42cda3e96576afbbc302 : FAILURE in 3m 55s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a3768376707447b39627daeea2f88446 : FAILURE in 5m 01s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/408d2b7a5d284c66a49baa8bad69b5fa : FAILURE in 7m 51s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e6800288950e41dfb041a5313512a8b1 : FAILURE in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd040a33490eebfd0780091e33bfb81ea2cc12cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:25:51.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62e91b6d4b8b4f4a82c7941dab2ef067\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/c6c75b3e8e90462da57efd8865746734 : FAILURE in 3m 52s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5a21d26c4bf14bd7a83182dc981cb888 : FAILURE in 6m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d04f5600f2ec44aeae2bc853e9ef8450 : FAILURE in 4m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e2a215f6a25e4ead8dd3570d2718161f : FAILURE in 3m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/23d7ad7e807146fe98d7b536fe2b5390 : FAILURE in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd76bfc348a54b10b5418d889788d6b7 : FAILURE in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e894dd79e2eb4d7387069acd5f6c94b4 : FAILURE in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b723ef1a82ae421fafcc0417d0016c40 : FAILURE in 2m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/642a10830ba64003923e44401ebe45a3 : FAILURE in 6m 39s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51fd8d6f8b9847e39651bcfc817ce188 : FAILURE in 3m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12f5e3f859da48c0b937c3c58766b850 : SUCCESS in 4m 54s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/853c1fb411bd47aa87eab0fbf9301abe : SUCCESS in 2m 56s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/256fb3e9e85846b7b2abdeb1226d464c : FAILURE in 5m 07s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7565ef40e6f4452ca36b6daf5fcb9718 : FAILURE in 2m 43s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/6def32bd3dcb445480e7fdc02b82c9f1 : FAILURE in 7m 30s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/500a6b9d02da49c3aed822c3bd801bc1 : FAILURE in 4m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/506199aacfff4c2790010ba59e66164d : FAILURE in 3m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/22b50d0d928140d2aa5b855fe832a587 : FAILURE in 8m 03s","accounts_in_message":[],"_revision_number":2},{"id":"242479c0801a4ba4b0b4f3dae312540158189a60","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 14:37:49.000000000","message":"Abandoned\n\nImplemented in another merged patch","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0f627e88c26d9fc61145b40f4604188e2607bf11","revisions":{"b5850e5df1543c229c89b6f8a6ba7862bbdb35e7":{"kind":"REWORK","_number":1,"created":"2025-05-08 16:51:03.000000000","uploader":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"ref":"refs/changes/30/949230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/949230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/949230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/949230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/949230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/949230/1"}}},"commit":{"parents":[{"commit":"e7bd46dad0b62ff151667be2a6f5c8a7be41911e","subject":"Merge \"CI: Fix pip mirror multiline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7bd46dad0b62ff151667be2a6f5c8a7be41911e"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2025-05-02 18:20:24.000000000","tz":0},"committer":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2025-05-08 16:39:07.000000000","tz":-300},"subject":"docker buildx via python_on_whales library","message":"docker buildx via python_on_whales library\n\nAdd another container driver using \"python_on_whales\" library, which\nwraps the docker buildx cli. This allows using buildx\u0027s improved layer\ncaching, and enables future commits to leverage remote caches.\n\nRelated-Bug: #2087865\nChange-Id: I911462285ec44d38e719b7b61b67800fbae97245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5850e5df1543c229c89b6f8a6ba7862bbdb35e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5850e5df1543c229c89b6f8a6ba7862bbdb35e7"}]},"branch":"refs/heads/master"},"0f627e88c26d9fc61145b40f4604188e2607bf11":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-08 16:54:23.000000000","uploader":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"ref":"refs/changes/30/949230/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/30/949230/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/949230/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/949230/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/30/949230/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/30/949230/2"}}},"commit":{"parents":[{"commit":"e7bd46dad0b62ff151667be2a6f5c8a7be41911e","subject":"Merge \"CI: Fix pip mirror multiline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e7bd46dad0b62ff151667be2a6f5c8a7be41911e"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2025-05-02 18:20:24.000000000","tz":0},"committer":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2025-05-08 16:54:23.000000000","tz":0},"subject":"WIP: docker buildx via python_on_whales library","message":"WIP: docker buildx via python_on_whales library\n\nAdd another container driver using \"python_on_whales\" library, which\nwraps the docker buildx cli. This allows using buildx\u0027s improved layer\ncaching, and enables future commits to leverage remote caches.\n\nRelated-Bug: #2087865\nChange-Id: I911462285ec44d38e719b7b61b67800fbae97245\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f627e88c26d9fc61145b40f4604188e2607bf11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0f627e88c26d9fc61145b40f4604188e2607bf11"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
