)]}'
{"id":"openstack%2Fkolla~949231","triplet_id":"openstack%2Fkolla~master~I6b33db65552abbf946707d92ce92042ed598efa0","project":"openstack/kolla","branch":"master","topic":"wip/2087865","attention_set":{},"removed_from_attention_set":{"35366":{"account":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"last_update":"2026-07-27 14:38:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6b33db65552abbf946707d92ce92042ed598efa0","subject":"WIP: use buildx `docker-container` build driver","status":"ABANDONED","created":"2025-05-08 16:51:03.000000000","updated":"2026-07-27 14:38:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e472811274475100b4920febde5737635d13e327","_number":949231,"virtual_id_number":949231,"owner":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-08 17:12:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"all":[{"value":-1,"date":"2025-05-08 16:55:12.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 17:03:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-08 17:12:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-05 06:32:53.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-27 21:30:08.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"CC"}],"messages":[{"id":"53481fc8a68dee28f536ab5f1aadf3dfba991d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2025-05-08 16:51:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8953b0212c23fa6e5e672a2f4f1e434912a446c5","author":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"date":"2025-05-08 16:55:12.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"96b862ec5b46173a00d4a9e48387ead7654139bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:03:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e47e97e7730346b5b3018be86d22f1d8\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/6390f1d7b216493f96cfc62c945405cb : FAILURE in 6m 10s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3d2da5d2f15348b9a096ed078a6047dc : FAILURE in 4m 18s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e4717de4e7874dcb9d14216e0e9563ba : FAILURE in 10m 12s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9bce27a417594ff796485b31efea6907 : FAILURE in 4m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a64e051fb94841d898b7d849498081c2 : FAILURE in 4m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/64d52cfbbd914013a525ffa2b7fa9ebf : FAILURE in 5m 17s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/affaee2035624b78bce8913a71dc85d9 : FAILURE in 5m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/1426898310ce47fc9949ffa78b85d7b6 : FAILURE in 5m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"02e6bd348d0972e049e5bf3d907fbf33cfaf52e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:12:30.000000000","message":"Patch Set 1: Verified-1\n\n(9 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c577390626144d828bffce3384cf7777\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/279e4963f9db4c229589d926834b4a34 : FAILURE in 4m 19s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/e5e422aa26384c43b35877e2cb7bbf0a : FAILURE in 7m 16s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/2b5685556831472999c86e028499aa86 : FAILURE in 4m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e549934fe62f4db8bead31d3522ba178 : FAILURE in 3m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/93b0bc59ada74834a6792f3cc6de9d1d : FAILURE in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57afd1e9b2dd403083ec5c5ef8329251 : FAILURE in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5aa7a709a48945428ad2277705ff6168 : FAILURE in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8ad2037095df401686836e1592aba9ba : FAILURE in 1m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b98aeffb5d93474ca9b8e43329377940 : FAILURE in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96f49d73dd7e48149dbee0221a6c85f1 : FAILURE in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a69a57cfcbf4fe49e7675408c10ad6d : SUCCESS in 4m 55s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/14a2d989840f4b529eca446eb4b52288 : SUCCESS in 2m 36s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/1bf198ec2ab0429e9daf144ce39124bf : FAILURE in 4m 43s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/70e4b10a16904a9e9652b527fb563e7b : FAILURE in 4m 17s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/7cb2baff46be45a59002a5f91fe240c6 : FAILURE in 5m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5f21b40d85bf40a0b58a48fc634ce6e5 : FAILURE in 5m 12s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/928f2c7e19e448cca2d00d342bfca70f : FAILURE in 3m 33s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/b54e02167f744905876d40ba435c1a36 : FAILURE in 9m 13s\n\nWarning:\n  Comments left for invalid file kolla/common/config.py\n  Comments left for invalid file kolla/engine_adapter/engine.py","accounts_in_message":[],"_revision_number":1},{"id":"e472811274475100b4920febde5737635d13e327","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2026-07-27 14:38:02.000000000","message":"Abandoned\n\nImplemented in another merged patch","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a244be9e4a483bc371bbf871ed93ac1acd59f163","revisions":{"a244be9e4a483bc371bbf871ed93ac1acd59f163":{"kind":"REWORK","_number":1,"created":"2025-05-08 16:51:03.000000000","uploader":{"_account_id":35366,"name":"Michael Sherman","email":"shermanm@uchicago.edu","username":"shermanm"},"ref":"refs/changes/31/949231/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/31/949231/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/949231/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/949231/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/31/949231/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/31/949231/1"}}},"commit":{"parents":[{"commit":"b5850e5df1543c229c89b6f8a6ba7862bbdb35e7","subject":"docker buildx via python_on_whales library","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b5850e5df1543c229c89b6f8a6ba7862bbdb35e7"}]}],"author":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2025-05-02 21:28:03.000000000","tz":0},"committer":{"name":"Michael Sherman","email":"shermanm@uchicago.edu","date":"2025-05-08 16:41:43.000000000","tz":-300},"subject":"WIP: use buildx `docker-container` build driver","message":"WIP: use buildx `docker-container` build driver\n\nDocker buildx supports multiple types of \"builder\", with the default\nbeing `docker`, and `docker-container` allowing additional caching\noptions. https://docs.docker.com/build/builders/\n\nWhen using the `docker-container` builder, built images do not\nautomatically appear in the local images shown by `docker image ls`.\nTherefore, we need to push parent images to a repo before child images\ncan be built.\n\nAdditionally, proof-of-concept for remote layer-cache is turned on by\nspecifying `cache-from` and `cache-to` arguments. The cache reference\nmust be unique or it will be overwritten, so we use image.canonicalname\nwith `-cache` suffix.\n\nThis allows builders to cold-start quickly, such as in CI. If needed,\nmultiple cache-from sources could be specified, such as when building a\nPR, pulling from the target branch\u0027s cache.\n\nRelated-Bug: #2087865\nChange-Id: I6b33db65552abbf946707d92ce92042ed598efa0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a244be9e4a483bc371bbf871ed93ac1acd59f163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a244be9e4a483bc371bbf871ed93ac1acd59f163"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
