)]}'
{"id":"openstack%2Fkolla~949444","triplet_id":"openstack%2Fkolla~master~I4507fb60d361ea504459a4b502d6f6c42a327064","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-27 20:52:25.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-31 04:06:36.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-31 14:49:41.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"37306":{"account":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"last_update":"2025-06-03 14:21:44.000000000","reason":"Change was submitted"},"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2025-06-03 09:58:16.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-06-03 14:21:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4507fb60d361ea504459a4b502d6f6c42a327064","subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","status":"MERGED","created":"2025-05-12 07:13:32.000000000","updated":"2025-06-03 14:22:40.000000000","submitted":"2025-06-03 14:21:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949444","meta_rev_id":"fa5e15cf20def5e0f1e6f571cccb9fb2d7ab51c4","_number":949444,"virtual_id_number":949444,"owner":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 14:21:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-06-03 07:52:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-06-01 11:21:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-31 14:49:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-06-01 11:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2025-06-01 11:21:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-31 04:06:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 07:16:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2025-05-12 08:38:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-12 12:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-23 06:21:52.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-23 07:19:00.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-23 15:28:50.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"state":"REVIEWER"},{"updated":"2025-05-27 08:56:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"73f5d23f5ffc18f3bf5698c8e33b2b0e9b44c7b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-12 07:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7ad830d090606d15d882e87b0401bf19a6a954a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-12 07:16:13.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"53a8e637ddcc900c24f33815c5cd64f8b36a22ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 08:38:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8c2f580f1c4ba191b0b073bc8c19d4\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/79876ce4ce814486a78b6e615de20aa9 : FAILURE in 45m 51s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9de3bdcfc43a4db48dfdf693c99fdfbd : FAILURE in 39m 08s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4e678392c6724a13a3d22d512fe27edd : SUCCESS in 1h 22m 33s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c4a3e4d3fad0478e937488c6bb90d5ca : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6175bc74cbb948c49fe6d816dd8cc9d9 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4992b64911e64bf6bafa63af5914da68 : FAILURE in 7m 12s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/711e965ed5e64041b70540c554557332 : FAILURE in 44m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e532c9e1950c46bfa4578964a8593f5c : FAILURE in 49m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d0471fa717478ccf7078d40756bf6143781685cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-12 08:40:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"141211025a993ec4726ce8208e7ad2f9266a7867","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-12 08:48:14.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"f236c376f57a5bdad20a30e714d4ae2540a95c3c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-12 09:10:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"346e2bb9c063fe57f0b86e5f17cf5a62cdbc8231","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-12 09:21:27.000000000","message":"Patch Set 3:\n\n(4 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"f2e34ddedc7c957a4ed0052e48f12078732c4a09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 10:50:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9d0f90064724c88ae03b7243731f0f6\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/391aa3186d2d4305a8b5479a59c5e109 : SUCCESS in 45m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a7018a9af4444968b2edf4a49abca73f : SUCCESS in 43m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/07c9064def32487e9e48c5d81db3d6a6 : SUCCESS in 1h 38m 31s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/85718e82df0c4e4e91d9e119d638f5dd : SUCCESS in 3m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/00c0f60aa8ec4f739c7f771c0b5da70a : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/043f9e1761954cd6894603e95cfda2f9 : FAILURE in 7m 48s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/9229ed3c5d954fc192cf33c593778ca3 : SUCCESS in 46m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f7eacc42de284d919a3e0bcf8bbff291 : SUCCESS in 42m 05s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a691231b4199ff8ce7ddd7fab4155454de464935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 12:56:36.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25179904a08e405a9dfd21039c4a0cf9\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/df1e27e4856f4225b3d0ad2fcc4100c3 : SUCCESS in 18m 37s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d80abc10be6246aa8ef6c8f49cf5af8a : SUCCESS in 33m 08s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5c48528a62464647852939e688c4b494 : SUCCESS in 14m 04s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/14547751512d4bccb12412471bd3fd86 : SUCCESS in 13m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e9bbf27385344235affd12ce27c5f492 : SUCCESS in 48m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/94e257fa3eff4b7c8aa5d8b49a58311e : SUCCESS in 46m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6b37143573a491a912d0015f69fac71 : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/582571ce31024c0b82d46bce670beee8 : SUCCESS in 1m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0965cca7c5cd438e872d7c0c8c2b3ff2 : SUCCESS in 3m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3752ddf3db30493e84b0bd55d64b107c : FAILURE in 5m 14s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66f717038f6441a5a37c9d9f7f34969a : SUCCESS in 1m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64d4792392d3434fb1d8c94178f368ff : SUCCESS in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef80b6f94fd34bc8ae44352fc6af05a7 : SUCCESS in 3m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/837ab97f36474e81845065a15482b20b : SUCCESS in 2m 47s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/a2f9cf1e1d4f4841a4f73342f0a58e44 : SUCCESS in 24m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/56d5ff3a45624c5c88d4ff0161bdad32 : SUCCESS in 1h 23m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/d023a0862af54939bc6deb977d06c8fe : SUCCESS in 36m 51s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/b40ca1c018bd435cbdd6769e1029a7af : SUCCESS in 1h 28m 38s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/590458f7ef834a6a8e1a9ed5abc4c6f0 : FAILURE in 1h 48m 39s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cf640c7b555446739fc1b357cf487952 : FAILURE in 1h 08m 45s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/36c6ccc3f5bc4583b7e278611a199a02 : SUCCESS in 1h 03m 11s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/e7ac886bf7a846d997105105bb6bfb0e : SUCCESS in 25m 34s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ab52b2380c2a4fdca7dfa40f3170e4bc : SUCCESS in 59m 59s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/497d584c95eb440cb34fb9a01f43a897 : FAILURE in 40m 30s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/64c90525a64c4f95bc0c4b0bb3e9272d : FAILURE in 1h 10m 53s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/ce4ac2acbefd414faf7c3718b98a376b : FAILURE in 23m 02s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/29a8c30270d0445988570fb894b4e033 : SUCCESS in 20m 36s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/65ba773d2c4b477eb118142113a2307b : SUCCESS in 27m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/14ef10dc042d46b4b48662e3f7395fde : SUCCESS in 1h 06m 01s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f38b1e85cf2d40a9961781333a2543b9 : SUCCESS in 1h 40m 09s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b76f7c388407442c83d24051474a0954 : SUCCESS in 1h 40m 08s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e45692425d93436ca11552180b31e622 : FAILURE in 1h 14m 50s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/49a0888981bb497894b5b100f1fa24f2 : SUCCESS in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/cc27da17c71046b3a8d3f4cc19ba99a3 : SUCCESS in 22m 00s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/b5037be357b9486f9fa03d7d21df42a5 : SUCCESS in 31m 50s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/16daefdfeb3d4eeeb87bd9640eb714f2 : FAILURE in 50m 59s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/ccffc3b389eb43c99d8eb9924098bab9 : FAILURE in 1h 02m 01s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/395e0e6f333d44de8fb8f66f34a286c1 : FAILURE in 24m 53s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"988097e1013f2808d175021c0eaa5195b3e580c0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-23 06:21:52.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":3},{"id":"779db5ff36074b3192e3790fdaf8d954d0875417","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-23 07:19:00.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b9a7d8fbb955a9031caef05e95cb2e2962547343","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-23 15:28:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"cbad2924e0a4067b87ef216c4e6976ee11aac9ef","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-24 07:21:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2c8fd0a322da007af182beaf40d4852ec294068c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 09:06:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d19f803b402e4716abedf36086594b10\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/1b4319ac4c5d4ddea668a38f9fcd4c64 : SUCCESS in 38m 50s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8d4fa73b3b0a4770b2ff3beb36692cac : SUCCESS in 1h 42m 48s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/768007fbef7b4c0e9ed3e67e43d9077f : SUCCESS in 3m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3584261a36f47f6bb20d930ef9f9382 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1ce52ce866274e6897c7d983fc271b42 : SUCCESS in 7m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/963947d778aa4469ab496c5a29594b31 : SUCCESS in 47m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3f2c4bc6eeaaff7ad798c4f782d55afaf53c539f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 09:54:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/115096bd405741f78d05af150acdc0fd\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/ab3d23c5f23347fd87f01e98f4a7d1b3 : SUCCESS in 22m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4c95f0da17424fa1ad6184c46709d4b6 : SUCCESS in 14m 46s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/8452ab051f2f4fd6a3c9df2b380b60b2 : SUCCESS in 54m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c1deaad07d0446428d3adf3d67072a06 : FAILURE in 1h 31m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff3a079d22604e14bdf24512610d9c97 : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d47f6d7b82d4be4993ffb5191872758 : SUCCESS in 6m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1edec32d03384b859f5893fb9c175332 : SUCCESS in 5m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0765658c5bf43c38bfff3a01619b89b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8e81af2cc9e4482a2cbafc8c13072f2 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de4e9b479a9d4c6ca15e8457791d026f : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32f5e17b00db4b4e87e8b0d713d67c92 : SUCCESS in 2m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/a3bc1d268aab4e60bc1048689c4dc433 : SUCCESS in 2m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2982699eed69462ab222f5248e94a953 : SUCCESS in 21m 34s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3b42390aff2e4c56af4208f2b4e818ef : SUCCESS in 25m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/de3d110407d44278ade3dedce31a4df6 : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/65e8f7a6b0804d16a9321aee50f948a6 : SUCCESS in 1h 21m 27s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2041dacae91c4a0396f69c287076c853 : SUCCESS in 1h 51m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/025956eb11d94915b4be4ba499b36cfa : FAILURE in 2h 27m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/3f80c3d2159441e696e1088c75cf07d5 : SUCCESS in 1h 10m 56s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/64e8f4f183214779967e8b486f4778d0 : FAILURE in 19m 44s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/933b29cefd584447b83d4b714e6f6c63 : SUCCESS in 55m 34s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e9725fe9fe2a4b2ab59b0e7b7b081389 : FAILURE in 1h 05m 54s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a9562417036b450ca4267483e22890c3 : FAILURE in 1h 07m 09s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/71f282aaec3a44f09692c5b3bfcae93a : FAILURE in 27m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b5b5f7494a1a6f6fae8e735db7fb51334cd7a674","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-24 12:15:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"43d504822c11694a43b4e34d0cc23223f9fafd78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 14:00:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89dd841f4af6406388848b05325477fa\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/9622a70f1ea84c7bbf7219081363bf1a : SUCCESS in 46m 35s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/45afd2b2af704022bdc5ccd97ab371be : SUCCESS in 1h 42m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/64b26652b1d64cc1ab22e9194a433a70 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c406ae92c2784022a51283b1cf6058eb : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6e526f3ee82e428d80c43fa3a7aa2bfa : SUCCESS in 5m 59s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/65b99187460e4957934f9c1b41e6fd5d : SUCCESS in 48m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f2aa2f17e738a6150adc40a39c81369161d6ed46","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 14:55:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d4b4130f2e264a678441982e0224ae3a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9196aa2cc86449699debf6fe73613769 : SUCCESS in 15m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ec62f4a0b83640849cce980d85d9a5fa : SUCCESS in 19m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2a9f29bfc4304b03881fb3b56aa9f5b9 : SUCCESS in 31m 40s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/17105bfbded14a6ea8e960ced67eb809 : FAILURE in 57m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9b45e88fac4d858b9e2156ff397f0c : SUCCESS in 2m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/851f12640a854b96a7c470b845a435da : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66846293b5464d689d7df47bc82d299b : SUCCESS in 1m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fd5098dddfc14fae93fd45f4183258ff : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73b2eabe9a124a06829db31dc713079d : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6277fb997fe4decbc89002bdc7bb816 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a17df9e6d21e4138a6e1c8d2ec6e3bca : SUCCESS in 2m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/e2d75ce577c54c32bbcce02eb1d9c83d : SUCCESS in 2m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a88bf0efddf74f9296130d2d11762a4e : SUCCESS in 17m 05s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/48af7914ae0340718b86175c77862a4f : SUCCESS in 19m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/279afc7c2f7f48cdb8b3b8b414d5782a : SUCCESS in 50m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b53c588541084bdf9ec824291aed345b : SUCCESS in 1h 14m 20s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/988dcff5fb2c4c3a986d6a58eb991407 : SUCCESS in 1h 34m 50s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c4c5c902d43a466eaf41af749f3b6e8a : FAILURE in 2h 34m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9745957a723e4268bee9cc4ad9416c24 : SUCCESS in 58m 43s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/88cc96b2211849a99ee512cfc8fd3fd7 : FAILURE in 18m 45s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6245c312fe924a96ae4f8de6c44e5491 : SUCCESS in 45m 04s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/f30b82afa5ee45548672667c6a92d7e6 : FAILURE in 41m 13s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/0dd9e253c27441c9894807e38360d0fa : FAILURE in 59m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/222eacea51a14dc78ea3e8cbf360011f : FAILURE in 21m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"acc9535376b9be8151ad9231bf85c1f266e735f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-24 14:59:31.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"ed2073bdd7f9e1b3b2b5407ba148ebb384f56b50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 17:45:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97d1db198ceb45fe9b3064640bcbf483\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/9002a012ae2143698cab0d16ff77a377 : SUCCESS in 22m 45s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6726b41d29c341739f4ce7ec4f7f40e9 : SUCCESS in 25m 05s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f499ca05bbe647958c7c5d835c4d07b6 : SUCCESS in 1h 11m 39s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/903ab5f0e80c436197109008054081f6 : SUCCESS in 1h 21m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c174fb074642c09c702dfaa655b42c : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c38ec819652463ba692f3c1d9d6c3f0 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6cc0fa7be97430e858ee2f91dd792a0 : SUCCESS in 3m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2966f820d71745578bb607f22879041b : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff2dd286d9714fc8adc51de2604edd23 : SUCCESS in 3m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/204f4f79b6be44728dc00cfcaec9e8fa : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/207bec7cd2ac4046867c524904a50655 : SUCCESS in 1m 49s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/16840e9410854974a004b29e4d0cd883 : SUCCESS in 2m 29s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a452653c295f42139ec71f099ac96f1d : SUCCESS in 22m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/39fbc2eb8a704efda1650c804c99b352 : SUCCESS in 27m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ea5d1b84f36f4f6b8aaf8eadbb7c4e3c : SUCCESS in 1h 10m 00s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/06f4654ec8ec4f90986306fbf8e481c6 : SUCCESS in 1h 43m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4e248899f6bf424ea8f184e11adac62c : SUCCESS in 1h 45m 28s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8c13440bd9964263a697fdb6e642ae92 : FAILURE in 2h 34m 00s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/40957d1c79124443a516549d25518511 : SUCCESS in 1h 01m 57s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/e9434c051c024df08e20d404c1da9f62 : SUCCESS in 24m 19s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/05e2051b21bc4d3dbeeb33d3b1035365 : SUCCESS in 57m 10s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/07bbe8d34fbe46038dd3307b2c3aa4f6 : FAILURE in 1h 03m 59s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/14d50bfd5a4d4e23a0ff32c620fcda55 : FAILURE in 1h 14m 17s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/09a6627ab3424090901c1b51f13f071c : FAILURE in 24m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"96ea241a3f1308390a5a096782791c3cf800e690","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-24 18:07:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a05f19b5eea641dabfeea3eec564b86d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/259abaaf242b4014bfd11893aa0ddcf9 : SUCCESS in 33m 15s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/74d71a714d8c4fad8e2398ebecfdcd16 : SUCCESS in 2h 10m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/13d0cd27142a4438a263d015ac1840bd : SUCCESS in 3m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be2e0571dbe24178a52941124e31c1ea : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d6c298280a1f479c855605c942502ac8 : SUCCESS in 5m 49s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/48dccd6488854d7dace69bd41c575e87 : SUCCESS in 1h 09m 15s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35ee81311499ab8d0bafd1fe49a9d15dc59326ed","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-27 08:56:42.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"45736a693c8aacc293ccfaa87555b7f49a2c1f30","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-27 09:00:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"834139291142c91bb497e3061e8a3e55acf033cd","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-27 09:01:30.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"518460db47ba7799b68d885ecceb12c2fab8112c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-27 09:02:13.000000000","message":"Patch Set 5: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"a4364b18ecb44334af0a79c46d00b8378e910414","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 09:02:14.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86809b7495484c868fdf14a1d4c49589\n\n\nWarning:\n  Dependency cycle detected and project openstack/kolla-ansible doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":5},{"id":"f2ede140a12d1e5ffed7f3cc909f94b9851e2799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 09:02:14.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1627237e4574749901693cd05466a7e\n\n\nWarning:\n  Dependency cycle detected and project openstack/kolla-ansible doesn\u0027t allow circular dependencies","accounts_in_message":[],"_revision_number":5},{"id":"098d4e3149dd8d8e74ca9078ef8a20a0cbff3a45","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 09:02:34.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e5d282a92583580e71c098aa61fcee6993c2a556","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-27 09:10:21.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"377a9d4100b76077b53ceb1354522c7c3618ed0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 09:18:45.000000000","message":"Uploaded patch set 7.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":7},{"id":"3da2fcdb16d73888edf630e1a22a26c0e05ab6f0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 10:09:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e802df527fa26a031cf264ed07efc77ac81b2d0e","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-27 10:09:40.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"8cb5c22e6674b9fbc0b77af8f6c83251b9634927","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 11:02:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bece43a58192499ba0ee5757bf1de28c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/72734491c2294a95a65921d55fc0f4b6 : SUCCESS in 1h 22m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cf35e6a1058b46238ca812ecdec58b1b : SUCCESS in 1h 05m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/857557e6d5294f51b3ce093d66b4a332 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/aad52ad3ed324d5990d8ee70cc365cd1 : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6dbc7988e6754c5fa97cb3c72b7f42e8 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/89dd089669e043c3a4408cfe09e7863d : SUCCESS in 1h 00m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ed1933f966968039187ab6cc9ca5a5760d32bbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 11:49:48.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8ddf269b88334f6ba094ceefc22476ab\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/eef454f2d8dd4f9aac4c886c1e8c17e2 : SUCCESS in 20m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/974a5f5534164677ac72be077a5f7282 : SUCCESS in 28m 10s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/14489e89aa8d4ded8350f30ff717d1a5 : SUCCESS in 1h 02m 05s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1b25ad4e977f47fe8cd9f1f4ee71c303 : FAILURE in 1h 40m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ef71629ab4e46ad9a4661a6fedc2105 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80bae6694f3d4b9c90191bc3a2e545e6 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cc8b36e40ad44a218bfd3b8573e9bb20 : SUCCESS in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2b16f685558543239965f31a45a676ed : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd674094c6244a36a0586267dd508296 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cb9e4354cc74d349ea25cf998d53c0b : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/46dc32820b334360a99abf37d9c0d723 : SUCCESS in 3m 33s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f577cbc826648eb96cdf5d0e089fb86 : SUCCESS in 4m 12s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/fc4a5279cb0946d692b7c8187883c134 : SUCCESS in 20m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8a89ef7191094c769d4408348c70aa89 : SUCCESS in 28m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f9073d00ae0a49cebcfa5ceffb7c419f : SUCCESS in 1h 08m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/72a4977b67fa47b7a17c848a27e773b6 : SUCCESS in 1h 30m 16s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/4386149d1d744619b1db553f7409d2f9 : SUCCESS in 1h 58m 31s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/268592e1b3f64875ad9296de08211399 : FAILURE in 2h 25m 53s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/84ea62ef0140445a98fbbeeda2b1a11b : SUCCESS in 1h 12m 05s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/9bedca10df0c444bb0514a57e4177b91 : SUCCESS in 24m 53s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/bdf4e72010d44072ac17aff703a6cbe3 : SUCCESS in 1h 00m 24s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/066cdb5d3d4f4ea2a6a1f18f9d5a36db : FAILURE in 56m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/de34edecab974f14875576f7fb6f53b7 : FAILURE in 1h 10m 53s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/e479e321f712459d8a64f45efeb6dd85 : FAILURE in 27m 45s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6d817651b36dbb85f0d88c3270e1242ac28a76e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-28 13:18:18.000000000","message":"Patch Set 8: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"3738495b50eb785415d6e0cf57d5ede4bce4cf78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 15:38:40.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8c44a87093a45ab96fcba00a174b774\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3a603963dec04187a6a82cc79f01a3ba : SUCCESS in 50m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/639b083ad7be419f99cf58669dc51053 : SUCCESS in 2h 04m 23s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dce72341974e4b308cda5c44257fb162 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41c31d59d76440a784a0442dbd496913 : SUCCESS in 3m 44s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/93e00f5e560d4bdea3095707fbf29103 : SUCCESS in 9m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a2768fb8976046f8a6f673a3c6c98db3 : SUCCESS in 51m 26s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"0bfdacf4a4f7aa8e68bdc8d6c68d0ae5ed87df27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 15:47:14.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/593378ebeed54eae9cd14e114a7da422\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f3d4049cde6747d481c755dd297096f1 : SUCCESS in 20m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ad6d4475bee6461bbfe0de2f28a0ebfb : SUCCESS in 29m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/17c8ccb369ba4615b3d5e480f61839d7 : SUCCESS in 1h 06m 08s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/35972c4489e14a47a653335045e07583 : FAILURE in 1h 47m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6002ded03b491fac8e858393ca8064 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b391051b22bb4ec5bea19cb7c79d3f70 : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a80c2484b4d14f868f7ca089512f9e4d : SUCCESS in 3m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/85c9b10fb4d04f899dacbbeff5d117fc : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9894787d9dc54f7c84f603bd05be738b : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ede17b54f945d8852550d9a579bfb0 : SUCCESS in 6m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8050d9bb72b94ac3a0efecf5219cd518 : SUCCESS in 4m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/f4b5a48fbcae466c82408dc667a7a7d0 : SUCCESS in 3m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/72dfc4f5551e4c62a171aedf2e479748 : SUCCESS in 26m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/48ef9d57beb24576a540e370c42aa49e : SUCCESS in 29m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4817715f970341a2ad0a9b3dffaf9057 : SUCCESS in 1h 41m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/37b4fd04ad6c44659c6d9487dc4b9c0d : SUCCESS in 1h 42m 05s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7fd4d3f474f94dd791787c14f7a8af9f : SUCCESS in 1h 55m 01s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/be324a022e764e919647fa0a01309a21 : FAILURE in 2h 20m 41s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d7f04a9044dd48049a52ca1c4791eb0b : SUCCESS in 1h 23m 24s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/51b87d4f5779423d8906f940c2d530f9 : SUCCESS in 25m 33s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ebc17aab63cc450d96b6db6043ab0840 : SUCCESS in 55m 41s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/e9abe8d703754815ab243a79e5cdfa2b : FAILURE in 52m 29s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/2dab864539fa46309c427b51424fe78d : FAILURE in 1h 05m 38s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0230de00e8bc48eb85d6bd1935d8a56c : SUCCESS in 33m 40s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2a56082fdf9ebf33d57a4f0863602fc229d98b41","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-29 10:03:10.000000000","message":"Patch Set 8: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"063c481a95e4eb4c71b8953450cc0df5ecae5eec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 11:05:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05873e887610485aa0c50880449679ae\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5fb61df98b2f4205a3aaf898f4b72c84 : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b356d670289c4b6ea8df9953d20a5027 : SUCCESS in 59m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/007ffcd654eb4ae1b3915329641e16fa : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23c4ac1e29944610ace7ec829bf864ed : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6680384268ec4338b4f6877eddcdf7ef : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/5cb21522d05c4f2cb5a018bcc343a9ad : SUCCESS in 48m 17s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"1488fff37d9c9d09f360b78862e5aed6aec5426b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 13:19:06.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/681d0c54c501458a8bbcf96a28446e53\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/bf486226a0a243a7a69d2709e1a1cec2 : SUCCESS in 14m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d74a8e4f8c3e4766b49f6dc4f31d0552 : SUCCESS in 24m 23s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/60b59bc7216d45c6874ec91479982040 : SUCCESS in 57m 49s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6aa99db127ef41c2a09f3a7de0442463 : FAILURE in 1h 24m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ecfd9bb95043b0beaefa7cb1358a7b : SUCCESS in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03ee234a3606481a97d3586669b80f7a : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/01b17f4e4e834e0b8e30d8b45aaad587 : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d8e170a6622e4779a7a0950af089b1ad : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6febe25af09a481085ff0f5294b7731b : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7240f5c9961c475aae83f5f5a161962e : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1fdb3376010f4b01af14ef6c116c24eb : SUCCESS in 4m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/144e9b3f3c9e4d809bb5164edfdc711e : SUCCESS in 3m 15s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/3427ecb037234ba5ab4a9f3e36b812f7 : SUCCESS in 24m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/9afc6c6d70504333bd790c8b80e41c45 : SUCCESS in 33m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/5ac964a63c9542308ffbd7cba8871273 : SUCCESS in 1h 03m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c200a6b1058e4eab9c05b1d11ca41705 : SUCCESS in 1h 32m 59s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/82580800b9e04ed0957cf0355530f132 : SUCCESS in 1h 56m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9cfb9d04694d4c0b8d51682d6e84e2f5 : FAILURE in 2h 46m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/0a08b965989648cead6fbfe82ac6a7d1 : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/a3c336de7928438aae13288b21873843 : SUCCESS in 28m 42s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/2da02af7cf694a218cd36b90cac3a743 : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/20d21338c4cc4778bab05d73bb74c2e8 : FAILURE in 1h 06m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/cc40312cef4e4500a31f89a7b8375384 : FAILURE in 1h 11m 05s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f0131497f1414f3cb09f019939205781 : SUCCESS in 28m 18s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"cb52c357b4247088f6b257f2761295b79b1931bc","author":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"date":"2025-05-29 18:18:35.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"70eafce3e293ecb4be773daec5982a44862a08c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 18:44:23.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"b45e74848e4cbb1e89b02e0594134c5f2f910a95","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 18:45:00.000000000","message":"Topic mariadb-fix removed","accounts_in_message":[],"_revision_number":9},{"id":"67862f7910b55dd0998a59e7f8731b2bc8ad03ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 20:43:11.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5407cd2321d14497983c755686abd37c\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/207877469b1f47989f4a06f12fcfa04f : SUCCESS in 28m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/936e76f16e2d438b9ae0ad8ab3a58a16 : SUCCESS in 57m 56s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b7c5b512144e4445834bafec889b780c : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/66532e7097f440cead173807ee63f772 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/08bf4f543a46434287c94812241e35c9 : SUCCESS in 10m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/9e9db343f4df4839ab605d3996660cc5 : SUCCESS in 53m 26s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"5ad5852227620931de5ae402be6d56c56b31bbb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-29 21:53:40.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":10},{"id":"f50bcbce33e86eb1ccb23847e48cdbdb6976cdc3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 00:15:36.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a84b2d3ff51446ebafd73e5786141d29\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/5ab6e8a37eb04da097a23a588e4a566c : SUCCESS in 15m 59s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/fbc9297f588f4a239fe228d14df35a5e : SUCCESS in 24m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d1fdafc412674d23b6fe3d1ab6ab4143 : SUCCESS in 56m 00s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9a097326377943d3bc50930d05fa9b5a : FAILURE in 1h 26m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/258be7a547d4464d871f415c7e408fbe : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5d7adf97221046619cbc6b86f62463bc : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/73af9a6adcb44756b14e1b98444dae22 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a5e0bfae0e2469689101c50e1c03788 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7a30783a8bf4b1da39b021d52627db1 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6998b3ff0b80430591f41ee3cf54e9b3 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b41c9737d74b48dfb9d016e9718066ae : SUCCESS in 3m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d29d0a1a141549deb1d2bdc0511d8f27 : SUCCESS in 3m 09s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/bbca96b729234f63ad21d4338f7ce9cb : SUCCESS in 20m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/246379e55c76410cac90645642361125 : SUCCESS in 20m 57s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/aad300c129bb4021968b45650b4773f6 : SUCCESS in 1h 01m 02s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/1967ba537749470e92b5004ca60855d2 : SUCCESS in 1h 28m 56s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/14aa4fbabce247a5a3063ea7841695f8 : SUCCESS in 1h 46m 53s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3ff00ad66e8a4675ab40c291a9faf666 : FAILURE in 1h 57m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5c3ffea8aaab41c79f86c24621794a09 : SUCCESS in 1h 11m 46s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/32724879d10e41108fe30ed54fdf4c00 : SUCCESS in 28m 12s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/0ef2167090c14c7d9d1ebf737f2bba64 : SUCCESS in 59m 46s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8ce6f06f8b7a414ca4e6928a95ab6e65 : FAILURE in 56m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/afcd1d5f10b34a85915073063851b617 : FAILURE in 1h 10m 07s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7dd49bf207634ea4a9682e65fb05361a : SUCCESS in 29m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5c5f84637361f34d34311bf07bba4f972ce0db92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 00:18:47.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75b43f6430c247a6a769207ac046cb64\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3836f1e130a84ba78ca68cf341f9d475 : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/59321da389384c3b9dfd1bf006ff1cc7 : SUCCESS in 2h 21m 58s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b6782d223eee4a33b5a84780d1d738aa : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94d0d7bfa6674091bef4ae05a2a83b71 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ade07876f79b4408ace45ad0ca7ed29a : SUCCESS in 10m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/be97f0f75f934dc28f27d7bf9d89742e : SUCCESS in 1h 15m 26s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"29b2b073629f3b2d18faa5e086ef774a11a0dcd4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-30 10:27:54.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"28b7dcd5ba45e38a4e563809ea13efdde3107081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:26:06.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/19f8ad3b6964418b82bdb92c2b2dc735\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6d717453e59c4929a33a12f4ddfc4244 : SUCCESS in 19m 12s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/43dc198605b1414fa7261e7b55cc366d : SUCCESS in 28m 35s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7784519c32bb42c589419d75b113c7f6 : SUCCESS in 1h 03m 02s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/1c38d2b0f2d447c1a85bb1f9b1deb326 : FAILURE in 1h 29m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf7408adcce4a928afb9f1d57a7475a : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2548e0cf627d4c0aab52e1be7309b83b : SUCCESS in 2m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34279f291da140ec9a22760ea2acc972 : SUCCESS in 3m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fedccc0f1d52479bad51b59ad23342be : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b0c5771cfc6452698ff98d47b7567d6 : SUCCESS in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/040ad8125442401781162038948848cf : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/051ee3bbe0464ca39453edbbeda56ed3 : SUCCESS in 3m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0a22c64cd10848adb643da4e882e19e8 : SUCCESS in 3m 13s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/9d31fcde8153402c8e6466645af31941 : SUCCESS in 20m 15s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/75b1f970b2c84d74acef25d3dd2b776c : SUCCESS in 24m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/7da619f8fddb4f19936206a7826b6878 : SUCCESS in 1h 04m 47s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/7d3367f8e1ac479baf937cf60ab6b6d0 : SUCCESS in 1h 34m 01s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/251e3617c9474d998a00226e84c28dc9 : SUCCESS in 1h 59m 00s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aaf40a31466c4c548335617efab9f5dc : FAILURE in 1h 24m 37s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/94338e8b250c420d93bb1ffcbfda8e7b : SUCCESS in 1h 07m 44s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c4fd9c23b84c4ee38d667c3da85f727b : SUCCESS in 24m 49s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/3b6b8ac878de4ec5830a92f913375853 : SUCCESS in 1h 01m 09s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/b0af9c901bf2421cb5262e65433df2b3 : FAILURE in 1h 06m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/35520706fcd04a3887a3e66b3722df6b : FAILURE in 1h 15m 55s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b08cda60d7ed4f369db0a2ea4a0c1855 : SUCCESS in 25m 14s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"0560800933c150cf9b73c3475d218853a331ef3e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:32:45.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c92af9c43434177bde6d184eaa90097\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7cfb1510374f47c39a048bfc4c6f0daa : SUCCESS in 50m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/3485efeefe9a4200a8a9978a41ac1679 : SUCCESS in 1h 58m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0c6d186fb5464128b0014d3143a2b7f7 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8e7ecf393b9445a8800c92ee8e713b5 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a40280b2dfb49ccbecb9f6ddbf9d69a : SUCCESS in 9m 17s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/b4995bca19654c71947f131b85f3bc8f : SUCCESS in 57m 21s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"68e87e6a06a9196a3d0059bb33a9a51f2f3dbf26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-30 16:07:58.000000000","message":"Uploaded patch set 11: New patch set was added with same tree, parent tree, and commit message as Patch Set 10.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"78707893037b01bc6d98262d5fae394c02cdf502","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 19:01:49.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a5117a75318425a880acab71eea4481\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/39177b26b38f45e4a782a88e269f3e4d : SUCCESS in 44m 10s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/677fdf3791b4434782295f126fd04872 : SUCCESS in 1h 47m 52s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9f64f22660034af0beff64ca98a807ca : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/412b02f0a0b9446182602974a2bdee06 : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c39d12f973644b44bb65b3c6a2fb3842 : SUCCESS in 8m 53s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/49da70c2599a46b1898b729dcbbc10f1 : SUCCESS in 53m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"503e00a38633f5b4dfe6954a32138538703e86be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 19:18:43.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/866d3b12434540749bda462fec2c2161\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e737ccc0b61d44f68b49c1fdf2c32acd : SUCCESS in 18m 18s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/745cd1b6e55f488092d8c02987ec3198 : SUCCESS in 27m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e483c2c3611640bfb3944657d90f4f64 : SUCCESS in 1h 05m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/288b2fd809364c059a2ebba825d5afa8 : FAILURE in 1h 32m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36a71c66f4b74f50baa693c19e836d30 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aff679e27edc4813a48b0c80bd251c07 : SUCCESS in 3m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/66aef19d75364ee1a03c8a9069d3981d : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57e9ed07538d487a8a817f61e7275b90 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bef99abdb8e463b81ae5d5b9fc6be3f : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a981fde29af24ae184f56608513ecff3 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/630ee096f63b42d585695d07589ad625 : SUCCESS in 2m 52s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/685a4a08f7bc4dbdb387d949fdc50afb : SUCCESS in 3m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ece73234b1544ff7bf59c79c215007df : SUCCESS in 19m 46s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/ac13bcde91964a58819357b501261bfb : SUCCESS in 24m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/28f30030fbf04c63a4c84e2d040eba8d : SUCCESS in 1h 09m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d780aeddc7474906bc699200f344da08 : SUCCESS in 1h 29m 57s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/9129086a081743889ed0177e950d55c0 : SUCCESS in 1h 55m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8d23a60a483c4869a79d20e17408faaf : FAILURE in 2h 31m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/df3620563fda4056b2bd4d4183c9d4c7 : SUCCESS in 1h 09m 27s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/074e34f2994b4e5bbe001451bdbb41aa : SUCCESS in 24m 57s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/c27bd1afa81c44eabb112c4bf1c3d068 : SUCCESS in 1h 01m 59s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/fd86adc3cf33426f929a7c509b162a69 : FAILURE in 1h 04m 37s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/eed759a58214430893d80aa56656e2e9 : FAILURE in 1h 14m 34s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2c31c64133204eb4bbc5c140cd95ae52 : SUCCESS in 27m 38s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"eb9357bc5a49cce49d38e3c761d94be850a8ac93","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-31 04:06:36.000000000","message":"Patch Set 11: Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"c80f7cc06de1ad3b155ce3b474fca8efee3952e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 05:52:52.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e13451e4ad6344ec8a761744e6a32bc6\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4c696b257fe149848ac02aa2f31cb8b2 : SUCCESS in 44m 32s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d9d6cd77c7184cf2a9390309af2abce3 : SUCCESS in 1h 44m 05s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d7a4719405f34541bff833aac9181846 : SUCCESS in 4m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b620da2665c246ba8dfafddbb7d1a492 : SUCCESS in 3m 30s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d5277bff0e3d4e35b082769d0928ac45 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0a205a1fad444a928923b14f02115064 : SUCCESS in 49m 11s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"70e916a8ab74fe538459815bc92ab77665406d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 06:23:47.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/38294e9737914761867473e7a85b2386\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b35e1e7fa4aa4a45b798ffc44f2d8b11 : SUCCESS in 18m 23s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0dc78efb08c1455ea24cb1ee7f4f2a05 : SUCCESS in 26m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f63660ff551547c7a4e53901d08ba0e6 : SUCCESS in 1h 08m 15s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/a0088c732c0e4cf6a2b802e40cbd7e70 : FAILURE in 1h 59m 32s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a5aa17c24024dd491c423b9839eaf35 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c58a515c7f249728402d3f880135df3 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ca3ef5d02694e2995a41d639fe24483 : SUCCESS in 4m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4003c249958045919015a489aa0858cd : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/625385c638d942e38d568ed89b693651 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f188919e8d34ac09943dcdbaee5c763 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd2bf72854eb48cea878242770a202c5 : SUCCESS in 3m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/cd5cd9c0e3454b7284d4023da386a6e9 : SUCCESS in 3m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/5eded6fbe330496e99b31550eee2720c : SUCCESS in 17m 25s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3f2e9658ef6d4819ae16159773c0ce27 : SUCCESS in 24m 03s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/ba95f2e4ff944192bf5761919ee289d4 : SUCCESS in 56m 23s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b6672c9a26424e1095f5a3ee080be681 : SUCCESS in 1h 32m 27s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/b46c3cd854294f898e2dafca2c20010c : SUCCESS in 1h 47m 05s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9e8d7dad5a7549bf834ba0ea915e5850 : FAILURE in 2h 10m 18s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/46d725c3c4644f08ac3d9a257099b22c : SUCCESS in 1h 17m 56s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/7188e6d78a8946b396eb2272618d4297 : FAILURE in 25m 09s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7eb171fa55e6423f98549f8ebdc0f353 : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/ceeb4e566e6b42b29ab690858223dc53 : FAILURE in 1h 02m 45s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/e97f46369ca34f9988430d1d2415ff7c : FAILURE in 1h 12m 49s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f560e489706a4d09b3078f3e2c76500f : SUCCESS in 24m 32s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"0435eb7d5c17fd97d80e22bf8c6a22053b5d55b4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-31 12:03:57.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"5f6f7a31b68bf6ea0d38e662a1b43415947e4303","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 13:55:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a51dd4aa8e0e4dca87a0e17d4207b6ea\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/fbd438e406b345fc83027b91ea03ca19 : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/eb2f7d1ceb004e91980b05c3a3a98c93 : SUCCESS in 1h 49m 04s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/96f761b5e93c4abc8997f1f50d4822b9 : SUCCESS in 4m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/affffd7048af4454b867ce6520ddc373 : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/44965701c133497e937cccb5f60008c9 : SUCCESS in 6m 06s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eda794e45b2b48b19a1244097bdc591f : SUCCESS in 54m 07s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"35478a3a57079b69080d2d8d13d13051f88d89e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 14:45:09.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25863fd2d36b4d419a26884367f06277\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/da1932cc830647b083c190f414f5d030 : SUCCESS in 16m 37s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/8f0663bf1ada45a382c7aa07ec09fc1a : SUCCESS in 24m 30s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/b909b09b1861476d90c1564de4d44f75 : SUCCESS in 52m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c9c31e1a8cfa4642be415d1328daf948 : FAILURE in 1h 22m 22s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8141f2d05cc4a52a8f805eb6479b443 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4cf756b6c0346348057a973422d4615 : SUCCESS in 6m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85f23b3ef6c7420cb955a30aabd3a904 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b769adc14f234781948c19c8bdf36ea4 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5bbf5582ef8549dd8b48396bd93c2013 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b53be2bc0de2446ab1923456f07b8b18 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf282962acc48cba6436b7a940cc10c : SUCCESS in 3m 02s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0f812880b7dc4453bf05a748aa7e00b3 : SUCCESS in 5m 36s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c4d5986ce07546d0aada7a191686fbe0 : SUCCESS in 23m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/f49b2f1bef644e3c8cb77f372056b7e6 : SUCCESS in 22m 02s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/97099451585a4d2fa0b8e6f138dcd418 : SUCCESS in 1h 03m 53s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3cc6b97e3b0b4eba8043b7ddafdab648 : SUCCESS in 1h 42m 25s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/6235c3fdd03a4979b322d00765ff9674 : SUCCESS in 2h 04m 12s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/13c5d696844949d88635caee1c3dc920 : FAILURE in 2h 00m 50s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/237c85f4463f42feb36c410761133c57 : SUCCESS in 1h 20m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/b5e552ccb23c4e5f926a70bf00af2a6a : SUCCESS in 24m 18s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/1252f0454ad6450f892e3afef6b4a682 : SUCCESS in 45m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/7a60d83f743f4fc987e10dcb057d311e : FAILURE in 52m 52s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/654f3d99db764c569902ab19b2812831 : FAILURE in 1h 14m 41s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1ad0db706f9d4da5aaa904126ed19b02 : SUCCESS in 27m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"60d20c3205e7dc4884c789de5c187aa284b993d7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-31 14:49:41.000000000","message":"Patch Set 11: Code-Review+1","accounts_in_message":[],"_revision_number":11},{"id":"2b3f6ae381628de715467a396657f3cba2327aa0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-01 11:21:34.000000000","message":"Patch Set 11: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"23b65f27bf29a86cfee7ac992a103716e54fd0f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 07:50:40.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"53e375a8751d115b7fa0e98212774834901e345c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-06-03 07:52:07.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6d4b29cb40486ab7a9f30c4d328485e8216d1b6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 09:54:09.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df5339cd32c9462a9cf196053a49e5c4\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/fbd67858eab94037acb2a1cf3b14c8fc : POST_FAILURE in 20m 25s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/6631975d88784f8f9cedef6451d1dfd4 : SUCCESS in 29m 47s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/778b0627fd7a4678bee101b96320beb0 : SUCCESS in 1h 02m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcfed7d1a20546ec916bc87976a304cc : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de19a57ce99c44c999968533b6842b12 : SUCCESS in 3m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3a7b6f45fd924c618e6f2cffdac6c133 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e58084fee44f199d2818c74eedbd36 : SUCCESS in 5m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60fd36ffd4f64706b90732108fa40c3b : SUCCESS in 3m 41s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d92e9033c27a4081937a88ba36506636 : SUCCESS in 4m 25s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/29cfdbd4c2a9455f97c84978cd6cd68b : SUCCESS in 20m 03s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b0b813eaacc24029bd67a8e7596bd559 : SUCCESS in 25m 59s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8bc592db3c7548d78b9cf40fa5a0ae65 : SUCCESS in 1h 03m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/b4b4d39b41cd4781ab5a9bb2b41a3251 : SUCCESS in 1h 47m 52s","accounts_in_message":[],"_revision_number":11},{"id":"fb9fab5b31f6a4934f5228b39789c694de097355","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-06-03 09:58:16.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"0e479888f033ff6cf7929c0e5ad8ec408e861397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 12:48:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71399c742b3e43d898fe3a7b3f0ae6e9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a1bc793cacc467c99d4fdcd672cc50c : SUCCESS in 17m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/2047091754fe43dabfe5c347bb0014f6 : SUCCESS in 32m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fc39d7d6808c46d3801f143060b1b21a : SUCCESS in 1h 09m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fc3505ba5d64414c8aca642a479178ce : FAILURE in 1h 39m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be79f798791453aafcd14c2db6acb4f : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a52943d389f459ab3e7d8b418e4957b : SUCCESS in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec216f4a6c5f41ca94dfd447ebca1dad : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e48ccd816cdb493dad7fd1c85dcbed8d : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fdecbceb7fe47488a18da5a6853392c : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69981ec179db49c7805d299661ec60b6 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/944191414a63432bad2b1613d111c19f : SUCCESS in 4m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/011db08b301949bfaa4868c439599d78 : SUCCESS in 4m 07s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/b69975caf9e54629bdf9e9985e8e7bf0 : SUCCESS in 26m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dd7be813540c440da180e94bf013689b : SUCCESS in 23m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1feb280d9ec8444ca95899704c751c24 : SUCCESS in 1h 05m 21s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/9b17fbb8d8384736bb08fe4f6563c7f9 : SUCCESS in 1h 35m 42s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1984e569e134424ea8865b4b26642858 : SUCCESS in 2h 02m 42s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/347315be177c47788916214d152de133 : FAILURE in 2h 29m 25s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5cb46bf01cd84b3f9041f5d22defaaa3 : SUCCESS in 1h 19m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/633f10070a0845dba0d4b955ab8d1a78 : SUCCESS in 25m 54s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/ee7ed943d0c04570afccf04e8bdfaa3d : SUCCESS in 1h 02m 21s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/20332676e9e84d14a5d41cfb2401c0d6 : FAILURE in 59m 36s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/d7245fbef88648849157e2ea456ad00c : FAILURE in 1h 08m 28s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/2703300ce92d4d66b9691e6e835f170a : SUCCESS in 27m 59s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"d6eae7a35734e630fcaebdcff6bb87fcad5f94aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 12:48:40.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"4ee092776e3b1296bfc95699f499436367bd83fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:03:02.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6edf90597a7f44ada1a497912f30e43b\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/22bcdb906b654509836e4759988e8761 : TIMED_OUT in 3h 02m 21s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/429223c3ab5544a9a0fd77ff56412619 : FAILURE in 2h 21m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/538bba383689434ca95e7edd47c6d817 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d643a66d3123473ca8ec29610591d921 : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9d46c64ce0d2422b8e10f076be8f864f : SUCCESS in 5m 39s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f01f59cb5ea94e4cb99c48479063be40 : SUCCESS in 2h 18m 04s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"58cdaef2444b52337a949684479f9b11427b63f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:21:44.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eea23cf8ed044b0abadf28a5485a528a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/95426f62000643ccb53f68074848b21e : SUCCESS in 19m 24s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e80ea903c794459cb6f29d522863b290 : SUCCESS in 27m 37s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/633d5cf4f95d4d5688edfc5007ae4925 : SUCCESS in 1h 02m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/175f64fe6bc94846b3f70968bffbd2ed : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc5125c85e3c4f3baeadb8691d040e29 : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2661be98a544a67baac471c891d41b2 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7cdedd97935454b9b1aab8a2211fcbc : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/553f4f36f34d46a1915fe144876c0a05 : SUCCESS in 3m 00s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/54fea0e89221402fa7926564d71abbc8 : SUCCESS in 4m 01s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/2201a34b96c54015becda2e514349f63 : SUCCESS in 22m 19s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8972f9e3832443398f005dae8af0b409 : SUCCESS in 32m 36s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c8872ffe42dc422c88dea665fe23ed5c : SUCCESS in 1h 12m 17s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/05f92a59c3f74d068a54b3e724517f3e : SUCCESS in 1h 32m 24s","accounts_in_message":[],"_revision_number":11},{"id":"f9f512cffb1811f294361585cbbc9b80ed9392c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:21:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"fa5e15cf20def5e0f1e6f571cccb9fb2d7ab51c4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:22:40.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e6816590c2e4881baa3ab3c6014ba52\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b898351a7c46d8afb6a5b374e6ae37 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/167d732fdbe74342bb61818aab4cc12a : SUCCESS in 39s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"bf86273cafbbff0ba3eb8227d3e3be1854866e5b","revisions":{"a9619197f87f86abcee786a7f7c86768a4db6a4b":{"kind":"REWORK","_number":1,"created":"2025-05-12 07:13:32.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/949444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/1"}}},"commit":{"parents":[{"commit":"a01e339edc7de6c1682a5b8fb19b053ec397210e","subject":"CI: Set upgrade jobs back to voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a01e339edc7de6c1682a5b8fb19b053ec397210e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:13:30.000000000","tz":120},"subject":"Opensearch and Opensearch Dashboards have been updated to 3.x version.","message":"Opensearch and Opensearch Dashboards have been updated to 3.x version.\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9619197f87f86abcee786a7f7c86768a4db6a4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a9619197f87f86abcee786a7f7c86768a4db6a4b"}]},"branch":"refs/heads/master"},"89a5e68e512537ab4cc1a83ed65a5b1ff30604f9":{"kind":"REWORK","_number":2,"created":"2025-05-12 08:40:59.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/949444/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/2"}}},"commit":{"parents":[{"commit":"a01e339edc7de6c1682a5b8fb19b053ec397210e","subject":"CI: Set upgrade jobs back to voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a01e339edc7de6c1682a5b8fb19b053ec397210e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 08:40:25.000000000","tz":120},"subject":"OpenSearch and OpenSearch Dashboards have been updated to version 3.x.","message":"OpenSearch and OpenSearch Dashboards have been updated to version 3.x.\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89a5e68e512537ab4cc1a83ed65a5b1ff30604f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/89a5e68e512537ab4cc1a83ed65a5b1ff30604f9"}]},"branch":"refs/heads/master"},"486ab1a06ddcb23619adce3a1f0090033fad173d":{"kind":"REWORK","_number":3,"created":"2025-05-12 09:10:00.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/949444/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/3"}}},"commit":{"parents":[{"commit":"a01e339edc7de6c1682a5b8fb19b053ec397210e","subject":"CI: Set upgrade jobs back to voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a01e339edc7de6c1682a5b8fb19b053ec397210e"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 09:09:29.000000000","tz":120},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/486ab1a06ddcb23619adce3a1f0090033fad173d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/486ab1a06ddcb23619adce3a1f0090033fad173d"}]},"branch":"refs/heads/master"},"24f8e9d05659a0ab44eeecf88e624e31afc49171":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-05-24 14:59:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/949444/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/4"}}},"commit":{"parents":[{"commit":"2acba2ec8971ad6882ff7b07f1236e978bead678","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2acba2ec8971ad6882ff7b07f1236e978bead678"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-24 14:59:00.000000000","tz":120},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24f8e9d05659a0ab44eeecf88e624e31afc49171"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/24f8e9d05659a0ab44eeecf88e624e31afc49171"}]},"branch":"refs/heads/master"},"57434f549f069f30db734aec78cb26ee8e5e49b1":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-05-27 09:02:13.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/949444/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/5"}}},"commit":{"parents":[{"commit":"2acba2ec8971ad6882ff7b07f1236e978bead678","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2acba2ec8971ad6882ff7b07f1236e978bead678"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-27 09:02:13.000000000","tz":0},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/949478\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57434f549f069f30db734aec78cb26ee8e5e49b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/57434f549f069f30db734aec78cb26ee8e5e49b1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"6deaaf332cff0d0a2d30ef9b02fef6a20e4287d1":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-05-27 09:10:21.000000000","uploader":{"_account_id":37306,"name":"Piotr Milewski","display_name":"Piotr Milewski","email":"vurmil@gmail.com","username":"vurmil"},"ref":"refs/changes/44/949444/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/6"}}},"commit":{"parents":[{"commit":"2acba2ec8971ad6882ff7b07f1236e978bead678","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2acba2ec8971ad6882ff7b07f1236e978bead678"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-27 09:10:21.000000000","tz":0},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nRequires changes in configs: https://review.opendev.org/c/openstack/kolla-ansible/+/949478\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6deaaf332cff0d0a2d30ef9b02fef6a20e4287d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6deaaf332cff0d0a2d30ef9b02fef6a20e4287d1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e2e8f6f3e55c8f804fffc441ea5b3781bb2d6257":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":7,"created":"2025-05-27 09:18:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/949444/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/7"}}},"commit":{"parents":[{"commit":"ee8c6bde8534f2e16bc9b4b528e6da3f958ba7ad","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8c6bde8534f2e16bc9b4b528e6da3f958ba7ad"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-27 09:00:53.000000000","tz":120},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2e8f6f3e55c8f804fffc441ea5b3781bb2d6257"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e2e8f6f3e55c8f804fffc441ea5b3781bb2d6257"}]},"branch":"refs/heads/master"},"e60c52a048559622b20d873488eec55f92a6078a":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-05-28 13:18:18.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/44/949444/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/8"}}},"commit":{"parents":[{"commit":"ee8c6bde8534f2e16bc9b4b528e6da3f958ba7ad","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ee8c6bde8534f2e16bc9b4b528e6da3f958ba7ad"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-28 13:18:18.000000000","tz":0},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/949478\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e60c52a048559622b20d873488eec55f92a6078a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e60c52a048559622b20d873488eec55f92a6078a"}]},"branch":"refs/heads/master","description":"Edit commit message"},"309818d55f1ef19dd72618209a83b0cf4aa8603f":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-05-29 18:44:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/949444/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/9"}}},"commit":{"parents":[{"commit":"a918799ce59b83c1d4721bed2df028c62a87c2cf","subject":"debian: Use bookworkm suite for rabbitmq APT repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a918799ce59b83c1d4721bed2df028c62a87c2cf"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 18:44:10.000000000","tz":120},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/949478\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/309818d55f1ef19dd72618209a83b0cf4aa8603f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/309818d55f1ef19dd72618209a83b0cf4aa8603f"}]},"branch":"refs/heads/master"},"638a8802e2578dab0dafd6fe603f4929a44247ae":{"kind":"REWORK","_number":10,"created":"2025-05-29 21:53:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/949444/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/10"}}},"commit":{"parents":[{"commit":"5b02f3e7ca34a2943a11134eab82acfa1d226a91","subject":"Bump proxysql to 3.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/5b02f3e7ca34a2943a11134eab82acfa1d226a91"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-29 21:53:13.000000000","tz":120},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/949478\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/638a8802e2578dab0dafd6fe603f4929a44247ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/638a8802e2578dab0dafd6fe603f4929a44247ae"}]},"branch":"refs/heads/master"},"bf86273cafbbff0ba3eb8227d3e3be1854866e5b":{"kind":"NO_CHANGE","_number":11,"created":"2025-05-30 16:07:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/44/949444/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/44/949444/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/44/949444/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/44/949444/11"}}},"commit":{"parents":[{"commit":"a4f68e18890fdc837230f20c6fcca2972746ab1a","subject":"Bump proxysql to 3.0.x","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a4f68e18890fdc837230f20c6fcca2972746ab1a"}]}],"author":{"name":"Piotr Milewski","email":"vurmil@gmail.com","date":"2025-05-12 07:12:11.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2025-05-30 16:07:48.000000000","tz":120},"subject":"Update OpenSearch and OpenSearch Dashboards to version 3.x","message":"Update OpenSearch and OpenSearch Dashboards to version 3.x\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/949478\n\nChange-Id: I4507fb60d361ea504459a4b502d6f6c42a327064\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf86273cafbbff0ba3eb8227d3e3be1854866e5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/bf86273cafbbff0ba3eb8227d3e3be1854866e5b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
