)]}'
{"id":"openstack%2Fkolla~949753","triplet_id":"openstack%2Fkolla~master~I28f474015ffd000a3c2cf4c461c3c2f2b8959b0b","project":"openstack/kolla","branch":"master","topic":"kolla_2025_1_reverts","attention_set":{},"removed_from_attention_set":{"35345":{"account":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"last_update":"2025-06-05 22:28:18.000000000","reason":"Change was submitted"},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-05 20:46:47.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-15 10:41:03.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I28f474015ffd000a3c2cf4c461c3c2f2b8959b0b","subject":"Revert \"[release] Use RDO\u0027s Epoxy packages\"","status":"MERGED","created":"2025-05-14 14:04:38.000000000","updated":"2025-06-05 22:29:12.000000000","submitted":"2025-06-05 22:28:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":944263,"submission_id":"949753-kolla_2025_1_reverts","meta_rev_id":"42e086cd7db1f39ab5f6755b1fe64d5018fe821f","_number":949753,"virtual_id_number":949753,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-05 22:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2025-06-05 20:46:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-15 10:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2025-05-15 11:19:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2025-06-05 20:46:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":-1,"date":"2025-06-05 20:46:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 14:04:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":35345,"name":"Grzegorz Koper","email":"grzegorzk@stackhpc.com","username":"gkoper"},"state":"REVIEWER"},{"updated":"2025-05-14 14:04:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-14 14:04:38.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 11:19:59.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"8276bdee29de7e30776da52f26c864485278ff30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-14 14:04:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"553ecb0a93a8f9c42f40ddc9a01e416aba08a969","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-14 14:08:15.000000000","message":"Topic set to kolla_2025_1_reverts","accounts_in_message":[],"_revision_number":1},{"id":"db2d209dcdd007db39f80403f046c1eda8fa66f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 15:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/718587b72a69492c8d5881c898da8015\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/8f96a7d4635d4d56853abe5fd076e88d : SUCCESS in 24m 44s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/ad880f9f458d458db3641c41a19bae4e : SUCCESS in 1h 06m 55s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3b5febe30df14d9c9a2be055897f9531 : SUCCESS in 20m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/879859cd5aa14ef6bb661d2424e56512 : SUCCESS in 29m 04s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/579fc1d6122e4df2b9f8dd13bb22d6da : SUCCESS in 1h 05m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/4f57cf51ef574789b58f4ec1b2edaa54 : SUCCESS in 59m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe6abfdd597e49aa9f071a5e3df45404 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/61cfafdc56324f03b9030a0a0a898d6d : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e6ca4fcd7064186b15b85a05d8ba068 : SUCCESS in 4m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4b37835b0cb84129ba50ae3d1a99ea82 : FAILURE in 5m 12s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2ba2691bbb1422290e101a7fc848d36 : SUCCESS in 2m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11db10ae9abe4afbb97338c5ce3554f5 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6daa870c78fb48ff946514bcba951c1d : SUCCESS in 2m 49s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3eafc24312544c428c1687edcba6684e : SUCCESS in 23m 46s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/2ecd1ed2fce847ee85f84eab7de8b4ad : SUCCESS in 1h 21m 10s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/10b10da738d34ff98c567c5b3ea45ca0 : SUCCESS in 1h 11m 16s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2bef09f3f9784593819bebf388d2ad53 : SUCCESS in 1h 34m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ddd6df3251d7494297cec1017fe3c38e : SUCCESS in 15m 18s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8a045904a0c04bd2990b08be2241b236 : SUCCESS in 25m 58s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/6e035db3b62541a9a4ca06fbd30160fb : SUCCESS in 1h 11m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3af4d2b7f91544999e33c2d5c325cfcc : SUCCESS in 1h 45m 14s","accounts_in_message":[],"_revision_number":1},{"id":"770901d794ed978aa44041c7e2c3c9cf7de1c13d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 16:43:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f6ccfbfd17e4a10ab5a24f1c7452222\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/db95489461c44561b6873d893c73fb4b : SUCCESS in 32m 35s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b391e4a6dccd4f388a2e4606ec069bb4 : SUCCESS in 48m 33s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d09b112d687d4d7e9f1dfd10e7e4da69 : SUCCESS in 1h 55m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6395af03495f4faba754932a176f167c : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a514a2182a6440e8f67cd1b1a084671 : SUCCESS in 4m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/565b71fe897d4688969c70b5c188ed9e : FAILURE in 8m 35s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/ebed694dc773435696dad69e9870ae70 : SUCCESS in 44m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/8d72d02085284253bca03bc5e0402455 : SUCCESS in 49m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b283ad19b343f80c8d5dcb588d8f4356bf17ac00","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-15 10:41:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f7ec885df980d8069a957630e0c14ac9d9d81a8b","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-05-15 11:19:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"37c76d79ee78fcd66cb94a92245c725b12b62216","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-05 20:46:47.000000000","message":"Patch Set 1: Backport-Candidate-2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"532c50c5c5489c35318c684a7d90ccdb53a5085d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 20:46:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7589cab339dbdaf499ec3999c5451cf9f6e16504","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 22:28:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8722a388f514973a594c6286481139e\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/96d8491166c14630966dc1e6d3df8a2a : SUCCESS in 17m 55s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/7968875a96bd440d86ed0092256eb8f8 : SUCCESS in 24m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/2bf7d7166977469f89140dbe87d55b14 : SUCCESS in 1h 03m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2334abd16d614f7988cf7a26cba3ccb1 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/789a5b8bc1484d8d8235b78965b1bffb : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60e67434c7fe4d66aace0f634051cd8a : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0240d06fbc7f4617af1c09277bbc3242 : SUCCESS in 6m 18s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/71b7795910904a4aa2e3fe9fb5172438 : SUCCESS in 2m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c103a84168d34a78b364abb125661ccf : SUCCESS in 24m 32s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6b7a9c3c808a427bac1d3774ffa53a76 : SUCCESS in 31m 26s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/941a8c2f7586412eb1390122698e241d : SUCCESS in 1h 03m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/d804a8bebe634d08aaaa500a2b39686f : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c982bce474fbb9aaedebeeb6129576b0e1922523","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 22:28:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42e086cd7db1f39ab5f6755b1fe64d5018fe821f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 22:29:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8c15cf7b14341cbb2e3943532dd9e6a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5807ffe7b4464df5b75590d262d491c2 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d21cab3a1e00686333966ff400bf5e0c3cf70aae","revisions":{"d21cab3a1e00686333966ff400bf5e0c3cf70aae":{"kind":"REWORK","_number":1,"created":"2025-05-14 14:04:38.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/53/949753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/53/949753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/949753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/949753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/53/949753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/53/949753/1"}}},"commit":{"parents":[{"commit":"64f171ccfa02bfeb9c92a641e186bd987f0099ae","subject":"[release] Use RDO\u0027s Epoxy packages","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/64f171ccfa02bfeb9c92a641e186bd987f0099ae"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:04:38.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-05-14 14:04:38.000000000","tz":0},"subject":"Revert \"[release] Use RDO\u0027s Epoxy packages\"","message":"Revert \"[release] Use RDO\u0027s Epoxy packages\"\n\nThis reverts commit 64f171ccfa02bfeb9c92a641e186bd987f0099ae.\n\nReason for revert: Let\u0027s switch to master\n\nChange-Id: I28f474015ffd000a3c2cf4c461c3c2f2b8959b0b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d21cab3a1e00686333966ff400bf5e0c3cf70aae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d21cab3a1e00686333966ff400bf5e0c3cf70aae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
