)]}'
{"id":"openstack%2Fkolla~949755","triplet_id":"openstack%2Fkolla~master~If888500a791c66a592f17aa2be1f3784ee009a91","project":"openstack/kolla","branch":"master","topic":"kolla_2025_1_reverts","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-08-14 08:01:06.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-08-18 14:16:30.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-08-13 11:35:12.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-08-18 12:43:43.000000000","reason":"\u003cGERRIT_ACCOUNT_23084\u003e replied on the change","reason_account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}}},"hashtags":[],"change_id":"If888500a791c66a592f17aa2be1f3784ee009a91","subject":"Revert \"Switch to 2025.1 sources\"","status":"MERGED","created":"2025-05-14 14:07:50.000000000","updated":"2025-08-18 14:17:32.000000000","submitted":"2025-08-18 14:16:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"revert_of":944672,"submission_id":"949755-kolla_2025_1_reverts","meta_rev_id":"63a012b749428618e82352a460bf18f682088849","_number":949755,"virtual_id_number":949755,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-18 14:16:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-14 08:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-08-13 11:35:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-18 12:43:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-08-18 12:43:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-08-14 08:01:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-08-14 08:01:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-08-12 09:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 14:07:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-14 14:07:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-14 14:07:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"},{"updated":"2025-05-14 14:07:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a1c5d07a6b0d7e460f2f1990d08bd0b1ccaad5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-14 14:07:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2736e16634f0761a82b919dc2cda09ab3f3e997e","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-14 14:08:00.000000000","message":"Topic set to kolla_2025_1_reverts","accounts_in_message":[],"_revision_number":1},{"id":"aa52c35e5901027da118b60309caa1b1cded3909","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-14 14:08:26.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"533234f981e9b0044cf35b7369d367d098c551d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-14 14:35:45.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f06623fee22deab9a8843091792907f0db58e1ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 16:09:54.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/480385a60e1b4fbbbfb64033beffd9b5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/66891f15926b4b238f6a471e61c85e9f : FAILURE in 22m 17s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/dc2e8e5b2da040dcbbf2c25899e7d8ca : FAILURE in 18m 31s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/b4acb7c4bf484328acc0e0be6b76d21b : FAILURE in 19m 34s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/dfa67cd1ce7b4ef1bb1b84015f374def : FAILURE in 26m 02s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/cf320a512fcf40399490ebe023cfc924 : SUCCESS in 1h 04m 56s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/44d2641f5736442883af4d4c281cede4 : SUCCESS in 1h 23m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156b219e2cac4f338cc04caab6b8687d : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3766879fc6154c3494e8d3f287cb62e9 : SUCCESS in 2m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d970e763b05147aaaa176d2c4349940f : SUCCESS in 1m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b2b2300c04144049b08ddbaedb20dafa : FAILURE in 7m 37s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e0b7d76be854ed6aa997c4399e9113c : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/237dd9592b324223b38cea6d1643190d : SUCCESS in 8m 03s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3b43f821bf5f4e028035316262f8a37b : SUCCESS in 4m 17s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/c671127aae3a451f90d011cfc758aeed : FAILURE in 22m 59s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/03ec6bafdf26499f84a95fda5b7fac4e : FAILURE in 1h 14m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/9606b18d61d24c1c8282607beeca0927 : FAILURE in 25m 13s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/0d709dbac130446d97d971c7b3c66e82 : FAILURE in 19m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d21ec0b5176241c6a95913cc4298adad : FAILURE in 25m 47s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/4e19c1441b9e455b85cd4f79c645f2d7 : FAILURE in 18m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/17d9a234b5cf42c19da59e1ab7fcdab5 : SUCCESS in 1h 05m 11s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/8347749e63dc46a6935277324d1577ce : SUCCESS in 49m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9ba721cc5014d4142b85df93488d1a39754836ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 17:01:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4845c816f6104077b5c97a2c43add038\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/697bba63d9824b6f9ebc4715f8465c00 : FAILURE in 47m 43s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/059024db081f42e6b6e7c9b267b49cc7 : FAILURE in 45m 06s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/64b3739c5a4746d1be513bd5f785c9be : SUCCESS in 1h 02m 38s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/be8c769b2c3e442a9662dc6af5a885bd : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f3f6778558141afa2d287500d745436 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cf0b0b5a6044b5194b76efde3fb80b7 : FAILURE in 8m 31s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/d199efbcd2854bcab7c8d5361a09ac8b : FAILURE in 1h 00m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/ebf0efb3b12247f0aa843dd21cc1f0c7 : FAILURE in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"310bbafaf74bc2ae71ec08cde8df317489781ad9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 05:23:57.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f4dd2e76dec9ab79226ad8e5f0f63df7151b8df6","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 05:51:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2d9bbfbe6c467174fdc7592dc74ddd335fe7202d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-22 15:57:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"e011472a4f04612455a6ffc38044d775bebcdc9f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-22 15:57:34.000000000","message":"Patch Set 3: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"45e49c40c5c8ee77c76739c9d1473971fd666191","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-22 16:52:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e67e285f168987fdf051e53279bc7ca7f610e9d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:43:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a761956a6554b8ca774912980bb212b\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f9cb9683bb6a4660891089c2e26f84ed : FAILURE in 46m 04s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c19d6198b2ce40d08f182dd25a40ad97 : SUCCESS in 1h 13m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/099bf7d3bc71440ba92390802facd279 : SUCCESS in 3m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d52011356a884e0dab6175e183f84def : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/890bc0af76dc440c9d75ade0a7eb4b5a : SUCCESS in 5m 26s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/87e6ad128ad44665a82b26fdaf580807 : FAILURE in 26m 08s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6e3c721d59c64d1e875ba9d29647463780053304","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 17:51:03.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/16ec8f530f284002a9ad3f71169668a5\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7a74ce7085f24a6080bade4d6545c7ce : FAILURE in 21m 17s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d961b1cbc3564e3a9afd686f17afba09 : FAILURE in 25m 36s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/edd0eb4e40ff4bf0b816de4588a42e08 : SUCCESS in 1h 29m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/444fa9c4007d4eb49ad715aa2405ce3e : FAILURE in 1h 48m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/750a9d7ee1a64ec7a9006d30cc71672c : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28c735efad0740eb9a7cfde6faac366a : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/95d85dd749724bb48f412cf6c71d36ec : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9d07132af3334c6b8f515ff722326795 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4eb859c0fda42fd88de89533e6f1a53 : SUCCESS in 3m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12fb8c377b5140c78cdaf1aa0fbc2d45 : SUCCESS in 7m 43s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/64d03cb192e14cc4a552caf9cf91fc06 : SUCCESS in 3m 17s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/d4dc006da5644fbfb47823fde26b8464 : FAILURE in 25m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/049cee5647244fd2b5330420e8c66807 : FAILURE in 23m 14s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c5c389564aaa4b92b6c483fe9ab0e9e5 : SUCCESS in 1h 07m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3ef2a5d4eb12451b93ae9b3b9388661b : SUCCESS in 1h 36m 30s","accounts_in_message":[],"_revision_number":3},{"id":"e1d953d1e5d55e0d56ffeeac88bbb5408943214a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-22 19:29:22.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"c7e0d6311502468541c9b6f327628940e0175076","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 21:07:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e538e01444f448988ae6390114ccf0f\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/de3a870a6ccf48e2aec8d947e6ad8ae1 : FAILURE in 19m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b6515f6d3c2443b7af247b5bae52c739 : FAILURE in 27m 12s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f5162185fbf442b5b4b59d0248276633 : SUCCESS in 1h 02m 23s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5604b21eef5241a3a6a97c37ed98e522 : FAILURE in 1h 28m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61e2d13d3aaa4d9185ef371640d7e026 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8eca3a57e840470f874717a00e773a1f : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/56e5b4954b3f4b49bf2bcec60cf7d831 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0eadd2644a64e65b353045844dbde8f : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4333cd0260f4c6fbb04e567b9195dde : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de3dc92f19a0421d8cda157bfdde27e3 : SUCCESS in 5m 39s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/588fb6d97db44b4192d59ecd94a953ce : SUCCESS in 4m 23s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/71dd784b467f49aea6b6cf2df1bc3184 : SUCCESS in 26m 50s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/5a5497342b4b4058b37949e63f38f130 : SUCCESS in 27m 20s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/26220e328a2c489492b7d1e6d99dc9c4 : SUCCESS in 32m 03s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/2ddc3d469f054ba9a1d865799d2aac90 : SUCCESS in 1h 19m 53s","accounts_in_message":[],"_revision_number":4},{"id":"ae88e647a6da1d1c8f4dece959f5f096be9aff8c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 21:12:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35cbbf9b6e8c4a14b3d30766d7a6a5be\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/290897e29e134ab9b3f2bde7ccaba775 : SUCCESS in 37m 59s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0007d21002be48de90304b16a0b526c1 : SUCCESS in 1h 40m 02s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/97f572f46d1b48e3aba0ade1e4962524 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8c9c34b1377844d2b35806ae1c96d2e8 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/627edec4dba9484692eb7ed378ffcb23 : SUCCESS in 8m 43s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/06ad83a44d06421b8378fa8723e02419 : SUCCESS in 55m 28s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"47aa7b70edfc1e191228dfe7c96c3b454ca00911","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-23 05:13:19.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8bac0bbdbbf48f969b8103766ab9d7eee64f8451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 06:18:04.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2894408c95f44c67ae14be53d69ed7a0\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/f534bc5f78ac412ba5a588a29e2ece8c : SUCCESS in 29m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/80d54866fecc404db5c4e861abd68915 : SUCCESS in 1h 03m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/346a4cc56b7f4ee7adbc5941f71831d8 : SUCCESS in 3m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/33c03fce89414c28a433eac9d013f551 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c5974ca5f39848d6abf05a9c9c392749 : SUCCESS in 5m 15s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/83782fa3c2024389ab08ccd3976555db : SUCCESS in 28m 26s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"bba2eaeb5866eb0dfcba0ee3c0668af3ae32d16d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 07:01:12.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/de0dd858d2024e0eb410c2254b3b7db9\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/6b9c8288f3414db79ee4487b589fd136 : SUCCESS in 9m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c35d926693bc4fb484c591bc7910d852 : SUCCESS in 19m 00s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/c7ef6d13708d40b5aea30717d527ecff : SUCCESS in 49m 53s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/263b2aea277345e0a5aecebaaab21384 : FAILURE in 54m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8c8c6fa9d84377aaf5989cea89b152 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a10ca9afb00430999a639df0e55a555 : SUCCESS in 2m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99cc058754cc45979025317183374db6 : SUCCESS in 1m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb2ca32726af490a876a55a667d89622 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4819129a937456f9ff5fd5f646af4f0 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d2d114fa8c54ee68a6dcfdb4e969118 : SUCCESS in 6m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/eb1163addac2404caa3b053f7b34150c : SUCCESS in 3m 53s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/af92874213364bea92685bc3540cadb0 : SUCCESS in 21m 01s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cff87ee5166b49808acca1f39d2658c8 : SUCCESS in 27m 43s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/197dbef32bd6490c9df684cc2d3ce33e : SUCCESS in 1h 03m 25s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/885d1f40d9b54bd98689dcf184a64c6e : SUCCESS in 1h 45m 13s","accounts_in_message":[],"_revision_number":4},{"id":"777958d3ab609b93b018e68c570a8656ad0c6ead","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-05 20:45:52.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"20638cfd812489e21cc2ff355cf70fc9eabe7a03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 21:15:36.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e980c0628894773974c3795c0bd3c74\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4c7690d8400b43daa6b9f5eac1851704 : FAILURE in 10m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0f205a3c94b7478285694861bc7a5552 : FAILURE in 23m 19s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/25c2248645724a21a9c23107f8364fcb : FAILURE in 8m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/180b96069c5b439fbd16ee362d350a16 : FAILURE in 12m 35s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24a489fdb92642efb368164cdb3fe4d6 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31d3ebb20c5e41968a7c82896b1c24c2 : SUCCESS in 2m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bee345eed670429f8ac496af058ad656 : SUCCESS in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ba6525c206644478f254820ea8a0cea : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/148ea50f8f554ba4aedce69d3d2e4f23 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79891cdfac144deb8d38891b68a7ed81 : SUCCESS in 6m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c317cad0846e4bd396dfecaababaaf32 : SUCCESS in 2m 11s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/04c0a56835c24c168e48f8b1c90d287f : FAILURE in 11m 52s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/022cbd74dd484ac892a1b90738dcb652 : FAILURE in 25m 06s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/64003adb9965465eb5315c023b494f8e : FAILURE in 14m 56s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bcec6bd30a19465e89cc1f4f7be2f33a : FAILURE in 10m 36s","accounts_in_message":[],"_revision_number":5},{"id":"df4e49470b78233fb6d8f6267f56d2ed66b12f72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 21:42:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d09d70dbf15488e9fea50fa0785723b\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/cb59e5abd08e447697ef0ca0c45f9504 : FAILURE in 54m 41s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7954d6cf2116471ba580f413234957f0 : FAILURE in 46m 18s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a631d42846124e3693ac3f3741516bb8 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/719624a33ec5437ba3024304c4245e63 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/237ab3b474c64adf9f2107a6619bfa96 : SUCCESS in 10m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a1b4b50a89ab445098243a3fb513170d : FAILURE in 41m 36s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fcb5f21ad019a92d809f992d2f1537bae81edd66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-09 08:18:56.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"fe58864ed178b90a90e7c57481d46a217937814f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 08:51:13.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a5c6ca8279c491ca9e8ac702a12e82c\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/94ed6c88a5d34315aa8f8eb7e5f7c935 : FAILURE in 14m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0efded6fb0604b68ab6f6a315a48f35c : FAILURE in 26m 42s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/734946b0a6054dc2887e6b78f3cef657 : FAILURE in 11m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/3efa3496945a487c8586ed60f48420ca : FAILURE in 14m 19s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa5cad41d544e329b89d901689527b4 : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac01c08f98074545805586f4643d60e7 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/496bcac575ba40cbb2bcc08337ad40ec : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/860f77c2ce72454987cb89e5ee7aaf8d : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a121c1a4c57424ead8ac7722bac814b : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d46309a745b9445f867e7c819ce7c3ed : SUCCESS in 6m 20s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b20e8c80c752423c806a8936091ba559 : SUCCESS in 3m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a5fdf0e2e37d4fce85420ccfb809921a : FAILURE in 20m 43s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/901ad8c27255477c8e2661ae39ed0995 : FAILURE in 23m 49s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f6842ff4cd7540e8b9a35833783af8a9 : FAILURE in 19m 16s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6661fc3348c44715809c7e0071af2578 : FAILURE in 19m 09s","accounts_in_message":[],"_revision_number":6},{"id":"cf6044da761f4b02ab12ad8a21f03b6f20232216","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 08:57:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce1f1b67a43345008e7e5d4fc366f4ec\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79d8b3ebb03049ad8816da6004a2da3e : FAILURE in 32m 54s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8da379e254d5413298660d85e84a4903 : FAILURE in 24m 37s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb4a139c07844f9b9174a0fb39a2595c : SUCCESS in 3m 10s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c4ddf1f4a99f4126b6e318c9a3b2dfb5 : SUCCESS in 3m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c520d3d1c1004141bfb32a7f6dda6bcc : SUCCESS in 5m 02s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c209ce6680b4477db9b4ad84c9f200b4 : FAILURE in 37m 31s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9b639ddb91c6509cd5c80967846e75909a846ff8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-10 09:33:16.000000000","message":"Patch Set 7: Published edit on patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"d2dd5c2f7527df1b270b0de3080156ab58319ae4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 10:05:13.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9d42cbc91a64506b1e3c5a5a5705d32\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/85bf76add03f47bdb706ca7f9da777d8 : FAILURE in 21m 57s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/37cabab669dd411d8fb104374d07126f : FAILURE in 26m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/62aa6e76aa59450eb53f3a3ef80e805a : FAILURE in 18m 26s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/06a5838c2b4c4129aa2b93a0fb218c90 : FAILURE in 18m 17s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c55e7e54cf0d4508a6d3a6cf043a36dd : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d779b12e849643c89a589e32de025b06 : SUCCESS in 3m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/62af98b79fc34aa299996be68c2aa2de : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8b9fe8d06834d46abb95af3b9fa3ebc : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cded03324a484ca49be7355e6c7a8bfd : SUCCESS in 3m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64822ec0dbb749e28b263cafeeed2ca7 : SUCCESS in 5m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4efcfaeb82844c5fbcb0697a492774b2 : SUCCESS in 2m 51s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/60746c7186ea4871bbcbb248c7447973 : FAILURE in 18m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b6e0eec8232d445c88ab5abef91a7d2a : FAILURE in 22m 17s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dc5d36886dad4353b29089537bf1c90b : FAILURE in 17m 48s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a1aa1ee472a740839e6b46a58d42eb68 : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":7},{"id":"32ffae5a734de3d7b3647bc16edb213882cffdf9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 10:26:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d151d17f82044e8b947705e360a162d1\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d8390af94dcf42bc95de68a5efecce2c : FAILURE in 50m 25s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8c5bb71ae98043a98964c884dc5e9142 : FAILURE in 42m 50s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/5f76657e266b404693c660f4bf94d929 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ed09e9a22745423ab8ac7af3ef88a985 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/04949dfb3a4742a195d187a5eb15595c : SUCCESS in 8m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/4bb3c029d26a4dd4958af12380d32d47 : FAILURE in 35m 44s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"17656e1442ef73ef9005fbae5eb736465c85df62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-10 14:46:44.000000000","message":"Patch Set 8: Published edit on patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"b9a85a63022262bb4b73ca2a65241ad35bf27fab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 15:59:08.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f0bfc9572ed4228a52de8f7e229c0ed\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d993fca775b447cf8f51f708b1d320c1 : SUCCESS in 14m 47s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/15b7b2433343406f8d6a8c9d09c5bc6c : SUCCESS in 14m 55s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d2d32c04f97b4fe3a918fa9c809c8d8c : FAILURE in 26m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/fb8c4d79c6e64439b97f5214b79fdd46 : FAILURE in 39m 29s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bbaa613ea2841e7b27f57afa1e64314 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/284e9fdd025d40d093b03e9cf2a750dc : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bdd4a2fcb0f84b358f3a6d5639ba57e9 : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8a4415e102c4b5aa9223c5db7ed3215 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17a3b446a97f4dc5a274ee97d003e9d6 : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/239794c156a8499bbc76ff87d986ac8b : SUCCESS in 6m 04s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/266a92cdcaa6458884d84224b63ae03e : SUCCESS in 3m 08s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/a16538240d8843398abda3aa31c92776 : SUCCESS in 13m 02s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/3c381dcac1a8478cad83995acfe2f5bc : SUCCESS in 24m 19s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/4f157b5803a44083a69d42c3200dc382 : FAILURE in 27m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cafde0b247be4136b7b5c216eca55c6c : FAILURE in 1h 09m 28s","accounts_in_message":[],"_revision_number":8},{"id":"7065fbb102b90438fdb90fe554c687bbb8c23d99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 17:12:58.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6251d6625c514bd58e2739468a9a1819\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/01aa6c6f8f9f472a9b2a729b4dea46db : SUCCESS in 1h 13m 05s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/b1c3210a07fc442682bde2893b6b9c7a : FAILURE in 47m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8a670c2ac22c48f4a5c7072498473f20 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dcf7751170a147e88688825b98cc9389 : SUCCESS in 3m 52s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/46dd9355d8364d3d93195e459a209da5 : SUCCESS in 5m 25s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c7829377fd624ced8bb17e660f3df9f9 : SUCCESS in 2h 10m 30s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"8465f1d2a9e32280450eb3ddd24027a989e3ebab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-16 13:13:31.000000000","message":"Patch Set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"4b374f8d72721f1d29a6448da42e6e46b2f7033e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-16 13:13:44.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"89cd0d654b36943f22eb99037a05219bd3f8165e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-16 13:14:07.000000000","message":"Patch Set 10: Review-Priority+2","accounts_in_message":[],"_revision_number":10},{"id":"4163b2984bc72cf24d3c8071fffe9f45247025d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 14:28:51.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccfee9cdb579429c9b576881e33d6431\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/00a610b2e21840279292bc43b24501b4 : SUCCESS in 10m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/5c1aa90e9f5149d1b8b67b7f00f87f14 : SUCCESS in 24m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/5ab3e1437ab24016a833f6bb35bdc690 : FAILURE in 28m 38s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/9976cbc7e9044f77ae0df442e820fb4c : FAILURE in 1h 13m 14s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d1f267eb36b4549be2891c9034c7e69 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62894a9d766444c29581c962b67503c1 : SUCCESS in 1m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c6579c0844bb483ea4220618b827d92f : SUCCESS in 3m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/01eb1884b6f44f85b6eeffa4f77660aa : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ab0348ab69a47069808374fc1c8c447 : SUCCESS in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b62b59c3e4a449da584d41c8009d4ee : SUCCESS in 3m 13s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/0c5e15498187419cb89d2f722cd47679 : SUCCESS in 2m 03s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ab354521f1834123acaf17878e036297 : SUCCESS in 23m 58s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cc406568ce1a457bbfbe0609152607eb : SUCCESS in 30m 08s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/c9c57bc8275f44fe9fa4d57e9b75ce2e : FAILURE in 25m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/828eda69a2634e62a204f333972d344a : FAILURE in 1h 10m 21s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/4e549ccaaa224b92aa92b3abdd0e8f75 : FAILURE in 52m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"c7092d4f62542244bf10a39d998720fdbda31a7a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-16 14:53:35.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e697037719147868db79b9a71ba259d\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a915ca795fc7487b8460b20d07b8f9f4 : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8f5878678ba544698697d917556d3f35 : FAILURE in 1h 13m 01s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fc451439fa704916820a92a10058c5cb : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/67cd04d70c3e4627acc5b8433bb05011 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b19b05b6bc15408eb6de9b463e883cd3 : SUCCESS in 7m 14s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f333f1c785a448a29422ea6669d55b22 : SUCCESS in 1h 22m 19s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"1ee0e46bba88a2fe22950310bca589d1b50024e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-07-17 10:27:19.000000000","message":"Patch Set 11: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"d4cc806c31fe7f4c0a02eb200a83912157bbf32c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 12:18:40.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/18722a5f62a441059b41d069046c7e71\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/8a8b3635b778449da9ceb22c54fa4425 : SUCCESS in 19m 08s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/e4200a0797ab43159844cb99de942dbc : SUCCESS in 31m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/bf7b86c3740740f7bd6bcde7b45a69a6 : FAILURE in 40m 18s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/2a218bb7f84e4c6fbe151067045b4310 : FAILURE in 1h 43m 24s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8d06b42e074e06aaef6805b7ea4a38 : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e5bd5f7abd34772b71a29551aeb5c6e : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c33b1fad666041c3ba885f8a86d11168 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6fed68e6bf2c4f4490139f1740b3a35b : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88356fef89db43e79ec8e71d3a058ec2 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9d410d9c0a347e9b84de04b836c9fc7 : SUCCESS in 5m 47s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5f8ccb4360844d2f942b077f7a68475c : SUCCESS in 4m 26s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/62496983bafb46c69629ce55a96a72d4 : SUCCESS in 22m 49s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/284d0cbc9a2049e8ac46def4c1a4f9f5 : SUCCESS in 29m 56s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e082c755add44820bc96d58559232849 : FAILURE in 38m 28s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/06b6ef4f7b804860805e8ead7b8438cd : FAILURE in 1h 21m 08s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/113cfbf8439847f6a32e439ce2bb0737 : FAILURE in 1h 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"08d4817e5612f1225adb45461b1c21a32fa948d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 13:33:35.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676f5d973a1847d28c281a2283eb7175\n\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d2988db3453492ab0a8ec542754070e : SUCCESS in 59m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e33422f1589a4f49b3a7a148b73c94a9 : FAILURE in 1h 35m 49s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0ab73449442a495fbf73e662488b2ad7 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/53a8aff81b6540bc9240016ff2af214a : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f3c6d2746306412eb4f7e9c5174e90c4 : SUCCESS in 11m 33s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/6579eb183c3249c0930923af20d1cf5d : SUCCESS in 53m 54s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"ed0ed4d6fb7aeebcfe52d5becb8ac57da6ef4774","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 09:01:04.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"3e8f260164473309e024802073778df38badd007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 10:23:05.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02fe59e9ad11421cb8a7932f2ff3226c\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/17014508420743f1879adf25dd72fd6b : FAILURE in 14m 12s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/4eb762b1b41742d7befbdca10f7f1d55 : FAILURE in 21m 35s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/27ffcb4021d24c2fa87381bb164d2750 : FAILURE in 20m 11s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/12dd6a8c0f4b4d9d8d5263903eebce57 : FAILURE in 22m 18s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f7768e66b0ea46578e40f6f59510826f : FAILURE in 37m 22s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/b6210d65228e4351bc9e2ce0fc5341d3 : FAILURE in 1h 18m 08s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cb43fa904de477eaea6d86f110000fb : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ba0e17088d449a694cb9f16abfa7b1a : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c0926dc7b90495aabe7df368a1da11c : SUCCESS in 4m 00s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f8a1d36e48a44e83a3b1766d3ec25c1b : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d758a494fc3649ffba5a95cefc61cdaf : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e82265a4ce1a48489852701b13519b1a : SUCCESS in 3m 22s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d6cbc5a4a1b84ce5a7c21bfb414ac542 : SUCCESS in 3m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/67e11556d64a47c99083996e0dc566d8 : FAILURE in 18m 28s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a752d9fb27a64c588d7afe81ed7d6b1f : FAILURE in 29m 12s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/da71e22860764b0190fba391e7e6f91b : FAILURE in 19m 40s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/51234e9abd754a2fafc35ec705b1922e : FAILURE in 1h 14m 10s\n- kolla-ansible-ubuntu-zun https://zuul.opendev.org/t/openstack/build/b334e4ff7f914383b023c4efe8e047cf : FAILURE in 1h 14m 13s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"e774a39bd2cc4dea8deaf2360242c35c81cd1a3f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 13:53:40.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b05d8290fd4058be11cfdd44bef073\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/b0827df47ba94528b4d182b854a33f20 : FAILURE in 1h 11m 36s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/691d14d1a4984e10b3079fc566ed922e : FAILURE in 1h 42m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/96a883483e8e495cad16bdfab3789b49 : FAILURE in 1h 15m 29s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/acd164ac253a4b71ad0b8be41afd7cdd : FAILURE in 1h 57m 51s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1b6845e5afd847039ba135b8252fad25 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a9be5ff6117f4a0c91e6e15a81732ff6 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8709f021a65a404b980ea7b26446f604 : SUCCESS in 8m 41s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/a83cbb54dd704f2cb8fe463554ea04cc : FAILURE in 1h 18m 03s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"70a7dce10f872287990aff1e91937621a9fb1c70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 13:56:20.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"a004288b00ae2c668898f47fe8512ddc2bad4186","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:02:29.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nThe following files contain Git conflicts:\n* docker/keystone/keystone-base/Dockerfile.j2\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":14},{"id":"e2f0776b9d2fd0e9ba30eb0bca9ba3218ab572c8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:03:15.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nThe following files contain Git conflicts:\n* docker/keystone/keystone-base/Dockerfile.j2\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":15},{"id":"60fef4a4f130a01f323f338bd67a61fd66cc8820","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-06 14:03:30.000000000","message":"Patch Set 16: Published edit on patch set 15.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":16},{"id":"7f41ad95e5f66e9842e8cb3bdeca35845e11f85c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:00:40.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02d956867d314ba684b642903aa68a3e\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/dc48a77fba474e759cfebd5e1b44084f : SUCCESS in 14m 56s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/e429225c96d14a7981f3f5749f8882fd : FAILURE in 51m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/3de83e33c0e04583a7bad36f50f005ff : SUCCESS in 12m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ebe0b60f47384db088ac0ceb77cc2de2 : SUCCESS in 33m 07s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/fd6719692c8b4cd5b9e9aba12286f12b : FAILURE in 42m 21s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/424187a757a44895857bca93649d59b3 : FAILURE in 1h 38m 46s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e9fd6a22874b1bab96e4fbb4de99bf : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/891682efdd1e423da63ae04ffc509511 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dada7720fd96441a963e4667202794e0 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f1de08d5b3bc49e6b08bf3704ada48c6 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8b12e658229412c976282ac23e24ae1 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f38cdf5f7c4d348a535efa0b1cf012 : SUCCESS in 6m 01s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/af7f994fcce74c4398a32951caca3eb4 : SUCCESS in 1m 47s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6cf00c54c9964aca83054df4900d2dcd : SUCCESS in 24m 08s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/a2f6e32b5f474d88a824ea3b9f8591ab : SUCCESS in 26m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/d31c09d052f04eaf83bcf2369a3feeaf : FAILURE in 38m 33s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/f16522ffc7fc46a59a55957f3668ddef : FAILURE in 1h 34m 11s","accounts_in_message":[],"_revision_number":16},{"id":"a9c24a1d5f9635efa3e6d50b9f8eff58dac3e0e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 19:31:05.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4a94faae7d14b50a01de3a18d8b8845\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/772d9709e7c647608e97db0eafc25ccf : SUCCESS in 1h 13m 54s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/242d37c8e8124517a165b67b74d66aae : FAILURE in 1h 33m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8e9698c697aa48d6ae4a27f5f06b1d66 : SUCCESS in 50m 55s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c4317cdad3e8463db1ce595953569f4b : FAILURE in 1h 32m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/68abf0ced1704bd5873bf6bc8804bf57 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/484ff18b61d24d1b86d9507712e8ae5d : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/07130834bdd747d183cf71478a69879e : SUCCESS in 9m 48s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/897a9b5b554d4faaa54d324b0d46ab53 : SUCCESS in 1h 01m 32s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"d329275aef3a7241266a8bfd2fbf1344b4578062","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 14:00:48.000000000","message":"Patch Set 17: Published edit on patch set 16.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"311516c5b65c46c4ffdfeb9bca4d8a7df668600a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 14:01:05.000000000","message":"Patch Set 18: Published edit on patch set 17.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":18},{"id":"c5a37b200176a1e47ae02f41cded08b894c203df","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-07 14:01:10.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":19},{"id":"054a0370a521cb01b5e99780c4036c27200c452a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 15:36:38.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66963717b56a4318a1eac03634e32bd3\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/5124aba0a2804957b5ab55e30b2e24da : SUCCESS in 15m 05s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/8b2757bdb68b475ebfec4ab37376de82 : FAILURE in 22m 22s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/f472adbd15734794ba388a48306edfa4 : SUCCESS in 9m 33s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/0239ffb03e5e4737ba23713a4a389fb0 : SUCCESS in 18m 48s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/549e9fcb66794ca986849a6b937fb56b : FAILURE in 20m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/7167842dbcbb41aa957218c1c6454f37 : FAILURE in 43m 42s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ae642b4ca348b1be0280ced2d7ce4d : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c878a9fe00e42c78309fd9f7834c82e : SUCCESS in 2m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/61b17c358cb04bab8e3dfdd62650f8c9 : SUCCESS in 3m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/09cd1b2b1f914f708e0b5b692fe88c9b : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/980329df5d6f4449aae99ba27561ddcd : SUCCESS in 3m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702162256a14483d813ae115eb3c0931 : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/47da97d803144ef190a8d96a25047840 : SUCCESS in 2m 55s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/6134f7a8ca9c43179d96657a0e01c348 : SUCCESS in 11m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/63e601ef004f4c77a29d94358774289c : SUCCESS in 31m 00s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/e5b34c4d6f9449249aee84c3631180a0 : FAILURE in 45m 19s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3f0af0276a274af5b5b44dbb73dd52c6 : FAILURE in 1h 30m 28s","accounts_in_message":[],"_revision_number":19},{"id":"b174b74f12b52da58e5d599689cfa800231ab153","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 00:41:51.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/708033070b8644c19939890891b2353d\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/bbd655dcc39c464d8424ead25f44943b : SUCCESS in 1h 00m 49s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/82c1183febbd479fa40548c44d570b4c : FAILURE in 1h 42m 44s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/ad8698d3134c4803b0e912d5066451ca : SUCCESS in 1h 01m 14s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/e32420a839fe4017be0b018b4966c086 : FAILURE in 1h 03m 13s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/49c741eb767e4d8dbeb4428e2ed1d16b : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c454e23f98b428291d8f1015f4ecbdc : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1c079394415147f693a331846ea5dd06 : SUCCESS in 9m 28s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/c27b9dc773674c91bea7d72dc071e066 : SUCCESS in 1h 07m 11s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"4329e8c2552976d9b3b2fb889f7a3c637e0ac771","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 06:23:54.000000000","message":"Patch Set 20: Commit message was updated.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"**changekind:NO_CODE_CHANGE** OR **is:ANY**\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"aac8a02ae3797f3775f787e51589bf29afe70e14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 06:52:41.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/956785 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"5a412bd77f2e72da66e19976413082232bb89db2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 06:52:45.000000000","message":"Patch Set 20: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/956785 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"d0269a024128573625f6435dd222f42531731de5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-12 09:14:29.000000000","message":"Patch Set 20: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"82fdf203f2b7cfa5f0c9f1a25b8d82b0081d4855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 10:37:11.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f0776934e294bf598396f6ea99356e6\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/dc1cfaf77f474593bbb2747a5a348b67 : SUCCESS in 14m 33s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/08c00a34f44f44478deb0f9a006da65a : SUCCESS in 1h 14m 34s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/08559a3371b3400b932fa9882ef40e9d : SUCCESS in 19m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/17f22a83d8e647caa7a2d02cf053c63c : SUCCESS in 31m 45s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f3cfd88651de49959fec9f231966dd5d : SUCCESS in 1h 02m 24s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/6e39c5073e8e4814abb753fbfebd73d0 : FAILURE in 33m 44s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29b59cf4d294b5086ed18e6f81761ad : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/381bb4ab68a1480dbc165a5261849f9b : SUCCESS in 3m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50e4831ac6be4dddb2d04b407735564f : SUCCESS in 1m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/97222f4159d54b05b12fa3a89479a8ea : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c4adb047547941368eca2a797d524fcb : SUCCESS in 1m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b6145c5738c4c52b512d38eceb3de48 : SUCCESS in 8m 07s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8914166def4c4377aaaf266eac1d3b69 : SUCCESS in 3m 22s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0db969666f8a4f19aa8578c2d2b09437 : SUCCESS in 28m 13s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/41c0b7973b8745c2b99a402af6a63792 : SUCCESS in 29m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3fb6d242663947d0a943cb4e3e389be6 : SUCCESS in 1h 03m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/5a02a1a875e24f9f8537261844b145ae : FAILURE in 1h 05m 37s","accounts_in_message":[],"_revision_number":20},{"id":"d1bd92a02d78314c4ab5c8d2049145de49c6f207","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-12 11:19:32.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/956785 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"82791b0197b551e9de989b27f52b7b96c0f18b41","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-13 06:53:01.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":20},{"id":"3c2135baee42ba794966f1a4143f70eb7be35ec0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-13 06:53:09.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"53f95f302a80cb3974c261df262fba74c6b1ec97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 08:45:56.000000000","message":"Patch Set 20:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/037b3182cbfe427084f3e04e3a30eca6\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/9573d946e15940178899e877dfff97f8 : SUCCESS in 14m 30s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/7c8ea09ac8df493db917d8aca9dccd3a : SUCCESS in 1h 01m 03s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/308a871661d24278add5e5918275f740 : SUCCESS in 21m 21s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/4f7b695e40904128af71c1abc8ca3459 : SUCCESS in 28m 20s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/55936f23d15b4c4296c2bc16be98519d : SUCCESS in 33m 36s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/0263e30b401f44959f85ba2e8c7a169f : FAILURE in 1h 36m 21s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46e8146a2cec4b84b3b6a7ad80739071 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb80b17aba55409b9c2ad221af5c0b64 : SUCCESS in 3m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0312bc1e8994754a6bc3ea61699ed53 : SUCCESS in 3m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/31b256d2cd8b4f4d9f647b5156aa6c6b : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01215b72b4d948d19210e08b6f93c029 : SUCCESS in 3m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e1fb4e790c04d2eb7083859fb1f21ce : SUCCESS in 6m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/dd58e442ca27411aa01db8d389ddeb21 : SUCCESS in 1m 52s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/ad5f77403781493b84d1be174a495315 : SUCCESS in 21m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/1b7dc17820fd438c9a1a7b28c9302ff2 : SUCCESS in 31m 22s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/713091ed97164e449e2215dfce7ad8cf : FAILURE in 33m 06s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/20428977597143e6be62498775c20d4b : SUCCESS in 1h 41m 44s","accounts_in_message":[],"_revision_number":20},{"id":"05b322641b12fede66f73a3e5d75d293c5bd61c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-13 10:09:04.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69fc79873fea40edaca433be8f31ce4d\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3b0f1fdc1e77485d809f3805d929bd30 : SUCCESS in 1h 04m 44s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/77137685785f4ca18b7bd9ff8619b064 : SUCCESS in 2h 21m 06s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/5ef4f23b04aa4c35baf5bb09b8b8e788 : SUCCESS in 1h 01m 30s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/79089e4329604a518a6ed1e3c639daac : SUCCESS in 2h 07m 54s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e75eea914e834e708c66f3701fe9ce14 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/121d3d0c9c4e4aac942b70635f45b0ce : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1916af24450c413d88e8a0f6482c5ad6 : SUCCESS in 9m 54s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3b0d25af4c894d22a2e557d7871a06bd : SUCCESS in 1h 10m 21s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"6167929d598f957a08dbf403d6243fe0dd000d5b","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-08-13 11:35:12.000000000","message":"Patch Set 20: Code-Review+1","accounts_in_message":[],"_revision_number":20},{"id":"0ca830b6ec17af0ea0da009e150c8f10cfd50d30","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-08-14 05:50:27.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2592837ddfca42064a6ff34c37edfaae793b0a55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 07:49:10.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4911333cb8b49b5bb27810f3cf57666\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/2113ee2dcd72477b949cb59969d16c21 : SUCCESS in 17m 08s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/bba393f0909649adb4f342491e6c53c8 : SUCCESS in 1h 19m 21s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/60d4b179011147fe9702997441415c7f : SUCCESS in 22m 27s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a46f6bf4c8964c19961cf004fa9ad4e3 : SUCCESS in 29m 57s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/11e6000dbe224de88238647f1d18336a : SUCCESS in 56m 54s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e99f550f856e40ceba85aa10bc5f7e3f : FAILURE in 1h 47m 31s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1de439c0c06744f6ad8e73e8934ac002 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e6d4350c4d64c0b820960d601bdac20 : SUCCESS in 6m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4781219854947da847d9cdba576a7f1 : SUCCESS in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b27cbfd824b946abbf44915c336e6761 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b427df07f7984920a53c10605519810a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69a6d9bf568848e8b3318eb7cbb5a898 : SUCCESS in 8m 09s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/b73f9a38cf8e42f8b01f5de08883b7dd : SUCCESS in 4m 06s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/8d711bc219e04a86a18bcb6e8bf5bace : SUCCESS in 22m 48s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/44526c880da3458080fd5c50b5b44406 : SUCCESS in 30m 07s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/727dd6f0ff5a456392178afec44d25db : SUCCESS in 1h 09m 13s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4e91dbd8a6ba4deabf532f068e09f87c : SUCCESS in 1h 43m 24s","accounts_in_message":[],"_revision_number":20},{"id":"e886fa059623879c70ca9f23b97f0947d74daf21","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-14 08:01:06.000000000","message":"Patch Set 20: Backport-Candidate-2 Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":20},{"id":"58bdd19e46067d6e6bebcb1a23ae1e5668032cb0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-14 08:19:17.000000000","message":"Patch Set 20:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/956785 is needed.","accounts_in_message":[],"_revision_number":20},{"id":"6c90686b5ffd30776dbe7fd17fd2562db5a0832c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-08-15 10:51:02.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"33c5803c4b0c420725e7d34211af827f2d9a4daa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 12:46:14.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e0d9bef529d44aab738528888cdf064\n\n- kolla-build-centos10s https://zuul.opendev.org/t/openstack/build/4846963ce933420cbe4b549b3b3b047a : SUCCESS in 13m 04s (non-voting)\n- kolla-ansible-centos10s https://zuul.opendev.org/t/openstack/build/f69624b191b44fd4976f565a9165449f : SUCCESS in 1h 16m 56s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/46fcfa97f21544c58c7a901e080d17d7 : SUCCESS in 12m 03s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ccd9071d75d74d83b85ce5c5aa92352e : SUCCESS in 17m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/f8b24e912f0c4126966b888e20929ca4 : SUCCESS in 1h 12m 04s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/bf24accbc9f44213a2fbf8b458674249 : FAILURE in 1h 51m 02s (non-voting)\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39269b9ad11b4170bfadc6c728d7ece4 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/462f09fb5e5f4f91b9a2d0c4ed81c938 : SUCCESS in 2m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9ebd2adaaff84aaf9ce4b0feafc49ccd : SUCCESS in 2m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f49b3d9e3f094b00a71697a17fa87719 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d3983fc00d34d98860651cc64b88d61 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5edc34ff75c743c79d7680f99af2ce2d : SUCCESS in 4m 51s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d0c78c0cf2c249f0b3d33ae1c54c7b35 : SUCCESS in 2m 44s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/eb09dbe41be14c15ae509e079819a64f : SUCCESS in 19m 29s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/8fe3613fb32a4e2fb7c09b18117de649 : SUCCESS in 16m 16s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8c4bc993bedf4a43a5d428de077946cb : SUCCESS in 1h 19m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a2ef436fd4394a82b562e37ee05b2bcf : SUCCESS in 53m 48s","accounts_in_message":[],"_revision_number":20},{"id":"4c7ff25e2dd2808d0ac337b4deb5bd9df91163e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 16:13:57.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e457d65e580243e19bdedffcef04d47b\n\n- kolla-build-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/f13532e311a544cca67c08e2d8af2798 : SUCCESS in 1h 04m 41s (non-voting)\n- kolla-ansible-centos10s-aarch64 https://zuul.opendev.org/t/openstack/build/3c1d1cac2ee8417498e4c2d0a83acb0c : TIMED_OUT in 2h 32m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/edeecb7f34284fd9ba01d31affea6dc7 : SUCCESS in 1h 37m 01s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/45db959ba6d94b608774267145ca4af4 : SUCCESS in 2h 01m 53s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9d3cea0aa58d4b71aaf8c762b0779cfe : SUCCESS in 6m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/19ae3c521528450b82165507ea606114 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f0119b4298e348948003a530be532fba : SUCCESS in 9m 47s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3477404d6de94c99b1ea4f9b01446c2d : SUCCESS in 1h 48m 38s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"eb8b7aec12e065a7f4eb9c0b716160207225547a","author":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"date":"2025-08-18 12:43:43.000000000","message":"Patch Set 20: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":20},{"id":"aceb0918758ac45010ae83fbd744dbf81169a61f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 12:43:54.000000000","message":"Patch Set 20: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":20},{"id":"c3ca1a12cfeb20598844275ad699e26ad5330ac0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 14:16:29.000000000","message":"Patch Set 20: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e4360164682402b85bb3eb45689bd9a\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/c5ae85837394491a8fb4c21ef198db68 : SUCCESS in 20m 30s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/9f82c150f8d44bbf922f7a4221859bef : SUCCESS in 31m 32s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9479dc6741a94edca495445d02121d84 : SUCCESS in 37m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd416c11ea9a4f9883140d7e7acc6bad : SUCCESS in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/891ff0b9053248858bed6b12ce72e916 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/75d31f5ea3054292be4aee4a8182e1a0 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a977b1e899b491c9066413e4cc9a41e : SUCCESS in 8m 08s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/36a8ceb785354ac48be7ea7fd5cd2f84 : SUCCESS in 4m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/061ffa9e8da2423e91fbab361b6345de : SUCCESS in 13m 55s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6b971dcd93e54a6c9c33d2bf710c1e97 : SUCCESS in 17m 25s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/eacf8b7684234fd4b57d82697ec936e4 : SUCCESS in 41m 43s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0182791f770f4b24af9f69d3247e6fc4 : SUCCESS in 56m 51s","accounts_in_message":[],"_revision_number":20},{"id":"269df02affd108f6d1539235e7890ae7754445e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 14:16:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":20},{"id":"63a012b749428618e82352a460bf18f682088849","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-18 14:17:32.000000000","message":"Patch Set 20:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7369bb1717474ba4a78bb4e20df4b0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d66b44846b478a85eb33a5673a0c09 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":20}],"current_revision_number":20,"current_revision":"87678d014f40e492edcc1eafe8a12fe68483a219","revisions":{"dc1afd0f94a1e8ebd6a815b4e7ed4eeb2685b992":{"kind":"REWORK","_number":1,"created":"2025-05-14 14:07:50.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/1"}}},"commit":{"parents":[{"commit":"0733a13d7be896dac4004212f9eef48a845b26c2","subject":"Switch to 2025.1 sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0733a13d7be896dac4004212f9eef48a845b26c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-05-14 14:07:50.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc1afd0f94a1e8ebd6a815b4e7ed4eeb2685b992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dc1afd0f94a1e8ebd6a815b4e7ed4eeb2685b992"}]},"branch":"refs/heads/master"},"0d791f7695de4861b6625514b70db12166ce5123":{"kind":"REWORK","_number":2,"created":"2025-05-14 14:35:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/2"}}},"commit":{"parents":[{"commit":"0733a13d7be896dac4004212f9eef48a845b26c2","subject":"Switch to 2025.1 sources","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0733a13d7be896dac4004212f9eef48a845b26c2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:35:44.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d791f7695de4861b6625514b70db12166ce5123"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d791f7695de4861b6625514b70db12166ce5123"}]},"branch":"refs/heads/master"},"a179f4827524f016eec401eb1704d3908cc0529c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-22 15:57:07.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/3"}}},"commit":{"parents":[{"commit":"ff373463c431635f4264490cf3a1442e3d9b38f2","subject":"CI: Drop centos9/rocky9 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff373463c431635f4264490cf3a1442e3d9b38f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-22 15:57:07.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a179f4827524f016eec401eb1704d3908cc0529c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a179f4827524f016eec401eb1704d3908cc0529c"}]},"branch":"refs/heads/master","description":"Rebase"},"dec20dca586d2020b17c199f1ef57fb9ad016b1c":{"kind":"REWORK","_number":4,"created":"2025-05-22 19:29:22.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/4"}}},"commit":{"parents":[{"commit":"ff373463c431635f4264490cf3a1442e3d9b38f2","subject":"CI: Drop centos9/rocky9 testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ff373463c431635f4264490cf3a1442e3d9b38f2"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-22 19:29:22.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dec20dca586d2020b17c199f1ef57fb9ad016b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/dec20dca586d2020b17c199f1ef57fb9ad016b1c"}]},"branch":"refs/heads/master"},"4657363463179e942d9f3b607329a75bd6976ecb":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-06-05 20:45:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/5"}}},"commit":{"parents":[{"commit":"b643e8f9fd3e9d8e881e112e53183db743f1e5d3","subject":"Fix ProxySQL rule sanity check to support username-based routing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b643e8f9fd3e9d8e881e112e53183db743f1e5d3"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-05 20:45:52.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4657363463179e942d9f3b607329a75bd6976ecb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/4657363463179e942d9f3b607329a75bd6976ecb"}]},"branch":"refs/heads/master","description":"Rebase"},"42a539fd7830953bcc42abc18e096d73f4ebad3d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-06-09 08:18:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/6"}}},"commit":{"parents":[{"commit":"a90a8893c8df1eed8d8409c70bd8177392290554","subject":"Merge \"Revert \"[release] Use RDO\u0027s Epoxy packages\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a90a8893c8df1eed8d8409c70bd8177392290554"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-09 08:18:56.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a539fd7830953bcc42abc18e096d73f4ebad3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/42a539fd7830953bcc42abc18e096d73f4ebad3d"}]},"branch":"refs/heads/master","description":"Rebase"},"665f255ade7acf5e75f39c6a986fefeda4f9faf6":{"kind":"REWORK","_number":7,"created":"2025-06-10 09:33:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/7"}}},"commit":{"parents":[{"commit":"a90a8893c8df1eed8d8409c70bd8177392290554","subject":"Merge \"Revert \"[release] Use RDO\u0027s Epoxy packages\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a90a8893c8df1eed8d8409c70bd8177392290554"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-10 09:33:16.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/665f255ade7acf5e75f39c6a986fefeda4f9faf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/665f255ade7acf5e75f39c6a986fefeda4f9faf6"}]},"branch":"refs/heads/master"},"657bd4055300e42dae6c5de6f5dcd9e2f857dd8d":{"kind":"REWORK","_number":8,"created":"2025-06-10 14:46:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/8"}}},"commit":{"parents":[{"commit":"a90a8893c8df1eed8d8409c70bd8177392290554","subject":"Merge \"Revert \"[release] Use RDO\u0027s Epoxy packages\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a90a8893c8df1eed8d8409c70bd8177392290554"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-06-10 14:46:44.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657bd4055300e42dae6c5de6f5dcd9e2f857dd8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/657bd4055300e42dae6c5de6f5dcd9e2f857dd8d"}]},"branch":"refs/heads/master"},"6ce11605027f0eafa0c1973c64cf0d7a2cc52e44":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2025-07-16 13:13:31.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/9"}}},"commit":{"parents":[{"commit":"a90a8893c8df1eed8d8409c70bd8177392290554","subject":"Merge \"Revert \"[release] Use RDO\u0027s Epoxy packages\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a90a8893c8df1eed8d8409c70bd8177392290554"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-16 13:13:30.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ce11605027f0eafa0c1973c64cf0d7a2cc52e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/6ce11605027f0eafa0c1973c64cf0d7a2cc52e44"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b166f6da7dc3002dc2f0f32960aacd2c86c22be9":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-07-16 13:13:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/10"}}},"commit":{"parents":[{"commit":"ba2b4bb56415b9bd7b690dcf077bb59cf4083f05","subject":"CI: Drop publishing to DockerHub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba2b4bb56415b9bd7b690dcf077bb59cf4083f05"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-16 13:13:44.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b166f6da7dc3002dc2f0f32960aacd2c86c22be9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b166f6da7dc3002dc2f0f32960aacd2c86c22be9"}]},"branch":"refs/heads/master","description":"Rebase"},"8cd52bf3455fcbe1bdd6c3dc4694f88967e7346f":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2025-07-17 10:27:19.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/11"}}},"commit":{"parents":[{"commit":"ba2b4bb56415b9bd7b690dcf077bb59cf4083f05","subject":"CI: Drop publishing to DockerHub","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ba2b4bb56415b9bd7b690dcf077bb59cf4083f05"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-07-17 10:27:19.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cd52bf3455fcbe1bdd6c3dc4694f88967e7346f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/8cd52bf3455fcbe1bdd6c3dc4694f88967e7346f"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2d32a42292163b2fa71e81556796ddc0f0903ee6":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-08-06 09:01:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/12"}}},"commit":{"parents":[{"commit":"ec25faade1581f8bcc62f8b7ae22a9f8b8d64f5b","subject":"send empty X-Registry-Auth for anonymous pushes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ec25faade1581f8bcc62f8b7ae22a9f8b8d64f5b"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 09:01:04.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d32a42292163b2fa71e81556796ddc0f0903ee6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/2d32a42292163b2fa71e81556796ddc0f0903ee6"}]},"branch":"refs/heads/master","description":"Rebase"},"b8bcf69a3b55fc649497d6e086bcd62d52378d8d":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-08-06 13:56:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/13"}}},"commit":{"parents":[{"commit":"63c4202d207b6b4686156bca81c84ca756513f3c","subject":"ironic: Stop copying rootwrap since Ironic dropped it","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/63c4202d207b6b4686156bca81c84ca756513f3c"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 13:56:20.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8bcf69a3b55fc649497d6e086bcd62d52378d8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b8bcf69a3b55fc649497d6e086bcd62d52378d8d"}]},"branch":"refs/heads/master","description":"Rebase"},"097e27ae28f76682a456f3783dae5f69d6655f6b":{"kind":"REWORK","_number":14,"created":"2025-08-06 14:02:29.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/14"}}},"commit":{"parents":[{"commit":"c5430ca576216d099ff39ce99f99534c69804bdc","subject":"manila: Drop copying of wsgi files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/c5430ca576216d099ff39ce99f99534c69804bdc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:02:29.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/097e27ae28f76682a456f3783dae5f69d6655f6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/097e27ae28f76682a456f3783dae5f69d6655f6b"}]},"branch":"refs/heads/master","description":"Rebase"},"791216dfd8ce78195f427b67cb3fe10613288479":{"kind":"REWORK","_number":15,"created":"2025-08-06 14:03:15.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/15"}}},"commit":{"parents":[{"commit":"489e0979af5e0c4c482da9dc844a1b03aa34c1b4","subject":"manila: Drop copying of wsgi files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/489e0979af5e0c4c482da9dc844a1b03aa34c1b4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:03:15.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/791216dfd8ce78195f427b67cb3fe10613288479"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/791216dfd8ce78195f427b67cb3fe10613288479"}]},"branch":"refs/heads/master","description":"Rebase"},"14cc53c83ad0cce5b4c3957f649cff73269e3db7":{"kind":"REWORK","_number":16,"created":"2025-08-06 14:03:30.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/16"}}},"commit":{"parents":[{"commit":"489e0979af5e0c4c482da9dc844a1b03aa34c1b4","subject":"manila: Drop copying of wsgi files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/489e0979af5e0c4c482da9dc844a1b03aa34c1b4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-06 14:03:27.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14cc53c83ad0cce5b4c3957f649cff73269e3db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/14cc53c83ad0cce5b4c3957f649cff73269e3db7"}]},"branch":"refs/heads/master"},"e886e37fc5d6cdf7ac6631d9eb3ebcd21d28c549":{"kind":"REWORK","_number":17,"created":"2025-08-07 14:00:48.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/17"}}},"commit":{"parents":[{"commit":"489e0979af5e0c4c482da9dc844a1b03aa34c1b4","subject":"manila: Drop copying of wsgi files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/489e0979af5e0c4c482da9dc844a1b03aa34c1b4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 14:00:48.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e886e37fc5d6cdf7ac6631d9eb3ebcd21d28c549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/e886e37fc5d6cdf7ac6631d9eb3ebcd21d28c549"}]},"branch":"refs/heads/master"},"a2f21427ac7c2b05fc028a0448011fe669489cc9":{"kind":"REWORK","_number":18,"created":"2025-08-07 14:01:05.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/18"}}},"commit":{"parents":[{"commit":"489e0979af5e0c4c482da9dc844a1b03aa34c1b4","subject":"manila: Drop copying of wsgi files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/489e0979af5e0c4c482da9dc844a1b03aa34c1b4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 14:01:02.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2f21427ac7c2b05fc028a0448011fe669489cc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/a2f21427ac7c2b05fc028a0448011fe669489cc9"}]},"branch":"refs/heads/master"},"0d3e443c09ed8e1c404fbd7424955a4b2637428f":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2025-08-07 14:01:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/19"}}},"commit":{"parents":[{"commit":"998f34be85ecf09f688346ac2ea711d1c7bb2340","subject":"Merge \"manila: Drop copying of wsgi files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/998f34be85ecf09f688346ac2ea711d1c7bb2340"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-07 14:01:10.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/938395\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d3e443c09ed8e1c404fbd7424955a4b2637428f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/0d3e443c09ed8e1c404fbd7424955a4b2637428f"}]},"branch":"refs/heads/master","description":"Rebase"},"87678d014f40e492edcc1eafe8a12fe68483a219":{"kind":"NO_CODE_CHANGE","_number":20,"created":"2025-08-12 06:23:54.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/55/949755/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/55/949755/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/55/949755/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/55/949755/20"}}},"commit":{"parents":[{"commit":"998f34be85ecf09f688346ac2ea711d1c7bb2340","subject":"Merge \"manila: Drop copying of wsgi files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/998f34be85ecf09f688346ac2ea711d1c7bb2340"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-14 14:07:50.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-08-12 06:23:54.000000000","tz":0},"subject":"Revert \"Switch to 2025.1 sources\"","message":"Revert \"Switch to 2025.1 sources\"\n\nThis reverts commit 0733a13d7be896dac4004212f9eef48a845b26c2.\n\nReason for revert: switch back to master\n\nDepends-On: https://review.opendev.org/c/openstack/kolla-ansible/+/956785\n\nChange-Id: If888500a791c66a592f17aa2be1f3784ee009a91\nSigned-off-by: Michal Nasiadka \u003cmnasiadka@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87678d014f40e492edcc1eafe8a12fe68483a219"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/87678d014f40e492edcc1eafe8a12fe68483a219"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-2"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
