)]}'
{"id":"openstack%2Fkolla~949840","triplet_id":"openstack%2Fkolla~stable%2F2024.2~I9579b26cc8131bb0769889470657275de0c99c4b","project":"openstack/kolla","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-15 05:14:55.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-18 17:03:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9579b26cc8131bb0769889470657275de0c99c4b","subject":"Revert \"mariadb: pin to 10.11.11\"","status":"MERGED","created":"2025-05-15 05:08:10.000000000","updated":"2025-05-18 17:04:44.000000000","submitted":"2025-05-18 17:03:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"revert_of":949062,"submission_id":"949840","meta_rev_id":"965fa485eb44872ab92f462badb8989da7d5d6f6","_number":949840,"virtual_id_number":949840,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 17:03:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 05:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-15 09:57:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-15 10:34:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-15 09:57:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 05:14:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 05:08:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-15 05:08:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 10:34:42.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"5a82a9f0315800e045918819505f0ecdb434a57c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 05:08:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ab943abcf66fceaff09f828d9557c3375e20a02","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 05:14:55.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"c3b84f9e240b7a7c3471e7ca402e8aea337dd715","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 06:20:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2748015adf65478db714b0a79274f7fd\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/aa1ffed94f1e415591377c1c06d46fa0 : SUCCESS in 38m 37s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/c7dbf5fdcc034c2e9c94490f99a0496d : SUCCESS in 33m 47s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/6047dec9b53d4813ae268d9a22c6d474 : SUCCESS in 1h 10m 15s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d6830788aaf941f99c34bc181a332655 : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2d8fc2a6eef499cade652a7ab35bb0a : SUCCESS in 3m 45s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/8de9a99478974aa2808034bce95da738 : SUCCESS in 31m 29s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/f6c4d79160b047479dcaefe16320c687 : SUCCESS in 48m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"00c0eba901aac8457475899c93dbf151f4a5e04d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 07:22:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5187ebb9e3e41299589ddb2e53f67e5\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/ebfd5cbfbd9a46fd96b60acaf4e2cf57 : SUCCESS in 11m 06s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/383075ec2a984d4b9440ff79ce576e42 : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a52003bc3a14e859ecbfdea9e62970d : SUCCESS in 17m 53s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/c685c78a8fcc4f478c23cda89bfd610f : SUCCESS in 31m 44s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e952a72320f14dc5a421c8e4d84d2a72 : SUCCESS in 1h 00m 09s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/f9defdf2f1a8426f9a7e3595fd33a1f0 : SUCCESS in 1h 34m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85130b1c1a8f470f95308fd274f1c2e6 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4741d93a9e214f3491db7815f0550808 : SUCCESS in 1m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a35b4996b2034bb69710bf2ecb5ec145 : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9832c1606e244ed8bb575c86d5da24c5 : SUCCESS in 1m 48s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3ab2a1d75f54207b3f9e5e5d29af619 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17a9a192d94446108689cb23db6ccf5a : SUCCESS in 4m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/00d79473e1f741b3aa51debc7cdcc61e : SUCCESS in 3m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/3fa9b560c22a4962be385ec1096826d9 : SUCCESS in 10m 05s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/8c7ef9f3120f4c40ad695887dbc4a6f6 : SUCCESS in 29m 39s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/45b3c15a73ab4b8ab7bca9307ca8aa7a : SUCCESS in 58m 33s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2185381554f04f3fb3b2f3f7d3daeaae : SUCCESS in 46m 06s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/eadba62899be4680aca3e4640cae899d : SUCCESS in 1h 04m 19s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/507f5d242ba04e0ea7ae0a0d9a48fd37 : FAILURE in 1h 44m 25s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/bb2e7202346844b9b82f0c3dcff31ec0 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/9747163e63c4445aa74039dea42079a5 : SUCCESS in 1h 17m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/4b1cc201a7ec4d44a2320b225c4d9e57 : SUCCESS in 24m 44s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/010b3d9de0f646939e265b9d69280e4e : SUCCESS in 52m 17s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/3ed2664027ba435d9db59450a34bc36a : SUCCESS in 1h 04m 11s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/39198f961da94adeb73f0bb397a9e5f7 : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/0b6c8ccc05284e2dad876773f30b8ea7 : SUCCESS in 28m 47s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/afa2f56279e743328b609d25b5a99c4d : SUCCESS in 19m 11s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/d787ffac0ab14184aad793ff7d957810 : SUCCESS in 28m 23s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/3fd13d9754994c3587ff9ce23fdb4965 : SUCCESS in 55m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/963430e1c3874432af6e8d6d39f84965 : SUCCESS in 1h 27m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/07f26b14a76345088a60493dd04a069c : SUCCESS in 1h 39m 36s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/beb8abee0a9d4fa0addfd7603ac9ecb8 : SUCCESS in 1h 19m 59s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/d08923683a204b3d86c0b29570e25867 : FAILURE in 22m 50s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/03e485b4d9b54c9fbc39af8e9245aac9 : SUCCESS in 1h 17m 01s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/ee36a8721d994713a44b5bce73e7e83b : SUCCESS in 22m 48s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/793ee847f9ad44139f9807784536261a : SUCCESS in 56m 48s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/677a1b78b8014627bfcdc27adb3a0c83 : SUCCESS in 1h 09m 07s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/b0b250f1b286448b8eb56fcf4641184a : SUCCESS in 37m 59s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/83157257befe478994bd40b8d31392f7 : SUCCESS in 27m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"715767a78346101ace6665d408cd0368c2343f9d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 09:57:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e6fe2568cbec44a04bfc4a2883d087f5c15aefad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 09:57:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"549fcabc93649a7551951be4f40ca5230e6c01f2","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-15 10:34:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"01fd53ffb00ba2aa95bd3361ac47858e320226a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:16:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b68c75def92845b68d217ee2cb86f129\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d14ffc6acc6d4ad19582dedb56a3b2dc : FAILURE in 15m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/d55d3b0cd5b04a27af6cbf7497edbc85 : FAILURE in 24m 22s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/784ed5db7a7b45e491d8c5a1aa084f02 : FAILURE in 14m 20s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/332265b1e54643938da9e5291b43caf4 : FAILURE in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d3c252d78e446d4a8d0f947c5c96a75 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb976e7610940c7813eb5ebff7153aa : SUCCESS in 3m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7b8c29a21f6045868e99630afb98b856 : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1cfe8c081e4a6297fe62421a22be71 : SUCCESS in 4m 16s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/95534b73e6464c75bb40889fdee3313c : SUCCESS in 1m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/91ec828104ee4763b0c424ad5d937a26 : SUCCESS in 19m 13s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/bd96c3a2b31e4eaba219655329103ff6 : SUCCESS in 23m 55s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/8bac14e822d348ac835f6ee62b14ad36 : SUCCESS in 1h 02m 10s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/cb1be83a995f432cad1c051bcb5697e4 : SUCCESS in 48m 27s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/f54ef4ca6f8543859642055236f008ed : FAILURE in 10m 51s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/cbc6edb3635749d48d54042aac4e50c2 : FAILURE in 24m 55s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/20f3c4a432364b9aa5a8e78f14c824e0 : FAILURE in 19m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3c697c39491c479abe9d6058926bc58e : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"68e264d53e38aac48699918b63d0af943c02434a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 11:20:01.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6047d244b8a093ca4d1a182b3671316c636e8a91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 12:00:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06c47f9730814ba4a2cf3f1059431e55\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/a70cb107246041ad9dfdfc1d62b20029 : FAILURE in 38m 28s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/74fc50dd6fc746548478d1672472ee2d : FAILURE in 37m 52s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f8b93cb274649df870eb5f53c0763f5 : FAILURE in 27m 13s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/053b15d89a044d9bac4eb471539fe680 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5a642184b42846379c91512304fc99d2 : SUCCESS in 4m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/70aa763089794d6bb1c48772405d80e3 : FAILURE in 26m 44s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/eab2298d9093466f94dc667c9358725c : FAILURE in 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"01a31f57c9b95c96e81b477d67554b2dc20c5921","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 14:07:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50cef89403134c47983ad054ed39b203\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/3af0f21349f645ffbec2d5166b7b0dd7 : FAILURE in 19m 39s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/d0676dbc9a464b0da47ef99894e12197 : FAILURE in 20m 05s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/226724c3325f4c4da060df3bdb1fcdf6 : FAILURE in 19m 35s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/af0b33b984d44a398e7e3ab450da3607 : FAILURE in 30m 28s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/adc9320bd4d84531873aecd383e6ed76 : FAILURE in 17m 13s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/58a1f3d39c524ad49ef38f60f7f81d17 : FAILURE in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4290bd022b4e42f0b87367e3d03cd898 : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edec7a8f14f847f58ced82c17de2af3b : SUCCESS in 3m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/26a188c868ec4bbaba6713d52444ccdd : SUCCESS in 2m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a466889d7934ebd80b77d5389210ae2 : SUCCESS in 3m 01s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9960fe33590d46e0afa6855007edf34c : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f9dd28d7fd14497bed6406cdb7b532e : SUCCESS in 3m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3505f4c66be44f23aabe589520ac960e : SUCCESS in 2m 34s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/f06868c8550440b0bfa1f0cc9af3d1e3 : SUCCESS in 18m 00s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1fce946f916d431093c49cb2ad018421 : SUCCESS in 26m 14s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/0cf8b29b78a6425ca182f2396af87e09 : FAILURE in 16m 39s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8631ecaf689c4e94bd0258e814620caf : SUCCESS in 1h 26m 55s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8644fde9f6454c0abff0a34153f7af9d : FAILURE in 17m 50s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7323d569979c4270a0c78eb8e799fc93 : FAILURE in 2h 39m 54s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/676eaf8d81f74fff91ddfe9cfe8cb51e : SUCCESS in 38m 18s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/ac220e1d3d7241e88035a282324888fd : SUCCESS in 1h 24m 30s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/59a4391d170c41fc84b860cc0e541958 : SUCCESS in 24m 58s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/438c599213cd4337943f5d2490df534c : SUCCESS in 59m 38s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/cba318e4cee74b4394d8c00f0711798a : FAILURE in 17m 13s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/8d3b412d3cc54c2597d9854364984dc3 : FAILURE in 19m 28s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/37e6cbbbd97548519ade909d6f3ef80b : SUCCESS in 19m 03s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/396d36572f2d4a3cb323873f5c771a59 : FAILURE in 20m 00s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/dee1b51825534ae3875dfa32d47dfa23 : FAILURE in 22m 28s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fbc61f1b2c7a4854aa2f81767853854b : FAILURE in 22m 29s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/a2b6517e69c14b0dbb2be7255cc65c91 : FAILURE in 18m 15s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/35b1f3b315b049aa9ecb5566befe9f9a : FAILURE in 24m 26s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/701d39731da249a08915bdf423b1794a : FAILURE in 19m 14s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/5d9de5e7c27c4fb9bf5c5fa6fca3955a : FAILURE in 19m 33s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/0de124a2ade047af8851a972a1ecdb1f : FAILURE in 19m 48s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/2d0c3decaf8b4bc58c683b4644da060e : FAILURE in 12m 55s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/7bf4829b3a4540b38648df338b429798 : FAILURE in 21m 38s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/8f9c43279b394e06a94a8fa0fcb31fce : FAILURE in 22m 20s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/74e64b27559147aeb6109737c3ad440c : FAILURE in 21m 10s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/dd2d4f182ba447c38acf38dcce15095a : FAILURE in 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b024f0c2adbb394782ee81e5ecfcf9d3e9358f74","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 13:17:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bb01c5eb32c856fb273b2b82b3db9a9af8b041d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 14:50:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1cbcdf82d13049968a781e3d074841ca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/c1be8c16de6d440289ca0a6d0e682e73 : SUCCESS in 35m 14s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a2dcc7ebcb814570ac1835922464e638 : SUCCESS in 33m 12s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/2b052144cd224837be429bf27dfd2682 : SUCCESS in 1h 31m 21s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/069f6dcdab334181b96abbc958390ec9 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/5d6535c9b7b34195b51b604df5481ff1 : SUCCESS in 3m 18s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/a3a05d9e87804ca9b2b5577d3d5dba2f : SUCCESS in 35m 36s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/49bae09d81fc4fd4be5d3607d3ce26cc : SUCCESS in 35m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"076653ecff567aedf8eec1c91ec5f908d84743f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 15:34:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/978de1105018444dba686242c1f8d67e\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/dff458cf70624b369b15af4bbb5299b8 : SUCCESS in 11m 32s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/5dacb6e1d77e4b259a657888f1f6c141 : SUCCESS in 31m 23s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/d42c4d257b78442d94af173c29e9b919 : SUCCESS in 8m 52s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/a49f5606280d477a9f4d99ef6c700d6e : SUCCESS in 14m 13s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/442e081c8ca94982a57bd503ff1649f3 : SUCCESS in 30m 14s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/e81814bdfc2240318e62d5cd33d1f4b7 : SUCCESS in 53m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abbd8b576311471493f770409e423353 : SUCCESS in 2m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a30622764b5a4f9bb0aae24fe9d45120 : SUCCESS in 2m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91e9006d20364e2c8743b29922054fc9 : SUCCESS in 3m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45a6d4162cac413cacf8aa0e5dbf8d95 : SUCCESS in 2m 34s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2938ed564ac427fae18d359156e6899 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c49188da886c45f798d755a7d820c986 : SUCCESS in 4m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/d3ac7210cec041e2adaf0b8aedb83d5e : SUCCESS in 2m 14s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8dbbcf25ad1e42619f4ebaa83e253532 : SUCCESS in 18m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/7cfe19d337a34ad4ade16dcf1dd93ebc : SUCCESS in 22m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/2437abe424964aa7af72626d04620fad : SUCCESS in 1h 00m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/70ffbf1f7ea14ed8b53aa1a6a733c6df : SUCCESS in 1h 12m 57s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5222242f3e754d0aa39f7bcefecd0e41 : FAILURE in 1h 14m 10s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1761a60e25404e5aa9853d7969536b29 : FAILURE in 2h 14m 01s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a128c362819747ee88a0ed02771e23f9 : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-ansible-rocky9-swift https://zuul.opendev.org/t/openstack/build/258c2638c6af41ccad2d2aa50e909f5b : SUCCESS in 1h 09m 11s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/8deb5b0f94f94de690eecc47237a37d0 : SUCCESS in 14m 02s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/c923249c04214dea9b02a57698978d8a : SUCCESS in 58m 58s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/9d2b9cdf34f8415ebdd5a3d7a7c81aca : SUCCESS in 36m 05s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f58c53baf67e48aa8b36e2c1e2222ae4 : SUCCESS in 44m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/79c2163fdf4e4a73bf2ad9feb3cd3462 : SUCCESS in 15m 17s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/40cb96a3a21b43c2a69446a8e62cf163 : SUCCESS in 21m 22s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/172587abd0944b08b6902a4ed470f0fc : SUCCESS in 26m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/dd5aa198ea434b4dacdf48b60259944c : SUCCESS in 29m 36s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/307a71f908b844d9bf77c684203f4bd5 : SUCCESS in 49m 58s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/34653e41680e40cda05957571bb88bbb : SUCCESS in 1h 46m 04s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/441d428d9d994f0da9e1096d3738966a : SUCCESS in 2h 16m 55s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/90fef27813b24e169d057a8071e18952 : SUCCESS in 1h 16m 04s (non-voting)\n- kolla-ansible-ubuntu-swift https://zuul.opendev.org/t/openstack/build/60e32599d5c548deaac5f98c9a8d1b07 : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/c6e6e64e005640998f0e49b4e9815ec4 : SUCCESS in 14m 24s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6bde229d2f404cb3b15b0ee56930fc7c : SUCCESS in 48m 08s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/d2825c6ac279438fbbeab5adfb101385 : SUCCESS in 1h 09m 15s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/33fed9495fcd4ecabeb27fb3c949115a : SUCCESS in 1h 15m 06s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/36bf3b74650d48c59c89ae575561c34f : SUCCESS in 23m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6479db489d08ef79e71e9fed7c67bc16bfae9716","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 15:34:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4d577df1902acdb9bd8fc7e8aeb4d8e33c5f597f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 17:03:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47eeed54a236403bbcb55d15da7711c7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/e066677934bd452c952fe028b8e948d2 : SUCCESS in 17m 10s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/59850a8ffc12469797babafa8e41a324 : SUCCESS in 20m 09s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/ca751ee1308a4aa4b94886ff396164b9 : SUCCESS in 47m 52s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/8fee4e6c4c5a4896b2b1d76f505856f0 : SUCCESS in 1h 08m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b842c500ea524830b81fb52021894ede : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b924a4afe9a4f8fa5e3a77bdf159829 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de8eb2fed8274f44916fb062ced753a1 : SUCCESS in 2m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d59077601fe46ed9c153b0df66609e5 : SUCCESS in 4m 28s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/8bd54e0cd2384c6e9abb2b7871113246 : SUCCESS in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/19d7b9aa7de340d58f6a7f853219a94e : SUCCESS in 16m 11s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6248e6695da84f0a8195ce9c77294c56 : SUCCESS in 18m 28s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/98a70075c7b046a99cf01366b1275915 : SUCCESS in 47m 00s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/f63d52ab933549d79f6ec023f1bca6d0 : SUCCESS in 1h 06m 45s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/0a3584ebaf51452d95a20400bb8cb04f : SUCCESS in 16m 56s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/6802be23d8f049178cd3999ab9c9314f : SUCCESS in 31m 24s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/1abb449cb73f43439b3ddce9130e92c2 : SUCCESS in 59m 09s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/0013a540447149d0a05dff9268cb6b50 : SUCCESS in 1h 01m 43s","accounts_in_message":[],"_revision_number":1},{"id":"d6da92bb0b4c67e16390cc9eb7b8f9912a53baac","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 17:03:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"965fa485eb44872ab92f462badb8989da7d5d6f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 17:04:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad5a9f2f7f564213ae1f4d7f72890c57\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f373e8fdadb499c8100982d89c95a47 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"895b6c4ce80714d925f40163a54d6eb556097625","revisions":{"895b6c4ce80714d925f40163a54d6eb556097625":{"kind":"REWORK","_number":1,"created":"2025-05-15 05:08:10.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/40/949840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/40/949840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/949840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/949840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/40/949840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/40/949840/1"}}},"commit":{"parents":[{"commit":"3d69a6369e4cab1b4cc46a3f8d7e4e22c09bbb52","subject":"mariadb: pin to 10.11.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/3d69a6369e4cab1b4cc46a3f8d7e4e22c09bbb52"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-15 05:08:10.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-05-15 05:08:10.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 3d69a6369e4cab1b4cc46a3f8d7e4e22c09bbb52.\n\nReason for revert: 10.11.12 is revoked upstream\n\nChange-Id: I9579b26cc8131bb0769889470657275de0c99c4b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/895b6c4ce80714d925f40163a54d6eb556097625"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/895b6c4ce80714d925f40163a54d6eb556097625"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
