)]}'
{"id":"openstack%2Fkolla~949841","triplet_id":"openstack%2Fkolla~master~I0a8cdc232e5b46cb47e4a3e1b0af7b2402103690","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-18 11:22:56.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2025-05-15 09:54:17.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}}},"hashtags":[],"change_id":"I0a8cdc232e5b46cb47e4a3e1b0af7b2402103690","subject":"Revert \"mariadb: pin to 10.11.11\"","status":"MERGED","created":"2025-05-15 05:08:37.000000000","updated":"2025-05-24 07:03:10.000000000","submitted":"2025-05-18 11:22:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":948997,"submission_id":"949841","meta_rev_id":"b3b85f83496365f0c674895b2acae94875cbdf12","_number":949841,"virtual_id_number":949841,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 11:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 05:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-15 09:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-15 09:54:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2025-05-15 09:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"rejected":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":-1,"date":"2025-05-15 05:15:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-15 05:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 05:08:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-15 05:08:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-15 05:08:37.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-15 05:15:57.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"20260791398199afa292b8b8d3ea87ed98c5ca93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 05:08:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0d795d363400ffd5fb58342ffcf2f96a09e8998","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-15 05:15:57.000000000","message":"Patch Set 1: Backport-Candidate-1 Code-Review+2 Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"49101fb471657ce4d418a42fc16886254933f783","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 06:53:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a7d13a6ea744d0f916b6572a78b2dca\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/148a12d1d6bc4dc4bbf52c0b9d2950cf : SUCCESS in 42m 52s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/04e087a0f23c48d89ad38f3eef8f32a2 : SUCCESS in 33m 22s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0d87067f61b742cb81ad7ff609b886cb : SUCCESS in 1h 38m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03a10a2f135a4b8f94f9c49c8644bae2 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b03ef8348fb540f981605a46ceacacb0 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/79344e62068f4bca8c4722d28394d93f : FAILURE in 7m 56s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/cc552af3263d433ea0d1086e62a36ced : SUCCESS in 41m 56s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/62b70610c9ae43a5bcc125f794d84ef7 : SUCCESS in 44m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b23f2a48293942bbdcc0a59f6bb75f72d6510442","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 07:36:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339a93e115f64b898db49f4dcd910c15\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/6347d50253c1448d9b34f2b1f0ba3f0e : SUCCESS in 25m 55s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/66ac6874e2974334aaaa23c254e3f6a4 : SUCCESS in 1h 00m 04s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/7fb52aaf8b084d6ab39c9d450dfd6db6 : SUCCESS in 19m 06s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ece3390dbc4f4bd187db5ea810c2b8c6 : SUCCESS in 15m 24s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/e328f588b4d442dea9c885ab1705be1d : SUCCESS in 1h 09m 11s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/076daa74491446a5ababfa7858d84184 : SUCCESS in 58m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e829b21206074387996462ce989bd6ba : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45aeead684c84e8eb6577965c0b3e6cb : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77bf4a49b5214caf849362b7de562f20 : SUCCESS in 3m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8cd28c673958418b8b6c3413b94d1055 : FAILURE in 4m 47s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7306286c8a2345428948f2b529393b9a : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d7cb3dacba043e2b767c75dadc7bb07 : SUCCESS in 6m 50s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/5b59ae94d0534582921aaf46a1729fd3 : SUCCESS in 3m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/8c17aaf9ae2a46238f265f4a3d581bcc : SUCCESS in 22m 38s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/dcc1810d5f25428aadb4a5265706bb2b : SUCCESS in 1h 34m 58s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b8fde34dab3d4813b1e30d977067a242 : SUCCESS in 36m 15s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/2408e4c8e87b40cca1e9c4aa3f5725be : SUCCESS in 50m 41s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/5f1c12e7ffe84b0eb774a35179a66434 : SUCCESS in 1h 08m 40s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f65890ff3bc944d0abcf1256b3dce7ec : FAILURE in 1h 26m 35s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/06e9a2e6cd9d4f78a82922681720cb1f : SUCCESS in 1h 22m 29s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/9f06c35d8526453583af5b82b3811d50 : SUCCESS in 26m 41s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/0abd1348c102454fa8a2551e63ee8a7a : SUCCESS in 1h 02m 29s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/14711d30b2ea4b6083bc975c8cbc4899 : FAILURE in 27m 47s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/a160a5392ad442b3945e95bee3463740 : FAILURE in 42m 08s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/229ff3923d9c48fc8ecc6e8e0557e806 : SUCCESS in 18m 09s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/04a04c0bbce04075b19d6c55eae124ef : SUCCESS in 13m 59s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/bc15728559804adfb60849b72919a148 : SUCCESS in 13m 13s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/551ec0e3ae034c9c9d79861c0a19ea94 : SUCCESS in 59m 31s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/71192a15405d4407a5c88eea8c3e243a : SUCCESS in 1h 00m 05s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/1119433f9288493081bb6961a0d18cc0 : SUCCESS in 1h 39m 20s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/513fc951707a41d5915e2d942e6fdab2 : FAILURE in 2h 09m 03s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/9105e3c810ae4fd4a6c2b552e0d6201e : FAILURE in 1h 05m 54s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/074c4bc2a35a4481a59b5ec5a7aa0c8b : SUCCESS in 24m 04s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/878738ee40c14f48b91c693fe4422214 : SUCCESS in 1h 03m 23s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/524fee279494492ca1cc201f8b779b66 : FAILURE in 44m 40s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/3521fff61167496683658203436e198f : FAILURE in 1h 01m 56s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4fa44417a6f04e5eacf8ffdc55feb43a : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c0e3c7cc7e203b1adf60f2607395bd90fc2e0f6c","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-15 09:54:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c832588bd50d41b781d87b233358c5fd31cad045","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-15 09:56:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"818a28f4ba140089e1749050a34b88cd1f0abfc6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 09:57:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"48c9242b9ab7607e46b4787a53dd28aa48064aa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 11:47:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a6dea2f9af3417eb5af578993aeb2bc\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/369aef23d06748e9bc8e376a7b16034b : FAILURE in 13m 29s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/17cf3ee5399649479622c129a6d3fc46 : FAILURE in 14m 40s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/7c8a7aaf5cf241048bd64514df2d1ca5 : FAILURE in 17m 55s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/5413770c6a0040ffa4b233d4642dc059 : FAILURE in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6264fd48067e4b51b4165d690b48546f : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e40f1c6a9ce843b7aac4b950f724aaf0 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9b075b6a99554a9c93ca6e93077e72eb : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8496dff07ac476fa15d48fdc18a3183 : SUCCESS in 3m 29s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/3eddf97362a648458bae54439d2a39db : SUCCESS in 2m 43s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/354ed7d37462465e952ae3aeb55f9174 : SUCCESS in 24m 01s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/886e2ea4415c49fcb5b3a7f86329f5b9 : SUCCESS in 1h 34m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b283b9354c3840d0a19bb5530b16478e : SUCCESS in 1h 04m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/8fe3a82be3854b6fb760f6da962ea5b8 : SUCCESS in 1h 35m 35s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/1f197ebf57e54b9787dab84d80a6420d : FAILURE in 20m 23s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/b1b3573e8be34223a8110123249f5b0c : FAILURE in 27m 30s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/420f929cea2d43088434c5a18ed59f7d : FAILURE in 20m 55s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/4784bae615004e1185b4854a03712e0b : FAILURE in 21m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a369cc02f09160b6c85e29ecafb4ed045e7d9b71","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-18 07:40:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67361fd1635f5b84aafd137156d488020d648645","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:12:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0807b8f306374a5c91dc896b5665316c\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/34b2e37c25814700bfe0985621dbf194 : SUCCESS in 59m 33s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/278baba48b8840fb91e1b0b1b05e141a : SUCCESS in 59m 26s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/8117c997fce946569fb6b68a30dceccb : SUCCESS in 1h 29m 15s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/41c3f7aa0734489cb1460436ce538114 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b9105ad4e7647738e4374d29838a880 : SUCCESS in 4m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/307ef0b3b91e4073a426ced837fc6d80 : FAILURE in 6m 41s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/eb1a80492a534c899c3771a4b31ef019 : SUCCESS in 55m 57s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/0bb8e3e03d2d4cb69fcd100e83348c0a : SUCCESS in 1h 04m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d0ea945397016222870e05c8c06ffc1e3ac0e1d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:40:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4228a92b71d74e3f9944e45e6c6fb655\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/733709b7b0be4495b9be713e73000d76 : SUCCESS in 23m 42s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/012d51cb80034dc7b85729e18e722609 : SUCCESS in 55m 09s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/343bd81e178143cb9b65bd53bc522d77 : SUCCESS in 14m 56s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3c0871909cdc4d38ae1584a5581a1b00 : SUCCESS in 14m 11s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/d072d72b29cc483f92b2a6c8285ab37e : SUCCESS in 35m 25s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/be56248997ae4062a15e3e840ffc8cdf : SUCCESS in 49m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39fc658ae4914756939e5fab47c4fec0 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bca719af2c64c05a3f042226f5b3d60 : SUCCESS in 1m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/19eed04cc38f4210ade23d7584cc8424 : SUCCESS in 2m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9db605499ef445548ae915805f6e9873 : FAILURE in 5m 28s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/afeead40c4c5444b900dd8f6a45c952b : SUCCESS in 3m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/010cb024077444798a82e4e5faba6dcc : SUCCESS in 7m 23s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/c7f503c447324b88a1e0a6751a3ce7af : SUCCESS in 3m 10s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/b35fe4afbb664efa898a548c6245531d : SUCCESS in 20m 24s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/1464ff993ead42589bc7c990d3d0bb2c : SUCCESS in 1h 02m 12s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/b5276eb31d3f42929ef7d207bfc50a18 : SUCCESS in 55m 01s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/d5df9396b38c4f529f6d7226d6195075 : SUCCESS in 1h 24m 14s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/7199423cf9a24008a9fbc96ad5b0be21 : FAILURE in 1h 13m 41s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/da67cffd2e604b25abb3d7ada2950492 : FAILURE in 1h 31m 52s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/a6a324e3cff24a87a41596b708ea215d : SUCCESS in 1h 11m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/2956fcb249784162a4950afee1510e55 : SUCCESS in 25m 45s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/92db24e155c6472aaf8020c2a121bb90 : SUCCESS in 54m 31s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/e4c9193f963d406ba497ea61c484aac7 : FAILURE in 31m 30s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/b39a1d5f7a324a6eae735edf7f41840e : FAILURE in 1h 12m 12s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/684e556d7c184d41ab6d519227b5fa95 : SUCCESS in 29m 19s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/c0478f0f499b4bf59d939e8c0bb397f8 : SUCCESS in 24m 24s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/64c297192b5748dda40747c271a0dd81 : SUCCESS in 26m 32s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8f044b98b29d44ae925506e209927750 : SUCCESS in 1h 03m 22s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/cab841464aee4105833b44d0c866c996 : SUCCESS in 57m 32s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/2f38c36fba894949be380f7fee5a7ccc : SUCCESS in 1h 04m 51s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c8b024c773bd41da8eba523249543036 : FAILURE in 1h 58m 58s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/de720d27b9714c5ea50b2f798660c4cd : SUCCESS in 1h 14m 09s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/8135eb8331084fcdaf628bd0066e43fa : SUCCESS in 23m 54s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/97bb9e80234142649f9a7225c8095288 : SUCCESS in 34m 27s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c4fb3fc1f4854f65b4c4d6f26c385de5 : FAILURE in 50m 56s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/fcc1a03aed144da48667012ff9dae938 : FAILURE in 1h 10m 02s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/f4e0aa7b9a634098842d6bc90029122e : SUCCESS in 28m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7d66fe041cf61ad7d7e2e7c8e8032862056b713c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:41:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ade05c025c8053c370e872ae5b7f3af4bb82818f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 11:22:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae7af4aecc3841088c2347d9bfa5d0e7\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/a25d057f8a5945d190a99b8df0d53d01 : SUCCESS in 15m 22s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/3d50cadc5dba4a8287854acc7b12a8df : SUCCESS in 26m 49s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/9670d7329dad4506ac6e519482233d5d : SUCCESS in 1h 03m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c0790d0b21464d0a9668b8f886ef646e : SUCCESS in 1h 23m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7ce433579424fc0ae18b3655c3d23ed : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06fd5a6a4c9045b9b73da8e2f0921e26 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/14425b71317849c3b49c9e9c026b8d06 : SUCCESS in 1m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa8abe9530534cb38229d02ca1376db7 : SUCCESS in 5m 15s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/ede2c29331ca4e35bf0b24e16653e2a5 : SUCCESS in 2m 23s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/4eee3404d68a4f8281cbc9ec6a9ce5ec : SUCCESS in 20m 09s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/6b98163ba37149e7a4bbd7a9b71efd5a : SUCCESS in 1h 13m 06s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/ec521b0597b8405bb99db0f1da69d9ed : SUCCESS in 1h 10m 29s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/298bd5e2106f4ee287f440b5bdb09433 : SUCCESS in 1h 37m 05s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/4f2c27e28efc411a8289739ee180a965 : SUCCESS in 23m 06s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/c6c7edb7f830430fa7e79a91a61a0b20 : SUCCESS in 30m 11s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/f70b657de5134c5e92a3479bf8602779 : SUCCESS in 1h 09m 12s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/c1dedf39d84a4d4ebba0e44bd8ae2033 : SUCCESS in 1h 34m 50s","accounts_in_message":[],"_revision_number":1},{"id":"20edb8d607f43fc42e8007c393c9bd137e0f00ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 11:22:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"826696ee7eebdf949022a988deb3342f619ee78f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 11:24:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/589e908ff4d74301b0cdb3e0fb182f28\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b431503846eb4c4798b9298dae124034 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1},{"id":"b3b85f83496365f0c674895b2acae94875cbdf12","tag":"autogenerated:gerrit:revert","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2025-05-24 07:03:10.000000000","message":"Created a revert of this change as I80093c342dc4403c88297c48cea976aa1d959e2a","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"431b4ae3e94f11a417d90229d58effb92fea4d8e","revisions":{"431b4ae3e94f11a417d90229d58effb92fea4d8e":{"kind":"REWORK","_number":1,"created":"2025-05-15 05:08:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/41/949841/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/41/949841/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/949841/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/949841/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/41/949841/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/41/949841/1"}}},"commit":{"parents":[{"commit":"535b2f270faf24cfa30b9aa1a896a62c5090f1aa","subject":"mariadb: pin to 10.11.11","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/535b2f270faf24cfa30b9aa1a896a62c5090f1aa"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-15 05:08:37.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2025-05-15 05:08:37.000000000","tz":0},"subject":"Revert \"mariadb: pin to 10.11.11\"","message":"Revert \"mariadb: pin to 10.11.11\"\n\nThis reverts commit 535b2f270faf24cfa30b9aa1a896a62c5090f1aa.\n\nReason for revert: 10.11.12 is revoked upstream\n\nChange-Id: I0a8cdc232e5b46cb47e4a3e1b0af7b2402103690\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/431b4ae3e94f11a417d90229d58effb92fea4d8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/431b4ae3e94f11a417d90229d58effb92fea4d8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"SATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"PASS","passing_atoms":["label:Backport-Candidate\u003d-1"],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
