)]}'
{"id":"openstack%2Fkolla~950138","triplet_id":"openstack%2Fkolla~master~I5db1b002fc05a642ef999bfc0093f92c3ff73e20","project":"openstack/kolla","branch":"master","attention_set":{},"removed_from_attention_set":{"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-06-05 20:46:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I5db1b002fc05a642ef999bfc0093f92c3ff73e20","subject":"WIP: Switch centos/rocky builds to python3.12","status":"ABANDONED","created":"2025-05-16 10:09:45.000000000","updated":"2025-06-05 20:46:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"70e7f947df1833db2073689e3361799fb4ecee38","_number":950138,"virtual_id_number":950138,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-16 15:07:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 10:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 12:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3402301ec9c5a91a3baecf4b20e5f1ca7d7d6cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-16 10:09:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3696f231135f198c45f0cc84b36fbcc654bd6b0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:56:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f601be1c00144fc0b5783eb10f29ebf7\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/1f2229ff14c04a48a7f6f94573d9eafb : FAILURE in 44m 24s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/0f238c3421e8419e9e05715634911942 : FAILURE in 15m 00s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/4289af99b62d4b6ba22469af0697aca4 : FAILURE in 14m 34s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/50796a0f5bef4e77a2ea508e611aed97 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee459602065d425b95322492fce7e6a2 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/8d189cc9dffe4ed7ba58dc697bc7e317 : FAILURE in 8m 22s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/3a0c2834f9b340268a337baa82fd8833 : FAILURE in 41m 37s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/3a0dafab2b584328b39ddcab21c182fb : SUCCESS in 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"484b937fc0e5f0b7f6897be93bc32e5eb866d33f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 12:34:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9e2396efda0426892ead3b76892342d\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/1f21b40275fd43bf9ae3efcc9e2c2ed2 : FAILURE in 29m 38s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/2f06e7d346314e70b620538befb8f74f : FAILURE in 48m 00s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/1adeaa41e94b45fdabe525be4e734fac : FAILURE in 10m 13s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/649e335e5eac489fba8ce4cf553c452a : FAILURE in 16m 14s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/527096be2a8842b383d7ae42ce74bc2c : FAILURE in 11m 46s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/227f604d244740fbbf0df67d21131545 : FAILURE in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85c4f275b41d4aa6a1c04bd27b2633bf : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/398f033a8b614a4ebcee574fc07bcdab : SUCCESS in 2m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/72bc0d45bac2497380de7b079144e84a : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5cebc4bdb74a412785a9117c3b56748e : FAILURE in 7m 20s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff0e5ad9fc54a479e4208fa75d391b0 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cacd1ace9c143ac8b453ecfc2c07870 : SUCCESS in 6m 57s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/49c3db4aa6c74cfb8fd47fb70f95a65e : SUCCESS in 2m 27s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/38a1826b6d1a49caab247ac0993b34ea : FAILURE in 20m 34s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/c36c17fea46c4d0390f520891a485456 : FAILURE in 1h 04m 31s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/53a1db61b2e74cc0b2f9998add9f8ca8 : FAILURE in 40m 19s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/23384930ab3b4f7aa06f3db6395bda9b : FAILURE in 1h 29m 11s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/8c6569456cd1467892948872ce1960c4 : FAILURE in 44m 56s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/18c0fd47dda44717bf83e4c2fc130928 : FAILURE in 57m 59s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/43c8b2fa83d14cecbaced2450c4bb484 : FAILURE in 19m 29s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/d29fd3d8154e4360aba79b747223de6e : SUCCESS in 25m 27s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/ec3f45a169be457f9ce37063788477e6 : FAILURE in 44m 02s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/07bad38c408f472a8d65f04543d95fb5 : FAILURE in 35m 24s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/9b01bd573a05458a8118e68919680024 : FAILURE in 1h 00m 52s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/8e9f140c9b9649dd80037a1f013596eb : SUCCESS in 33m 55s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/23b7439bfbd04832b3310c0495bf6f21 : SUCCESS in 22m 30s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/2e0c1084328144da89a31bac99f93c86 : SUCCESS in 14m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/76326f5cd656405abc97fc4a2f9cccf0 : SUCCESS in 1h 09m 44s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/3718a488d6d840bb9d80afe4ce96ccde : SUCCESS in 1h 38m 38s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/7d153b20dc724f92ae1ed1aa5a7877f5 : SUCCESS in 1h 54m 15s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7054bf8d85474913abc30c12726f7c47 : FAILURE in 2h 18m 52s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/a796a1b917af4a96b49ecc0c66908277 : SUCCESS in 1h 24m 30s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/35a696f37af94086a8112a3030028092 : SUCCESS in 26m 39s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/54fd267f4ad04d569c3a4528ed58ca77 : SUCCESS in 29m 03s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/c56a92a996d248e88fba210fe68748db : FAILURE in 1h 21m 23s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/39ae6da85d154d9fa44fc3e08ec36b0d : FAILURE in 46m 46s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/43bdce9a30ab4d9c9aefb4ed2c00654c : SUCCESS in 31m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"64070bdd2c21087020e4b68c3e9fefdb7089f76e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-16 12:36:16.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f394894646aeb50a3263e3e422d0dfae1e7083de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 13:33:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a916caf3e754d33a6e5bec32c26ed0d\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/49734871a14d46d5be5c025720ffcaf2 : SUCCESS in 38m 09s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/127559f2726f4cffbbc344533c8f2681 : FAILURE in 17m 37s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/d1e64bdea95448c5a710ada8a726fe96 : FAILURE in 17m 09s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e0813e88395746dd8bd597ca2d9d5e51 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6286ad446d5542508ddf3471b1827631 : SUCCESS in 5m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/18092585378742b38255793bcbc1f478 : FAILURE in 5m 42s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/23f87671e2cf49ff82402e27d4ccf134 : SUCCESS in 54m 30s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/01615eba561345ada3ebe1e45979f2b0 : SUCCESS in 54m 47s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"fb9274a738cbe06483c6176bf20938d7e6111bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:07:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d318868824b1401186b90a61b0ab45da\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/9b51d169b2354c7da2dc74bba20cb917 : SUCCESS in 16m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/b5d16a5bc95b45d2ac6bc3a3871e1fa9 : FAILURE in 21m 49s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/876df05fb1cc4db7a3bf9f125a1c9c12 : FAILURE in 8m 02s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/40a6239daa824e86911d29565324f3c7 : FAILURE in 11m 50s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/4c1c71dd36994047b11b01d9bc1995df : FAILURE in 10m 43s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/c56dd9aa99ea4127af1eef31a3f27c49 : FAILURE in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ccb1e7a550c4fc2af4544106c3f5da4 : SUCCESS in 2m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a5df21cae414a8f8d44c78be62dcbd7 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e5ea5451ff94f5c86cd5649f9802cfa : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e559ab862a544c0ba8d53d247d1a65ca : FAILURE in 5m 57s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/570b1360e3dd48e5bfdb62c0dc0f45ad : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb53222d873844bb99e120e7bcd9cc50 : SUCCESS in 3m 19s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/6c308ae02dab435da6a634334b45e99a : SUCCESS in 3m 02s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/5f0984b02dc44f348b13124b99bb6e53 : SUCCESS in 22m 44s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/85baeec9e22b4c4d90d52c8dc6781dae : SUCCESS in 52m 34s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/400ce138fbfd4b6a85389f63fba60812 : FAILURE in 1h 03m 22s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/a3a4942e6b7d48049fba53ae1bcf3609 : FAILURE in 1h 29m 52s\n- kolla-ansible-rocky9-cephadm https://zuul.opendev.org/t/openstack/build/a1c05b03d0f3403d836952f10017ef7f : FAILURE in 1h 05m 04s (non-voting)\n- kolla-ansible-rocky9-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b52cd845fcd5418d8f963ff3cf3f3d1a : FAILURE in 1h 04m 51s (non-voting)\n- kolla-ansible-rocky9-ironic https://zuul.opendev.org/t/openstack/build/5e6b53c675cb4581b3d58992bd68260b : FAILURE in 45m 17s (non-voting)\n- kolla-ansible-rocky9-mariadb https://zuul.opendev.org/t/openstack/build/f0bb05697d5c488ca3f6b85c1c0750a3 : SUCCESS in 25m 28s\n- kolla-ansible-rocky9-masakari https://zuul.opendev.org/t/openstack/build/52b13a94677e489f93f33c50c5b6a7c4 : FAILURE in 46m 30s (non-voting)\n- kolla-ansible-rocky9-octavia https://zuul.opendev.org/t/openstack/build/a291b036c4124f9fa70eb75e25b19fdc : FAILURE in 44m 49s (non-voting)\n- kolla-ansible-rocky9-ovn https://zuul.opendev.org/t/openstack/build/f1e932fa50f442b7a12c526ddc7abc29 : FAILURE in 1h 02m 25s (non-voting)\n- kolla-ansible-rocky9-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/7adc059ea8084881b2887b19113f2b9c : SUCCESS in 32m 10s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/179a70fc5b164d46b6afc68f68f4c088 : SUCCESS in 20m 40s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/955883fa38e94171a961267ce39aff59 : SUCCESS in 27m 47s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/8ace722ca0d24af083795cf5f2f34eac : SUCCESS in 1h 12m 34s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/6994d643a3f647368098f78fe251d17f : SUCCESS in 1h 48m 55s\n- kolla-ansible-ubuntu-cephadm https://zuul.opendev.org/t/openstack/build/5c8d2dc9a85447219ce0c33aa435ed75 : SUCCESS in 2h 00m 06s (non-voting)\n- kolla-ansible-ubuntu-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8d868431a7e946c2bf06b657fe04f366 : FAILURE in 2h 26m 24s (non-voting)\n- kolla-ansible-ubuntu-ironic https://zuul.opendev.org/t/openstack/build/275d7193b5c44565b1820301173d3e15 : SUCCESS in 1h 07m 33s (non-voting)\n- kolla-ansible-ubuntu-mariadb https://zuul.opendev.org/t/openstack/build/862e9454c56d4c268116ccb7a000c3ae : SUCCESS in 23m 45s\n- kolla-ansible-ubuntu-masakari https://zuul.opendev.org/t/openstack/build/6f160726d5d243d8a889c4f83a3708e1 : SUCCESS in 55m 04s (non-voting)\n- kolla-ansible-ubuntu-octavia https://zuul.opendev.org/t/openstack/build/3390f0148bd848e29e0820d381c3157a : FAILURE in 25m 50s (non-voting)\n- kolla-ansible-ubuntu-ovn https://zuul.opendev.org/t/openstack/build/a9af968adaa44bfa8fed34c65778cced : FAILURE in 1h 08m 42s (non-voting)\n- kolla-ansible-ubuntu-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/826289bfae9940a58760ae6df3ad3b11 : SUCCESS in 27m 00s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"70e7f947df1833db2073689e3361799fb4ecee38","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-06-05 20:46:25.000000000","message":"Abandoned\n\nimpossible","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ed86ba95bb2b0f337d9fa461c7e12cbc7f0395cb","revisions":{"b6606c647963d1c0cb985a0c564df154a5a70361":{"kind":"REWORK","_number":1,"created":"2025-05-16 10:09:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/950138/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/950138/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/950138/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/950138/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/950138/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/950138/1"}}},"commit":{"parents":[{"commit":"d430850a90485c308bb20817b61ea87671c52a53","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d430850a90485c308bb20817b61ea87671c52a53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-16 08:13:54.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-16 10:09:42.000000000","tz":0},"subject":"WIP: Switch centos/rocky builds to python3.12","message":"WIP: Switch centos/rocky builds to python3.12\n\nChange-Id: I5db1b002fc05a642ef999bfc0093f92c3ff73e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6606c647963d1c0cb985a0c564df154a5a70361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/b6606c647963d1c0cb985a0c564df154a5a70361"}]},"branch":"refs/heads/master"},"ed86ba95bb2b0f337d9fa461c7e12cbc7f0395cb":{"kind":"REWORK","_number":2,"created":"2025-05-16 12:36:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/38/950138/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/38/950138/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/950138/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/950138/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/38/950138/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/38/950138/2"}}},"commit":{"parents":[{"commit":"d430850a90485c308bb20817b61ea87671c52a53","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d430850a90485c308bb20817b61ea87671c52a53"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-16 08:13:54.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2025-05-16 12:36:16.000000000","tz":0},"subject":"WIP: Switch centos/rocky builds to python3.12","message":"WIP: Switch centos/rocky builds to python3.12\n\nChange-Id: I5db1b002fc05a642ef999bfc0093f92c3ff73e20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed86ba95bb2b0f337d9fa461c7e12cbc7f0395cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/ed86ba95bb2b0f337d9fa461c7e12cbc7f0395cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1","label:Backport-Candidate\u003d-2","label:Backport-Candidate\u003d2"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
