)]}'
{"id":"openstack%2Fkolla~950288","triplet_id":"openstack%2Fkolla~stable%2F2025.1~I69227a0548037411ab81890072aa779708cb85a0","project":"openstack/kolla","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-19 13:37:49.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2025-05-19 11:59:15.000000000","reason":"\u003cGERRIT_ACCOUNT_22629\u003e replied on the change","reason_account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},"23084":{"account":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"last_update":"2025-05-19 15:30:39.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2025-05-19 15:30:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I69227a0548037411ab81890072aa779708cb85a0","subject":"[2025.1 only] Revert \"bifrost: bump Ansible version to 11\"","status":"MERGED","created":"2025-05-19 09:34:58.000000000","updated":"2025-05-19 15:31:42.000000000","submitted":"2025-05-19 15:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950288","meta_rev_id":"b5e7b5b13f644df1170f5bc99eb897f137ad92c8","_number":950288,"virtual_id_number":950288,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-19 15:30:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-19 13:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2025-05-19 11:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2025-05-19 09:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2025-05-19 13:37:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2025-05-19 13:37:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-19 09:49:35.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2025-05-19 11:14:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-19 11:42:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-05-19 11:42:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2025-05-19 11:42:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"REVIEWER"}],"messages":[{"id":"7ab844e904ad2218f1d8e8a611d33e7311b08325","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2025-05-19 09:34:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4165382d6c8413e12249b5cc9729be1851b3dda7","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2025-05-19 09:49:35.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3ef978a9f329b1d9938cdbc92bd876440a9e0a81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 11:14:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a96426d2a6d3481f888fb94c10a3ad3f\n\n- kolla-build-centos9s https://zuul.opendev.org/t/openstack/build/354c2126a23a44c3887a2a96b5195ac0 : SUCCESS in 24m 30s (non-voting)\n- kolla-ansible-centos9s https://zuul.opendev.org/t/openstack/build/f7be6fbfb7864de8a7384b72e2cdc250 : SUCCESS in 55m 28s (non-voting)\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/4a43a94616a848fe8a1ae9f0b2c46b15 : SUCCESS in 18m 05s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/ea601e6adf0947cc9a0d4da8479a6573 : SUCCESS in 18m 06s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/0816ecb1069a4f2e9d8f0e0869db68eb : SUCCESS in 1h 03m 33s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/03600549760a42cd990229747c5ea6fc : SUCCESS in 1h 30m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff4026ef58f4e6aa5f9fa43298180ba : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63fa9158bef44e32b1474bba20b91484 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e5933fe88b6d472e9320cd8175d06fda : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55da49737b2741ddb45ce76b7ac37ff2 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2e030855d5a438d9db1c0fd73eff110 : SUCCESS in 6m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd81bdc3943c45b083175b5efe5df851 : SUCCESS in 3m 44s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/403c8221e00149968abf6c44eafb57c1 : SUCCESS in 4m 00s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/2c8cf2c3bb7e457183919d8e6a017a90 : SUCCESS in 23m 37s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/45d41bc7ddf245bcbf399b8d7baf4cb2 : SUCCESS in 1h 32m 23s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/a145e47a54a9430682969bf0e040dc55 : SUCCESS in 1h 02m 57s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/498d13784c314cffa8b408b8eea98cdd : SUCCESS in 1h 31m 24s\n- kolla-ansible-rocky9-bifrost https://zuul.opendev.org/t/openstack/build/adc394dac4ac476a8222b57a759605d7 : SUCCESS in 28m 53s (non-voting)\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/e8bd47ee76064e9aa99863096f002f4b : SUCCESS in 23m 41s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/0320db122da44fd4b241427745b1d25b : SUCCESS in 21m 53s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/fdbf33a60d9f4011aefcf103fa7d3e9a : SUCCESS in 59m 50s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/bd3ccb27493b46a38587c6db297ca342 : SUCCESS in 1h 28m 40s\n- kolla-ansible-ubuntu-bifrost https://zuul.opendev.org/t/openstack/build/15f27b39c3494fa48f6c55ac9fe45cb1 : SUCCESS in 23m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5b7f3330a8628f4e61c4a8a93f1ca345fbc5233f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 11:25:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b501c46eb3fc439a8841a78342c0d49b\n\n- kolla-build-centos9s-aarch64 https://zuul.opendev.org/t/openstack/build/941cd29d4f424bffa7da4cfb36f8179a : SUCCESS in 34m 26s (non-voting)\n- kolla-build-debian-aarch64 https://zuul.opendev.org/t/openstack/build/7d9d758343b84171b22c0115ca19dc04 : SUCCESS in 34m 27s (non-voting)\n- kolla-ansible-debian-aarch64 https://zuul.opendev.org/t/openstack/build/a080db75eece431eb50acd73ee5787f0 : SUCCESS in 1h 47m 56s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b58475380ed4b439a63f89306f95219 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9740b6d970bc40a3bfc8406d6df663ee : SUCCESS in 4m 15s (non-voting)\n- kolla-build-rocky9-aarch64 https://zuul.opendev.org/t/openstack/build/0558b2e003f140538829b21903456385 : SUCCESS in 47m 31s (non-voting)\n- kolla-build-ubuntu-aarch64 https://zuul.opendev.org/t/openstack/build/e816be918a9f41d599d187bd0c40dde7 : SUCCESS in 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bed103a188a26ccca4ebb4183121b6f196716f2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2025-05-19 11:59:15.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9ffb5bae2cd318c49d55707437b7d180d30a7f15","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-19 13:37:49.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"deee19393a1a35c6e0773621661721c5286c38c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 13:37:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0533859a639bce53365325fb3b54de3753e57d78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:30:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f9790d9053e44f5a5dfe27270035e92\n\n- kolla-build-debian https://zuul.opendev.org/t/openstack/build/215a833d02604306924fa3ab173a25ab : SUCCESS in 18m 36s\n- kolla-build-debian-podman https://zuul.opendev.org/t/openstack/build/b5f2699fa28840db93b94d2c9c7346ea : SUCCESS in 25m 53s\n- kolla-ansible-debian https://zuul.opendev.org/t/openstack/build/869ab9ff2f4640d389057f12659ba3dc : SUCCESS in 1h 01m 35s\n- kolla-ansible-debian-upgrade https://zuul.opendev.org/t/openstack/build/feb35a7aec854b25807cddf54ba01bc3 : SUCCESS in 1h 35m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/468c54753efc4993b54a82979e9f7b27 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae72019f5a564f38a69e4219e46193ff : SUCCESS in 2m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cd41ca26dd1344718fcfe5c8d5fcbe7e : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1395c5f7f058493fb8474e636ee56d95 : SUCCESS in 7m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63ec0bd9b3264167852387a3f0764e54 : SUCCESS in 4m 35s\n- kolla-tox-genconfig https://zuul.opendev.org/t/openstack/build/4cc2f52bca2d4519aeeb1d23f06dfd7b : SUCCESS in 2m 07s\n- kolla-build-rocky9 https://zuul.opendev.org/t/openstack/build/575dc82e85e142199bf6334deab8c1ac : SUCCESS in 27m 08s\n- kolla-build-rocky9-podman https://zuul.opendev.org/t/openstack/build/12bca3d9c4384ac28d155ecaf2332d57 : SUCCESS in 1h 10m 53s\n- kolla-ansible-rocky9 https://zuul.opendev.org/t/openstack/build/f70f3228779549b48bd8859e0e1a2a42 : SUCCESS in 1h 09m 27s\n- kolla-ansible-rocky9-upgrade https://zuul.opendev.org/t/openstack/build/4149b290639345be9ff03bcfcb6ec055 : SUCCESS in 1h 31m 54s\n- kolla-build-ubuntu https://zuul.opendev.org/t/openstack/build/7894497b64894f5da7a08f9488e7cde7 : SUCCESS in 13m 10s\n- kolla-build-ubuntu-podman https://zuul.opendev.org/t/openstack/build/e8ba1bfb60d14ddcb4685bdfb0046c0a : SUCCESS in 23m 38s\n- kolla-ansible-ubuntu https://zuul.opendev.org/t/openstack/build/25f6241f2f7941a6bf8270a9f08e4713 : SUCCESS in 1h 10m 35s\n- kolla-ansible-ubuntu-upgrade https://zuul.opendev.org/t/openstack/build/fc94e8b5d7844f3d8362e5df9bdfe588 : SUCCESS in 1h 36m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fe609dff91ca9fd3a25d8dd3f6d272ef28c34d51","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:30:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b5e7b5b13f644df1170f5bc99eb897f137ad92c8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:31:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/171ffbe2de0f4e7785b893da07d404e2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6457c115b5ab4831b9e3606c0df9761f : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a73a14d04bb0497dba456ef15fee1656 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d9621f295ec8773f630edbdb9f442021a583f62c","revisions":{"d9621f295ec8773f630edbdb9f442021a583f62c":{"kind":"REWORK","_number":1,"created":"2025-05-19 09:34:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/88/950288/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla","ref":"refs/changes/88/950288/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/950288/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/950288/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla refs/changes/88/950288/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla refs/changes/88/950288/1"}}},"commit":{"parents":[{"commit":"74cd69056f2741c0020757a3d4fe65eb07916303","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/74cd69056f2741c0020757a3d4fe65eb07916303"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 09:34:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2025-05-19 09:34:47.000000000","tz":120},"subject":"[2025.1 only] Revert \"bifrost: bump Ansible version to 11\"","message":"[2025.1 only] Revert \"bifrost: bump Ansible version to 11\"\n\nThis reverts commit c8fdb5550b15f98ed8f331f8d4b88ce150ccf3a5.\n\nReason for revert: Bifrost is not yet fully compatible with ansible\u003e\u003d10.\nThe master branch should soon be fixed [1], reverting only in 2025.1.\n\n[1] https://review.opendev.org/c/openstack/bifrost/+/948245\n\nChange-Id: I69227a0548037411ab81890072aa779708cb85a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9621f295ec8773f630edbdb9f442021a583f62c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla/commit/d9621f295ec8773f630edbdb9f442021a583f62c"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1 OR label:Backport-Candidate\u003d-2 OR label:Backport-Candidate\u003d2","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
